)]}'
{"id":"openstack%2Fkolla~886514","triplet_id":"openstack%2Fkolla~stable%2Fxena~I84c7633765749a5e9ced983505ef290c32154ccc","project":"openstack/kolla","branch":"stable/xena","topic":"cinder-volume-nvme","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-21 15:03:21.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-21 12:31:51.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-21 15:03:21.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-06-21 12:32:39.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-21 15:03:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I84c7633765749a5e9ced983505ef290c32154ccc","subject":"Fix cinder-volume: add missing nvme-cli rpm package","status":"MERGED","created":"2023-06-20 15:48:27.000000000","updated":"2023-06-21 15:04:20.000000000","submitted":"2023-06-21 15:03:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886514-cinder-volume-nvme","meta_rev_id":"706d1981423a7454e8ea2287b999659e9f22b1ac","_number":886514,"virtual_id_number":886514,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-21 15:03:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-06-21 12:31:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-21 13:36:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2023-06-21 12:32:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-06-21 12:32:39.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-20 16:03:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-06-20 16:03:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-20 16:03:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-20 16:03:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-06-20 16:26:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-20 18:58:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-21 13:36:32.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"c31ee5dac85114dd8eec524211f4166e81e7ae14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 15:48:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da279045ced33b2f39c1fe083610ceb6a43eeb59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 16:26:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85c3b8deb80c4ba29146eb604d946c4d\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1ae8829e9e244ce687a4b668e380412f : FAILURE in 33m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2839319c5fac490fbdf0972649676f29 : NODE_FAILURE Node request 300-0021497154 failed in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/56629713e113477ea08c4d8f9f3f55a1 : NODE_FAILURE Node request 300-0021497155 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0849a2ec32c344acbd6c457c242a3c69 : NODE_FAILURE Node request 300-0021497156 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cef28fdeb2d242d2a6c8bac850c97efc : NODE_FAILURE Node request 300-0021497157 failed in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/03b6d6d640714693a9ab99619180b1f2 : FAILURE in 8m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d06d562053b8d658d7c32d2081e223702d29fcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 18:58:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97077b32261c42d6b1f5d4c456cc68ec\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6949f01747104f49860a2700412b5109 : SUCCESS in 33m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/84e71e37387746a387be60764a1351f3 : SUCCESS in 29m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e93da354554747eda60d047067128cd6 : SUCCESS in 1h 31m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b17600b84f69408b8de106a8d4f52078 : SUCCESS in 1h 25m 29s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/578b5cb0a72d48f2af691661e3dd6eff : SUCCESS in 20m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/828749e45bf74454bfebc4b0f68743f5 : SUCCESS in 18m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2eaef8db9200412281cfa4079f33f83d : SUCCESS in 1h 02m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23b95e36a2b74671a3e1075000adf9bd : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e444cd449e3485ba2ccea0984ab517c : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50b1469d20594142b0d1a97a7fda8e50 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffd71db45dbf45798c406350a0400f95 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9176c110256242a7968bc6719f4a7568 : SUCCESS in 5m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e288a3972e874a2bb3a3d92a5f772626 : SUCCESS in 21m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0ebb0fec877548f0946a7a424cea8298 : SUCCESS in 22m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a3336e1b2b234b73b8e0be2a96562623 : SUCCESS in 1h 03m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/00df92f15dcb4c9bbc9408495a16e7bd : FAILURE in 41m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5df3a813b2984ce8b3d452b49b59af3e : SUCCESS in 1h 11m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2f1e31e4171ff8aa24db29000e87ea727767902e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 07:53:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"24ab3baccaa150389d73a46cbbe547d6e152eed4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 08:40:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51ab9b4ca93a4bfdae69e73bce96f509\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c9c4c31560e44435980a67a2c2e607cf : FAILURE in 41m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/51fb8f3c1fdc41f69d32294ae9865537 : SUCCESS in 30m 36s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/08c667a3c990471c93b377603f3cc8f4 : NODE_FAILURE Node request 300-0021503781 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/30aea19c4e9f4c24838e445ba9eebc74 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/10f9c4fc062d4403a21dfec349b624f3 : SUCCESS in 3m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/747e1dc903744cbf8a343ebbc0c3a216 : FAILURE in 6m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a98549e0f5670539f1251e7cfdd3905c8897dd5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 09:13:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a5683f277468f7c6f6448eba469eb0ac3cbbebc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 10:44:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b784ab2882549b39203e4c8722957a8\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/489faa21911e4d3db7828ca4741517f1 : FAILURE in 33m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/64a3a29f4cdf48668a4ebb15c2ef4029 : SUCCESS in 31m 21s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/992165d386a54729a53ed819f87c6e0f : SUCCESS in 1h 27m 15s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a4bd2ce212145f1a799d16683c018ab : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/872b9679af9049ca8e6f82803932b55a : SUCCESS in 4m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3abf0f8f92bf454c841176cbda47d587 : FAILURE in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b5038620f15131dcfe8598c0dd6e7d6d65fe1223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 12:22:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cea28dbdbafe4beea38f1d545e455bc2\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2028fd5c305e43eaae4b0593f8e69814 : FAILURE in 2h 56m 30s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/63fc212c219f4ed484e6418bb774661f : SUCCESS in 25m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8c290038a0834a758aefacd8cf66f41a : SUCCESS in 1h 33m 49s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2d564a90695d46b3a43b4f4d9488f66f : SUCCESS in 1h 17m 29s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a1056fa7e8054bbd9bfdb3cc8eb68bc1 : SUCCESS in 13m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/665139cad970490b91d0865b059d734f : SUCCESS in 20m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5c1a0d47bc15463e888c8d81f1762f96 : SUCCESS in 1h 28m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d199ebc70b5d4a61be40be11159253c0 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90d45320af9f44d2b02663f0fdfd35a0 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb6486835b8545d4a3a9029340dafbb2 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f7953d4f27646b895d502ccd9f98fc1 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dc4255a777f46a2bc7c297a6ac88f30 : SUCCESS in 4m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/792ab697fbc74a3dab3ed5b1f1bf162b : SUCCESS in 21m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a99ba8e6761747f2a9afd43c1a2a0cba : SUCCESS in 40m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b4d95837b9cb46a0b2b08dae9507d71a : SUCCESS in 1h 11m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/29d5d570331540f99f435b31a5056406 : FAILURE in 32m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dea8c1c931204a67b6a99effdac650f7 : SUCCESS in 1h 12m 50s","accounts_in_message":[],"_revision_number":3},{"id":"2f72fe81876410d994abb18eb2d704e213f141ba","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 12:31:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a39a7b2dd8aa0b6895cba5cb8e4739ad5657bfc7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-21 12:32:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0ab9bb71c0a7d713fbeb20bee84d835b5130cd1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 12:33:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"23ca44e7632fc7722a448fa288f40d23f40b5e47","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-21 13:36:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"48af8e4e8f9c1bd462a29fd4827614858307b253","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 15:03:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c23c0623bf94ca0ab22be00bf2607e2\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9f5796b21fde450d830051599b841efa : SUCCESS in 33m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0cce86cbe27740e3bf85481621f11037 : SUCCESS in 1h 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc91046c1e114ddabdf75b591d651aa5 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e44eacadb777483b9deb8c598397dce8 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f697ca1a46614f89a76b91de1372e184 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd6ca006501c498e9f8079400183c1b0 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7eb2de9e83fe4391a5c5dd8a42098607 : SUCCESS in 22m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/489f8e692ea54b6e9694f23260e7bf5b : SUCCESS in 53m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cb91388eaa5e457cb6bed293249cba25 : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":3},{"id":"2712eb5c0d35b0375116dbd5e4f3eeb3d16544fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 15:03:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"706d1981423a7454e8ea2287b999659e9f22b1ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 15:04:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc6e3a6d88bb477383d0cbf3ffd280c0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42bb83363778477ca95d5b66e45b3cf9 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fb41d37463c53526fedfd8bb99d558c23e3a2031","revisions":{"fc12177a388feb118e34c24137d7c59957cbc137":{"kind":"REWORK","_number":1,"created":"2023-06-20 15:48:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/14/886514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/886514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/886514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/886514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/886514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/886514/1"}}},"commit":{"parents":[{"commit":"1e0cef81af47d915ae1230e92661bcf24b010515","subject":"Disable build of Gnocchi containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e0cef81af47d915ae1230e92661bcf24b010515"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 18:14:11.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-20 15:47:44.000000000","tz":180},"subject":"Fix cinder-volume: add missing nvme-cli rpm package","message":"Fix cinder-volume: add missing nvme-cli rpm package\n\nCloses-Bug: #2024390\nChange-Id: I84c7633765749a5e9ced983505ef290c32154ccc\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 11f90f3746f6affd34885019a82eb94d3c098ee9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc12177a388feb118e34c24137d7c59957cbc137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc12177a388feb118e34c24137d7c59957cbc137"}]},"branch":"refs/heads/stable/xena"},"ed326c6b05df26a34ef0f592562ae0cc3a906882":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-21 07:53:38.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/14/886514/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/886514/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/886514/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/886514/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/886514/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/886514/2"}}},"commit":{"parents":[{"commit":"9802f3faebd6a80e3a2dcb30af4fad5e52da84ac","subject":"Fix cinder-volume: Add forgotten lsscsi rpm package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9802f3faebd6a80e3a2dcb30af4fad5e52da84ac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 18:14:11.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-21 07:53:38.000000000","tz":0},"subject":"Fix cinder-volume: add missing nvme-cli rpm package","message":"Fix cinder-volume: add missing nvme-cli rpm package\n\nCloses-Bug: #2024390\nChange-Id: I84c7633765749a5e9ced983505ef290c32154ccc\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 11f90f3746f6affd34885019a82eb94d3c098ee9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed326c6b05df26a34ef0f592562ae0cc3a906882"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed326c6b05df26a34ef0f592562ae0cc3a906882"}]},"branch":"refs/heads/stable/xena","description":"Rebase"},"fb41d37463c53526fedfd8bb99d558c23e3a2031":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-06-21 09:13:37.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/14/886514/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/886514/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/886514/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/886514/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/886514/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/886514/3"}}},"commit":{"parents":[{"commit":"1e0cef81af47d915ae1230e92661bcf24b010515","subject":"Disable build of Gnocchi containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e0cef81af47d915ae1230e92661bcf24b010515"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 18:14:11.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-21 09:13:37.000000000","tz":0},"subject":"Fix cinder-volume: add missing nvme-cli rpm package","message":"Fix cinder-volume: add missing nvme-cli rpm package\n\nCloses-Bug: #2024390\nChange-Id: I84c7633765749a5e9ced983505ef290c32154ccc\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 11f90f3746f6affd34885019a82eb94d3c098ee9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb41d37463c53526fedfd8bb99d558c23e3a2031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb41d37463c53526fedfd8bb99d558c23e3a2031"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
