)]}'
{"id":"openstack%2Fkolla~886515","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I84c7633765749a5e9ced983505ef290c32154ccc","project":"openstack/kolla","branch":"stable/wallaby","topic":"cinder-volume-nvme","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-23 10:54:33.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-23 10:54:33.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-23 10:54:33.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-06-21 13:10:45.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-23 09:45:22.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I84c7633765749a5e9ced983505ef290c32154ccc","subject":"Fix cinder-volume: add missing nvme-cli rpm package","status":"MERGED","created":"2023-06-20 15:49:25.000000000","updated":"2023-06-23 10:55:31.000000000","submitted":"2023-06-23 10:54:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886515-cinder-volume-nvme","meta_rev_id":"19cb3d8e805f8af20644b0a0502b371e7960bb2c","_number":886515,"virtual_id_number":886515,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-23 10:54:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":1,"date":"2023-06-21 13:35:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2023-06-21 13:10:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-23 09:45:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-23 09:45:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-20 16:03:58.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-06-20 16:03:58.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-20 16:03:58.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-20 16:03:58.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-06-20 16:27:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-20 18:26:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-21 13:35:54.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"3157f39291fc7675b4e420db9972350466dbc92e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-20 15:49:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba08dc3eac1e52f607b165eb4dd2f7c28139840f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 16:27:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb55b83f632542ad922ae8c9ebabd356\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ef903ded0974301b10c1d5aaa15448a : FAILURE in 36m 57s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/aceb770d6083444ea6a8a19947e11a2f : NODE_FAILURE Node request 300-0021497180 failed in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cd9805496f2a4795b1655c83db755751 : NODE_FAILURE Node request 300-0021497181 failed in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1991a512f18c4239b63facb1ab3dd440 : NODE_FAILURE Node request 300-0021497182 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf6cc484a9d0cccfbbb83cbbd0c7a5525c76ccb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 18:26:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce23469f8e6943b2b838a1613a29a9f8\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8079b0adf85043aba758b0f3d9537058 : SUCCESS in 32m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6feae8dc326b4222b8c568929c738d6c : SUCCESS in 36m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bf5b7c8db44643a5b42a91891ca6517c : SUCCESS in 1h 09m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/be90676eec9340f89207bd3aaaeaadb6 : SUCCESS in 1h 01m 50s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d136d9b11b2549e8b64e7f2df307e574 : SUCCESS in 21m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/965a589e71234c9e934492745f1ed146 : SUCCESS in 16m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a15e8e63fdc1434b896fa711dd3c1edb : SUCCESS in 55m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afef6eaeca79486fba3a163e5c753b4e : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b633bcd51d44cb0beebf349109ff89c : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f95ee7beb9c445a9dab005ef21d9801 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eff2466f5ed947da87862221baec7f09 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a870c72e24a249a5af80e64672d3aa2b : SUCCESS in 6m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1642a3addefd4736a6d000ed391b4a84 : SUCCESS in 21m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3b129be5ac9c4c0897337d0526bcd15d : SUCCESS in 31m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/93b03dd7c2e34995a13f4f859a055121 : SUCCESS in 49m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a90df0af84e4445da808775e9dda04cc : FAILURE in 41m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2eb3cb8fbf882e74802c178c9bb7250d4946a197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 07:54:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"971c6cea5e7f9ef0459a6fd41aaa264d6fb7c72f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 09:25:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/601f5d247db345b7bbba8e1cee1a641f\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c6cd4ec7e904c66a3358caea8b4a976 : FAILURE in 41m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/14ed3e2f5c93422da06fe1f22e0a0510 : SUCCESS in 37m 10s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c9a6c528489447cf88b5cd75319c7ccb : SUCCESS in 1h 28m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c3922e5324f54e4fa28f4f68f7ff8434 : FAILURE in 7m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8724329a8a8edd6cf0c28d4dbea25cf2a8873fc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 10:04:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ac25774510644b987bfd49c875a3e9f\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8050e58f4f384ced9db6d1f9c9c28b2a : SUCCESS in 33m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2d37cfb5a7404fffa32d8a286a87e6b5 : FAILURE in 1h 06m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f991d3e8f5014526b18218926b21a674 : TIMED_OUT in 2h 02m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b94f3ef7408f4117a3174eea2b9c9b17 : SUCCESS in 1h 03m 40s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9ec5c97a952c47caab5b03029a484a0b : FAILURE in 1h 10m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ce2cf4b2d3e44c9e81ec0c26084e143d : SUCCESS in 31m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1b409221f502454f8bf8aa12d84371e8 : FAILURE in 1h 06m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fceb36a922934986bd99fcfe67fd0d03 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b349d8cb8c564be28bd32032fa40b827 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e455f37b637f4d5c84a58561728399b4 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26f63f36d1e2492dae97a26330d0f53d : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/602c52214c2046ef9b083373ee6684a5 : SUCCESS in 6m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/80a303d3de514267b4da7b9c2c819330 : SUCCESS in 19m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dfba1fd02cb3485d955894a6d25146bf : FAILURE in 19m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c1f70118dfc48b18bcaba65f9c12150 : SUCCESS in 1h 01m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4fc0091316774db39e2c3230d1f8de3f : FAILURE in 35m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"94a8de33dd24aafb5d1b9da898eff26c0da6a9a0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 10:05:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"610785828fd2b2ad883a136967ea2ec474cd39c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 10:36:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d51bd19f79584da8a942baa3d07d2a59\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a838a500d5684023805afff385ec987e : FAILURE in 28m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a16703e40be14a01bf9c79f700f9d144 : SUCCESS in 27m 21s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5de0c96c26054250ba59dd27435db1c7 : NODE_FAILURE Node request 300-0021505510 failed in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/be2f793cddc545429e1f28c9c37795e6 : FAILURE in 7m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a7d5a944b0116241d3eb9b3508a01f619ba886da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 12:55:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf4cae1fb07f46d4b3deba931a404351\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e572d326a8a04a80be70d1b2fc7c0a02 : SUCCESS in 44m 04s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/618d7d303bb648afb1999a3faea9e08b : SUCCESS in 42m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d5a135468b734cd29870180694a11120 : SUCCESS in 1h 02m 22s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b03b1b587d764103991b2c1d0b8670e2 : SUCCESS in 59m 33s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e55f1e6996a04ac28c981be271b7b5f2 : SUCCESS in 27m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3ddfa83e10a74d1b8bab98e424b82f57 : SUCCESS in 22m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/13fd50b1243146eb85ab81190efa75e3 : SUCCESS in 1h 04m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9406bf01d9742e582f9142ffb843e8e : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa470a2188544b96ba2950f519ceb3fa : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8778b70df6d74844b0f6a203ddae79d2 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a270979ca5b492c93433162ba8af05c : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a06b7a566934d2ab843bd0374995ed4 : SUCCESS in 5m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a0d81c38f5fd4cf288f3f9328bb30195 : SUCCESS in 23m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5a343f3dc27d46acabafa10b0e82ebe6 : SUCCESS in 32m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cbd163e1303c4390870dd2ac1762415a : SUCCESS in 1h 03m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0f346482fd214bcca8833687e54241f4 : FAILURE in 42m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1f672e15e2015e4adf9b88da631edb07b6b91f5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-21 13:10:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8ebc7e8f5b2f26b91a962d0b600f7476902f51a6","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-21 13:35:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"eb7a899a8c2b0d5065511b0e4a07f1d2b1621056","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-23 09:45:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aa164bb08fb9ed304543bda2875c49519735202c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 09:45:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d3b9be709b64a2de299d683ccd5f4efd734aaff3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 10:54:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd3d5a5ccf174aae95a947b6164354a2\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/92200791fd464457aa7b346f27763fca : SUCCESS in 33m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6fc159d92da24516afb110335801e8c1 : SUCCESS in 1h 08m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/792a3a0bd7414c3ab1a28e92ca582b59 : SUCCESS in 12m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab223318ba5e455785f80565abafc6a8 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4827d52578ef4e4c9994b6c1921d6cb5 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b5a4e0191c4c7aaffa4a47337e9060 : SUCCESS in 5m 12s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/450cf7e27cfd405db0fe184ca7e54a63 : SUCCESS in 18m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/79211d780f61409ab0cfa571f93d5cc4 : SUCCESS in 1h 01m 16s","accounts_in_message":[],"_revision_number":2},{"id":"a0573ec5cd861f960976e2d0a7f48752aaf5050c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 10:54:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"19cb3d8e805f8af20644b0a0502b371e7960bb2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 10:55:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61d667f7fede4738b5c61d9c78db35e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e754aa246e439daffac51bdd3c22d3 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6c6adfc5013353345f91564721c18d0d0f7e8948","revisions":{"b11906b7c7eb88fe0a6d78f8c5d5944b5d3fa5d5":{"kind":"REWORK","_number":1,"created":"2023-06-20 15:49:25.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/15/886515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/886515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/886515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/886515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/886515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/886515/1"}}},"commit":{"parents":[{"commit":"e315994346eb557a5312e954b2812423f5db0ecf","subject":"Disable build of Gnocchi containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e315994346eb557a5312e954b2812423f5db0ecf"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 18:14:11.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-20 15:49:10.000000000","tz":180},"subject":"Fix cinder-volume: add missing nvme-cli rpm package","message":"Fix cinder-volume: add missing nvme-cli rpm package\n\nCloses-Bug: #2024390\nChange-Id: I84c7633765749a5e9ced983505ef290c32154ccc\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 11f90f3746f6affd34885019a82eb94d3c098ee9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b11906b7c7eb88fe0a6d78f8c5d5944b5d3fa5d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b11906b7c7eb88fe0a6d78f8c5d5944b5d3fa5d5"}]},"branch":"refs/heads/stable/wallaby"},"6c6adfc5013353345f91564721c18d0d0f7e8948":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-21 07:54:10.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/15/886515/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/886515/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/886515/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/886515/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/886515/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/886515/2"}}},"commit":{"parents":[{"commit":"4a326f91b764bf3b72c14615936e6f433fcae63b","subject":"Fix cinder-volume: Add forgotten lsscsi rpm package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a326f91b764bf3b72c14615936e6f433fcae63b"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-19 18:14:11.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-21 07:54:10.000000000","tz":0},"subject":"Fix cinder-volume: add missing nvme-cli rpm package","message":"Fix cinder-volume: add missing nvme-cli rpm package\n\nCloses-Bug: #2024390\nChange-Id: I84c7633765749a5e9ced983505ef290c32154ccc\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 11f90f3746f6affd34885019a82eb94d3c098ee9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c6adfc5013353345f91564721c18d0d0f7e8948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c6adfc5013353345f91564721c18d0d0f7e8948"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
