)]}'
{"id":"openstack%2Fkolla~886547","triplet_id":"openstack%2Fkolla~master~I008e32434e5046e3d7d2bd0cdf08ff3a37dc59b3","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-21 09:37:43.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-08 08:58:09.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-09-08 10:35:20.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-09-08 10:35:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I008e32434e5046e3d7d2bd0cdf08ff3a37dc59b3","subject":"Add python 3.10 to setup.cfg metadata","status":"MERGED","created":"2023-06-21 01:40:26.000000000","updated":"2023-09-08 10:36:26.000000000","submitted":"2023-09-08 10:35:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886547","meta_rev_id":"47b95c0423e24448017cf875e52f466b4e9fb514","_number":886547,"virtual_id_number":886547,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-08 10:35:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-21 09:37:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-09-08 08:58:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-07-13 07:31:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-06-26 07:50:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-09-08 08:58:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-21 02:02:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-21 02:57:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-21 07:09:56.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-06-21 09:37:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-26 07:50:54.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-07-13 07:31:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-09-07 10:13:01.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-09-07 10:13:01.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"42c589a6af62df352fee15a54ce8d3c85edeee90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-21 01:40:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"419d6516cd0cf046d5767b95ca4dd81fb08d5c9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 02:02:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5ffd4dd036c4826afe7b73c1d5c76cd\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/01524e171c9348e19b8729913ddbe300 : FAILURE in 18m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e45459ed5ee1412f81940e2c1de55e75 : FAILURE in 18m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2146487c12484f61bf89155972060000 : FAILURE in 19m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39facc69427b4404883a4124c359cfae : SUCCESS in 2m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e78d311b71a24eaebf0e951e6d1168d5 : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9af663f4990b4d3590984f2bba11c03e : SUCCESS in 2m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b0f60cb2f5cb4a6fbe6cafa5f9a5c57f : FAILURE in 19m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/cfdea6cfabf747c8837db45b91a64c43 : FAILURE in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"14cc902b8d20d8eca1721f493b2ce225280c2181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 02:57:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/264b256c59a6400ea3403febdb25f50f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6ef7a85b503e4bd7914a241942afd271 : SUCCESS in 1m 52s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/910e071921a445579a39de3ba47b07f2 : SUCCESS in 20m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/057783bc8dcc47a1af2a36aea9f52930 : SUCCESS in 1h 03m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/16dfc6d3322b4c89b7a88fdc6bb10b19 : SUCCESS in 18m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f25770005255437c9cd7b1181b528980 : SUCCESS in 1h 08m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a9e6668184d545c2a2546d7c68e3e3c7 : FAILURE in 57m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64e63e54c41049e390db82c278d6dd11 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36ad8f14f9df47bdb52c680eeabdc879 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f713f294ab084e878dd595bd85a911cc : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b10af8e5875b4c10b8fe682cf5649462 : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbaa08f7592142f49e2557969cbf1202 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d91a1dd415d14d1e8d7aa0772799eeb3 : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6d051e50f87e49a7899eddbba46c2811 : SUCCESS in 24m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5026161e513b49afaaa559e1e02398e5 : SUCCESS in 1h 07m 47s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/50c699ad8f4d4bbaa3ca6dd93c935445 : SUCCESS in 1h 12m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/50bdecc026194594bc5fbbd06510e71a : SUCCESS in 23m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bc20f923f44647c0aab59c73620525ae : SUCCESS in 58m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dbe2a4485eec4ab9b8f57f5eee9ff2f4 : SUCCESS in 1h 02m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4021cc8a348a1189d61706fbb4512ab7ae5595e5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-21 07:09:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7aaaaf641774951074a1c6ef8aef08aaf51cb3e5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-21 07:10:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b48830e8c72f4fce451d77f1fe962ed050196c1a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-21 07:27:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a6b5320dee928ebfb6d289821128e148172da47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 07:32:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1636c4e1e1444fbea7adbcd17fdd34a9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2e6a39bc60464e15bfca1d603043339e : FAILURE in 16m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5beb2b7fdb554f18965d072202b6a870 : FAILURE in 18m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d83234b97909494abf46a4b9ef28b833 : FAILURE in 16m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bfa91ffdf82b4edea095a5543204cdf1 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8061b613fd949aaa56214e5621bc02d : SUCCESS in 2m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/196690210c9c4fb29a700b1b75bc9bb9 : SUCCESS in 2m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4b69a6c5d4324dd58c67e577cd2ecd56 : FAILURE in 15m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4bee8c856ce24c89b7cc92ce215e02ac : FAILURE in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c6989a33da529f2f139a15fc2fd64b6d1da3c91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 08:32:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/185da408268a47138c99ae678414f007\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/22d1cddcc6204028b5d5842f3dba4728 : SUCCESS in 3m 50s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7c65b3cdb89949f18530ae3638c5f65b : SUCCESS in 18m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b35d69a6493c44e2b7f472493a22010a : SUCCESS in 59m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/017bb092e48740fdb801616b43aa91c5 : SUCCESS in 15m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/19a543369e6845bab751ba17f052b837 : SUCCESS in 49m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/16489326e98b4b5e901efabe3712a241 : SUCCESS in 1h 17m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08a6b1e4f7f84130a1750c2857011a94 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e57cd59501f54486a8d7aecaa8ca2d1e : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/816b9bc5aa7a4c65b8a45558fae82a74 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/937a8d1039ad49b38a8f842b13dadd9c : SUCCESS in 3m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a248b11884104d3c8432352cfb313431 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d25dac0c86d424995901756c91222ce : SUCCESS in 5m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/87f7ba4832804c0f9b385a41dbd6adc3 : SUCCESS in 22m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/036f32e02ace41fc9da911bc7445c30a : SUCCESS in 1h 07m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4e9049e813f244838bdaf3062f8b5a5f : SUCCESS in 1h 14m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/40da11bf1fb44211abf79d84c1fc39fc : SUCCESS in 15m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b8fa13578369456a908cc08e9219185f : SUCCESS in 53m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/93031f94640b407499c7dba7982cb078 : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":1},{"id":"1b58d4c34ab2ecd0592d4d46af63c1d47cefb2d7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-21 09:37:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d41c9c9e34e67eb79c2512849b731473109ca4e7","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-06-26 07:50:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4f5df557f188a0f9568001ebc7ecd1d4217ec195","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-07-13 07:31:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4f20b620a3695ad170c8423c34a1721a8cb2b39a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-08 08:58:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0bed6f515b16d858aee7b9fdf051a460232772d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 08:58:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"08f5be5a8913210155c97d0cb440a45703857f0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 10:35:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5289716b71b442da421c3ba28d65664\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/41365668fb95459088d8c0c63a32e0c8 : SUCCESS in 2m 25s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/99aab7fd3d7343daab3dbb8b6e5cb4a0 : SUCCESS in 20m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0e312c47f50145b8b33738f52705e854 : SUCCESS in 1h 00m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/151e1f28aad84710a47898cf0a9bcba9 : SUCCESS in 1h 01m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ebdc16c20ff49d4824eab58efd91af4 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cf0b512c41b4ad48fe044a707bd0574 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8dad2e297be42e3876521bae5b7ec80 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68923312317e444888fda211570988f3 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83627b30dde345c3a7438506b133093b : SUCCESS in 4m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/00d514779dd148cf9900ec5b72980172 : SUCCESS in 19m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/94f08fff4b114dfaac96bff40a79765c : SUCCESS in 1h 10m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/acc9b976319949aa8aade0fafffe09af : SUCCESS in 1h 30m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a44afdcff07b4e9ba2f1221007d54be2 : SUCCESS in 23m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4369eb1f46c448228da8a00e5cc5e16b : SUCCESS in 1h 05m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fc6a1945efc046328e4d7a1d40e1fd90 : SUCCESS in 1h 19m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f43c5bc3a8c7f17289c9568114f474dbe20e9d17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 10:35:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"47b95c0423e24448017cf875e52f466b4e9fb514","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 10:36:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6adf8a508e494b0abda50b0e7ae2f9f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f0a42f56c574842bd4c60aad178574b : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ccad439a4b85276a132729860834f0f178a51e3","revisions":{"8ccad439a4b85276a132729860834f0f178a51e3":{"kind":"REWORK","_number":1,"created":"2023-06-21 01:40:26.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/47/886547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/886547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/886547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/886547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/886547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/886547/1"}}},"commit":{"parents":[{"commit":"11f90f3746f6affd34885019a82eb94d3c098ee9","subject":"Fix cinder-volume: add missing nvme-cli rpm package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11f90f3746f6affd34885019a82eb94d3c098ee9"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-21 01:29:56.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-21 01:36:55.000000000","tz":480},"subject":"Add python 3.10 to setup.cfg metadata","message":"Add python 3.10 to setup.cfg metadata\n\nAdd python 3.10 to setup.cfg\nrefer to https://governance.openstack.org/tc/reference/runtimes/2023.2.html\n\nChange-Id: I008e32434e5046e3d7d2bd0cdf08ff3a37dc59b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ccad439a4b85276a132729860834f0f178a51e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ccad439a4b85276a132729860834f0f178a51e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
