)]}'
{"id":"openstack%2Fkolla~886672","triplet_id":"openstack%2Fkolla~stable%2F2023.1~I7e289bd5ce52319604717e0b319d0642a8e598db","project":"openstack/kolla","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-22 07:35:13.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I7e289bd5ce52319604717e0b319d0642a8e598db","subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","status":"MERGED","created":"2023-06-21 17:17:19.000000000","updated":"2023-06-22 10:36:34.000000000","submitted":"2023-06-22 10:35:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886672","cherry_pick_of_change":886574,"cherry_pick_of_patch_set":4,"meta_rev_id":"49889dbd56cbb3f12e5ea7cb5afc958c624aea73","_number":886672,"virtual_id_number":886672,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-22 10:35:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-06-21 18:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-21 17:41:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-06-21 18:54:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-06-21 18:54:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-21 17:41:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-21 18:40:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-21 18:54:40.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-21 19:00:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b7f044851a3c1f0860d8cad20dc79163e42729f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 17:17:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"31d7c79fb6932a3c32f6ee303dd78eb995d3779b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 17:41:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bededc4b41f397b116dfad8b7f3fba77e7b956e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 18:40:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f85f50638249433fba46df0327d9ce64\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/139b1ec4a11a4597ab08dc368d52961b : FAILURE in 31m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/874007228a2943b1bdf269d1824c7774 : SUCCESS in 28m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/69a3dd99860646aba4ed5e73ee4b607e : SUCCESS in 1h 21m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2c729a4892c454192dfdebb855c28db : SUCCESS in 2m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/64b23430652b4915b49a0f0aed2a9efd : SUCCESS in 2m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c6ed992e626e4b738b656cb94366525b : FAILURE in 30m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0c5fd5ad6bc24a60a1f2be954295f017 : NODE_FAILURE Node request 300-0021509686 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ecb10debc3bfddea84418b0c162f350660352956","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 18:43:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ca2be9ccb7d95bd3b57e03fb24ab88a708699eac","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-21 18:54:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"532996971386f6d9b48eebb569e8e840299b13ab","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-21 18:54:45.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a517305045cd8c65fdc2691a81dfa8800951c043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 19:00:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/687e7c9edcd148c78d06528ea99c4847\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3b582af3dc2e40a6a942059af49eaba4 : SUCCESS in 56m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/dc4c58425ec54906a400d6eabcbfdedd : SUCCESS in 1h 01m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9f3aa24908c04e03be9f61b3450273f5 : SUCCESS in 18m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/aed50d09d4b94865b12703236b9bf4cc : SUCCESS in 57m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d1af5ec1c9764a56a81d87094b6bdceb : SUCCESS in 1h 05m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23e278f12ab148488dd5d8bf1c306554 : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/436f8cd572f94afbb1b940a7c0f57392 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bed3c496dfc648a3a22ea26bb4bad960 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc5aa1d634224efd8118c61664f15ad9 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/280bd46928f24c038acdd2842bd5d0cc : SUCCESS in 4m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/526d8aae649443d78470677e0e69e044 : SUCCESS in 23m 09s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3bd023ea058a46caacd6e3278c150aeb : SUCCESS in 1h 04m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3a01b6c5987a4e10a74f1c2daa2f2126 : FAILURE in 26m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8490bbc21b0d487cafbd3a6358547565 : SUCCESS in 33m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0d2558d9a4e7442093f1c1ee930fce14 : SUCCESS in 1h 00m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b7daf43d276a4f0ba076de3f19ed6b04 : SUCCESS in 1h 33m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9952ef28393c8c38ec27f0d3fd6b055206a53a65","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 20:21:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dfa1bf88cdf3e8429a78586e9a9c6834ac405aa4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 21:58:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/546ce2d512294b81965f67247825e676\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f271402703e644638751eb6c996efa7b : FAILURE in 29m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b1bef1065e4a49d996207d9433dcd9cb : SUCCESS in 25m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bed6fd3cb2c443af8c14e9cfe55a1804 : SUCCESS in 1h 36m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c82ee2e3f8a470a8ef0d21e0b7d4da0 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8b210a6d03b042e38cda622a1afd5c21 : SUCCESS in 3m 10s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cf079f52150949088c39ee4bcb7be182 : NODE_FAILURE Node request 300-0021510397 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2c4c3152f2b240fa9b7b15ce88fabba5 : FAILURE in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f013c2403b96221759dfb7991452e1b0637a123","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 22:06:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/131c5c6215ca4d0e9b789dc73c6e5699\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3396764acc6e4c89acbf9f6a004e2faa : SUCCESS in 23m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/18eeea0fcb4e4299be6357fa5146e14d : SUCCESS in 1h 02m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5eefb5722ee743f2a532e2719d11b2ec : SUCCESS in 19m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2300cd247cc847b3adb539b4efb221e0 : SUCCESS in 1h 07m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5dc52c0e40c844c08a8eb93a766f8b7d : SUCCESS in 1h 03m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd304577b0964c3a9680c3d58fb1de9a : SUCCESS in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c5635dde9204bc7ab76b96d88a97777 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ba829ca63f04e2f8705555aaff42061 : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c6f576ab2ba4877b14eee38fe9647d0 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/613da72f657645338d436d9c248d7ec2 : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b4fcb7febe3e4d62809062b487eb3144 : SUCCESS in 31m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e52a00dc668d42afacc5468269ecb261 : SUCCESS in 1h 04m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/83a45c63e76a442abe9382523d1b2ba1 : SUCCESS in 1h 12m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6fcc888e36694f4ab4592fdc31bbe2db : SUCCESS in 23m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9669ab2a8f5a4357902a58a0c3c8d0d6 : SUCCESS in 51m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e2e6cba123514fed9524d943feb2767c : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":1},{"id":"038e9d45cb4f392382e4c9fef30dcad634228a7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 22:06:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a542df64ff1eae7e02bb337faca270cb9791410","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 23:20:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/62055232be3042ecbef5effbe124a8dd\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/500d4666f11b4dd3a58aee4e8fa79ad9 : SUCCESS in 16m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/24c3eb31a63b41a2a192c0b313be7934 : SUCCESS in 54m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c457d648346047c79b993f3a5097ae30 : FAILURE in 51m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56063a6a77b241338a253b1e9860ef0a : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/637cec0ffae247c8af376f54fc287c32 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f760ffa4a87d473f985d501315d188f6 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44026c7246b43f48628d43471d618cf : SUCCESS in 4m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3236384db4cf47db91fa07efe554c5c3 : SUCCESS in 20m 19s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/48b75cc2c6d04d778504edad43317482 : SUCCESS in 1h 04m 47s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8831b775d07c4ba7a6c725ab6a7268cb : SUCCESS in 1h 09m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cc0511e808304058b4a5819fdd95141c : SUCCESS in 16m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6d031e82fa4a4215949cd1e634e29631 : SUCCESS in 1h 04m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/99f0d8b048d244d0a51e64ce9f61bef3 : SUCCESS in 1h 07m 27s","accounts_in_message":[],"_revision_number":1},{"id":"43d6ce0bf4ca2622e17af5304b1dd93470dbdb64","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 23:38:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01096989422f36ce9a549160ac3e44faec64ab27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 00:57:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/051a3ceec9d642d99d0abb9ba5ab2698\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c10ad0e611864495980faab5c02d1f5f : SUCCESS in 21m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/91f842bca2044dd790466fb26145a107 : SUCCESS in 1h 08m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2c98212d568c490891281f5641bd75d2 : SUCCESS in 17m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a1e47b0d9d06497fa6069dbba3de700f : SUCCESS in 44m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f0aeb72ac9e64f8a9a639081f751f5be : SUCCESS in 1h 01m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa042cb83fae4a088eb262b4a4b3cbf2 : SUCCESS in 2m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac403f830d90451383f4a51d07fb33dc : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52fdb0c2036c46fc9cc044481c69a0ea : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a512882f5ba413488adb92705f86c2a : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c85f562aef154ea5856c13cbaa20a455 : SUCCESS in 4m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/30e7d62120f642b2b71348358f636b57 : SUCCESS in 13m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1f00349deecf4036a0db2ca574d6479d : SUCCESS in 44m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a08379274146428291505f303403ecfe : SUCCESS in 53m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5897d90ff1b44943a0db22c42fef5b6c : SUCCESS in 15m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/617a83c507c9469b878f5cdcedeb396b : SUCCESS in 1h 04m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f8f288bb09ea4816925d7409d573e7c8 : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":1},{"id":"62778620615cdebaee81101a1d511619715ee23d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 00:57:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4883b626df73ac314ab9b937255e0e301b10700","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 01:51:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/997424fc8bd74e4ba5ff719487d150e2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b13f6ccd8d344cc3ae09f1ca8a4a300d : FAILURE in 18m 20s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6d1f42ae19734303aa1cb3f0f999126c : SUCCESS in 17m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/79312f8439334b0ebd0b337a52f02d03 : SUCCESS in 2h 09m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/12977799889f4dc4a6e5cf2215999891 : SUCCESS in 2m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/76612c7ebc0742568f0545e78aa41f61 : SUCCESS in 2m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0604d2a56c034a228820f074c417a3f8 : FAILURE in 15m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0f268c602d024729877190ee16f57636 : FAILURE in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"742c25679253b119778229203eddd2c88e076de3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 02:21:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38f08ce045e24ad2b0ccb28bf473bdef\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/65972206d76140d99f50598176a0a43a : SUCCESS in 14m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1f44f29969b544dbadfbc87fed7c108d : SUCCESS in 45m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/504422ce0c9e4b6c850b6e8d239f9130 : SUCCESS in 49m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/156f70dac0e5434c91f4c99df55e1285 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ac41bb35ba5408f915621053ce5ce31 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d7c0fd05b8746bd9c93d9173a9dfc34 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd1d85fcb2d47d8b83c6242b3cddfca : SUCCESS in 4m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b4a0efabd646411eaf524e10f5fb41e3 : SUCCESS in 21m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0fd45265590442cc959362ca2c5f92ac : SUCCESS in 1h 04m 56s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2ceb107b225543e3b6da87e09fa89c88 : SUCCESS in 1h 19m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ebf6cbdc00f44c8f9d94d33af26aa36c : SUCCESS in 17m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ec9f8b371f9a4d2589818961d49d59df : SUCCESS in 41m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1637b5878ef5492b838cb19541886b95 : FAILURE in 42m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1e1957b658497557b83476ad0258f23a37b1e153","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-22 05:10:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a130b23819f617a336e1796f88a8fa41e7629446","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 06:28:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdcd5ffe8d0b402a9b26e89c6113bbd6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/bd92027413cf4210970926586e138736 : FAILURE in 15m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/229ce59cea2143eb995f6c47e3a1dff5 : SUCCESS in 16m 21s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6e6a7ba37bd440cca254086cf14567a3 : SUCCESS in 1h 16m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b210e1a242b40eba3739a685a28e284 : SUCCESS in 2m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5d406012f96341a4acb70afe6d0b920f : SUCCESS in 3m 35s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/898c49915dc3402981a172df7abaacdd : FAILURE in 17m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8b236f8e7ba24dfa9e8d15fbfa83004a : FAILURE in 5m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16590a6c12d209169957bc824b95f9027d2e504d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 06:43:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b117201f42444ceb9c9205ec7367b262\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b89cf14e248b4f20972f87d42aae7925 : SUCCESS in 1h 06m 19s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/49b4b4bafee242008739edec1d7fcb2d : SUCCESS in 59m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/48b4150c6de24b31a39d4169159cf698 : SUCCESS in 22m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/815b040c01074610a39663d1a7993b6c : FAILURE in 30m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4b8d08b935ea4714925d610a4295f33f : SUCCESS in 1h 18m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07daa895325d421eb6a549a9825c0763 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f010bed8ba3547a9a8f2d9a7e1887905 : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29a301550ce64bf0b24aeeda2d2f32e8 : SUCCESS in 3m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7813090e6f2a4adfb8f558612b81e04d : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b29d9647cff549edb5a24365b847537c : SUCCESS in 6m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3a936043a14042f39fd4bd93a4646643 : FAILURE in 17m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/445bbcdf41cf435c9730599e4be29ccb : FAILURE in 27m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1c2a9dba9ad542ef88eebf31c90d60de : FAILURE in 25m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/47cfb499291c4e399e5b6ea6442489ba : SUCCESS in 27m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4937b18654b8416cb46ed4e10015e36b : SUCCESS in 50m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dd284f20eed64524aa39ab84a142fc94 : SUCCESS in 52m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ae8444d00e29130d95183721bc80f34bf21f1644","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-22 07:35:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"909452c4b5389411822a5c9e0c4be09c24e8190e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 08:57:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d72dfa7dd1954dd2ae497077bcac4bb9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/27d5454283bc4e0982969564159e89b2 : FAILURE in 24m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ab5093f28e484b5da0fe82a8a658023e : SUCCESS in 27m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d29c96f7cdd34342aef5127a4019cf51 : SUCCESS in 1h 20m 41s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cece2ca4903a45c3aa0621d8e5ddf4c1 : NODE_FAILURE Node request 300-0021514754 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/91b8a897f3164e90a503a50a46136339 : SUCCESS in 2m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8ccfe536564644dd82b27ef15a07f5da : FAILURE in 21m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/15aea3282d1d4d69b651cfcc1e23251d : NODE_FAILURE Node request 300-0021514757 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97d1559e0a5c9e99db525935ceb077e96c3bba5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 09:09:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce8be602a69a4ab6b1d54333bd976fa8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b72e4abf62a84155a960ed5c52a199be : SUCCESS in 26m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0cb8cd55e303474b99d2fcec4e8f5636 : SUCCESS in 1h 13m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3429056106fd43a2b61e3e6eb199a335 : SUCCESS in 28m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f2798145a78d4a0ebd3af52d22fd90f0 : SUCCESS in 1h 05m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/64a06c658787445092218ec420e4163f : SUCCESS in 1h 23m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d846e73d9277420783df7c3b1ab25e8a : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee805398f69847cfbbbbc938735d9a9b : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65858622377b484984efabe53e96c979 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7baf0526da464e6598d28df13e47b759 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42f4dd38afa74615a1a45f06528a9104 : SUCCESS in 5m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a00a7e4692fc4959876abd08d2f2a42b : SUCCESS in 31m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/03889feed1d0487e935daad9415b420f : SUCCESS in 1h 20m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f79869253a9f469db2bf397cceb768b2 : SUCCESS in 1h 13m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b83b5405633f4be6bbcf1f902737c99a : SUCCESS in 33m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/50f583a915324ee29e736d3cb1c48b23 : SUCCESS in 1h 08m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/decc323e5ab343a98fe2c45d9ddccbef : SUCCESS in 1h 25m 24s","accounts_in_message":[],"_revision_number":1},{"id":"89fa20c1fe62e224985487489d72b506893d9990","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 09:09:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06d244624f9c282974064c61aa788058a9e4975b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 10:35:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac802f210dbe4e05bf9b6662e8f3c8e0\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/755ec13cbaaf48f2970b8ed19e70c480 : SUCCESS in 18m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a29a9633a255447a957cfc89a41e8471 : SUCCESS in 58m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9ba2ad05e6cc4ed5a93a37b5f954697f : SUCCESS in 1h 06m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b23f56f87a50408a818e90f2e7c683f2 : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ca2c14035d24a43b70e3491ed31b5d5 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8cef3bc7bc224780bd49c6d608b6eb87 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c75c29506ce41139d296360bb8eb161 : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5e155c98fee74580b467901042b7803d : SUCCESS in 18m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3d4ccbd665bf4180a61c123d94eb24a9 : SUCCESS in 1h 12m 47s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/65781804a86a4b82bf91ae18ffe0d924 : SUCCESS in 1h 08m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0edaa244275c4e8b991b2a531b86e8d3 : SUCCESS in 22m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0ce83f3278f041eaa773be3c99c90ee9 : SUCCESS in 1h 10m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4d45ecfede66453083a16df5cbd2c68f : SUCCESS in 1h 08m 22s","accounts_in_message":[],"_revision_number":1},{"id":"75d85cd05a730562d325c6d1cd07c9892a129c09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 10:35:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"49889dbd56cbb3f12e5ea7cb5afc958c624aea73","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 10:36:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90a55b5dcc0a486b9e649d37fc1a6a6c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d0c5607167641f4b1e4fb2e3c52a92c : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2c6b01e96f4f355455fa56f98d4d15fd1a3dd48","revisions":{"a2c6b01e96f4f355455fa56f98d4d15fd1a3dd48":{"kind":"REWORK","_number":1,"created":"2023-06-21 17:17:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/72/886672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/886672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/886672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/886672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/886672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/886672/1"}}},"commit":{"parents":[{"commit":"f4600a14974163fb6e1a7b2de398185f01744e6f","subject":"opensearch-dashboards: Fix permissions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4600a14974163fb6e1a7b2de398185f01744e6f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 07:53:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 17:17:19.000000000","tz":0},"subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","message":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu\n\nRMQ team is now maintaining erlang-24, erlang-25 ppas, see [1].\n\n[1]: https://github.com/rabbitmq/erlang-debian-package/discussions/33\n\nChange-Id: I7e289bd5ce52319604717e0b319d0642a8e598db\n(cherry picked from commit 0e881148d0036f27b95c5a27459e527fb3b0476d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2c6b01e96f4f355455fa56f98d4d15fd1a3dd48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2c6b01e96f4f355455fa56f98d4d15fd1a3dd48"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
