)]}'
{"id":"openstack%2Fkolla~886673","triplet_id":"openstack%2Fkolla~stable%2Fzed~I7e289bd5ce52319604717e0b319d0642a8e598db","project":"openstack/kolla","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-22 14:25:47.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I7e289bd5ce52319604717e0b319d0642a8e598db","subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","status":"MERGED","created":"2023-06-21 17:17:34.000000000","updated":"2023-06-22 18:02:36.000000000","submitted":"2023-06-22 18:01:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886673","cherry_pick_of_change":886574,"cherry_pick_of_patch_set":4,"meta_rev_id":"69cb97e13e7d81c965cefb704148d0a94c582127","_number":886673,"virtual_id_number":886673,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-22 18:01:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-06-21 18:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-21 17:41:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-06-22 10:36:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-06-22 10:36:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-21 17:41:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-21 18:42:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-21 19:16:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-22 10:36:59.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"991b90ffa6ffb5440d0072e32f01de424951cec2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 17:17:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"abc4e992b1c5f2a60445f3d76ac8705827b52a7f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 17:41:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b7bb556b202bbeaea437ef0e8030735fe2ce857a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 18:42:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2324928a2d024ade9a030ccea9da4725\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/28890fb07a234511bd6e81b71c1975c9 : NODE_FAILURE Node request 300-0021509703 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2645c55633d24fe2926b788ef8aec2cd : NODE_FAILURE Node request 300-0021509704 failed in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8bb9ef9b7ce44d87b3b9d7101178c617 : SUCCESS in 1h 23m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/134febf058c04577ab0cf8a970063c38 : NODE_FAILURE Node request 300-0021509706 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/291c6bda1b0646ff8fa646a45be3f007 : NODE_FAILURE Node request 300-0021509707 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c77da2a5c012463cab6393b9b03fd895 : FAILURE in 33m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ed18bb23da4f4fea8ee3e82e1e13cde8 : FAILURE in 6m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f286877dc446b75885f77dc921a81feccdd6ea8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 18:44:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d1e139e5f945263b72707a7e53fc9925d827c637","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 19:16:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8dbc52903b448999b100052cb02a539\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/310a4aa18b564760882c9c2c67e975e4 : SUCCESS in 37m 04s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/d451ad377f5946f4adfc7f7bf46a7ecd : SUCCESS in 1h 02m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f8ca2870f58943b395e73865aa11be32 : FAILURE in 34m 07s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b41eb698b65b46db985ba4d4e5f7e3b8 : SUCCESS in 51m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b4a9ba0bed2a415e821a32800d13a896 : FAILURE in 1h 00m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d2a34213eca4a6b94dcb01575b1cd76 : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c93e079fb030437692fac7e99dc34fda : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/243155d46ec849ba9ec40d9cb98bbb65 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c34bf2763beb433e89bc462c6a2ad99e : SUCCESS in 2m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57a4ddaf779846d8b4fe3394498b6a1f : SUCCESS in 5m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a82231a5bcd54fd182eaf91121fc5005 : SUCCESS in 23m 44s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/380451ca03644c0088036125de49b28f : FAILURE in 42m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dff032568bcc4fe3a616c926d2dd350e : SUCCESS in 38m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/64583b15aa3940d8b4f305a15ef62cbf : SUCCESS in 1h 12m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/67794239c13e4dc1bcf8c13be5b9eaeb : SUCCESS in 1h 45m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d99a1ab4a902174034fe9c5116611272233946f8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 20:23:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1130f7d98783ea29f44c6d4f39f7df817c518abd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 21:43:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b12a2795900479d9985f84caad061c8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/96c89c430cae432c8d5e34023d5b87a2 : FAILURE in 25m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/487dee6f66fe4ef18cd7313a5882be77 : SUCCESS in 27m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3f75a6ad3a99469ab82c7ca0c0cac2b3 : SUCCESS in 1h 17m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa38779ceb974cb68cf25bc6273d5c89 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a768322b18940949fcd986e6dd3ef37 : SUCCESS in 3m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3dae8496a033416b949b9f00a4970a62 : FAILURE in 23m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/53f3662dd3c046d1ac03de5cff2c471d : FAILURE in 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b4bb88b3201a4b94f77389c7e1dff6fea240354","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 21:49:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77c767acdecc4af39a9fc23b7e9dd0b2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f4c8db25bc4844a1a66acb5bc3e790ee : SUCCESS in 20m 35s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/61809f5ee1da49b5a64d6edbd4002b61 : SUCCESS in 50m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9419d1957ed04feaa3abad37408d378f : SUCCESS in 17m 38s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2afe0229255541398a0ce3731ca0719a : SUCCESS in 51m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/57714a694cb947749a16e251252f2faf : SUCCESS in 1h 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf17d08b9a23484bb684073022158c88 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bbd7e8e319f4ec5954138f9f2b684a6 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a61ee3336c5469c8a7798dcadb9c0b7 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65bd17a7a3ce4101bf2ff55e97310691 : SUCCESS in 2m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50f371af047848e9b0acc1310cf97cd0 : SUCCESS in 4m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5f3d362c84264e9580db619fbade58c9 : SUCCESS in 28m 51s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a22cce41d5e7485d99dca65326b980c7 : SUCCESS in 49m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/283d02659790487390517d710fc23606 : SUCCESS in 17m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/816c17ac334e496ca109c464fb1a45f8 : SUCCESS in 44m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/00fc243708a1495b962fe2c98c5de8c2 : SUCCESS in 1h 04m 44s","accounts_in_message":[],"_revision_number":1},{"id":"709f8e2d53a5e576b523d3cb5ab32bdbdca3f416","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-22 10:36:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7433fdd1478b3c72194f11fdf81b6c615ad1ada7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 10:37:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a177fd39b3dcbc34e8d14e3ca8274b369e2c9694","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 12:30:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef869d184c0f4566b3dd70efc08ccd87\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f1b3549968924a88a258242818070b82 : SUCCESS in 31m 24s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/18488a94bcd34847b9c0811eba412852 : FAILURE in 41m 12s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f8955b274f5c4319a190c1dadfbbf0c4 : FAILURE in 37m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27becf2f624844d79719675e1bd67f8a : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fe0c60b7e2247a6abe51473ab52955d : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ed19ba58e9744b6a475a18e21e7d286 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3546066e89142ca8f42c3ece39bf659 : SUCCESS in 6m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c47f1c86cb90446e906d8e40d0a12e15 : SUCCESS in 21m 30s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ced77a88926d4c9c9bab66e3caa8a980 : SUCCESS in 1h 02m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/de42478d61d847c68541bff6f0ee16af : SUCCESS in 41m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/284b3ebe3c5b4ab88b91002abd387bdb : SUCCESS in 1h 17m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/21345a9a6ff64553bf03097701c4e44a : SUCCESS in 1h 40m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d6b6331547953447d5d07cb34d4a82875f876e53","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-22 12:36:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63f3c2757220070264957112e8e6cbe14caf32da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 14:00:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3013ee0f2134bc88ac65a6eb0fef357\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9b89bfe654a547808bdf7cd17cab2c87 : FAILURE in 22m 42s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0e7e981b9c0e491eac23e07d19ed7f54 : SUCCESS in 21m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d0c53198f81b4daf9414b503623c6386 : SUCCESS in 1h 22m 26s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f61108792c54419984eed896bef4829 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b20fab0fa0864b0db157c3561619d304 : SUCCESS in 2m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/17070c2813ca4afc8f012787525497f9 : FAILURE in 27m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/83f848764198405eb25ca94bfca029d3 : FAILURE in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90579753b0da1d41356acc77c8c2cd2bae467968","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 14:18:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4386a441fcb4b3fbca8ec96d08e46c2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b0ccc88f736e444fa53273d622b825fe : SUCCESS in 23m 18s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/a68c7ea1aac644e5baf16754160d4c84 : SUCCESS in 1h 07m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/af057aa86fbb467e8d847f01f8fde211 : FAILURE in 22m 44s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f1e852af726c4a36b0d6ef29b0b0c7c4 : SUCCESS in 1h 11m 27s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f3401452450e4c9fad4e86b1bdc9f9af : SUCCESS in 1h 22m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab20529f6b1b4ab6b7ed916f041572cd : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5ba920ad8244908a8ec8fffbfdaf278 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc2539575a93424383a44f01f9a8ad86 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8cacefe305b04e568573a061d4ca1b27 : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff062d5d29e1484983796174adbf68e3 : SUCCESS in 6m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a1d44d39924b419c9ee1ac84ac206140 : FAILURE in 30m 44s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/0d8356d15de8443c8abb6f2d739b9888 : SUCCESS in 1h 11m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8d78ee9fd6ea4a9589ac35ac689de46e : SUCCESS in 14m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/54ab3d3699d14c938d272e0d06883419 : SUCCESS in 52m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9f64de94f8c1481280f7037a7fbb46e6 : SUCCESS in 1h 18m 09s","accounts_in_message":[],"_revision_number":1},{"id":"efee08ffb014b046355ad8a2d87bdb42fbdc32a9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-22 14:25:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2da492b21ba8247cfc4064d0ecf7a378870340d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 15:34:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d3f016b218145a889d872786518133e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/42d5658bf24e4554ae5ddc5bce3badb1 : FAILURE in 15m 06s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5f7075f91bb84cbbbf0ce736b394891b : SUCCESS in 55m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6e6e31b6fa8e49a6b520822a270d1f00 : SUCCESS in 1h 07m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7164c0a01b9b4fb9b978ba09dff8bdd2 : SUCCESS in 2m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b4af52d1c3314facae8a74c29d9736a4 : SUCCESS in 2m 59s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0e0c87e8005e425689900d9813a68d02 : FAILURE in 16m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/072c253ab35540b181d4af7f78b8e87a : FAILURE in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e53729b167fb0cfb83897c33f26ceead968d7b69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 16:27:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19cbf43a5cdd4d6387a514f0b6bbebdf\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/748d19de76334be0a504cd4fa8ff9c36 : SUCCESS in 24m 16s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/6f5839c346934a90bbbcb971e2246bf4 : SUCCESS in 1h 02m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fcac715ec00b4484a161678f685f175a : SUCCESS in 21m 31s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/da8479b0ca25464abcf54216e6b93950 : SUCCESS in 57m 14s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/13a025b6d0414b77ae540c8579494040 : SUCCESS in 1h 57m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c530bb6c62e2455dbeeee7b378ae3e3b : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cf979a8a43349e9a7a10edaf17ea48b : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e121231ad39e495da3d38e48f9418439 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65714c46789b429098faff5132638456 : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7e418485665480e9c50e3a1b2bef99a : SUCCESS in 5m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4d19a150f92442e0945b7616c9ba9d11 : SUCCESS in 23m 26s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/eb10c86ba9394386bdde8c51b82b2876 : SUCCESS in 1h 08m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d087b981e67b4472913b21dc691c8872 : SUCCESS in 14m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/610f622e6ee64150bc0032656e17a1e9 : SUCCESS in 52m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cff1465ef3f24e7ebbcc759aa6905e3d : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9aa5f68daa6d0ed6213283872a7053d0522110aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 16:28:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a56d544aa90808e0024c53996f0cbb97d4eaabc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 18:01:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41b9ce4e803345e2a4f09a647e705488\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3347847a3b0a48aea8ff533b2860d54d : SUCCESS in 26m 18s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f48423c4508d4b719f2396620800dc63 : SUCCESS in 51m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/04cd4be09ebb469a90e6f67ad16b2c4e : SUCCESS in 1h 24m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d9812efa2664d87bd9a6ce5d7f1d9eb : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/352ac56f3bbc44db826b0dc58f098d4a : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d30d1bb6282a4ea68106ceee72d06070 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a299b90bf7f54a86af929fcff4f0d5c5 : SUCCESS in 5m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4676bdfcfdcf4cb48b527b69e75ae4e2 : SUCCESS in 23m 42s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ac6f4f2dd53540a290a7721d6060bdf8 : SUCCESS in 1h 00m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cf3e88ac2baf4d23a9265675af5738a5 : SUCCESS in 20m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d7259367332430190b1d563b0cfe4dc : SUCCESS in 46m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c399a0c819f64e3e884daf1230a32b81 : SUCCESS in 1h 09m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d2b1c45d80715586f11ec871f0a9cb621a9516d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 18:01:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"69cb97e13e7d81c965cefb704148d0a94c582127","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 18:02:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ca7284e11a848f5beaf2c07602fbd30\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e6083c2f76f41ca9077d165e45c8d9f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e83e2a739cdd412dfa1abc9e45218187be31af2","revisions":{"0e83e2a739cdd412dfa1abc9e45218187be31af2":{"kind":"REWORK","_number":1,"created":"2023-06-21 17:17:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/73/886673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/886673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/886673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/886673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/886673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/886673/1"}}},"commit":{"parents":[{"commit":"85c764a653b6cf8da92ce55ac9615068683f5212","subject":"Disable build of Gnocchi containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85c764a653b6cf8da92ce55ac9615068683f5212"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 07:53:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 17:17:34.000000000","tz":0},"subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","message":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu\n\nRMQ team is now maintaining erlang-24, erlang-25 ppas, see [1].\n\n[1]: https://github.com/rabbitmq/erlang-debian-package/discussions/33\n\nChange-Id: I7e289bd5ce52319604717e0b319d0642a8e598db\n(cherry picked from commit 0e881148d0036f27b95c5a27459e527fb3b0476d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e83e2a739cdd412dfa1abc9e45218187be31af2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e83e2a739cdd412dfa1abc9e45218187be31af2"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
