)]}'
{"id":"openstack%2Fkolla~886674","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I7e289bd5ce52319604717e0b319d0642a8e598db","project":"openstack/kolla","branch":"stable/yoga","topic":"bug-2023668","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-07-17 12:31:39.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-07-13 11:44:00.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-07-17 10:55:06.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"hashtags":[],"change_id":"I7e289bd5ce52319604717e0b319d0642a8e598db","subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","status":"MERGED","created":"2023-06-21 17:17:52.000000000","updated":"2023-07-17 12:32:40.000000000","submitted":"2023-07-17 12:31:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886674-bug-2023668","cherry_pick_of_change":886673,"cherry_pick_of_patch_set":1,"meta_rev_id":"a0a14360e31f0f2c79e189827a035574393d4da0","_number":886674,"virtual_id_number":886674,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-17 12:31:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":1,"date":"2023-07-13 08:59:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"value":1,"date":"2023-07-13 12:06:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-07-13 11:44:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2023-07-13 10:32:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-07-17 10:55:06.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"all":[{"value":1,"date":"2023-07-13 08:59:07.000000000","_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-07-13 10:32:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-13 08:57:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-13 08:59:07.000000000","updated_by":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"reviewer":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"state":"REVIEWER"},{"updated":"2023-07-13 09:12:43.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-07-13 09:38:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-13 10:32:12.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-07-13 12:06:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"b966ff273ecc005569f73e09d677711dba1d890c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 17:17:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.\n\nThe following files contain Git conflicts:\n* docker/base/apt_preferences.debian\n* docker/base/apt_preferences.ubuntu\n* kolla/template/repos.yaml","accounts_in_message":[],"_revision_number":1},{"id":"75cc8554bb8252278f04b7d99d1b58de6dbfe4e9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 17:18:54.000000000","message":"Abandoned\n\nfocal, not needed","accounts_in_message":[],"_revision_number":1},{"id":"16aff56154a6e34af36cb9324705ce8d4f6a922d","tag":"autogenerated:gerrit:restore","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 08:10:27.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"8b52632d6691347f3c306105238b85cb8af2297c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 08:15:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42a9770e401db186fd3b5891b0dc59711f372534","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 08:15:23.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"f43b641d283b7452bff09fcec8c53aab19a99e5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 08:57:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca45a74d7b12493988a1ee3bc0a8eb4b\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b5e247a8c16c440ea435670c66be9344 : FAILURE in 35m 39s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5aeab8c6586c425eaf8fda58efc8164c : FAILURE in 21m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e125b707ccbd4759806001d8bd11e9c1 : SUCCESS in 34m 14s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/78fa5dba9c2b4b69a56214f715a5d143 : FAILURE in 24m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26ab79c947ed45c0bc255f183beb0f24 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e065c8a9496447aab54ab4d1900c66d : SUCCESS in 3m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ae3db63e7f64fdfac6e5f2f4bae44d0 : FAILURE in 39m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f09f9b29d5d3d775956908cc1dcdb0eeaa7eef92","author":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"date":"2023-07-13 08:59:07.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e0b3371cc1a0ba9ce8493456e7a7ebb5944650cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 09:38:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2fa284c977e4fc69e8926dae41d3713\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ac5b1022039e44f7adcf150844cec1da : SUCCESS in 46m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e586be561c8e465198a2d0647abd0c72 : SUCCESS in 45m 58s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/a82931b99a5b4a6c9416b2996f62e9e8 : SUCCESS in 32m 25s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/285db1bd4bb8483a978a69ad4639cbaf : SUCCESS in 1h 10m 51s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1736c0533a6341dcb7f0a4f7b1a20534 : SUCCESS in 1h 15m 24s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/cf3631b2791f41a5b4733e439fb9d290 : SUCCESS in 1h 05m 25s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c5205bbcf8ba40dcb926f89c5de2e784 : SUCCESS in 1h 22m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/34131dda2a354144873c4848a0c47a73 : SUCCESS in 26m 46s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2a910a46974d4936aafc9828ce3b8db7 : SUCCESS in 30m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/682c056cab404592ba10bbf3d2d315fc : SUCCESS in 54m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3a0e3bf7526a4c54bab5623df346e721 : SUCCESS in 1h 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f97312a013f4078bf85b4ae6aebbbf3 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2b9d96c690c4f6aafdd89edf43d3ae3 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acfc391e9e0d424c9c031834051f486b : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52465e41dd5c4d93a557115bcb700442 : SUCCESS in 4m 45s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/0003784eaf2445aca601888b95d6162e : SUCCESS in 20m 13s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/26a4cfd32f094955b39016b1b984ad8b : SUCCESS in 1h 04m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d16464643ea4433ea8097b095ceb9a63 : SUCCESS in 23m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/60523182a88941baad480e8759df0eec : SUCCESS in 33m 31s","accounts_in_message":[],"_revision_number":2},{"id":"d4af1debab7038643fd2cfb78c6dee89123fd973","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 10:32:12.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"18e45fc7bbb2d91ca517a8e46a6c175a4dfa6a7a","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-07-13 11:44:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"aaa8845c48fc1dadfb1f9d912d8b1ce522eef982","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-07-13 12:06:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2d8ecf3363cefddf4a0d65255f53b1dfdfcfe12e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-17 10:55:06.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7e1004e27f6646b38d514e0a48df9ce47c8ba6f0","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-17 10:55:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"77644b15c80b40222ac1046372670eee661d74a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 10:55:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8cc2d30d866c134afbfe3b5b966c7a6bd63b86f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:31:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b38b49bd36f842839868e514ec5c97ae\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/559bbfb1a8844065a01967339fdc6821 : SUCCESS in 41m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a1817f45200e40058e6d6677e9847852 : SUCCESS in 1h 08m 14s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ecd9603a98394402a46680263f5833b5 : SUCCESS in 1h 28m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0ad40748dd9f4763a122ecbf89c3a38c : SUCCESS in 31m 18s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/281ef3fb6fe047319d8b69ae9eda9f7d : SUCCESS in 1h 10m 13s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/29aaae4333604750abebd2c9a10b564d : SUCCESS in 1h 08m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/753314dae9cc4a91846e27d980c33b9b : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb4f1e2e0ce942068c7497c6e4356d67 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b343023fe19c4072a7bca908ff52e755 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06160a7af9504342b8079d9dc87160e0 : SUCCESS in 4m 50s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f65350511a38439e84317794cb7f5daa : SUCCESS in 1h 13m 42s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe9af3aa338c49f1b7b2b6750e3ef8e2 : SUCCESS in 23m 57s","accounts_in_message":[],"_revision_number":2},{"id":"56c0d33bb35ef6b25109c8d6dd8aa7c2eb851448","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:31:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a0a14360e31f0f2c79e189827a035574393d4da0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:32:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efc1202d032c47f4a8d21d406efecbbd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff44a39468f14b98a05b242ee3072d13 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d49cca54922507c040dd1003e628938b1aaf5cf1","revisions":{"32fb02f36ee7c9e492248eb5445b7ee5bae3af6e":{"kind":"REWORK","_number":1,"created":"2023-06-21 17:17:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/886674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/886674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/886674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/886674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/886674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/886674/1"}}},"commit":{"parents":[{"commit":"75f9186288e6410935d19f430394410289ecb63b","subject":"Fix cinder-volume: add missing nvme-cli rpm package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75f9186288e6410935d19f430394410289ecb63b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 07:53:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 17:17:52.000000000","tz":0},"subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","message":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu\n\nRMQ team is now maintaining erlang-24, erlang-25 ppas, see [1].\n\n[1]: https://github.com/rabbitmq/erlang-debian-package/discussions/33\n\nChange-Id: I7e289bd5ce52319604717e0b319d0642a8e598db\n(cherry picked from commit 0e881148d0036f27b95c5a27459e527fb3b0476d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32fb02f36ee7c9e492248eb5445b7ee5bae3af6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32fb02f36ee7c9e492248eb5445b7ee5bae3af6e"}]},"branch":"refs/heads/stable/yoga"},"d49cca54922507c040dd1003e628938b1aaf5cf1":{"kind":"REWORK","_number":2,"created":"2023-07-13 08:15:13.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/74/886674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/886674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/886674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/886674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/886674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/886674/2"}}},"commit":{"parents":[{"commit":"482ebb999756bd2ed3fade8380a9a7fd4d3c1c0e","subject":"Update support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/482ebb999756bd2ed3fade8380a9a7fd4d3c1c0e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 07:53:13.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-07-13 08:15:02.000000000","tz":120},"subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","message":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu\n\nRMQ team is now maintaining erlang-24, erlang-25 ppas, see [1].\n\n[1]: https://github.com/rabbitmq/erlang-debian-package/discussions/33\n\nCloses-Bug: #2023668\nChange-Id: I7e289bd5ce52319604717e0b319d0642a8e598db\n(cherry picked from commit 0e881148d0036f27b95c5a27459e527fb3b0476d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d49cca54922507c040dd1003e628938b1aaf5cf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d49cca54922507c040dd1003e628938b1aaf5cf1"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
