)]}'
{"id":"openstack%2Fkolla~886675","triplet_id":"openstack%2Fkolla~master~I4659bcd48ea32294b3ee84af9af94fae32b38bf3","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-21 17:26:03.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-23 08:52:31.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-21 17:36:18.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"34411":{"account":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"last_update":"2023-06-21 17:36:04.000000000","reason":"\u003cGERRIT_ACCOUNT_34411\u003e replied on the change","reason_account":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-23 08:52:31.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2023-06-23 08:52:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4659bcd48ea32294b3ee84af9af94fae32b38bf3","subject":"Revert \"rabbitmq: pin to 3.11.16\"","status":"MERGED","created":"2023-06-21 17:20:03.000000000","updated":"2023-06-23 08:53:30.000000000","submitted":"2023-06-23 08:52:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"revert_of":884677,"submission_id":"886675","meta_rev_id":"2e86a5555cd0caca6e60dd6df401636ac1acafb9","_number":886675,"virtual_id_number":886675,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-23 08:52:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-21 17:26:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-06-21 17:36:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-06-23 07:10:06.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2023-06-22 08:46:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-06-21 17:36:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-06-23 07:10:06.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-21 17:20:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2023-06-21 17:20:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-06-21 17:20:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"},{"updated":"2023-06-21 17:20:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-21 17:20:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-06-21 17:20:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-21 17:20:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-22 08:46:00.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-06-23 07:10:06.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"56b6f9ed3d262677f0406f5d03e7a513c7efa090","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 17:20:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e23be2adf0763e263f177ead3ea8c31bb9f50d0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-21 17:26:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0e937745c7178967f2c51f5e3c3dbd887074def2","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2023-06-21 17:36:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2e6e33e889830a823a937c03ad84c3cae6a6719f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 17:36:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fad23f293e1e64b98b00291a64684e139d89f3a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 17:52:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92f0487b633341328a44c193e3013bbb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9323ad8d45db4c199fc24dbf2adb4114 : NODE_FAILURE Node request 300-0021509751 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e03865dd3299444e983825e1857dd861 : NODE_FAILURE Node request 300-0021509752 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/60194dff53c5432da4725484a7ba8461 : NODE_FAILURE Node request 300-0021509753 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/19e443ecd9ad4f6185468bc5a76aa0cd : NODE_FAILURE Node request 300-0021509754 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac952a2cf3c44901a7d7e95935cfda7d : NODE_FAILURE Node request 300-0021509755 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6d30e75509274807a0c557e338d25b29 : NODE_FAILURE Node request 300-0021509756 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/811d867c077046aebdf107f6e7c3e915 : FAILURE in 28m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/85255af90273468aa650f53ab10c7f87 : NODE_FAILURE Node request 300-0021509758 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a299db408a1f8dd281286a6e9a90a276635894f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 19:45:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03ce7b0b95a04a9d9b4850b0f1129160\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/51ed7de1163144e19bb8bd13a226a820 : SUCCESS in 29m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/66a38f2dc75d4d5da8061c26f088b2fa : SUCCESS in 1h 38m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e9bbbf5f680a41bab109c37285055c6e : SUCCESS in 16m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/036fec887b854e6c90b53ce69dedbac8 : SUCCESS in 57m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2412d96c53864cc3a465657688c09f1b : FAILURE in 47m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f66a2bfe5c843aca0032a0ab7faaac1 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13442f9648d642d0a988939367c48d58 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fe2026fb3ec44aca25b0eddd0117f21 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b8437a5c4c7480f8ad40d95c7ad9c2a : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3025db8b83f441de84f54c8ddd03e7c8 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0076f1f493a4c2881dc3e30efd8b49b : SUCCESS in 4m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/28dbc1bf6937427a96848a4f71c81503 : SUCCESS in 22m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ada87340abbb48788017193697893c36 : FAILURE in 25m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4dbc734623774d1c916dac5e065fc852 : FAILURE in 32m 52s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/fdda7997829d489cb3d13de0bb75befc : FAILURE in 33m 39s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a9aebecee8304312b969ab48f766a93f : FAILURE in 35m 47s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/1440c61b55b94f7d911982d828343155 : FAILURE in 1h 08m 35s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/1c5a84dd0a3b4b8bbc648b71bfe30e8e : FAILURE in 1h 03m 58s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/8ba6f5d5a62b4dadbb358b9b0c6be421 : SUCCESS in 1h 22m 49s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/812c263faa824deba5aa8fc84f3254ac : FAILURE in 33m 22s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/67e0b7f23d04454493f71f2444018ef2 : SUCCESS in 1h 02m 55s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/b786d96547e14f3bb4480b15b05aac11 : SUCCESS in 56m 06s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/7fdbb5ffad904c489bce3d15548d7838 : TIMED_OUT in 2h 03m 53s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/aadbafbb826d4e0e886b4f2ae7a900d3 : SUCCESS in 44m 03s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e4a2a032308c409a8bf3195cac931210 : SUCCESS in 29m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ef550174632c42b6aef550427301eb63 : SUCCESS in 58m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/df20b0cd8d144baaade0b0773e593212 : SUCCESS in 1h 05m 02s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/03ff5cb48284450782a1207b9482a806 : SUCCESS in 1h 33m 14s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bf14a23c25ee4efca988af2fee7db769 : SUCCESS in 1h 47m 41s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/3a000b331ff44714a506c730a543819e : SUCCESS in 1h 35m 08s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/b8b79831495e400dbf0d59491db00f65 : SUCCESS in 1h 08m 35s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/4f38ef24cd4b4f6eb2ccaf999f98780e : SUCCESS in 1h 13m 21s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/fb9999d360e746d3aa483e8533ca3736 : FAILURE in 36m 14s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/1cc6949a1ae042e5abcf99e3cf0be4f3 : SUCCESS in 58m 09s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ab9cdcc67dd740708b26f09277059e61 : FAILURE in 1h 17m 18s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9630997dad324ec7a77aabeada00517a : SUCCESS in 1h 52m 18s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e08d5d7534794b85aebee3421fcdef4d : SUCCESS in 29m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9309ba529b30d20e404ead55303eecf7357026de","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-22 08:46:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fccd627cc721542144e299119ffa9c4801bf6d13","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-22 10:37:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8cbca65e6ed8ce236c617a01dc1c1c2c591bccc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 11:50:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a597142ea6a34074bba1291e768559ba\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/493344cd7db54f9b96cf4f4f2cb03f61 : FAILURE in 55m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1ec671dbdba14e4ab4232e83b99ea64e : SUCCESS in 20m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/de756acf1e2d43cc8312d3f0637ce476 : SUCCESS in 1h 11m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/82606cea0aec422faec7dc5e101f3a29 : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/822352032a1d458c91ba0997d257bbc2 : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2d1a6c8ac40f43b6876c9a3dff96c666 : SUCCESS in 3m 24s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/fb1310480d404f108f3e5de6df9122ee : FAILURE in 22m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d011eb607fd74545b76e271e80b2b65a : FAILURE in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20587f2825a9d641865ef5c1f78522d6964fe432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 13:34:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f8e1632d82647e991093d853b14d6f2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1abced753773481f9a48d3f7b512ba90 : SUCCESS in 44m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/963c6f5764e54f6eaac69cf990e20b8b : SUCCESS in 58m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4dada858df614f76ac16fd655e99314a : FAILURE in 39m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c5c4825985cd42248dd8c18bf777246b : SUCCESS in 1h 17m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/51f71bf9b0064a468071b2577aa07696 : SUCCESS in 1h 32m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373a7bf944364eb7b2691f85f4e61c91 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91f202026dd54aa5aec6897a6cc8a965 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e02da773418447a2858f1bcd3b8d0824 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b1f407b4b854b128e3eb774e88cb1ee : SUCCESS in 3m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26c54274595a4a51b30e96f55cafbdc5 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ebd76399a7b4041a3311757c1878f60 : SUCCESS in 4m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e284d56d26f34edca6d49c834c6bea07 : SUCCESS in 46m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/58734aacc7f54d3a8bb486e3567c150b : SUCCESS in 1h 26m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3cfb5cdbda8e40dcb2e1c5df50401677 : SUCCESS in 1h 11m 15s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/0dd6d4d976aa47f49613895617a3b089 : SUCCESS in 1h 59m 40s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a42420390e03415499703534c53ef100 : FAILURE in 1h 06m 34s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/450f7c1fbff44448ab8e333a54a40f14 : FAILURE in 1h 10m 11s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/c023b8891fee4bae83233b5692c200fc : FAILURE in 1h 00m 53s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/fa0d4ba005734131a5bcf5c6f9903be1 : SUCCESS in 1h 37m 19s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/ff567c9ef9cd4f46bac24509049f6c40 : SUCCESS in 30m 48s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/a78b987a4c5041838ec9c37eedc2d5b0 : SUCCESS in 1h 08m 29s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/1b64ad710a354fefafa0d461670cd5cb : SUCCESS in 1h 03m 00s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/4b1724d41a1e447486040065df8cd6d9 : SUCCESS in 1h 26m 09s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0e40ddf4480145f88e9f215317f3d575 : SUCCESS in 33m 51s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/10d77decf7444147a1a1699aeca36b9f : SUCCESS in 23m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/39cdf1a52bfe454e9fc6703c6dad54a7 : SUCCESS in 59m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/22b7e8ee8bea4215a748f4757621c52a : FAILURE in 51m 40s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/69f0be7e99e8452eaca43f06f7dd3afd : SUCCESS in 1h 34m 31s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/54b37f507f994a449fc9b6feadc57e4d : FAILURE in 2h 06m 04s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/dba54524f9894033b7adc46b672a0547 : SUCCESS in 1h 44m 38s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/19bd8071a2714371831ce5aee6124094 : SUCCESS in 1h 05m 38s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/3f5e1de4f125441a96134499d209b83f : SUCCESS in 1h 12m 12s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/70b3d147a5064fef849475ac34952d9d : SUCCESS in 29m 13s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/a1e257a568de4e41b73634934496bcbe : SUCCESS in 1h 14m 08s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/60140ada0c854e1eae30a7564b5e260f : SUCCESS in 1h 13m 08s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/af6d41d75ce34aa69dcbf4e68e04d6ef : SUCCESS in 1h 25m 30s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6cbfdbb6eae14add8ce7a43a389aea7c : SUCCESS in 32m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b4d303e4203782aa15211c8f90dab6eeb73a249","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-22 18:20:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38c076c45ff81c5818ebfa68c2dc38694efbd46f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 19:36:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecbbedbf0b9c43939102bbade026452c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/279fa982f0284b84b0646cba6cbaa818 : FAILURE in 16m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/83eb7f28d4e14dfe860f9ea24d3987aa : SUCCESS in 21m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1347a363a98d41a8ab5e56c589251693 : SUCCESS in 1h 11m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fa3f391e7944857bc6336ce012351ed : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fad61532da3e4ee3a8d424a934b2f696 : SUCCESS in 2m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d660d810b2a14b60994f5c121d864b32 : SUCCESS in 3m 40s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/dae3fa01fc024efebdb85b6823717c89 : FAILURE in 18m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5e99fa70b16745769a081cb3bfef0673 : FAILURE in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db7ffdc9c05e5e64f90206ae285ddc6cc08a7ee3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 20:37:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad348e166d264ce597bd5ff96e552642\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0e5e7a67e77348cda652a5f21cc457aa : SUCCESS in 20m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1fa9ed2c04a144c0bc586c304fcca824 : SUCCESS in 1h 06m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b1436eb7827c44e7bf73677d39eb3a61 : SUCCESS in 21m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6964156ca0194ef4a75a28e8b03a9c1a : SUCCESS in 55m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a26333ff5d1f47a898e3c0eeee25264a : SUCCESS in 1h 23m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a76cab76a334126bb14da93aa086639 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81cca56375e1431899d8031d32882685 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63cc3979220d4670bb950e29769058a8 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75294d22717749f0867d4f295575d3f4 : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e6dc1a62eae4f7383e203dd4856d70d : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3cf891d31c4d23bce7ac1d9dd74032 : SUCCESS in 5m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/576b7984a14f466fbc2030d8c571dd3a : SUCCESS in 14m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d1a0df4823a14c2fa40005d92b5bd9d1 : SUCCESS in 1h 17m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b44b26b2db2e4c4cb16a8a4627c2d722 : SUCCESS in 1h 08m 15s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1260108663b940ed94a7f94fe6563d9a : SUCCESS in 1h 36m 09s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/86ca83e22aba48a696892b09a99daab6 : FAILURE in 1h 14m 13s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/5208d5af61454487873225ee2db92aa7 : FAILURE in 1h 20m 33s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/64897a3c85d74e31b51cd4a269623f32 : FAILURE in 59m 00s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/88ce9ea18e1f46b0a8841106b43543d5 : SUCCESS in 1h 40m 22s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/1cf197ceb127408ea5eb0c278f7fa672 : SUCCESS in 29m 57s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/77a7fdd053ce413789e791de7cbd9082 : SUCCESS in 1h 13m 34s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/4e5b54a445ee4ea4a42ba3fe58582221 : SUCCESS in 1h 21m 43s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/63587d60e4e446eb95b9d2552565b4c7 : SUCCESS in 1h 46m 53s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f676534e03804cba9102524e42ee9c81 : SUCCESS in 29m 57s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d853687266494e5cb680ea89181e4125 : SUCCESS in 25m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c8f4c1b489c14cfe80655e8690447843 : SUCCESS in 1h 18m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/eafc5fcae9fb49e792bc60d78356e132 : SUCCESS in 1h 22m 02s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/2e11520b252041ccb6c20e0246fb8d44 : SUCCESS in 1h 28m 33s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6e36cf7fca21478ab753c89d68e7a2c4 : SUCCESS in 2h 03m 52s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/27251ed84a91483eb75d8d0a85015296 : SUCCESS in 1h 29m 37s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/9eecf15702ea43f5b5800901adad8b58 : SUCCESS in 1h 26m 29s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/f3caea07e24649e7a65cbad91fcfa775 : SUCCESS in 1h 14m 28s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/55231a5bbd2b4a19a3b74be22b53e8c7 : SUCCESS in 34m 25s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/92e45f87efca4bb1b057a97e7274d2b1 : SUCCESS in 1h 07m 25s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/9f7d9d86dee74ca6ac282adfb0dc6be6 : SUCCESS in 57m 18s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/706e01cbe2ac40688cf80d943d6ad189 : SUCCESS in 1h 25m 26s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1512450c54334d8fb2bb234d6a2c5993 : SUCCESS in 37m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b69487c90274f970017432e691ae1254d27e23a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-23 07:10:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f96ee2faf0dcbc5992021b7047e354b27b5129e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 07:10:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e413fd7789b6d0662b8a496b886778e832d8f76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 08:52:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a652ecd04084137a7b4701689b9145a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6575de47005c461482358ba79c7ec895 : SUCCESS in 17m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2b8b8cd551c24a1b9a102c6351e709b8 : SUCCESS in 1h 09m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7d65702d0b0d4c419245285645979af4 : SUCCESS in 1h 12m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5043ef3ae83478eabe57ba100b8ef2b : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1177f7502af940128a38c82069549ac5 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9612704a94e44e29b3c6b0a18876242a : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e0b82825bd34a4c9d250942ff242c6e : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb1bbec828d401dbf30f4f86b88475e : SUCCESS in 7m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/723b55d9d8654c8884f72d629b8dc962 : SUCCESS in 18m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a1ae2ab8a53b43eba1cdf67a877857be : SUCCESS in 1h 10m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/80d43e7369b54d96b0c2e2c2fda14d45 : SUCCESS in 1h 08m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b89c8dc97c13456590402b8ab5312ba2 : SUCCESS in 43m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f33d0588421645788a30bba3c224f461 : SUCCESS in 54m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9c91371b20bd41ccaf476e0bdd2e46ad : SUCCESS in 54m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ebff26d5762911218378a0ae2243dfeff5efaecb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 08:52:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2e86a5555cd0caca6e60dd6df401636ac1acafb9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 08:53:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/435242bc44134d50a25fa2faacd9acc3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/181b32f97d4549368b12ae7e340462cb : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1dba21d012884e2d101b635787c7233fb27b060b","revisions":{"1dba21d012884e2d101b635787c7233fb27b060b":{"kind":"REWORK","_number":1,"created":"2023-06-21 17:20:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/75/886675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/886675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/886675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/886675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/886675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/886675/1"}}},"commit":{"parents":[{"commit":"600ed4f149a1cf542f54d7fc5005030f53c69a83","subject":"rabbitmq: pin to 3.11.16","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/600ed4f149a1cf542f54d7fc5005030f53c69a83"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 17:20:03.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 17:20:03.000000000","tz":0},"subject":"Revert \"rabbitmq: pin to 3.11.16\"","message":"Revert \"rabbitmq: pin to 3.11.16\"\n\nThis reverts commit 600ed4f149a1cf542f54d7fc5005030f53c69a83.\n\nReason for revert: 3.11.18 is out with a fix\n\nChange-Id: I4659bcd48ea32294b3ee84af9af94fae32b38bf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1dba21d012884e2d101b635787c7233fb27b060b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1dba21d012884e2d101b635787c7233fb27b060b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
