)]}'
{"id":"openstack%2Fkolla~886687","triplet_id":"openstack%2Fkolla~stable%2F2023.1~I4659bcd48ea32294b3ee84af9af94fae32b38bf3","project":"openstack/kolla","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-23 11:18:43.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I4659bcd48ea32294b3ee84af9af94fae32b38bf3","subject":"Revert \"rabbitmq: pin to 3.11.16\"","status":"MERGED","created":"2023-06-23 09:12:13.000000000","updated":"2023-06-23 12:36:33.000000000","submitted":"2023-06-23 12:35:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886687","cherry_pick_of_change":886675,"cherry_pick_of_patch_set":1,"meta_rev_id":"4a00a62751c61703164fad59f0ad6c8302de2e24","_number":886687,"virtual_id_number":886687,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-23 12:35:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-06-23 11:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-23 09:42:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2023-06-23 11:16:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-06-23 11:16:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-23 09:42:12.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-06-23 10:32:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 10:34:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-23 11:16:33.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"850eac2d797f298aa5d069899865f75278d94b08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-23 09:12:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"933984396c9e723a16b50164406a3cdd90a3a2e5","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-23 09:42:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"64d61886796d3f5ad2228538bd1191b442072ce7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 10:32:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d16a30498eb94916a98229f7db458ff7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/dd28543d535e409db4d48e1a8c54be85 : FAILURE in 16m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b3d9b5ffb88b4254864e15a7a4f6dc73 : SUCCESS in 19m 31s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bdd56aace3e849a1ba3cf9a3b5b0933f : SUCCESS in 1h 19m 21s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7cdcda524d74a21b7c716044a2b6e69 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c7c34b3b6f14243ad37d694570c4ced : SUCCESS in 2m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5044c07814dd4625810457b2ba136df5 : FAILURE in 17m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/53ed82c73e8640959df922a2cfa5908e : FAILURE in 19m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65366c01d8a1f522f3ed50ad8e0b5b338b2f7a6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 10:34:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3249975afd84a929a2ab26827910c25\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e86679503ecb4b12ae6ef7211077e8fb : SUCCESS in 31m 46s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/22ea4e6b9d84467194181b0f4ed681df : SUCCESS in 1h 05m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b3414d812f8948dc89f544bb65a4fa11 : SUCCESS in 18m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0649b3756e204dfc92a21ec050578a2b : SUCCESS in 59m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d3675567cba44e1da2a9815e8b4e1394 : SUCCESS in 1h 01m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab787c3d925b4cf5a23be6cba5e0f6df : SUCCESS in 2m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06ff9c9cfdc446edbe5effdf65175eff : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81eb1bfbccf54316b3ae4656a5458d2e : SUCCESS in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1287ba2b037495eadf180fc6c44cc2f : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d1abc082c98465aab38f6678097815b : SUCCESS in 4m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a5b5ea119ea3472ea9a2953151033e5b : SUCCESS in 21m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5396fc14984944c1aa547b88de9aefb5 : SUCCESS in 1h 03m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/51863691841343a9ac613ad5a7b0b1d9 : SUCCESS in 1h 17m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7e479053d1ee47978ae79c80e3a6a2c6 : SUCCESS in 17m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8034a90862394339bc835ba0666f9ae9 : SUCCESS in 54m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3abfbe14ecee411b80ab6294fcdb2a78 : SUCCESS in 1h 06m 55s","accounts_in_message":[],"_revision_number":1},{"id":"86bacc177ee47b1aca96cf45c2ac985a5109a85a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-23 11:16:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"82c4cdeea29bed979144df1b8c100970054103ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 11:16:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"046a3708e09f7b7999289ed680e6c12ffe173e59","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-23 11:18:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"daf0248b6e4cdb7927bdf43fa3095ed81e3801dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 12:35:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab1d60d1365f43889e8319502cc4f8d2\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/56bfc30f1ac0408187146a21027f0e0a : SUCCESS in 26m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2576ac83760d40d8b1d0c34c81e69b83 : SUCCESS in 1h 07m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1a982919dfd24f26a312431dfa9e86a3 : SUCCESS in 1h 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdfcf0a49b134d0fa32f7e2ef76ec387 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f526eba08ff4de1b09cf62434d4d558 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d2eb6fc4a024825a59e028f598ea526 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7351ee148db04865a7cda5c7b9c7bae4 : SUCCESS in 5m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6fd193c9ed2c4de49b30a901be7101a5 : SUCCESS in 20m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/88ccd1fb8487400aa0b731e52e3490c7 : SUCCESS in 56m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c2537bcb5fcb4da8bd2874e56d333aa4 : SUCCESS in 1h 12m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a1d371229c77488f9d1aa0405585e009 : SUCCESS in 26m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5a34d68ba26342fe801c9c2efdc6397d : SUCCESS in 1h 06m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fac65474ac9c43629aa563a7394df099 : SUCCESS in 1h 09m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ffe03c518e2bcf0ab8d611a2b1a74d3fbb5a48d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 12:35:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4a00a62751c61703164fad59f0ad6c8302de2e24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 12:36:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4e4cb333cb847cb9455ac42f24063d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d86d53cc414433ab1b5a0433adfaf3 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44a81974e609f1c4054563c0f2cb6cded7731826","revisions":{"44a81974e609f1c4054563c0f2cb6cded7731826":{"kind":"REWORK","_number":1,"created":"2023-06-23 09:12:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/87/886687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/886687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/886687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/886687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/886687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/886687/1"}}},"commit":{"parents":[{"commit":"b9d1b69e04498c88037bea80e4a2809f1dd0091e","subject":"Merge \"rabbitmq: use erlang-25 ppa on Debian/Ubuntu\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9d1b69e04498c88037bea80e4a2809f1dd0091e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 17:20:03.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-23 09:12:13.000000000","tz":0},"subject":"Revert \"rabbitmq: pin to 3.11.16\"","message":"Revert \"rabbitmq: pin to 3.11.16\"\n\nThis reverts commit 600ed4f149a1cf542f54d7fc5005030f53c69a83.\n\nReason for revert: 3.11.18 is out with a fix\n\nChange-Id: I4659bcd48ea32294b3ee84af9af94fae32b38bf3\n(cherry picked from commit 1dba21d012884e2d101b635787c7233fb27b060b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44a81974e609f1c4054563c0f2cb6cded7731826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44a81974e609f1c4054563c0f2cb6cded7731826"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
