)]}'
{"id":"openstack%2Fkolla~886741","triplet_id":"openstack%2Fkolla~master~I2ebc8b0e35c67bf893deb89ea7fa0146dedd4640","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-09-11 20:04:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2ebc8b0e35c67bf893deb89ea7fa0146dedd4640","subject":"build: distutils.StrictVersion() is deprecated","status":"MERGED","created":"2023-06-22 11:40:38.000000000","updated":"2023-09-11 20:05:04.000000000","submitted":"2023-09-11 20:04:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886741","meta_rev_id":"f93f606259aa3e8d562bf1954007ae0088c24d59","_number":886741,"virtual_id_number":886741,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-11 20:03:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-09-08 08:57:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-09-08 06:34:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-09-11 18:19:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-09-11 18:19:14.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-22 11:55:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-22 12:15:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-08 06:34:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-09-08 08:57:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-09-11 18:19:14.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"774d8d184923669e93d7de4603430a2fc70278ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-22 11:40:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a94e43ccb2c4dfeacedd2e105ce458b4a83aded5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 11:55:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dcb22521aa647f6b93f86fb1394d40f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/11bc1e5377054b618cf9d77053157707 : FAILURE in 3m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/73c42d5c53cb4ff6a0de6f5c7602c0ea : FAILURE in 2m 41s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ec6f4a366bb547948ac1070fcf333b79 : FAILURE in 11m 46s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a943656097d74779b924d2bc7cd6d972 : FAILURE in 2m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75cf6a0531ad408a923cfd2a26ad9f5d : FAILURE in 3m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cdf3c16a2491435fb9f26783dc3ace50 : FAILURE in 3m 34s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ff0125b6ea7246829c36cc35f06fdba0 : FAILURE in 3m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8ee5bea45f7f4feea51e130aadd47f60 : FAILURE in 3m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06c5e23b334e8e5d34023f3342f097acf27a315c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 12:15:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af51c6591098481c85ed9c528a84e90f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/81a32ce7d45b4a489f793f496a2b94a0 : FAILURE in 4m 53s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/323b4c4c966d45d784b2009166a0914a : FAILURE in 10m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5dd153f034b44fd1a011fbc7a6523d9f : FAILURE in 4m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/387e0f5f704a42a7ab7db6b4d04f0f62 : FAILURE in 13m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/113b06e6f08b4818a18d92a77d13ea72 : FAILURE in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98160fc1d5644f289999d9a79ab69a0d : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28595663d87845d2af0dc0670f295074 : FAILURE in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d39a785d9acf4a9d9ad61c28b4fea6f2 : FAILURE in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e3e5dc08bb84413afd8ebbd75ebe4ac : FAILURE in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f887c3ee266948599c93c8db9b0fca6d : FAILURE in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f84df84dd6144bbb8eb9f81b0589c883 : SUCCESS in 4m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c530909b4c1449c1ac109f65fe58a811 : FAILURE in 3m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5a7dcadd64e44d849c5f375a3989aa81 : FAILURE in 11m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cc478b85b53f4bfa96b216d056797c29 : FAILURE in 9m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3fe5ea0bfd3a43288fbfafbfb2128346 : FAILURE in 2m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/294dfe3e17ca47248b747125492f2fa5 : FAILURE in 7m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/edd1f8eb84ff4a48a4c8f5422e169d40 : FAILURE in 7m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5b087825572d8418b0dcbb61bd98aa409a1d6551","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-28 16:40:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ae0d95e0d344280a15f73362a202eba11b64ee7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 17:13:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7ae9f939cb74f039844c1c0b036a6db\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/886bad10a6cb45ea8559d05975b21e80 : SUCCESS in 24m 37s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/720aec2d82734c9988a067686d8de97a : NODE_FAILURE Node request 300-0021566830 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/58ddecfbea294a7996e08f411b2ec5e5 : NODE_FAILURE Node request 300-0021566831 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/708c6bd1ac10467e95b1b14d07605817 : NODE_FAILURE Node request 300-0021566832 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5204cb68229e49c9b5c72dc96b49beb8 : NODE_FAILURE Node request 300-0021566833 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45cd90408f634a02abb0ba5966dcee45 : SUCCESS in 3m 16s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ee808d04606a4ab3a9469af294b9c0ee : SUCCESS in 31m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b057ad8805ae42508e75d592f710e74a : NODE_FAILURE Node request 300-0021566836 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7da750786cef58f57c048a6d943bb15f44128554","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 18:27:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/671d78ee04cb4c87a0d7ba12625e8d22\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bce1a45fb5b74a34b7d9daa8d52aad4f : SUCCESS in 26m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8bd0677ffa944995baf98b41a42a2b5d : SUCCESS in 1h 25m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ea92a9987cef41cd9b00ced8c54d44cd : SUCCESS in 24m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8949f61b68c94fddb45e4e3110ae0844 : SUCCESS in 1h 20m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fcf088fb0a2449ff89c3363682f7f7ff : FAILURE in 27m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b0e54f785ca49ddb128f26cfc45aa77 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff0f82d58c1c4406b4d4e33ef1889770 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d4a9b7165d744a889dfd399a1758503 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79db397323e3465aac9fa0a5602f6076 : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25dc34ab76fd4fc49f114d7504de026f : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b9c1d7e74c34bc085b7b931b61bf2c5 : SUCCESS in 5m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/27ce2d634b5640dfb60fd0fa2fb95165 : SUCCESS in 18m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/477dd29a56994514af0c85afd5d31c42 : SUCCESS in 58m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b9d7d2036f794284855bb79627814621 : SUCCESS in 1h 18m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0cf16dcdcb6e43eb980ffad9f13fc25e : SUCCESS in 24m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f14c92b098804d449ab0ee204c4c0319 : SUCCESS in 1h 17m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6867ccea992a4c14bfe41d7ef082db36 : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":2},{"id":"7d122476748a11520936fc86641c39e315161d4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-09-08 05:45:27.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a861034d81de7f74b7f3058c0e8ad17c032005af","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-08 06:34:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fb4321abb0a1193be53e948dd528efc4addc1b51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 07:02:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/291069f8bb434298b9b75fb9245964f5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ad85b8ec4c3f444b8f2288b8ebc2510b : SUCCESS in 28m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/10015596ca4447d1a4d7c27a5ce021f1 : SUCCESS in 26m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f303973bd5d04564801271a4579093e8 : SUCCESS in 1h 16m 08s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d9d1d63b42f41bca51008c6aa83d7db : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/acc124dc0ac646718e9a8474ad394900 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/540eb5eeda5444f4b1b1c4da043612a7 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2426c73d6fb34825b0a55a761ada7acb : SUCCESS in 4m 13s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d58756abc2dd4c8aa3d9b336e04b34f8 : SUCCESS in 58m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8eacc3023e4f43a5a747a7cb6b9c6c74 : SUCCESS in 26m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2baa47c37f482be9095241454bc99c00c7b0d450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 07:20:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5da1aa108e7744ec9d623a1be448ffd7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e1db054b3d744d73a6f9533fc939e45b : SUCCESS in 21m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3724d2ca8a774ae0b0cc3de016d5bbc8 : SUCCESS in 50m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/709d743658e8426399e0cb8af3695fac : SUCCESS in 20m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/06a11f6d885942d1bdf42233a920f5cf : SUCCESS in 1h 04m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a577ffa8560e4f2c9b134ed58ba21345 : SUCCESS in 1h 17m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31c76bde86cd4cc19e0a47b8f43f195c : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77b53bd3ed664204b60c61ff22a4eabb : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae4a972943084516a30a311e7d5592a3 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2da669e9ab7348f480e22e07e17b804b : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40c81ae620fd4e1e905955a7d8b20511 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5a19e17e9274914b6894c8d7971a0c8 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b449723a2f2040589fd3a66af31a6a57 : SUCCESS in 6m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2c0585b576ed4a27ad1b6ed7488ea785 : SUCCESS in 18m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/836e56e3a9f54a50a20e38cbad6115a7 : SUCCESS in 1h 08m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6a651fc84dd44b78bd85981823dfd755 : SUCCESS in 1h 28m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/965b0283914b474798132a80023332f6 : SUCCESS in 17m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ebf326fcab79481c991465e9242acbc8 : SUCCESS in 1h 03m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/48a6ed9df0f94e5cb641a444f23b851a : SUCCESS in 1h 08m 41s","accounts_in_message":[],"_revision_number":3},{"id":"fa83632bbb4805b81b49be5515248aba2f9510b0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-08 08:57:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cfa0922376b8cf77e3786b8b0602d8455d2dc7c5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-11 18:19:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"685789b5c845b5a4959735286e7ad0a5fffea869","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 18:19:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7ef732276e3a65f1507ba36d38ebd833d1dd0b57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 20:03:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acbc9a7969b64ae190b9eff4692b5917\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bf8508c92b1f40d0908b5627681e0fd1 : SUCCESS in 25m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/39c43fd2a0f445e790c8e6f4f20e7ef0 : SUCCESS in 1h 06m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e3d7eba4988b43a98acce6c76957a249 : SUCCESS in 1h 27m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f4cdb743b574488a779f156256b4984 : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e222d493593430db475c5a2022f4be8 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7173b3be18f14897be9d6ac2691ff323 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19b41c91085f46fcad02f47e2e470d0a : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1c9266a782b474d83b28f04165d6935 : SUCCESS in 4m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/55ea80768e184bda8a73e4083051a8b0 : SUCCESS in 16m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9770eb73352f4b3381edb30b4d837b9e : SUCCESS in 1h 04m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b7a93d8b32bf45ad983cda1849eb5364 : SUCCESS in 1h 38m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fb45839b3a784ff099d3782fb2e58380 : SUCCESS in 20m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7a1dfb4dd9924f0598895d84d8fc6ead : SUCCESS in 54m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fedbd8d3ec4c40eaa685a388980e5fe9 : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":3},{"id":"522ec95cae3833ccd93b636d81cc8335de2a3ff4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 20:04:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f93f606259aa3e8d562bf1954007ae0088c24d59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 20:05:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9232a8ea7294da19576a520ffb2cb8d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21cce569e26c4ae9a3bf27f23c777dc4 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fd5ec1ca18eeec65e4d0b6d345a3232d2159ba51","revisions":{"b4595156d4f19f0a67dab9bfee65526f2cc89371":{"kind":"REWORK","_number":1,"created":"2023-06-22 11:40:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/41/886741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/886741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/886741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/886741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/886741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/886741/1"}}},"commit":{"parents":[{"commit":"427c8b66aafe65d051afd15f5cb82b60298860d5","subject":"Merge \"rabbitmq: use erlang-25 ppa on Debian/Ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/427c8b66aafe65d051afd15f5cb82b60298860d5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-22 11:39:59.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-22 11:40:32.000000000","tz":120},"subject":"build: distutils.StrictVersion() is deprecated","message":"build: distutils.StrictVersion() is deprecated\n\nkolla/tests/test_build.py::MainTest::test_run_build                                                                                                                                 [65/1834]\n  /home/marcin/devel/linaro/kolla/kolla/kolla/image/build.py:117: DeprecationWarning: distutils Version classes are deprecated. Use packaging.version instead.\n    StrictVersion(docker.__version__)\n\nChange-Id: I2ebc8b0e35c67bf893deb89ea7fa0146dedd4640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4595156d4f19f0a67dab9bfee65526f2cc89371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4595156d4f19f0a67dab9bfee65526f2cc89371"}]},"branch":"refs/heads/master"},"bfb0777f612056204bca77cfa55525dd2be4d5cf":{"kind":"REWORK","_number":2,"created":"2023-06-28 16:40:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/41/886741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/886741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/886741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/886741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/886741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/886741/2"}}},"commit":{"parents":[{"commit":"70f74eb64101431e23d56c6a7df96d7aab37ce2f","subject":"Merge \"Revert \"neutron-mlnx-agent: pin pyzmq at 21.*\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/70f74eb64101431e23d56c6a7df96d7aab37ce2f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-22 11:39:59.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-28 16:40:34.000000000","tz":120},"subject":"build: distutils.StrictVersion() is deprecated","message":"build: distutils.StrictVersion() is deprecated\n\nkolla/tests/test_build.py::MainTest::test_run_build                                                                                                                                 [65/1834]\n  /home/marcin/devel/linaro/kolla/kolla/kolla/image/build.py:117: DeprecationWarning: distutils Version classes are deprecated. Use packaging.version instead.\n    StrictVersion(docker.__version__)\n\nChange-Id: I2ebc8b0e35c67bf893deb89ea7fa0146dedd4640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfb0777f612056204bca77cfa55525dd2be4d5cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfb0777f612056204bca77cfa55525dd2be4d5cf"}]},"branch":"refs/heads/master"},"fd5ec1ca18eeec65e4d0b6d345a3232d2159ba51":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-09-08 05:45:27.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/41/886741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/886741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/886741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/886741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/886741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/886741/3"}}},"commit":{"parents":[{"commit":"67596360587a35e8f07d5da5ea678e97fd3c135c","subject":"Merge \"Use latest tagged version from gnocchi again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67596360587a35e8f07d5da5ea678e97fd3c135c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-06-22 11:39:59.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-09-08 05:45:27.000000000","tz":0},"subject":"build: distutils.StrictVersion() is deprecated","message":"build: distutils.StrictVersion() is deprecated\n\nkolla/tests/test_build.py::MainTest::test_run_build                                                                                                                                 [65/1834]\n  /home/marcin/devel/linaro/kolla/kolla/kolla/image/build.py:117: DeprecationWarning: distutils Version classes are deprecated. Use packaging.version instead.\n    StrictVersion(docker.__version__)\n\nChange-Id: I2ebc8b0e35c67bf893deb89ea7fa0146dedd4640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd5ec1ca18eeec65e4d0b6d345a3232d2159ba51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd5ec1ca18eeec65e4d0b6d345a3232d2159ba51"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
