)]}'
{"id":"openstack%2Fkolla~886832","triplet_id":"openstack%2Fkolla~stable%2F2023.1~Id6db7ffb22b3f63e9a406455d52078b76071672f","project":"openstack/kolla","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-26 10:43:26.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Id6db7ffb22b3f63e9a406455d52078b76071672f","subject":"opensearch: drop unnecessary workarounds","status":"MERGED","created":"2023-06-23 16:00:43.000000000","updated":"2023-06-26 14:17:07.000000000","submitted":"2023-06-26 14:15:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886832","cherry_pick_of_change":886467,"cherry_pick_of_patch_set":4,"meta_rev_id":"56aec8470bb0c458af86a47efbe213ea0cd17f87","_number":886832,"virtual_id_number":886832,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-26 14:15:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-06-26 08:17:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-06-26 08:18:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-26 09:35:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-26 08:18:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-23 17:17:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 17:24:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-26 08:18:22.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-26 09:35:06.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"be56f44c769cce516556059b1e0a49055b04c32f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-23 16:00:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a46aef73e3c8ea2b66160fea2f831a7c379229d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 17:17:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/669f61b9a6424683a3ca6bced86f2fc5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/095d9309434c4678ab8569e2f84ff7b7 : FAILURE in 29m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/937901c0e0e349359b711119dc825168 : SUCCESS in 30m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dd0e7df5a84f479fa609392404a3ac8a : SUCCESS in 1h 13m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a63d8e5025204f8d868c20b6c1ed47ba : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7d9b478ea0434675ab75167859aaa303 : SUCCESS in 2m 55s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/95d502d2183d4f258f62f1fd48227a98 : FAILURE in 28m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a62b286108f8422cb133b02337aa9cf5 : FAILURE in 24m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc140a47fc7d7411708bd956e1212d289853fd59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 17:24:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdf4314c48a246a39355604f989fe10f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f66f8f8556fe4ff2b6cc35238db29e4c : SUCCESS in 23m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/36100910045f4c1fa609004aab3f1d06 : SUCCESS in 1h 12m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/66d9ff2fe8b342e49d8d6158cb131107 : SUCCESS in 13m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1a92e42aa1e6411c9931b33e36e5e72a : SUCCESS in 55m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/97ae1c9d0bf84a5b9383896dc3d88656 : SUCCESS in 1h 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69b14d4c1d9842b39318729e54a8eb7f : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41afb8026e9d4e5faac56b3004c30191 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bdf31c80905b48128f72592299fff463 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92fc16f5f3e74ddab747c0c707518132 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6baacc05d2244a6a2c310782a40646b : SUCCESS in 3m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5389d988f35b4ee89fdea7bebcdbd0f2 : SUCCESS in 21m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9a9a924f513f469ab1572a64813c518a : SUCCESS in 1h 07m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e53697e623d74be68f3bfd559c5bf4d4 : SUCCESS in 1h 16m 49s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5f02170404d84af5870f49f92c2295da : SUCCESS in 20m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0e65552397f6484dbd1dcf74059b78d0 : SUCCESS in 58m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b72beb9ec160464b9f2bd31134096d9b : SUCCESS in 1h 00m 34s","accounts_in_message":[],"_revision_number":1},{"id":"bf3301a28f35d688e05bd4327cb31f33525ddd1e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-26 08:17:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"10dba6a2c89975eeb4760b37fef77d03e7e2a791","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-26 08:18:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"79d5fd8c7909bfed50dd74f6c349c9cca025740d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 08:18:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c685a8ae8491537c8b9cb98375fd4c5188db8e9","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-06-26 09:35:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9a949c988b22fc17391acff766d1dbce87709480","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 10:37:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8ab823ab6444255bf9fd2981b1d8f94\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/69984150bae64d63a187b741f887ab6a : SUCCESS in 19m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2771652ea91c4b819782032534fbba4e : SUCCESS in 1h 01m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bf1c4266d71441908b006afd40849e0b : SUCCESS in 1h 08m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8c17421d14c42c38f7399f98890fba0 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/483785d777c34a09b1f66667fc6b2c5d : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df232a0e6e55436186c45a9194f1edb6 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f7bdd7e5fe4d8b9e6d88daf161f542 : SUCCESS in 6m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4606d94d5c784a98a2f0128ad98bbaa3 : FAILURE in 19m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/164f9064f8434e7f9629aec380637280 : FAILURE in 18m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0ddd44c755e345e890556f9ec3044660 : FAILURE in 17m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/77cbbc03b28d4092a40751e810692b77 : FAILURE in 1h 14m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/06c14f3d923447d6aeda6c0b1a67c337 : TIMED_OUT in 2h 01m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3da04c1192844f2c9c4f31451ad98216 : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f73a58efc1d685ae0e898404bb3cbb8c3fe391d1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-26 10:43:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d041242af006d89f7cda42727f37cdbb0e6e1ab8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 12:20:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1066417443a14ac0acb0e7bf449152e4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cf6b475b395445f3aa198ad3f3159728 : FAILURE in 1h 05m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/98067969ccf1414a9c20f371f5e1f717 : SUCCESS in 34m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/af9ad39e0d3141b98420e20c1455712d : FAILURE in 1h 32m 01s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/972d8f7a58ba458caabcddbae3430484 : SUCCESS in 2m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8608c2490c46473ab756d650e76d01a9 : SUCCESS in 2m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9d4b81cf60cb45129c5b08b3571611e6 : NODE_FAILURE Node request 300-0021543341 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6107070ffdfd4df583eefb36ac5f5338 : FAILURE in 27m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2eae56cf609389083b02e1bf299e86eeeac1002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 12:28:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5e1ac8592ae461a89180cadd5763411\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d79f4a10b5ba4ca8bb6f629e0c5a2dba : SUCCESS in 26m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/29fadbba06d84470bb7e95f8feba20f9 : SUCCESS in 1h 19m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0f51cc9775584bbcb3dc436dd8d512de : SUCCESS in 22m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a8ab374066384eecae90d6dd557c0e1e : SUCCESS in 1h 02m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8413f3273e9a4f95a33d35c97b6de4f4 : SUCCESS in 1h 09m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca5f00d3d4484e038e76a8a47a6b582e : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e25d76ee02bd44c7a88772198bcc8968 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce31f2175ad5417e920f4ca796b869d7 : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46e05f1790914d6182b19f32d951ec6e : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b0deca1a86748a88aa130df03553b46 : SUCCESS in 4m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/756415bee3794ef896ab01dbfcb39921 : SUCCESS in 24m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e9756259cf564bc7b0f652e0ea4d2589 : SUCCESS in 1h 18m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4e02e9d610c1418d999e074f5dde0981 : SUCCESS in 1h 26m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e9439a2bc2ca4117922eb480270a243e : SUCCESS in 29m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1bec9b2359f6428fa5e3a6200d617f09 : SUCCESS in 1h 02m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b3a8eb93a8074ecfbb29748e0f4fd355 : SUCCESS in 57m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d54a060ec898f79c3b1a4394db9b56c59e46a9d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 12:29:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7bf795efad01f69510592a56f96dda676b226680","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 14:15:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/267d18434c5b4c34a01a3bda02357165\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9549f416c73141cc9cdca79bc7171577 : SUCCESS in 19m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/258a665dce7941b296157b240b84c43c : SUCCESS in 57m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/748b15b1b4d24b74b52003186e8e57e4 : SUCCESS in 1h 33m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b74aea845c064ff4895b48947f25e6f4 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e31f1bcffb924d7899887ad75e7202c0 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92e99b4c078a4e869a773af08c0ad52b : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f7b189cbaa04d8bb7b5f9a497dc2508 : SUCCESS in 6m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cdc8229007bf487ebde53ddbbb46c037 : SUCCESS in 43m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4d14dcf7a219499f901af755f4ee4d7e : SUCCESS in 1h 28m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/70b94462623343a7bf32102e44179b91 : SUCCESS in 1h 37m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7b7f123926bb4417b6c9eefae753c44e : SUCCESS in 18m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/85cf6fddbe1d407e8bb43524b0f6554e : SUCCESS in 1h 09m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3aa643565a6f490f998043a4f1af86b7 : SUCCESS in 1h 12m 32s","accounts_in_message":[],"_revision_number":1},{"id":"81682888351ece4fa44c58d6668689a3a2c98a79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 14:15:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"56aec8470bb0c458af86a47efbe213ea0cd17f87","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 14:17:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e24e01cea6940a9963fa6ce1ab9c82d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcf37eb8aaf144a0b30ab5367dd7de50 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f43a0ed4ea136d18f8415495b9adc70e96b92c41","revisions":{"f43a0ed4ea136d18f8415495b9adc70e96b92c41":{"kind":"REWORK","_number":1,"created":"2023-06-23 16:00:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/32/886832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/886832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/886832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/886832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/886832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/886832/1"}}},"commit":{"parents":[{"commit":"44a81974e609f1c4054563c0f2cb6cded7731826","subject":"Revert \"rabbitmq: pin to 3.11.16\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44a81974e609f1c4054563c0f2cb6cded7731826"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-20 13:20:38.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-23 16:00:43.000000000","tz":0},"subject":"opensearch: drop unnecessary workarounds","message":"opensearch: drop unnecessary workarounds\n\nChange-Id: Id6db7ffb22b3f63e9a406455d52078b76071672f\n(cherry picked from commit b33167972d6e17b2eaab85e72043e3b4efa4b789)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f43a0ed4ea136d18f8415495b9adc70e96b92c41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f43a0ed4ea136d18f8415495b9adc70e96b92c41"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
