)]}'
{"id":"openstack%2Fkolla~886833","triplet_id":"openstack%2Fkolla~stable%2Fzed~Id6db7ffb22b3f63e9a406455d52078b76071672f","project":"openstack/kolla","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-26 10:43:34.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Id6db7ffb22b3f63e9a406455d52078b76071672f","subject":"opensearch: drop unnecessary workarounds","status":"MERGED","created":"2023-06-23 16:00:53.000000000","updated":"2023-06-26 14:01:18.000000000","submitted":"2023-06-26 14:00:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886833","cherry_pick_of_change":886467,"cherry_pick_of_patch_set":4,"meta_rev_id":"4b5e785a68eb92d506b18ae7d20b6d5134528aa4","_number":886833,"virtual_id_number":886833,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-26 14:00:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-06-26 08:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-06-26 08:18:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-26 08:18:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-23 16:33:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 17:25:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-26 08:18:27.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"f88f573e08b3e1354f411bb8fa0da4b06e650285","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-23 16:00:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b255b93b185c0e413f0054cba6fa5530737c8a28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 16:33:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ac5ca9fba1f4ea79ddb4ee832f3dd12\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cbfc49ad074f4ee5b075d00d28e2ad60 : FAILURE in 26m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/294479f8e2074a74bdfa84bb6a89124b : SUCCESS in 30m 18s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b9ef9a0962c4ecb9d1a4880efdc6f3b : NODE_FAILURE Node request 300-0021528748 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b3f94be7434438d804f13ad69d430d9 : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d0f5ab61a234894a35b99fc5fd34b2b : SUCCESS in 3m 14s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/81d53cece5ae4500b43aedb12403a0e4 : FAILURE in 25m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9983ea0fdb9c47bcb438c2f87ab7f8e5 : NODE_FAILURE Node request 300-0021528752 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52b51a37d2cfcb3d573e3f142dc2ba66740f5b61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 17:25:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61553b9db5f04d29a9d793a6ae0c63f2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/647150f8bc134eb8977081b04753bd84 : SUCCESS in 32m 58s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/3b6b10e7283149e7ac7b4699312ab262 : SUCCESS in 58m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c3b60fe3b4ca4b7db1691b41bb617d60 : SUCCESS in 20m 29s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/04344dd231654aa1ab1ec69fa7e53c4e : SUCCESS in 1h 01m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5503b3fc952e4fd4b52ef02a1cba65f2 : SUCCESS in 1h 19m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b40c64d8416459ea83f3fc9b1a761e7 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5ee993fc27f4b83ac6114ca84d35029 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e55fe208c82422988778ef090b70438 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99115fbe73ef484faa0c5eeefe2c3b09 : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e6760441594204852334d47f774bd0 : SUCCESS in 4m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b84970df1f934c7197b01b28c0ec2040 : SUCCESS in 23m 25s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/07d27fb7c09c40caac47dcf5cb5f6e00 : SUCCESS in 1h 08m 26s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ab781e9e98d54516aec344fb3e8b823d : SUCCESS in 14m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e4aa9cff47fa4b50813b7777fa567b06 : SUCCESS in 48m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a2e7416621134afaa3a008c6ad4dab57 : SUCCESS in 1h 10m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1fa7fc1a7f5a5c1b5b1863b782ecfac61a2b64ec","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-26 08:16:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4634eac258a96505bce2fa22581385dc532bd4c6","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-26 08:18:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9cb5d255ef61dde443570c2ba8bf7f10914b8a13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 08:18:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed94e0aa8e7433d74ecd51ab7565071dd1e80cd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 10:36:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca022f43b0194b1a9af400e123e8e85b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7ea46f7655b6491d83b7ef1ee4044514 : FAILURE in 31m 45s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/54b32f3f27a64463ba2d452d1c9ed88e : FAILURE in 27m 32s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2599740909e94f128f244000ba5debb0 : FAILURE in 28m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/132341b962814833b0a179cada0b22ca : SUCCESS in 10m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37eb628587124ad2b4d43844bfacb57d : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a103e9613f4344be9d8d8e2b027aedfa : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf3ee75368ca498999b0ab05c4a0f4d5 : SUCCESS in 4m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e936993b662948b79ff84988265e0d36 : SUCCESS in 1h 38m 31s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/743f70da605348e39c8da3ed59d8af55 : FAILURE in 22m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/144506fc83734753afc144228efdde0c : FAILURE in 44m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/011e80de82054f8187e4806e4ccb9336 : SUCCESS in 46m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9b58d69407974b209e51ea8721a17ea6 : FAILURE in 1h 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0d12c8e08cb11bc88cdd24e3db41a178c3c8baaf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-26 10:43:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd2ba95694e789ec861e1d66746698f018c8c029","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 11:20:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cf96093e17d46f093c0b3d21538c0bd\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3becf84f15de46cc914d31ae8ab75fd7 : FAILURE in 32m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/24386bd0f4f34ed19f0ee95e8c82b072 : SUCCESS in 29m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d068754eba404083b75d8a950ace73a5 : NODE_FAILURE Node request 300-0021543361 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a8fd1cd1e9d44059a1e81a0755b0a6a : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5abcb0f3fbde40ff88897252d08b8acf : SUCCESS in 2m 47s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e0c74b12f03a421f8e125a214589880e : FAILURE in 30m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/890b8e5362e84ecd967f41db985d7d66 : NODE_FAILURE Node request 300-0021543365 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72883d00bd579fe13ed38bf34a461d2f5a04ccd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 12:22:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b934764098e4bf3bbb358df20052ffa\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1f21d4d928d5497eb18c81365ca76073 : SUCCESS in 33m 36s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/06f39d9624704d44968435436292d020 : SUCCESS in 1h 18m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d8e7306064d549b487aca4a1374b2709 : SUCCESS in 19m 54s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/18ff9c4650db4c3aa88ae48ed96c3a9a : SUCCESS in 1h 11m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/15c4a0bc985e4652808b548473282a76 : SUCCESS in 1h 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/306bea6661534abe951087fa5a5cbedc : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c44152c8d144ddda735ec111a1f94fc : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c22a464d4c114999a9b3068f2cd2c344 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8801e99d251d44c09fd4a835a66e3aaf : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0871b2987f44124bb7649f942cdb3c2 : SUCCESS in 5m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0f5d9ba70a824b38b61a663bc490aba2 : SUCCESS in 31m 53s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/4b1b121bf69c44c5995fd8807374a2f6 : SUCCESS in 1h 17m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e3344b9e6f4c47d7a215d8d5b9628ef1 : SUCCESS in 27m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/613224b7f56340f98f06e5851ba849e5 : SUCCESS in 54m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aeb373d8707f4bf0be81ac0fcd32228e : SUCCESS in 1h 08m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d9dcc8e7978c137413387f85a10b062bf770b5f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 12:22:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fdf9908a2263cb48c2210d4d6547c748181b75b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 14:00:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ba439511c4144a7a537a6bd4c591e52\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7e5b56b508c84688b9d9f3b308b78af5 : SUCCESS in 16m 57s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/55f72fd9c2b2492cab8960f6ed45643d : SUCCESS in 1h 25m 01s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e5794f74cbd744b8a9a24be44ee368be : SUCCESS in 1h 01m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e0a879e54db4d22b2036ac7b5fc0ba5 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/313e28fc5fb74e7095dc7f130cedf08b : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dadaef8cd6749a5853ab7c08eb797af : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/162f5eff4c8f4e91bc1f258cd4ecb3d8 : SUCCESS in 5m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/674d44737a8b4e778829fed69e7cce34 : SUCCESS in 54m 36s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/2516492cea094b67a61a1c7422fb22e8 : SUCCESS in 1h 28m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fce41274a24f4b0aa1bea17a5781fa87 : SUCCESS in 45m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/73dbbee031a7498fa4bf95f475ced05f : SUCCESS in 1h 17m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/96f0d0de7e0a43149b0bd1d2c7773636 : SUCCESS in 1h 02m 23s","accounts_in_message":[],"_revision_number":1},{"id":"22e6dd50d89bce1d87ca71d469d89236a04654f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 14:00:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4b5e785a68eb92d506b18ae7d20b6d5134528aa4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 14:01:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb46d2f2fe824063b32bb4175fdb403b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35932a92095e44379d1cc61fe0592b69 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"530490a7153b8189b24e4cb8f3acb96a5f40d1c1","revisions":{"530490a7153b8189b24e4cb8f3acb96a5f40d1c1":{"kind":"REWORK","_number":1,"created":"2023-06-23 16:00:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/33/886833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/886833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/886833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/886833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/886833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/886833/1"}}},"commit":{"parents":[{"commit":"ee8c40c860b15977037e2fd79c5ceeeb2afe9a74","subject":"Merge \"rabbitmq: use erlang-25 ppa on Debian/Ubuntu\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee8c40c860b15977037e2fd79c5ceeeb2afe9a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-20 13:20:38.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-23 16:00:53.000000000","tz":0},"subject":"opensearch: drop unnecessary workarounds","message":"opensearch: drop unnecessary workarounds\n\nChange-Id: Id6db7ffb22b3f63e9a406455d52078b76071672f\n(cherry picked from commit b33167972d6e17b2eaab85e72043e3b4efa4b789)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/530490a7153b8189b24e4cb8f3acb96a5f40d1c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/530490a7153b8189b24e4cb8f3acb96a5f40d1c1"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
