)]}'
{"id":"openstack%2Fkolla~886964","triplet_id":"openstack%2Fkolla~stable%2F2023.1~I9f40167fed9412b3f33f82e2f97f353275431292","project":"openstack/kolla","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-30 08:20:25.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I9f40167fed9412b3f33f82e2f97f353275431292","subject":"Use erlang-25 from copr on aarch64","status":"MERGED","created":"2023-06-27 12:02:03.000000000","updated":"2023-06-30 11:09:41.000000000","submitted":"2023-06-30 11:08:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886964","cherry_pick_of_change":886948,"cherry_pick_of_patch_set":1,"meta_rev_id":"41e4deffc5462973c49bf397095c69579cd24e18","_number":886964,"virtual_id_number":886964,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-30 11:08:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-06-29 07:49:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-29 12:26:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2023-06-29 07:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-06-29 07:56:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-27 13:40:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-27 13:43:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-29 07:56:25.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-29 12:26:37.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"21d29107045df3cf6af388492bf79f8761f2e68c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-27 12:02:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f8049e89ca90da7d09fecb4024d8c6fb5b9b10b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 13:40:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b158a08e0de44604b41f3928ca48083b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e123ccfc45754279a4d0a1f613403314 : SUCCESS in 48m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9499871b4fd840c08a74008be6b2d504 : SUCCESS in 49m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f829139376064f41ac1cc8e631b01023 : SUCCESS in 1h 35m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e8a50ceecd04b89b858be147d76b1f6 : NODE_FAILURE Node request 300-0021553498 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bead40cb84a04b7e9092d3dcdfa131e0 : NODE_FAILURE Node request 300-0021553499 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0c47dc424b1444bf81248a4611fcde5f : NODE_FAILURE Node request 300-0021553500 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/02ae5c213e5f4b7b95b49d0f6f474c8a : NODE_FAILURE Node request 300-0021553501 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c052074b35444e802b349dcde0e8551b3e4cd43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 13:43:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e96c93cb707e4696a5c48b62d30fa169\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d49673cda01d41ec969d467b3f9d6112 : FAILURE in 49m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/cae3ea1f178c47eea91832a69b8f6096 : FAILURE in 42m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f10a1446b0b5434e99d9a3d745b4aaa8 : SUCCESS in 19m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/eb818dcdf40c40c3b444dbf15d586d67 : FAILURE in 29m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/87fac6afb0b2420b95625f0f8c441e4a : SUCCESS in 1h 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f0ec12c3e35459d9e5647abcef3c3c1 : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e289c55b3341465186b4d0b88e205144 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/219ddec0362a461e9fd4e104780c8de8 : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a66bc125bb4a4ee1b71dd9eeb9219fef : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15e214ded5184b9d809ee0f35993cade : SUCCESS in 4m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/859fbfce984a42acba2f3c826ab544ec : FAILURE in 14m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dd55dd71aef8480da30f54168214f54a : SUCCESS in 1h 00m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a2b6d1111bef411fba562c0cf412bff4 : FAILURE in 24m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8206e28c3ba942fe89635da6b1fb7737 : SUCCESS in 33m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3b281f82eef240b7a6e4acdfec5eff34 : SUCCESS in 1h 02m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/05429f04450e49008bee2e74132eb577 : SUCCESS in 1h 14m 43s","accounts_in_message":[],"_revision_number":1},{"id":"7f0163984d0adceb15a2417b68902851bd6e1f02","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-28 19:51:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"541ddf4df9f6ff933c3f37a52fb7ec7fb141021a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 19:57:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daab836371164fe5acadd87ebf1c4928\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8b2d4083a9e5415faa978229146dccd5 : NODE_FAILURE Node request 300-0021568562 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7124ace350e844c3ae9dfb5a1e8b56a6 : NODE_FAILURE Node request 300-0021568563 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a434beb25aea4b5e9a28e06d496c8a85 : NODE_FAILURE Node request 300-0021568564 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9be4dc57d014bb2ae19a45ff0a33c94 : NODE_FAILURE Node request 300-0021568565 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ae89259073ec4f40a51081da88a60e42 : NODE_FAILURE Node request 300-0021568566 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0e020452468844f3b73e49ef84a5a360 : NODE_FAILURE Node request 300-0021568567 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bd601bb8bec34c4dabd58796f93f490f : NODE_FAILURE Node request 300-0021568568 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"257facdb85b0eb875e8ebe58c9d294a778e2474c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 21:37:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef15f1fb080f4c99a00f1ca77de9ed62\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/df46adf5ab2047629b873d3c45e8ab7f : FAILURE in 43m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8cc4bd6b81e2463b97d2a87d080d4d47 : SUCCESS in 1h 38m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ab4cf89615c2457292339c86de484c74 : SUCCESS in 25m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b7421ebec4ed442cbeb921d2a3adf02f : FAILURE in 22m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4ba2bcda839345a69065901853dbfb8e : SUCCESS in 1h 00m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d603586b610a470c884715f2d9f522f9 : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f9eb4057b934e98a7566016ab2dd372 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a7674fe47d3457fbf60c069865fbdf3 : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7cc58ca5bba4abf82e422f072e4280b : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad73d6f2d09c40b4876eb2b202ea468e : SUCCESS in 4m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fbb0234622704ba39a3bbd83b6d04d42 : SUCCESS in 25m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b4cc0e37520a4883aae4537c52dc14fc : FAILURE in 20m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b887d74f0e124f9184f695edffe6e044 : SUCCESS in 1h 07m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a213e08c40d749ae9ba321d48f5f333e : SUCCESS in 19m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4d00dedc186146a8acd035487d3120e5 : SUCCESS in 56m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/455fd06e0cd64780b8272f37145047b8 : SUCCESS in 1h 10m 42s","accounts_in_message":[],"_revision_number":1},{"id":"404224ab1672f5f55ae9ae2ed43cf93edb988a46","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-29 06:22:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b0da4429a8eb1825a879d60972f5f473206a04c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 07:48:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6228354e5c3c4caf83adb097dd20fd35\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f05ee31f719e46cfa5b332456274079f : SUCCESS in 23m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e78424d7fb7e4197a1ba2e7c0bda9016 : SUCCESS in 1h 02m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0e63c47bb1e84a448a2f2c504f4da5f9 : SUCCESS in 27m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f9b1ff677c004a6db27612a9566fe117 : SUCCESS in 1h 09m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/86068e7d331e4b21822107ad113d52bd : SUCCESS in 59m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a68b90e9c4974c47a0077d96a8467232 : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a11550628bd46f682df841315f24fbc : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d93cccd0470a4fd6972ffeec931e1737 : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b343849fdb794b34afa991002a4aa35e : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c18981a3a09481f9ccf81e8da4aeebf : SUCCESS in 4m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6a7eeb1dd34344858e5f576bfed4eefb : SUCCESS in 25m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2400dfd1fe034ae8a150406df912b3b7 : SUCCESS in 1h 07m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/648ac7fad1ff4f088154bd8188fa614e : SUCCESS in 1h 19m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e05a662952d943718fa78a247a1a1a12 : SUCCESS in 33m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6615baaa74d148d8aae9d07aa2318aba : SUCCESS in 56m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/28eb7bef6f2046459aca8fcfb29e6b84 : SUCCESS in 1h 08m 54s","accounts_in_message":[],"_revision_number":1},{"id":"559ad5849e2c5b830b231a859b6ec75eb5e30137","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 07:49:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e6656c3cd634e56a3a64ff4b11f8e82\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8d26c65e1dd443a5b06968df5512eca7 : SUCCESS in 31m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3fef672b6cd643b6a418c29b8e5553e9 : NODE_FAILURE Node request 300-0021572964 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5d104fe171d547e99e7a75a3f5b01f3c : SUCCESS in 1h 25m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3862983e78964e2d94ff72121c2e5314 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6eae6f9c730447e5b2a29e53de8dfbeb : SUCCESS in 5m 12s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a9a07b41a62549908945621822d24a75 : SUCCESS in 30m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3f69bbb64e454efcad4385a21aa29000 : FAILURE in 35m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e751e7267d3a6cb28d8bf1e5c2b96d37abe9d37","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-29 07:49:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ea3a0fc864a19214ae4e40e3999a7054ed29abd1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-29 07:56:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"440721bc3d549be5ee914b9fd36fc03d9581a357","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 07:56:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b8047cacd1f1904436577158a3e62897ad25381","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 09:33:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f05e4acc79e40bf843fb39e26fbbde7\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3e96cb83dc0f4f62b445eeff670193c1 : SUCCESS in 27m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3a08e0d87aaf4a9daa41c9eceeb1a75e : FAILURE in 38m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6001bcb09ea24cecbfd0755b4224b891 : FAILURE in 36m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44cddbaf5c364e1390aa9af21d0583e1 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cf30e2c297d4c7597a6211b21fb9271 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0166b671badb4de3bbd974512d0546d6 : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5764a0f0ec114734a5c60cd02fa8c9d4 : SUCCESS in 6m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/56694b17189e4e83b8c65386fa9bb6f5 : FAILURE in 34m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7134b6185018405eb56c9850f765c9ee : SUCCESS in 1h 01m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6fb7994cf24d41f0978fe1510ae1ccf8 : SUCCESS in 1h 19m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/79dfaa95026140319af890d9c312c71f : SUCCESS in 50m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0ce4d72bf6b14a3a8f0aa61a73f8bb5d : SUCCESS in 57m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/febaedc57c8a4bdd8e1ae8ebc7eaf41c : SUCCESS in 54m 11s","accounts_in_message":[],"_revision_number":1},{"id":"fe42df673facdb7937127b63781ca566081ac9c9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-29 09:49:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f69466982a0b8fedfac8662a05bbbb6280b1344","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 11:00:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a213d9fa87fe461d8f309fcb59b9e4a2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e5a8af0c001149d0ae1ada600a7fe5ca : SUCCESS in 28m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6c40a5d25c3b4619941db048362f8ff4 : SUCCESS in 26m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/855d92f2182c4409bd8d055e72083110 : SUCCESS in 1h 09m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01281ccadffc4b32a24c5bfead5961c9 : SUCCESS in 2m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1a4b364936c94baf8b2c496685513568 : SUCCESS in 4m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2aca7c2f4eb14581ab3a764c50e49dd9 : SUCCESS in 29m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a340c62634fc4d868918c9b90b93e092 : FAILURE in 1h 07m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2fc30d78de5867225515df83d4e536a202c4bf29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 11:55:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/713542fbac464e438a3d1a46f0f77569\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d3a818ac89094d58aabb252c22c06e43 : SUCCESS in 27m 53s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6974b5798df24de3bdaf3df6f0edaf20 : SUCCESS in 1h 10m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d262625978f1422781bf1441c16afd9c : SUCCESS in 16m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6ec4dba8558b42979488313745438d3a : SUCCESS in 50m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/10b9874a03d74abe8590b7922c4a1a81 : SUCCESS in 1h 01m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5a1f591b56d471ab45894f91d9c673d : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43692936089540aeabf6019610a28149 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4627477b1be3488c83fa142c8771b156 : SUCCESS in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc516043ae324a508156532a14ed21d2 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1a2a439f0a1458f91cf1610eceab8a6 : SUCCESS in 4m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ec114ae15d964b5da6a5d2c8d538eff9 : SUCCESS in 16m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ac9bb3e758a54daeb9e9c1a0fdc4848f : SUCCESS in 54m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f823c7797ee241dfb9afd3b7cf1aa8e8 : SUCCESS in 1h 03m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/980e90fa0af44b7182d661e9f548f51c : SUCCESS in 23m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8da29cc431e247bb9529c190a7cd00c4 : SUCCESS in 1h 11m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f17a819202ef414fb8036e60ca204776 : SUCCESS in 1h 10m 57s","accounts_in_message":[],"_revision_number":1},{"id":"52d4be25162ffeb15365985a5f6e3fa01934ac2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 11:56:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"55f0e7eb7f1a7edcf2f14e6f99b4b7ec5faebddb","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-29 12:26:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fbc2a656fa54f845f24424c0e6cec0f20ff3245a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 14:36:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2eff789ff424016b399562e7c8d83e3\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b17e9ce2926042788aeaf17ef681436a : SUCCESS in 15m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a705fa2a240e46a49dfd5ccfdecbbf66 : SUCCESS in 49m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3a7a3892b1d743dc9616428e5ae1f7c6 : SUCCESS in 1h 14m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f702f66e3d074dec9129142c020c1ea6 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9be33039f624c54b11a69ab82419887 : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4dc642b30c34322bba99c59f8810cb8 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1239885c31f1416ca59a25a15c92c8b1 : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5d9db86725fc4753ab5731e29874c625 : SUCCESS in 19m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5843d2ea85444e87bb89f9eba04e6412 : SUCCESS in 51m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/64fa76f2b22a4dc2a4f9c49b55b13c49 : SUCCESS in 1h 22m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d0636cd3e40c4fef86997b4ea41a5f42 : SUCCESS in 14m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e58b6e561a4b497da32a613e1380fda6 : SUCCESS in 54m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cec6f692353c4a4397dca76e81272812 : TIMED_OUT in 2h 32m 43s","accounts_in_message":[],"_revision_number":1},{"id":"cf2ed929b79916878a50ee71bebdc454bd7e1571","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-29 15:11:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9e7bc7a22444e754597e4da87612d25ed7489c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 16:29:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbbd9efe5dcc46eca8ff76d4de6e5deb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b701415fccbc42e5a627deb733badddf : SUCCESS in 18m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1e039e14ed524a409d114f2ad5dc3d52 : SUCCESS in 16m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0e234edb419b480a8b8d4b5313f0979a : SUCCESS in 1h 16m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe4581eb7ab34b12a41dbb59213b8ef9 : SUCCESS in 2m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/40f2645f8e1f4668aa40447fcae3dd62 : SUCCESS in 2m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7f5f6ca2fd3744a497d5b82cd7937a56 : SUCCESS in 18m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ba38bc7e6880494e93f02ef46710d62c : FAILURE in 16m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a7ad1176b92a1a2b5b055cd4e66253e77508882","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 16:47:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b5214c445214176a529e416e4160034\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/dbf68e947e7d44a5a7bb6b20111a2136 : SUCCESS in 34m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/66c58fa7501b49a69f8fe15ecda802ca : SUCCESS in 1h 15m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/90630ce1d9554c93862531cb1c11ef63 : SUCCESS in 25m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8f2793f7317e431cbd9cb9511794299e : SUCCESS in 58m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2dfc18d0dd7043528eb844f3dd0b7287 : SUCCESS in 1h 26m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30ba07ef21274125b63f27082aa60f88 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03ec0e2f68344538aa9d539b49f8b4c3 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec11500c9c744a57bc752e5b18e550fd : SUCCESS in 4m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48c25eb32b044ace93f3d2e2f24be106 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baf783799f2f450d92dce989579c9437 : SUCCESS in 4m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/291022477cce466bbd7845f0870a2c28 : SUCCESS in 35m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ea05d9c6e3524c9f89bd5f5626173ab5 : SUCCESS in 1h 18m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/701c17305c3e4666b304944fb3448db3 : SUCCESS in 1h 28m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a6874f81656047e2b865b5ca1524b4e7 : SUCCESS in 30m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dc947efb85984157ad6a9aab89efecd0 : SUCCESS in 1h 13m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f5644b2b9ce94801b634ad41a213fadd : SUCCESS in 1h 13m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7c9b3305150256704bd6d150da1f9864d9279ed7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 16:48:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f5721aae6489b22b0297207fb3057b1605aadc44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 18:17:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3054df6f7fe24a9294b516878ed26bb5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c220dab105f24de6ac56791b8544fe0a : SUCCESS in 25m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cdcef89dcfa44183ba6085e621d47ca7 : SUCCESS in 1h 12m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2a816198aad14f7181bdf400dd5a4bbc : FAILURE in 58m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5b3a2441c53431ba7101cb7abcbafdb : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2948a6b06db94bedb8dbdad413ffff0c : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d31c67a94864ac3b5f166abe2344630 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6510ffed4f344ccaa665d0d761808eee : SUCCESS in 4m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e19ce3e52115432f821a7e169e6c369c : SUCCESS in 18m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1ec40a80e02b45659aac4f099093f3b7 : SUCCESS in 1h 21m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fd7d817df1644b1e9c612e5e13c5df3d : FAILURE in 52m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d0d251f3fb534a2fa7996e0349a6c404 : SUCCESS in 31m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/177a27f0f6ec407fb1936175bbfb916f : SUCCESS in 1h 16m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0d96cf7dc4e84f5fb601935a964f62c8 : FAILURE in 50m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d36ef054dd835074f090ad79810fe9b1ea5f7788","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-30 05:30:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03e9955f87a452904faf3e9957d79ebc13418674","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 06:46:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/305622c6911e4ef88546b66103cb0661\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/dedd89888d2d4b9d9e79c719c7c1b733 : SUCCESS in 32m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/fe41f78f60b146e78988ee9ac961f4f3 : SUCCESS in 57m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/631acddf96e84c0d82ea6b01e514a857 : SUCCESS in 15m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6ea6fffec8a54f669651a2d0d5b66371 : SUCCESS in 59m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/af0c5628df1b4aa5b1c3df222e64094d : SUCCESS in 1h 05m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c550693cab824e8ebb6260241341a4f1 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74b6d8309be94d8d8b01f1f4302f486c : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4dc91417d804a388a494e7b3c4211ca : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e36fbb2d78b347fb821ee12aaa2e96ba : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/360b884acc024d22a89fa2c0c4347de7 : SUCCESS in 4m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4ba24bb0742f46089790edc0b14762d1 : SUCCESS in 20m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ef68fabbafc1421f8443f4dbdf304a2e : SUCCESS in 1h 03m 56s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9fecaa016ca64b53be3e50940a10e396 : SUCCESS in 1h 10m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8bb9655fe8434a7c87962f8d2a685c68 : SUCCESS in 27m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6c1790bda3044ba88b37ac1d686845ad : SUCCESS in 54m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/40b45f8b84f44e3e8522dac81e772039 : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":1},{"id":"70f8b243ceba45a772c49bd82256ca68cc04cec5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 06:46:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"674ecf011f708ab58a4417b7a5b950c85959f783","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 06:57:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abfa12dd818c413db074155aff253caf\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c4a1916a21e24c75937c37108b9a2c8b : SUCCESS in 23m 42s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/edc9a0094fd34d30b203a35740793492 : SUCCESS in 23m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e5a91b071f6c40d48cd102aec72d9c8f : SUCCESS in 1h 24m 08s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b5fd817fae947e1a3e9c2a992f4b7ff : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e59b53fbbca94b8494201e1b5df6c152 : SUCCESS in 3m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/61cfc64edba746a3964cf9c159e37cb5 : SUCCESS in 28m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/cb7910a4662a41e9865dbe7cbe9599b5 : FAILURE in 1h 09m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52e14143de04f3fbaecfaad67f5e2d15cd435f9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 08:11:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/102fb5cc2b8043958375ef466202b50e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/85ac64ad1e0e497db07418bb5b2652f1 : SUCCESS in 18m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/59b47fff157645d49183e06839109648 : SUCCESS in 1h 06m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f30267f353cd4396a819fcaeab40d8e1 : SUCCESS in 1h 07m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f3a40493f09409b8a06ac23c31812e4 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d739cd7912c4db2a158a0473efeea39 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb871f4d7f84460b9409fa3ff23645b5 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3ca37ab1d2040e884f5e2a14a39ee16 : SUCCESS in 5m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c5f955b6e90b4a1a94c03437da3f47a5 : SUCCESS in 20m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/56cf6024e8db4699927b113db8c69541 : FAILURE in 51m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7507d4c0367b4f768db8229252e27229 : SUCCESS in 53m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/34037e0f52d44bde8704ccfd1a78fa1f : SUCCESS in 21m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/58fa59198f67405386297470e096cbcc : SUCCESS in 47m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6f1b85aa22d94a28aca92fa0125a4f4d : FAILURE in 50m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3e271a4fd3476e45059fa93e5b7a79c90bd641c5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-30 08:20:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f861760139120756f9d5ef9344a5e5a43014d1fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 08:50:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/369d3ecd8cc048c5aea6509dcc0e1a7a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e79bd924dca94a5aa43b49ae25e96dbb : NODE_FAILURE Node request 300-0021583090 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e908e4e6f0ef41b7bd7d2fc898b1a6d4 : SUCCESS in 29m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/11fb77adf72f4e80a7c95d937be349f1 : NODE_FAILURE Node request 300-0021583092 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70e93ef6e6e8409dadd0e08a16332f81 : SUCCESS in 2m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c7baf23472504b1ab5ddd3f41a0d13c9 : NODE_FAILURE Node request 300-0021583094 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a1af9def7a0c4eb788b6088b953476ba : SUCCESS in 25m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f3a06785e2d2460bb33383d65676c63d : NODE_FAILURE Node request 300-0021583096 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"371d5b7f2bf2aadb7e07a250305a0313fa792391","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 09:58:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ac354563ef5457fabc55c9e546c6144\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2be983a577424b9086b63ec26a5ef1be : SUCCESS in 34m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0d8a185566974f33b89e8518fd8ea4d6 : SUCCESS in 1h 22m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6839706cb262406489122032b6909762 : SUCCESS in 25m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4565826fbae048e59117f666c2053568 : SUCCESS in 1h 00m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0055d7fd81cc4e239feea7190877f621 : SUCCESS in 1h 26m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9a6e6d78a304af8ac4f7ffb04ff7c8c : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8abb6f44c5f941de8e6553ab9fcea906 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84465d4ea61440d686b6b1a21d026eb0 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31d93ffa79ce42aa9d17b28f567b44bd : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aa02dd2e0544f09a39a9c6a9a11fbe4 : SUCCESS in 5m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a8506c4e71a944cb813ddceaf974832a : SUCCESS in 23m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1b42013313554fa185829081d2ac6e50 : SUCCESS in 1h 13m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/244b36b2f7174736b2520bbcfbd5c814 : SUCCESS in 1h 20m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/baee1f6729bc4247ab8172d1e7fb4553 : SUCCESS in 26m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/58a53cd79d664c519b6e0526fa235c02 : SUCCESS in 1h 03m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cee80c6c60a84ba7969a7801da4a73d0 : SUCCESS in 1h 24m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e23a3821c53cc6ff5a30fbf9765f6c0f36f6e07a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 09:59:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb08a1d3343277dfd14618b0a696d691d1e2f7fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 11:08:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed8739ab304f43adb143f8b236ff61bb\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c75cfad2288e41cc9dbda3e5911231dc : SUCCESS in 16m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8e2bdda41b824c9e8b8bcb3ad5a806f6 : SUCCESS in 50m 45s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/eca27eead6c3469a9aea2b8f40923cff : SUCCESS in 56m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2044d957e02c49559293dc0be496ce41 : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b78dbc21a1c46ee9c97512fa086b173 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c41bf9927dea4811ac68db34f7f675bb : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/519c2f6981414ef6b5368b52fb92a620 : SUCCESS in 5m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0911edd0feb84055bff1091bb8ffc572 : SUCCESS in 16m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3b8ddafa9c1b413189f1944f70fcaa13 : SUCCESS in 54m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cc1eb6b74bc84f2985fafd23766cbba8 : SUCCESS in 1h 01m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/96294383f6e549039de8695bb9fdf9ed : SUCCESS in 14m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ff02f426aacd45bf8fc5731d347abb72 : SUCCESS in 59m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2dc05dea468b4431bd7ef3322d2c0b85 : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":1},{"id":"6dce12d3cae7f7af34db30176eef48db892fd164","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 11:08:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"41e4deffc5462973c49bf397095c69579cd24e18","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 11:09:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6b101a43975448a87c2111b4153eb92\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b664107fd10749ac97e9bf18910a8f92 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"91a4079d84cec7ea384b301b5ba0cb79db495503","revisions":{"91a4079d84cec7ea384b301b5ba0cb79db495503":{"kind":"REWORK","_number":1,"created":"2023-06-27 12:02:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/886964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/886964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/886964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/886964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/886964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/886964/1"}}},"commit":{"parents":[{"commit":"f43a0ed4ea136d18f8415495b9adc70e96b92c41","subject":"opensearch: drop unnecessary workarounds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f43a0ed4ea136d18f8415495b9adc70e96b92c41"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-26 08:36:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-27 12:02:03.000000000","tz":0},"subject":"Use erlang-25 from copr on aarch64","message":"Use erlang-25 from copr on aarch64\n\nChange-Id: I9f40167fed9412b3f33f82e2f97f353275431292\n(cherry picked from commit 51675bb232f0c4826ff74351b68c1e16a1893545)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91a4079d84cec7ea384b301b5ba0cb79db495503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91a4079d84cec7ea384b301b5ba0cb79db495503"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
