)]}'
{"id":"openstack%2Fkolla~886965","triplet_id":"openstack%2Fkolla~stable%2Fzed~I9f40167fed9412b3f33f82e2f97f353275431292","project":"openstack/kolla","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-29 16:48:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9f40167fed9412b3f33f82e2f97f353275431292","subject":"Use erlang-25 from copr on aarch64","status":"MERGED","created":"2023-06-27 12:02:15.000000000","updated":"2023-06-29 16:49:56.000000000","submitted":"2023-06-29 16:48:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886965","cherry_pick_of_change":886964,"cherry_pick_of_patch_set":1,"meta_rev_id":"d025e19cfcfdaf08e4eb7f2f42c8ffa03397c90c","_number":886965,"virtual_id_number":886965,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-29 16:48:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-06-29 12:47:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-06-29 12:49:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-29 12:49:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-27 12:55:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-27 13:41:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-29 12:49:14.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"2c13944df27b760fae2a6b8de67109c717455a48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-27 12:02:15.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"22799a59e74d88d0f97c17b6c12ae1b9f37ddd1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 12:55:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae360bac6cff4a048137fe65e61aeb64\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/42017eafe7c441738e204fa55442c84b : SUCCESS in 48m 36s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ba4c7f328b8b4e28b56f016ce98de6da : SUCCESS in 48m 24s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9e369657e47d41a3ae5a0cb1fce7e544 : NODE_FAILURE Node request 300-0021553520 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81e41c17e2834e5ba75291ac4818018a : NODE_FAILURE Node request 300-0021553521 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9fb41b892050406ebe0b188124688582 : SUCCESS in 3m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c602884500004e9b910e9659d78a842b : SUCCESS in 48m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4365bf7de932446082ad11ba18af5c56 : FAILURE in 45m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"154809d5200490465bbb9d9c6c4ea2ee0b9057d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 13:41:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3687b8f347e74624bc5be8e0f6e723db\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5823729356dd4554ba24c8b46e2d4a07 : FAILURE in 1h 10m 55s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b6c92e924c8d4f80a76ae2b72c2da7e6 : FAILURE in 41m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a0252d3182574b30acb63d44b21b69fd : FAILURE in 46m 28s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/33bc4b92090947b08463e26e0e642241 : FAILURE in 31m 38s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1a5d5ee41b9346b294f9c3127470a7e4 : SUCCESS in 1h 12m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d37b63417ae540fbbd4b8c3e0d8a779b : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32b23333428d4127bec806168809e725 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25390995a6a14b30abc36811e441f11f : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d8943f12af145e0bf3e839f28cd6db9 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b5eb3d2ee34778a68305960cb7969e : SUCCESS in 5m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/71bdce71212142529e98a29b91c5e3e7 : SUCCESS in 27m 20s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/c25d3fb79a384255a08ccf9eef858f73 : FAILURE in 24m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/09eeeaa9ecc6458f9e04787f74069318 : SUCCESS in 26m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/39f5c7510eb246449c9f1d53e2627e5c : FAILURE in 27m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/916f016f4d3f4fc08b76b7f7559ffa18 : FAILURE in 24m 37s","accounts_in_message":[],"_revision_number":1},{"id":"53a4adc281d36bd4de5ce2198fa89819fb6bca5d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-28 19:51:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9597424a737c99b7492b58b334c2f7d382a56ed6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 21:06:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f67280584bc42fa9c8abf3b89e7a720\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6afedc6eb5bb49589bc81ee74dd38cf6 : NODE_FAILURE Node request 300-0021568540 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c1b0db8ace7e40679e50ad4cee35d6fa : SUCCESS in 33m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9fd7616f30f943278c615d17f3798280 : FAILURE in 1h 13m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c452581dce774d47b74445d73eb71a51 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb916685f8094d69a8a3c4a23488fcab : NODE_FAILURE Node request 300-0021568544 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/94fb309269a4407da4a0f79c0ab84a6d : SUCCESS in 33m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5d717036fad443e8b6952e098390bea1 : FAILURE in 31m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2905b0b9167a143b5881afdc708ddfb48924a9b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 21:17:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f7aea1f1b274cbfa92160148d63d4f8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2dc8f35a27dd4dc4868f42cfd5bab74b : SUCCESS in 46m 27s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/c5d4ed15ae0f47c088d761f29100a3b7 : SUCCESS in 52m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/186e9a48b3e24b4283ab55f7bb801a7e : SUCCESS in 25m 22s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ddc9facd9aaa489f92058a44085227de : SUCCESS in 49m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b39c4196d0d84c06a8a5c6662802bd25 : SUCCESS in 1h 05m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37eac49f7990470d86ad85efb37bd18c : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93969936b0484ae28bc8057fa65ade7a : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ee5f5bdb69f48089a4c803931dc5382 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3117ed789ca5456dbe0bff76b1b0b946 : SUCCESS in 2m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed5afeda717459e80edd01d66261dc4 : SUCCESS in 5m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c4541421675147e09fa04ce4c19d6ac3 : FAILURE in 22m 56s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/d11cd1ed9a68434a9bd889b5353b7570 : FAILURE in 20m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7f5f07b77d984274964070e35279e1ac : SUCCESS in 43m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8392a8e8f8ab40fc80be8e278fe92d0d : SUCCESS in 1h 18m 59s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fc5cde559a4940c8a63e00dc8164c0aa : SUCCESS in 1h 12m 00s","accounts_in_message":[],"_revision_number":1},{"id":"2deec9e7befaf0e898a6f4df2677914473b8307e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-29 07:49:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac46650c7c672811f3d7795b9c21fd7f5816d4d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 09:14:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4235fec75b6b4edfab246674e4086e2b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6df6b0c561c9404ca499d645ac3f466c : SUCCESS in 21m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c8d8801f35fd4eb0ab3dc8280de9267e : SUCCESS in 22m 44s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e2a38aff810843d5a6c4f4a66c7332bb : SUCCESS in 1h 24m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fea70d0b23ab40a4a486457fde7f528c : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48dbab3fe6064628b49297233d73bd01 : SUCCESS in 2m 40s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b1d8468631874913b173e55d275ce0e5 : SUCCESS in 22m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c7f1d01de53a46febf9e922217377719 : FAILURE in 19m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"791116ffade22f1dec54c0b8cc7b58a75192f89b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 09:50:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/136bfba59e7045c6a266cfd9a6881d11\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/49e633c420b24ae484619a6e24361a8f : SUCCESS in 25m 15s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/f9935b9c91774c42b883fd59ba6c7a87 : SUCCESS in 1h 07m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a290da35e0a24fc485d822f7880c6b51 : SUCCESS in 34m 15s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/77bdff4cce09429ba3cc46444a3cc3ac : FAILURE in 48m 47s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b3e72ea4cd694f7092deebfa91d432af : FAILURE in 45m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69e49ef8664d4925bf963485915b5028 : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efc32b00138a426baa9b6cd3666beea0 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/daaf7cc68f934ad585f8c99b843fde09 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b24978d0d4474e619b42d5ef3812ccf1 : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56efe6bc17a8431f84f66aa8ab3af7d3 : SUCCESS in 5m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/100a6e6f41ee4d88bdff033f5f7d4e9f : SUCCESS in 23m 39s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/22e9aa2b6b574cf1a7c87a4496945fed : SUCCESS in 1h 40m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3b0aae18a8194394ba60a276127c4566 : SUCCESS in 18m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f4316e0511d475e8fd65ae2d1b6f23a : SUCCESS in 1h 28m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5622d2dd200d416ba6133f3baf23d425 : SUCCESS in 1h 52m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e62166d43fb01153f3cee563b44a2aa756505123","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-29 09:56:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ea980e66516a7d8964ec14041e1619b78c73821","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 10:20:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29bf850a1fd641e380c607a3a61f35d3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a80fb059dd2348f0bc33bfe9a8aae89c : NODE_FAILURE Node request 300-0021574922 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1abd77c094ff46fba27ffeba13e5ff62 : NODE_FAILURE Node request 300-0021574923 failed in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5b42cc74dac64e59996d9a6ec2415647 : NODE_FAILURE Node request 300-0021574924 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eea2b826c8684fcf80cfd456aa98be9c : NODE_FAILURE Node request 300-0021574925 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/62bba45167474ba7a0429bfe4c1b575c : SUCCESS in 3m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/70119fc9ecb54c72a4e693a7dca94aba : NODE_FAILURE Node request 300-0021574927 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5c38ab2ae91e4ed7813a0d4583c64ac8 : FAILURE in 23m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"116db9534cfc0df9f492e7b59c246a0457959cd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 12:32:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0c44c8dc4714ed3a9876366adfb1d63\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/488dbdc2061c4ef9822c54240c3783d5 : SUCCESS in 26m 27s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/9f3251a8667843b7bc772ca038ac9802 : SUCCESS in 56m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5b0ce4022a634c399eef16d5da0899a8 : SUCCESS in 1h 45m 36s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8254f80906b64d6eaab073ed89e51b7d : SUCCESS in 51m 39s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0d8b192d8c95424d93e6f7677eae172a : SUCCESS in 1h 22m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2a31f29c1aa40c890ad24c00a149446 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4966531c173e4adcb0121d83d9e97e08 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af0b7d3c023741119ade405c457da1be : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c61d5b5b35354cc0b77c95adb527209d : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d28abe479e7e4c00b0ee4b4111e3d090 : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/45d11053859b48cca263535b0e3fc708 : SUCCESS in 31m 24s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/eba29860b4af4798abb47be77b912044 : SUCCESS in 53m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b2703bb035eb4b408d006709f7a69d99 : SUCCESS in 15m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/31c4076ba2e24465bfeefe60d67de7d6 : SUCCESS in 44m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4054c6e33dde40658b61a38fd77a3486 : SUCCESS in 1h 03m 05s","accounts_in_message":[],"_revision_number":1},{"id":"2f0a967c8fcf6533961d3469a0701f090841ea8c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-29 12:47:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"61e3d3741238dff44f731c897b54c3584616d597","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-29 12:49:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1f8266d44100c1777cbe14885c98828755e89c22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 12:49:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a0c39c9c292aff55e0955adac6abb381fa44c3ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 16:48:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e05f6fb6d85430fad9c4e8474415152\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/86dbd3af7417482f9adb51d9977081df : SUCCESS in 20m 05s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/20060756ccb84c87853f811290696973 : SUCCESS in 58m 26s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f65ec9d956994986b0e21aee7feeefb6 : SUCCESS in 1h 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f4dcf2ad05b44a1ad97d5c3f73772fa : SUCCESS in 8m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99134988828c45eb93ac6cca4fddd899 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/287f4b13a6b64009b03e2868c6b64813 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4effa47e585640a1871221d9ce639782 : SUCCESS in 10m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3bbf667d3e4349db80e09aac94122614 : SUCCESS in 1h 12m 49s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/bd441dcac92a4d318aae7906ece1df36 : SUCCESS in 59m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/30b141c5024d4af9852a42cd010fd655 : SUCCESS in 19m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/87e658e6190648e78d07d09c345f93b8 : SUCCESS in 47m 14s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fc4a7614b64f488090079f8b64e88b07 : SUCCESS in 1h 08m 40s","accounts_in_message":[],"_revision_number":1},{"id":"8f377ae728c120c2c2bf315292086c4c69c13354","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 16:48:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d025e19cfcfdaf08e4eb7f2f42c8ffa03397c90c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 16:49:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3434bbcd87774f9ebfd356a125ea055a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba0b68f962440fa81d98817a2c61b14 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"13aaca13e460aff768aca41d3b7f4853bb2c82ca","revisions":{"13aaca13e460aff768aca41d3b7f4853bb2c82ca":{"kind":"REWORK","_number":1,"created":"2023-06-27 12:02:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/65/886965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/886965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/886965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/886965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/886965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/886965/1"}}},"commit":{"parents":[{"commit":"530490a7153b8189b24e4cb8f3acb96a5f40d1c1","subject":"opensearch: drop unnecessary workarounds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/530490a7153b8189b24e4cb8f3acb96a5f40d1c1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-26 08:36:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-27 12:02:15.000000000","tz":0},"subject":"Use erlang-25 from copr on aarch64","message":"Use erlang-25 from copr on aarch64\n\nChange-Id: I9f40167fed9412b3f33f82e2f97f353275431292\n(cherry picked from commit 51675bb232f0c4826ff74351b68c1e16a1893545)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13aaca13e460aff768aca41d3b7f4853bb2c82ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13aaca13e460aff768aca41d3b7f4853bb2c82ca"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
