)]}'
{"id":"openstack%2Fkolla~886966","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I9f40167fed9412b3f33f82e2f97f353275431292","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-08-16 07:03:26.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-08-16 09:09:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9f40167fed9412b3f33f82e2f97f353275431292","subject":"Use erlang-25 from copr on aarch64","status":"MERGED","created":"2023-06-27 12:02:46.000000000","updated":"2023-08-16 09:10:40.000000000","submitted":"2023-08-16 09:09:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886966","cherry_pick_of_change":886965,"cherry_pick_of_patch_set":1,"meta_rev_id":"abd90a773085facb9782ba23a768f27a57c17cfe","_number":886966,"virtual_id_number":886966,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-16 09:09:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-08-16 07:03:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-09 15:58:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-08-11 04:01:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-08-09 15:58:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-08-16 07:03:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-27 13:03:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-27 14:23:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-30 07:17:20.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-07-13 07:28:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-11 04:01:32.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"8f438251dcd95432924ef585d3b6781b61b3e72a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-27 12:02:46.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"93fe852d4396744940efe1345905af26c05f9e52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 13:03:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d128ecdd8524b21b3953695ed569e04\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c78aec00a3a4fbd9ef23cab1564dc7c : NODE_FAILURE Node request 300-0021553554 failed in 0s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9bdec577466d4e31b0da2ea1abf9ee4e : SUCCESS in 58m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8fd8529708bc4e349107103e712ef842 : SUCCESS in 51m 03s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/48a797d15bba41a4ba4d1bc5a2270c5a : NODE_FAILURE Node request 300-0021553557 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/708da4c431f040ba9dfda8f2e055384e : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/587c14991bc447449fc18f33a0dc316a : NODE_FAILURE Node request 300-0021553559 failed in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f432bb2c585f4f60bb5bf918c42d74fe : NODE_FAILURE Node request 300-0021553560 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1af4247c9996d7184e75fd3fbcb6fd6cba7e7bcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 14:23:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d04e39275464f3fb214351fd7c441e1\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8b3bd55f8dae40e1878c5d3914aa8ab8 : SUCCESS in 43m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3af088016d9345bdaa8e6dea75753959 : SUCCESS in 1h 05m 36s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/1c0cf229a66f436badc06fc1990bb629 : FAILURE in 36m 27s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ffff55c43f1244b5b99a6b2664f81b8c : FAILURE in 57m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f34dca0171cd4fdbb7e731774e5367fc : SUCCESS in 1h 17m 38s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/a10bd130c6f44a7982e69bab8b1dc3f9 : FAILURE in 38m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/09f9d422d85c48d4b5d1c93cab4537be : FAILURE in 57m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/83c7cf94f8404c2181a4a86360da1d14 : FAILURE in 19m 34s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cd5d9796ca334275899dae65888dfe2c : SUCCESS in 20m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3b4b0ae0451047e8a1a1fa176973cee5 : SUCCESS in 1h 46m 27s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c62ee9ec71e749e4a715de63f4d1e1aa : SUCCESS in 1h 09m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4580ff643524904afbca0cbfdc5c0f8 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5df02a73369d4de1a125b35998ca6e06 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b62402644fd241f4aea74f402080d4b6 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f8fe3906d9b41bfa2cc3e71d78631aa : SUCCESS in 10m 44s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/2e31f75e0f6c4ddcba50015153239708 : SUCCESS in 23m 52s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/b97735716f3d41c5aa66ed91a4a36a81 : SUCCESS in 1h 07m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cdd3a7cc941c4834a4bb67b089cd51bb : SUCCESS in 30m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/31647fd1c5f341ea9275990f55698d90 : SUCCESS in 38m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0970a377c914ea6645b978c407b7aea063d548d3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-28 19:51:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f901efa7bc48d7a15eb2095000b57535f702b43f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 20:27:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8324528cd97b4c4e8b9dd4354c2f704e\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/81887c9007d249ba9a41248a833e69fd : NODE_FAILURE Node request 300-0021568589 failed in 0s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/546c9b76c9e84e0ead369a44b93e0263 : NODE_FAILURE Node request 300-0021568590 failed in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/86b29caac37043d1aebb89c14aa032c6 : NODE_FAILURE Node request 300-0021568591 failed in 0s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9b2e81381ba4491e81b162c29b027567 : NODE_FAILURE Node request 300-0021568592 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7178e7e5cf0f4d3294dd8261a26e4d20 : NODE_FAILURE Node request 300-0021568593 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/17f6967ff3404f7099634a88c0d0802f : SUCCESS in 4m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6231a7a1216d4161b1e54362785a638c : SUCCESS in 34m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"611281aed14e40c94669b64081afdd537e396978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 21:43:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f1cd822dd904eb8a05343e2f83840b3\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/16680289ce2a4e81a44dfd43e7df5b1a : SUCCESS in 39m 37s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d2250372e9ee4f0c8b952722f415040d : SUCCESS in 1h 00m 01s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/d715de18ece24e2690eff6229be5a762 : SUCCESS in 59m 59s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/095f683c9b934ccea510f9477858d2c4 : SUCCESS in 1h 38m 42s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/54b16ecde7d1491bb0813df469870393 : SUCCESS in 1h 18m 43s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/e7121566fbd54195bb27b8ab98c15806 : SUCCESS in 1h 35m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e5bb5cf8261e4360a870dee467b74bf5 : SUCCESS in 1h 08m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b4232afd6ab440eeaf3fecdd1ed69abf : FAILURE in 20m 40s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3efde46171a14c51abc3911e61e522f9 : SUCCESS in 22m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8584fc812ff744a697519da46f7f1245 : FAILURE in 24m 29s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c2ea01a88fa74bb5a425f2adddc72ba8 : SUCCESS in 1h 42m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f488be5b2bbd49bd935164f41a4bf4a6 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/053e41ae56eb47a9acde1839ee3ed207 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/045e5c28709a449ca6c63a47bb4e69e9 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb7e97d8841e4754bbd72ad58ce8b3be : SUCCESS in 8m 41s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/6b65ee08c6ef42a68e36af2f5780b145 : FAILURE in 21m 33s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/8064fc5ef6de418eb2e839c8869bc479 : FAILURE in 26m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e6c40cf5d5f54499bba107aadc06e36e : SUCCESS in 24m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1babdc038d054ead9c12d2c5530a3fa5 : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b9ab208d9789da9f5c808aa852566d99df5ddc3c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-29 06:22:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0df5f49a9c324c1d04f14fbacab131e471c1dde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 07:35:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b09de0d5c3247b3933c47f763d69ca5\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/56fba21f07de48f889ba8f4c2d0d90be : SUCCESS in 41m 07s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b5d2555b33d841a795df0fb8a0f92843 : SUCCESS in 32m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/05afbc2cdc954ddb9997104d35fe47f6 : SUCCESS in 33m 41s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/905ca67639f44ec7a013168f88f8be7e : FAILURE in 1h 11m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c7e4c33ae4c494098e49894bca5ab46 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87ab028698a64810b38aaeba79b91aaa : SUCCESS in 4m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a6d911de8f424144baeb6eb5adb9ec0d : NODE_FAILURE Node request 300-0021572994 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb10c41a364d71a242a64aa63709b4c04491a34c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 07:59:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9719f631a7646aa8bd8b84f72f1de33\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a96aeb7f121841bf950a283ac19331a0 : SUCCESS in 33m 57s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7677ad3f5f8f449981839721d7fb761d : SUCCESS in 36m 45s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/f5c25d1e9b7f4add8273008f013f3744 : SUCCESS in 28m 37s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7f1b9344be804d29a8fa66ecd21e5b53 : SUCCESS in 1h 15m 48s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8b7c7a6135c844508a512828e8c49a04 : SUCCESS in 1h 10m 05s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/dc7925f5bb63487ab1c53a185426e958 : SUCCESS in 59m 33s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/73714b7a5f86439c893ae5db48197044 : SUCCESS in 1h 35m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/42e5def71963406f88d7ef835feceef6 : SUCCESS in 18m 43s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3b2a98b0164c4cc0bfbf9590bd9bb70f : SUCCESS in 15m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ab6ae581036c4a04b43e8e9580305495 : SUCCESS in 53m 19s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/196aa405d580431d8b68c94f1ce78311 : POST_FAILURE in 27m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/011a6cab594f48bba444cfbbab2c31a4 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/239437c8d6844f249564eafaab96672f : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/000606088db545f4b6f40032c33c22dd : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ce4ac30418647bbbc19079004e210d5 : SUCCESS in 6m 26s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/57412bd0b4354a00ade6215be86844df : SUCCESS in 24m 17s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a89ac8d33597498088a4b80e4e9dc257 : SUCCESS in 1h 07m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c157737fb48f49f99d86995f00fd20c0 : SUCCESS in 29m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7a61145768e490685ce60deb08deb4e : SUCCESS in 37m 01s","accounts_in_message":[],"_revision_number":1},{"id":"28810e7542553f75edf40b053fbc83e6623e3dc8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-29 08:03:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db62f224b5d4575cd78a525f950cdb2375055aa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 09:17:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/251789f4003842889893275f10fbe58b\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/05651be8aaa04ce18cdc58017cc41a8c : SUCCESS in 40m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fce554757e22419cbe252c5160b356ff : SUCCESS in 38m 26s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/76e8e78bc78249548dbfb6274ce78a2b : SUCCESS in 37m 09s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/983d2fc4f29e45359a58e0b075d6cf21 : SUCCESS in 1h 10m 51s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/793e5f1db6d444bbafe59803e2ef9e69 : SUCCESS in 1h 10m 37s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/a597aabb93964e35a11a58cd64c56299 : SUCCESS in 1h 11m 14s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d56ceca586b04d20a1e0fb5366cee6b4 : SUCCESS in 1h 08m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2ee431b99b254b659aab9065e3d53fc6 : SUCCESS in 58m 01s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4011c3d084994f5eaba3803c4e45ee5c : SUCCESS in 19m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/30b51a6c90ba48c0861f601c9c71d854 : FAILURE in 49m 09s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7797560b50974300b4eb42d05139b76b : SUCCESS in 1h 08m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e822a46ea1d54b8aac90d8e4e8ea963b : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f78f495439a64e099ce69ae8906cbe24 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1993c8f3ba3247fa91ac395973dc6c84 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/872f0be5c8c8459cbcaa4a670057ca90 : SUCCESS in 6m 22s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/290a67e4601d4c44a357f1b9b4dcbc16 : FAILURE in 29m 19s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/08b0992ae1f94ffe8ca8ba835a7b05ea : SUCCESS in 55m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/65a0fba226f245d99145e44c0f690ac8 : SUCCESS in 23m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b406c0e6d3b3401b82d2f38434f9063d : SUCCESS in 26m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5ec6cd8fbe22220cdfd30c2cd772202816ceb125","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 09:23:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b20d74b81b9c49d9ae9364527fd934a0\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea0de79d108b4a1386f783ed8e9f6fa0 : SUCCESS in 25m 19s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f1501932dc3e444c84a087f33874b92c : SUCCESS in 1h 17m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4ae06bc1ee314bf8843e780b600e03c9 : NODE_FAILURE Node request 300-0021573722 failed in 0s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/934a417f9f86490ba355719d1a67ad4d : NODE_FAILURE Node request 300-0021573723 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4207e5ee82e4c8d8f9e7ffe00cbb3e1 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e72cc02355e445a9057ef2bcf87ae36 : SUCCESS in 6m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c46a8a579adc4d78b1ac36b64d686fa1 : NODE_FAILURE Node request 300-0021573726 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a556745d09b54d993415b07e723f8d41aa2cce29","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-29 13:38:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e47e287eeb0a3e7518a6abe2d0a4226916f849a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 14:24:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c9c7ea570024939bb1a62c36190c61f\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c1282b4469e442485f204e2ce334f29 : SUCCESS in 45m 01s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/845913cb45da4360b1bbeb93a932cc9d : SUCCESS in 34m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d2f1cc86dada4bffb9c5250c0a5e344d : SUCCESS in 39m 09s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/607e55a3e42d41edb954df81a95ea0d8 : NODE_FAILURE Node request 300-0021576718 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/198f68989de249308c0522b9a96b971e : NODE_FAILURE Node request 300-0021576719 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/165102816877451f8911a2ca8614c824 : NODE_FAILURE Node request 300-0021576720 failed in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/aa1490e89dc94f8b9dfef70065622b4e : NODE_FAILURE Node request 300-0021576721 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25db50fe30b651297693d26d0b002379ed3a3e8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 16:40:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd99e49ca39645beaf9e9e605ad50167\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3ac024ad2aef45d1a3f5aca5dd9d15d8 : SUCCESS in 43m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/dc2cfd0d3168496bb09004c386340aec : SUCCESS in 28m 49s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/1030c8c31151467380de1d5fd70d291d : SUCCESS in 23m 02s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/69d69c86f2fb458bb7dde5cb4c50aaeb : SUCCESS in 1h 08m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/594781d439fe4e82a1cba1e3abd879b7 : SUCCESS in 1h 16m 40s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/50f4e26c0fc2475788746e4c51d248d8 : SUCCESS in 1h 05m 28s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c74f55fe023d4e45aa8b9d5274809d6e : SUCCESS in 2h 28m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/881c6cfcb2ce4de1819a30caf53ebf65 : FAILURE in 38m 33s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9e901339851045e99aa88906aee74a3b : SUCCESS in 19m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cca4bd26b40a44b09e45020708c2c501 : SUCCESS in 1h 02m 42s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f248279f47354400a8968cf96521846b : FAILURE in 51m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cba095d1dd0843e39cf2e3f3834ff5bd : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64b0d9e3587c439a931689bd7197eeba : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acd9f9cdd9a1471c9bd838ad01628862 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7024d8f5940f4f2a86dc3deafb17572f : SUCCESS in 6m 03s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/52b827bb0f7e414b9c407e46708e97f4 : FAILURE in 36m 33s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/862deb1bc0ec4836a85358948b2d86cf : FAILURE in 33m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/600b19e1d6944f21a93f5ba8dc0d1f15 : SUCCESS in 29m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7604cbf3b41a4dd69bd98c16cf80267e : SUCCESS in 43m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d60656f4b52cb7edf007004ace43af393db449e7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-29 17:09:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee7a6af330f2930832f92fc87f3d660374835a2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 18:33:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afeb77bb35814a25905e7815152b4097\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c14bca2e6c6c4df28484934f2225bcf5 : SUCCESS in 23m 28s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8916f12d5bee40b888a4ee638ccdac75 : SUCCESS in 18m 45s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6cb533235dad4d4b9a0e4c9bf35d2f93 : SUCCESS in 17m 52s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6195cb740cf14d3f8116e85536b21a51 : FAILURE in 56m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4079fcc67e614602b0c82f967d9599e6 : SUCCESS in 2m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0cbdb8cefb240bea6f3f42cb0de4dab : SUCCESS in 3m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/49bab692b8294247b47b471769fa1d46 : SUCCESS in 1h 21m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"811643fd94ae36f966612a3c6ad106546f2ad27f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 18:51:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae8006c2d8ef470182a6a48c9dc68926\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/bf5340bfe0d040d494327c67fe6d7309 : SUCCESS in 38m 31s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/01eb4db1b372463e91cf9a5e99ae7b56 : FAILURE in 39m 19s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/9e27c839939a466291c50960c8021652 : SUCCESS in 29m 45s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d24fa97fd84c436bbf3e791160376e4a : FAILURE in 52m 35s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5b24d5b71a974ae6bc2b5b9ebad212d4 : FAILURE in 55m 59s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/5ca5f2ae69784de79adc073cba54bef5 : SUCCESS in 1h 16m 33s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f1a289befb414841a1820b52d8ecee92 : SUCCESS in 1h 25m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/03e30992734a4c82b00f959fc961365a : SUCCESS in 28m 20s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b7ebedf0601749268a9193228caa2866 : SUCCESS in 22m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bff8509e2cb64bc6b91d4065e1159fa3 : SUCCESS in 1h 21m 47s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ecebe47ca1524d2fb905c3e0f7fc7a63 : SUCCESS in 1h 32m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac51d1b3db843f0a277f5edc3e75976 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd54803618924d59873b59d301e4bad8 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79459401dbd14d7f99058d1e998f10b4 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05fbea1c74744883895f7d044294fd5c : SUCCESS in 5m 41s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/26ac91388bb44a80bb11df72dbdde2da : SUCCESS in 50m 12s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/38fa1a75cf16441e98cb1d8b87310b8d : FAILURE in 48m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d39aa3fea2bf44138a3a2832145dd565 : SUCCESS in 24m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/08ec8b3a4da14b0c9e21036d13635fb2 : SUCCESS in 26m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ba32a3d4a3cb356a1774acf67340686590121ee7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-30 05:33:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c8b483aa9a5b1f704ed51b1bf9ac22acdb26c89e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 07:13:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8085ac486784339a06a126445e0abcb\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b8fb4e37364a45adb29131e00919ef37 : SUCCESS in 31m 58s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2a760c09a6b64178a0fc4bfaa81545b8 : SUCCESS in 33m 01s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/92233d5c1e5240289b5124260694fe5e : SUCCESS in 25m 00s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8e1bb3f910f54fba9d5338456b590521 : SUCCESS in 1h 10m 13s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1996fd0a499c4364a81f0a8451581522 : SUCCESS in 1h 10m 11s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/02853a8659c84deea5b11c6c9874def8 : SUCCESS in 59m 33s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1ca1a786f23649ce99082a1914c5256c : SUCCESS in 1h 33m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3cf8098b85da446cafb3a0ad92efb014 : SUCCESS in 24m 55s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/649974a6aa8a4e1381f78ab97960d427 : SUCCESS in 22m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/29c5eef9bacd49dfaf8d46cddf446382 : SUCCESS in 1h 09m 14s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/869016666b9040cd88d485c056439202 : SUCCESS in 1h 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/783aab47b6c049fdad892a865b18611c : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/375c5a18cf8b4db3b1977c2040fd2ab1 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4922abbf2ec4464d9fbbaee172a465c4 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9933dba60854194a80f1b83f0d4eeef : SUCCESS in 5m 22s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/89ef88e0195a4a6088b88bed1c41d207 : SUCCESS in 20m 51s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/cdbaec52434046a4bf38c8394703136d : SUCCESS in 46m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a735d47aa4e9438d8d0b1bef7c0dc3fb : SUCCESS in 19m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc531218c0bd4221862ca8775b537602 : SUCCESS in 24m 43s","accounts_in_message":[],"_revision_number":1},{"id":"993abac2991b9797dceff55bd56833ff1e229bdf","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-30 07:17:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7012cd56e810341fc05f2f59461743ceee87fd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 07:17:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98e7f14f681dfeda1890ff1555004d99e754bbff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 07:33:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/011c71e9007d4b838ea84228b6a0b579\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4500a3ad3bbe4d1b937e6b75a79a03c1 : SUCCESS in 27m 59s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ddd82eca45b144a997c70f86b565aef1 : SUCCESS in 1h 03m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef7a3462bd074341a4ae3150a332f6c5 : SUCCESS in 23m 30s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c8e4c954e6c647869c7a03ea8c58a728 : SUCCESS in 1h 57m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9edf517240e4feba87f3c444afe2392 : SUCCESS in 2m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cfba0b0a495d4171991acef3031a0bad : SUCCESS in 5m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0aaa3334d1cc4cab912536c48ea02c66 : SUCCESS in 22m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d1bac710469b28f6c173c8b4d171f2b3be999d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 09:00:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc385424822b4e8eac916b6b62dee238\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d19590624b8b47acaaa2e3aacc5db0ff : SUCCESS in 24m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4f969110b9874b75a9e0d4d40acf6e72 : SUCCESS in 1h 08m 29s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/04df28a7b7ad498ca2680c8f6be6c410 : FAILURE in 44m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d2678ecb307f4a37a53fa0c790761402 : SUCCESS in 17m 41s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6cbad67554ad4909afab6f94bc49efb2 : SUCCESS in 46m 27s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4e28593f99544528a67d236e548ed181 : SUCCESS in 1h 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbfe71cf0b1f47608c0d72c2a05a21c2 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17d902f7526845e099f5429cdb4150b6 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07aed42a90cb4137a14b14f73f143a38 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24dc3f73aff04f51aca313f410fca364 : SUCCESS in 4m 33s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/9f68a03574ce46e2b8809c46df722c26 : SUCCESS in 1h 06m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/da4a7819ac2143238cd5a55c803ee767 : SUCCESS in 25m 50s","accounts_in_message":[],"_revision_number":1},{"id":"12dedc5594aa68866f0bca9abed964422fcafc09","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-07-13 07:28:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fded58a7cf03b90ece819d3c2dca9353c0bf42cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-09 15:58:42.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0c579950f3e20d248c512a46f67efc06912cf8a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 17:43:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ede27537b1184fdaa3ae99d23926851e\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3685a41824a54642a41559450b974086 : SUCCESS in 37m 33s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3507a923b8d3474aa9becdefb1f5c36c : SUCCESS in 41m 09s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/5a8b392b3aeb4221a9abe06a812b70fc : SUCCESS in 28m 08s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e03fec097c544ae3bf8a15960094b486 : SUCCESS in 1h 30m 07s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e0c7c50de46a4652880d1b4cfc604329 : SUCCESS in 1h 20m 45s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/385d840e3f8146e38811ca11cebfe124 : SUCCESS in 1h 19m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2cedf5c10e5f4d8198ad4d33df8c1424 : SUCCESS in 1h 10m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9f49bebfff3346339d1458c33f023188 : SUCCESS in 16m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d6e066b76e6e4b059f6ddf93413b72ef : SUCCESS in 16m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/511bdf87d0c94e1ebe85184d63c84c00 : SUCCESS in 58m 24s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/89b749e9dbc04da5a24c80eaaa13fedd : SUCCESS in 1h 07m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21e59cc18ec646d2ac5334e87c232bd7 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85b116fa88b647b686841becbe41907c : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77520061b3734aaea36c578f5ca29ff9 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5d9a38850af4a2c85104fed4a6939a3 : SUCCESS in 5m 55s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/3742409539c34db8bfbf3d1a5a4dad9a : SUCCESS in 21m 14s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/383a0e7491b047c8ba7003afd76359bc : SUCCESS in 1h 08m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2c7f74bee822452f8b4040c06aa5ce3a : SUCCESS in 18m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3a74797c3764aa4a27a863ed1230db4 : SUCCESS in 19m 51s","accounts_in_message":[],"_revision_number":2},{"id":"17eeade4bf03bb75ccfa898cfdfea3c949669499","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 17:59:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/771eccba716b414f962405732fd55a59\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c0b64a5472604029a1cfa921dd262bc6 : SUCCESS in 1h 34m 45s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f38b40a322c4c078d514672a295b6d3 : SUCCESS in 1h 07m 59s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d3f3c1ccea83458194d8600a47f506af : SUCCESS in 22m 52s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d04f8933413948f08907843595154824 : SUCCESS in 1h 55m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03db3293a684428b9f440c34341ecc99 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f36ee2f4342415ab024cec67fc85929 : SUCCESS in 4m 00s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/995cce30a19f4cd4b14f5735018f3236 : FAILURE in 21m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ee183e20fc191cc3a017052e736039c42368eb8","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-08-11 04:01:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"00eca7721de52582519fe68895755445dfbeb643","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-16 07:03:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"532728649c51cae8233553c1b297963925eea769","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 07:03:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e30191e8a1ee5c18bb27f5ad3aa328a4cf3cecb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 09:09:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d3f38e32b3049459eaa6206e9e9e715\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f05cbe6bf4474ff2b98dc988ba4e5db8 : SUCCESS in 30m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7d4e3d168c2748e5b40e72251bc89656 : SUCCESS in 1h 10m 49s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/136d8bb30dd94c9788c0b0e9e820d9d2 : SUCCESS in 1h 35m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fe44f708c5ec434193a50a6ec1c6ef9b : SUCCESS in 21m 45s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/643d3f7c7b614ae38b1d3fa88301a89f : SUCCESS in 59m 28s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/db37aa4110f048438462c33eba5bb73a : SUCCESS in 1h 38m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a931bddbb22427ba5dd683b5b1971d5 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82a72f8052554678bfa22222b2287c4a : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16b57063f8564d8a83e161b1c9db4a2d : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f888c0cc288e48d08c9f51b401953983 : SUCCESS in 6m 00s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/4a7c953ae8414f6783c25dda6527903e : SUCCESS in 1h 01m 41s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/472a35952a5d4c759e42c1506e4ab017 : SUCCESS in 24m 34s","accounts_in_message":[],"_revision_number":2},{"id":"18efb416387528c0db786b7572f5697f2f23fdcf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 09:09:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"abd90a773085facb9782ba23a768f27a57c17cfe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 09:10:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c5ffbe12e524f5bb21d1fdb7cb4ff39\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6ed8b59d62f4e3883ca0ba7d21b23e3 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ce2b14e5a6833d261e3ddeacd7cc8a1d18fd769c","revisions":{"dbeff225895210f39655cbfe14b4edbda12b2e27":{"kind":"REWORK","_number":1,"created":"2023-06-27 12:02:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/886966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/886966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/886966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/886966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/886966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/886966/1"}}},"commit":{"parents":[{"commit":"75f9186288e6410935d19f430394410289ecb63b","subject":"Fix cinder-volume: add missing nvme-cli rpm package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75f9186288e6410935d19f430394410289ecb63b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-26 08:36:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-27 12:02:46.000000000","tz":0},"subject":"Use erlang-25 from copr on aarch64","message":"Use erlang-25 from copr on aarch64\n\nChange-Id: I9f40167fed9412b3f33f82e2f97f353275431292\n(cherry picked from commit 51675bb232f0c4826ff74351b68c1e16a1893545)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbeff225895210f39655cbfe14b4edbda12b2e27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbeff225895210f39655cbfe14b4edbda12b2e27"}]},"branch":"refs/heads/stable/yoga"},"ce2b14e5a6833d261e3ddeacd7cc8a1d18fd769c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-08-09 15:58:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/886966/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/886966/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/886966/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/886966/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/886966/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/886966/2"}}},"commit":{"parents":[{"commit":"d49cca54922507c040dd1003e628938b1aaf5cf1","subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d49cca54922507c040dd1003e628938b1aaf5cf1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-26 08:36:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-09 15:58:42.000000000","tz":0},"subject":"Use erlang-25 from copr on aarch64","message":"Use erlang-25 from copr on aarch64\n\nChange-Id: I9f40167fed9412b3f33f82e2f97f353275431292\n(cherry picked from commit 51675bb232f0c4826ff74351b68c1e16a1893545)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce2b14e5a6833d261e3ddeacd7cc8a1d18fd769c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce2b14e5a6833d261e3ddeacd7cc8a1d18fd769c"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
