)]}'
{"id":"openstack%2Fkolla~887347","triplet_id":"openstack%2Fkolla~master~I61b70fb4e12ba03b96e79004e735d2ead0f52319","project":"openstack/kolla","branch":"master","topic":"rework-letsencrypt","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-10-30 09:51:05.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-09-26 13:03:32.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-10-30 09:51:05.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-10-30 09:51:05.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-10-30 08:14:02.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I61b70fb4e12ba03b96e79004e735d2ead0f52319","subject":"Rework letsencrypt","status":"MERGED","created":"2023-06-29 23:30:01.000000000","updated":"2023-10-30 09:52:08.000000000","submitted":"2023-10-30 09:51:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":74,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887347-rework-letsencrypt","meta_rev_id":"edce5c94c049466ea9ac3461b0527af521d86045","_number":887347,"virtual_id_number":887347,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-30 09:51:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-09-26 10:35:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-09-26 13:03:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-09-22 08:29:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-09-27 07:15:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2023-10-30 01:24:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-10-30 08:14:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2023-10-30 08:14:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-09-20 13:21:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}],"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-30 00:53:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-30 01:03:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-14 08:44:58.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-07-17 10:25:21.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-08-16 07:37:34.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2023-09-20 13:21:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-09-27 07:15:24.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-10-20 10:15:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-10-30 01:24:59.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"80ef48737458b2d98bd95b1ca4217b7ebd2f2f52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-29 23:30:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d05aa969506a50b8e2f767a50c556dfb5c00be0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 00:53:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea8f87a46b8b4d7ab31099a0f4498e94\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6842ab3ccd9d46a38c1be1c5e6fbdc67 : FAILURE in 19m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/71a6e4541cfe4d6c89088158a96f1a92 : SUCCESS in 20m 38s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c3e0871b1eb04563bd3aea35f8afb648 : SUCCESS in 1h 20m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8cf9fc2d02740689d3c0dc82ccb95aa : SUCCESS in 2m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d103f918e2f4e0f8b5cc6c0d5684889 : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c6ffe91e2154ba4bd6db360929b4a4a : SUCCESS in 2m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/46a1f4fc56de472b82d61f219e8c0dd8 : FAILURE in 17m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b660c4a140754c2785eb72e2af2a1406 : SUCCESS in 13m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5abbeed288db3c50ccec06cdd2f3b809e7a8e3da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 01:03:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45ea19e0737d4f0db84b2f90c9fd7f51\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/799c73c5c8f6406289b08fca81ef426c : FAILURE in 20m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/752867c8ab8e4fb4b53cac25e6c18026 : SUCCESS in 55m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/23c4ab7f0a094ef3abd0a7da105ddc57 : SUCCESS in 18m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/79fcb71340694b039b83ea480a2d3f3c : SUCCESS in 49m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/238c8c6d9f1543a982d8aca1f58f0e3e : SUCCESS in 57m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca57997e0bed44a4b5cc51f01763af9f : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6524f7147e1948279ee52c0f1296fb30 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62eb3fe5a8754a24991f75da7bfa6cd7 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e398f9f205444912aab0403ff4987ceb : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3557a7d8e0d94dc2b0ff101d0f077f19 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f0dde5c6224431a95a2602339c7d0f6 : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c7e2a78201154ebd87eb06b8ad29bea3 : FAILURE in 16m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ac8d192c189e4d99b89c03a798593cb1 : SUCCESS in 1h 08m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9ae183a637534ec58ce636d5a31d799f : SUCCESS in 1h 26m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/928d5ebebb7141708101018180b15f09 : SUCCESS in 24m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1ab27debff8047938cddd38aa69bb38e : SUCCESS in 1h 01m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/32d9321c433d4cc7b6252f19eedd1169 : SUCCESS in 1h 06m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e6c9102b43199b01f83a382d4c6270b693ac8775","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-30 07:30:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5d364f8e637051f87aaf3f9bc8ce291ee6bd0dab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 08:50:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c8465bb47484091bc25b8cd44d1da7c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/11d99f82de644ef28285523e8c09326a : SUCCESS in 19m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bc417b4828a646d29998f82b1b89a709 : SUCCESS in 49m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/97ac3038e817435b8dc4784799951462 : SUCCESS in 14m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d9ebe8e2c5554d48bf8e1d926fea419c : SUCCESS in 1h 07m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ac23d57c87cb4ee2b647b45dd95304d3 : SUCCESS in 1h 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb7797996dc34daca1f99293d041c2c9 : SUCCESS in 2m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01d4d251ee2f4e1fb20ff94e2f3727cc : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1491a2fe1c244dd93e4d3cb2766e565 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d36071e037e94136824f9879ede07bb8 : SUCCESS in 2m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e106ff5aa7dc43d3ad56030dd49a8f49 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a62147aca8f74b9aa22f6ee903fcbe84 : SUCCESS in 4m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5b0478b58dc2453686274736eeb63ab6 : SUCCESS in 18m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f45e40365b8c4a688e28e62bbdc07079 : SUCCESS in 1h 03m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c0c63afff0d244c7b9e658e1b5518873 : SUCCESS in 1h 15m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ced59222756f416b93ddd690be7397d7 : SUCCESS in 19m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/715d454e2a7c4b76b60c135a575747be : SUCCESS in 48m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d122e6ae4fbd4214bd137ef4ca06fb1b : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":2},{"id":"a2aa5a5a7e2aa4f084dfc5743336c65d6a4c421d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 10:20:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a9d730e92b043d996ab0e8af976edad\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b6fc63d42a6648b9bee81e70ae4ddbce : SUCCESS in 18m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/19a546f06ed541139896e18a7ea3d16b : SUCCESS in 21m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8858fe5ca5d9473084006a769a8cc752 : SUCCESS in 2h 47m 33s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/90fdf2486b134ad89854de68715db9e5 : SUCCESS in 2m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/50eb98bcbbc043b692dc18d59718e612 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2a6c51f5dd244fe3a94e1aaaccbeaff1 : SUCCESS in 2m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5cfb533aceab4f6a950c904cebca4549 : SUCCESS in 21m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/faea63aadca74c3a8825a9961ec292f8 : SUCCESS in 16m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b09177316a32d6194ee06dc1640c8d6e703c4653","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 23:24:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a2de8499aa451f6acfe24a04fa23b1e80bdd6663","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-14 00:29:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8a6a41c820c035fa146694dd32160f4de21b56e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 01:46:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b3a3229872b4fbabb847e1f327c89e5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/177af6f2af594f9aaf2a8357123280f7 : SUCCESS in 18m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/07b46a3e20d8418eb1a872644caef735 : SUCCESS in 17m 37s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e7f23ad099374c29aef330ab721574a6 : SUCCESS in 1h 13m 15s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae3cbc3d79b340af8120178be020a56d : SUCCESS in 2m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/93e33528a0824ca99b0ca47e9800dde2 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6c3858b3105c4611825da5a9a48f0acb : SUCCESS in 5m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d780814d5cfc4347a2fa043c87c06a23 : SUCCESS in 17m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/16a31397eaec45a29ece14e5999880a3 : SUCCESS in 19m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3321fcba790c28b186774cd2939034b31efae5db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 02:10:49.000000000","message":"Patch Set 4: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ac5a75cbb2b42d49074c4a0e64bb3a9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3427d95369364481a7fae12ceeb33a4d : SUCCESS in 20m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/eedbcab63f3542dd90a6929685fe2520 : SUCCESS in 1h 04m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1f6e6b5db95246d28a36a18e8e98c1c4 : SUCCESS in 19m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0c16545b4c7845a69a359c92c44f0659 : SUCCESS in 1h 02m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c0e8a3aa44ef4a54a3646374ea309dc9 : SUCCESS in 55m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25d0fd27f284400190038e3f203419cf : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a617f4e9e1374ad58420a115bfe730b8 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d573c432d425410b8740234292e6fa69 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3654595329824ad6ad17849f86d850e6 : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c51c51ea9d704272a01effdecb2dc70f : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1adf9b838bf4af5a304941be4978763 : SUCCESS in 4m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b8fb5b56c3a74d4ba95beced728605c5 : SUCCESS in 14m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/77043adab5784c818e82d137a14c1b7d : SUCCESS in 55m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/83483acf38ca4eb8853be1beb92c6977 : FAILURE in 1h 40m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a4dd396c9b6f4f95b7dd646bd3d1ccc9 : SUCCESS in 23m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/651cea7204a04d2a9b88b7b1a5d3abc6 : SUCCESS in 48m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8b881d960ede44fbacb71c3f55201b23 : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":4},{"id":"7979f4c36eef51e3b4a535613864fbbf5f50fa20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-15 10:39:01.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"46f90db5a599b5b372004439f107ed517d93bd08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-15 11:51:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bfaa22a978f4527afd1805531f22985\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d2b0ca05c95e4365a02f564ca973731c : SUCCESS in 22m 22s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6258414534a74393bd9d501738f1981a : SUCCESS in 17m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2a101a4b4baa44e3b5115bbe58c9bef3 : SUCCESS in 1h 09m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/828db65c704448e5b9b713a4ab005c0d : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b49aa512b0724c0e9d7e5d5895f05d91 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6a956742ffd14ae6bda68ca491f30f90 : SUCCESS in 3m 14s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/147ce678278546348214b1b9abaac59c : SUCCESS in 20m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4968ce35070c4716b6da704a11d25fdb : SUCCESS in 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"995ffdea6148967e8dbd23f81f0b223281f7bfcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-15 11:54:24.000000000","message":"Patch Set 5: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/932563a5ef6e471db6c71d96675b5ae5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0006d1236f3c4ef490798141eea74420 : SUCCESS in 15m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f2816c8ee7a743018eb4703a526f95ed : SUCCESS in 44m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/756d0b80962c4d5c9bfc7ac1d7f12a33 : SUCCESS in 13m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/836f39a2824e40d0abb0e73c54de6142 : SUCCESS in 42m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/faad3e2680a740f9bd0318008fb5b8f1 : SUCCESS in 51m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3129d672fa5949aebaa3a8b4aa463999 : FAILURE in 2m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42202b5aa11b4179a045b3468f109a96 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75936d3068fb47a5818101c78ae1270e : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/350351792a2c47df9d25ebdf19ece94e : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81f6182b9355429e9289f8d8c7efa9f3 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e8a74744d96455fa2da90450aa8072e : SUCCESS in 4m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/946f58d8e7024431ad5a2823448762f5 : SUCCESS in 19m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9f7f2c0258b648078f580fc7cc8d3469 : SUCCESS in 1h 07m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3849beefd499471baa1edcc6b18e090d : SUCCESS in 1h 11m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f3b50269f5664cbcbc1ea0ac49dfd7f7 : SUCCESS in 17m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e487574aaa2e4b2f91a7034c52d3a0a6 : SUCCESS in 42m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0ee6e3b0723140cfbc1c787dd8c22b75 : SUCCESS in 47m 39s","accounts_in_message":[],"_revision_number":5},{"id":"a629ab7973f4ffdf920f073725f14aa046be1f4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-16 17:06:12.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"7f5fe4c5169ca3a761891060a52db001b54ebcd7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-16 18:13:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f10c8bd57b7647788cd12eb84cd11e04\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1b683d98ac1c4d13b8841553ed78a48f : SUCCESS in 18m 30s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7eb6cafcab61480cb4bf25da3f548ccd : SUCCESS in 17m 32s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/278bbd0b1c3f43a79fd4a09d483724dc : SUCCESS in 1h 03m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77204efe106549bca5b6cdd09f1972db : SUCCESS in 2m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/108d6c2ac9684e1c96bde7ba95adc687 : SUCCESS in 2m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc24859f98a649078ed5eb89772ecbdc : SUCCESS in 2m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/53163267d6214a6d8911fd3f01e1cd24 : SUCCESS in 22m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8d75d3ad2fe84981834b31b3cbd408f5 : SUCCESS in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e518f414cce5ca794aa279433f981343d314847d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-16 18:31:34.000000000","message":"Patch Set 6: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/030602ccf7204626a07edd8c7ba2c071\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/034e41f44e1d499991d9f39fce5aee64 : SUCCESS in 15m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/91a6a02172cc4111a264030f29cc4a14 : SUCCESS in 46m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/881562dac19343eabecc977e92026b47 : SUCCESS in 16m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/937080e74a0c4b2194ce98768113076e : SUCCESS in 46m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9d6d731aadfe4dc9b2b296696c74c899 : SUCCESS in 1h 03m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/073dbaa798dc459687fab2f48b6108e5 : FAILURE in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d168a34899554a71a531bcd206482e47 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b49dff3c43e46dbaaec5b4d7c65e709 : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0932c5556c248919d0119af4cd8ff47 : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/384c68d80ae94ccc8cdffc4b163d6f59 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08ddeee903b5441bb708df5051447c98 : SUCCESS in 4m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/21f035379b25428c8d4cde6621fb8835 : SUCCESS in 18m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/28687c62f9544e74a66d73e272e568a5 : SUCCESS in 1h 02m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a0da67f844b4465b8f243d07c3d2bc66 : SUCCESS in 1h 20m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/82403cc280964387bdfa00d60153236b : SUCCESS in 15m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ac5119b9397b4df9aa88b0ac8cfa7f40 : SUCCESS in 45m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cc89559eb4c5417caa0d4baa7c5d8231 : SUCCESS in 50m 50s","accounts_in_message":[],"_revision_number":6},{"id":"5e04390ccc2bad5f9d79d1b3ae4010f4b50c9a09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-17 10:22:18.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"1681f709cb685985e9a79ed9553e8a5c24342e5e","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-07-17 10:29:16.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6b6b11ba915ea7c850c0ae6834f02672181b13e5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-17 10:32:32.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"2da6d8664c9047d3f289f1e7546ae798be3a12a3","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-17 10:33:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"36463ff647cb618769dddefc826410bd54cfe11c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 11:51:58.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4681aed77d6a4d098f5e47695f31fa2e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5c835f11d7d34a37baf146724417c3dd : SUCCESS in 21m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5433241a39e9407a8615a8555fb48d1c : SUCCESS in 55m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7ff8ceab990349069ccef9cded0c8ab1 : SUCCESS in 1h 28m 41s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7688b6a8b35845dba67c640e28e5df3c : SUCCESS in 2m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f8f1437c7be4c44bf9ab2845418f889 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/226ce4149ef7405bbcd5355fab335e97 : SUCCESS in 2m 24s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/40bcbd674b5f44c7a337f7247d322a29 : SUCCESS in 21m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7d7334ae20834b4eb5e6caadc7b8069f : SUCCESS in 15m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8f02fc37b906d015703563f3479c017cb111b892","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:31:41.000000000","message":"Patch Set 7: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e76721681b14bacb5e3013111ed1fd0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c45f0f7a423544cc9998f75cd8d9bb33 : SUCCESS in 21m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8a2f0e787d5e4daa952315499822b04e : SUCCESS in 1h 02m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ce39f9d855124f979f36e3b242d7ca79 : SUCCESS in 14m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/89b5f750ba5c4851a39226af7364ae71 : SUCCESS in 54m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3ea20909e50d453d8df48a1bf6da8bb5 : SUCCESS in 52m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f9f3660676a4724bf33c2f0ee9ac83e : FAILURE in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e82512ebcac4b84991521d9e743ed56 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d80767eca834117b36fc8e3e6963c73 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ee3606c17cf40228fef3d8b36b5a75a : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/585c0187b1ca4c149d9b7c1eeb2c3313 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed5c003309d4c37a0d19cd9a15899a0 : SUCCESS in 4m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6041d8071bc84475b0be5283973035e9 : SUCCESS in 13m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2dcb22e8b8d6435b92e5fe66ae7809d8 : TIMED_OUT in 2h 02m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3faf31ced7734b9ca3468c87bd93e4a4 : SUCCESS in 1h 06m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/40cbf4ae591949dba01504d21917694d : SUCCESS in 21m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/572322665892411e8ca89b4966601c46 : SUCCESS in 56m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/856f599517e84ca6b8c1852e7348df62 : SUCCESS in 1h 00m 34s","accounts_in_message":[],"_revision_number":7},{"id":"19bdf0184c9407cea03836a0bc4f24e666356a66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-17 15:42:27.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"e7d14c4d335900f72c5d76d707adf4765a485654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 16:56:23.000000000","message":"Patch Set 8: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5057e7aaf2b0458e8ec9a6a1b5de8bd9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/81f4496423f842509a0a2290a36f52b5 : SUCCESS in 18m 58s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1c93b76c994445098b1fc02980211dfd : SUCCESS in 1h 01m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b6c2a8b3963e467a9f824808983c09e3 : SUCCESS in 15m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b335f0b6a834492ebe5d61180064312c : SUCCESS in 1h 02m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5002e7d207d646fea872662aa66e9705 : SUCCESS in 1h 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5110d6ec6f14390890e9ca968150c8d : FAILURE in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/030e5644a1e049078bcb966a84848932 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73e88a659a7f4ec1b2d2f948fedc5d4a : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d299e54a8b614fdaa5c0bdf6d77cdbf2 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d7ec910d1514d1c92eee627e932fb99 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be5e85dd6e584141844057fce553065f : SUCCESS in 5m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/267de6dd70564a3b9880947df7e0f8b1 : SUCCESS in 17m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3ec2cea37c48499db7fd9c8c3a4aecd6 : SUCCESS in 1h 07m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/51d6ce55779d4562b8ff10fa79799269 : SUCCESS in 1h 01m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e68c3ca99b63495394b04e38bf216eb6 : SUCCESS in 23m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c76c5d2052854b29b4c0e06d9195d689 : SUCCESS in 49m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/603064d648f443bfb09549bd79e8c710 : SUCCESS in 1h 06m 09s","accounts_in_message":[],"_revision_number":8},{"id":"06428a25982a18f2ef06b6c0966974cc9cf7f04d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 16:58:03.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f67bc0857b2d4e9fb0c7b984137b1a3d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cb90f1a34f204e58a208e3e38dc3a1f5 : SUCCESS in 15m 29s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f69320d18bac4c709f1a59c3ca3da10f : SUCCESS in 15m 00s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/231b81a51f244d9e88c86a065c237cf7 : FAILURE in 1h 12m 51s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c6811f790384060a68b14c1f10e38cc : SUCCESS in 2m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac3038a5bf0d4d769bbea8d308131684 : SUCCESS in 2m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/25fd4a4d61264b0cae31020f34726d1a : SUCCESS in 2m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/db354a0041f144c6adbb874c193834f0 : SUCCESS in 44m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6bf041b1f8d9441eb8f496f3bd3499a6 : SUCCESS in 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"772d38b0ac3aa69b87690d7377885cf2087dec47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-17 17:15:52.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"668ee993f1011a4e7a384ccbe59041a5c1fbaba4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 18:31:58.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4be6432758e945a0b47f59e2990538d5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8fe85b62861344feb3f7c704db9a43e5 : SUCCESS in 20m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/17e9184afd5946d2985f037e81760044 : SUCCESS in 19m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/56a4d07a3f6d4cbfaae14389b63824e9 : SUCCESS in 1h 14m 46s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1a8de0405cd4ee684ef675a6e46cf80 : SUCCESS in 2m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af0a5d1b59374b0f8775bc4cf977e537 : SUCCESS in 2m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6572fa23da474a1b9887b5dcd556eddd : SUCCESS in 3m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/094e8c757a8a4c2fbff43c65a5db55f7 : SUCCESS in 19m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c1a18394c3194cd4900048f10af7441b : SUCCESS in 17m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"54d72b05f001ad2ffe5369499dccdfa9e5e2f515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 18:40:04.000000000","message":"Patch Set 9: Verified+1\n\n(1 comment)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00af6258b6dd43399b74fd62d182e000\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2b6ad87ab93e4ff2ba09b051c8084dda : SUCCESS in 20m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d8734c5e121f43ddb41317eb16d00fd1 : SUCCESS in 58m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/736937d029444442941b046b10460061 : SUCCESS in 15m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a2fcb68d654f47c188d174bc2e0a3035 : SUCCESS in 1h 09m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/858203471c214ed3acce08e902c60310 : SUCCESS in 1h 00m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c13ee9117aaa485eb5633bab4afeb699 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33ad39c8008643d78d51a555a320af55 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3279a22ec0ae465b9f5b33221ebbd7e5 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2f949e09bb0421e98f337cbdaf71d92 : SUCCESS in 3m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/667bf3deb79747719d2b2265aed5e7f5 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7a8545f937b4cc791c4ad1d862c30bf : SUCCESS in 4m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ce15ef8ee14c409987cf72c94fdbc55f : SUCCESS in 17m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/751442021b8f4a399c05719f9b435b59 : SUCCESS in 1h 05m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fe5e7a40e59944d2b7eb5cbf9e451291 : SUCCESS in 1h 16m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b679444faf1047eaa0cdaf3f41bba024 : SUCCESS in 23m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d903bf81a01c4269b017745dbe396bfb : SUCCESS in 58m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/78904b8842b249b7b686d7de749e76d0 : SUCCESS in 1h 06m 02s","accounts_in_message":[],"_revision_number":9},{"id":"2c6d3f061b5c7d6728155aeb75f0c2815fe8e54f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-17 19:31:36.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"ffa3e2bb933d47555900d92584a73a04736ab2d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 20:45:21.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dca84fc2fdf4cba976c8209af6d2f36\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/960a2bde406543c2aa08cf6ce38bb9e0 : SUCCESS in 19m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7e067ddf9f694884bba6c0616fbbaa9d : SUCCESS in 1h 00m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3fcaa4d997f647fb8bc484e84999ce77 : SUCCESS in 19m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/14c085f1babb4cd5a080086572d64ead : SUCCESS in 46m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2733697e9e8d4f8995024c63ca9b6b66 : SUCCESS in 1h 08m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5ae44dc6a15459e9eb1b06704e45392 : SUCCESS in 2m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d475d00fa2b4810b15d0600da015ad4 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de2a417422384c6ab0606c45c7a770a9 : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/459bfd6574be420baa1657211b4156b4 : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50d7e99776814cc594118ee6f8d5bd7f : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c55f0e49e0494e04ae97300228cf4d5c : SUCCESS in 5m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b21a7291f0694232ba1222a484170297 : SUCCESS in 19m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fb89940ba0ea4fb79ee19a608f88ab73 : SUCCESS in 1h 03m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/637e4a7027294d9c80622bd0c405e754 : SUCCESS in 1h 03m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/77d67d16b7cc41a9a1cbd5da78ee05d6 : SUCCESS in 12m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/35941888f8e34a0e9ea9c22a2ef5d4ea : SUCCESS in 52m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/56fdfe7bf454462f9e2c6d92f22a7d7f : SUCCESS in 56m 26s","accounts_in_message":[],"_revision_number":10},{"id":"fe252e3fa342d5ceb2c16dc792eeb302f0036407","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 20:51:28.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01a21f06c2234748b05b2e29b5292c2c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/71f3fe2d50d94751bee1e771f9d75e0b : SUCCESS in 18m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5fcc7dad981b43f0a48c01ca8aabddd6 : SUCCESS in 15m 49s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2ea329cb2435472ca723cd858a44b80f : SUCCESS in 1h 16m 39s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/97aaba8446b34d889a0366e42652a22c : SUCCESS in 2m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ab8c8e4852d4d25b6e78a75322a4d74 : SUCCESS in 2m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45d6706df2a246b29789fb156dd69201 : SUCCESS in 3m 10s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d0b66ad3375b443392e47f9b6839178f : SUCCESS in 16m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7a160cb49747483c81757773e6f4ece3 : SUCCESS in 14m 24s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ab1a8912832b1d4eed4528fd299d846a4b0a5f45","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-08-04 14:04:23.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"4b3f1da8ebe46e5cde9c03991bd30091c457bd5e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-04 17:10:33.000000000","message":"Patch Set 10:\n\n(6 comments)","accounts_in_message":[],"_revision_number":10},{"id":"a64628e17981bd9feef593eb071f45ed1afab627","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-08-09 17:12:27.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"484f43ecb7a3540729655c4277afcb8a45f7a753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 18:45:07.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffd07cc0785d426d941172ad358cb61e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f88947fbbf104a428592318eb8e268c3 : SUCCESS in 19m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/14cd0597be1643e181f3c09a8b054b5f : SUCCESS in 1h 13m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0c8d3cf6f269468b8dc0179804a1a394 : SUCCESS in 19m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/460e1ccacf364ac3af72aa2871602696 : SUCCESS in 52m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/39bdda3485d049c89d4acee453fd5f6d : SUCCESS in 1h 05m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e8b084e433b45aa8e2a441793283391 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01aad178eb59405481f9735c4dcafc88 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94484983874c4fa0b3adcca784128456 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4795548f9a54158aa6793923869bc43 : SUCCESS in 3m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fb5b7eb42ea47349695e5b728ad9d0c : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3892907dbac54288ab6ff3db4a3fb4a3 : SUCCESS in 3m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d075618c82d9405f981695543fc344c9 : SUCCESS in 19m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6dd8e3030d1f4b74a63ff9aefb31cd77 : SUCCESS in 1h 12m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/41be8f51aa1643bc80aea04f4989b32f : SUCCESS in 1h 26m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7c85b8171b12409782e7dec43b217f41 : SUCCESS in 13m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fe004b4ecd454c05980f1f621d62b21a : SUCCESS in 50m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b8942c2a4d4445d5b9bc543c7f765c51 : SUCCESS in 1h 06m 36s","accounts_in_message":[],"_revision_number":11},{"id":"87d6e28d32e89f733fc74975a6f71becdd6d775d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 19:18:50.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd7e8b24094449948f3c6f3b23a64b7b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/92f6b71a5fd84844bf9aca559b07306c : SUCCESS in 1h 13m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5b12588519c443caad0527ff5ad965b7 : SUCCESS in 1h 07m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f1450383a02d498c9d26ec3d65867aa6 : SUCCESS in 2h 04m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/117e36f8def8479e9424b45f7a048aca : SUCCESS in 2m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00116fd047ef4585928a5f1117eea07a : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4639e3cbd09d443ab68b62ede996f576 : SUCCESS in 2m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b55ba1f54ce24f26904c29d46d806ad4 : SUCCESS in 19m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2007d97aa66040118f0e60094ff04040 : SUCCESS in 17m 42s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"4a67bee34001547266faa7b613b6462fb0019812","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-08-09 20:09:33.000000000","message":"Patch Set 11:\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"27492c4c13e3984168dfde86513659e4286300cb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-09 20:15:10.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7ee2c8b39fec801b9773487f6e416d49f3fb48c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-08-09 20:18:29.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"811c19f70de29fe8bc8a56cc9e065d84aa68f383","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-08-09 20:18:39.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3f436297636b689a9dbb5d2429d2247687203eac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-08-09 20:24:13.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f3e84de05e95a4f4392d821d9426981f1b0f499e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-08-09 20:24:35.000000000","message":"Patch Set 12:\n\n(6 comments)","accounts_in_message":[],"_revision_number":12},{"id":"28382cd8816058dd8d3981e02937426e63298758","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 21:30:55.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da42ea8e74194915a8a15468d17e1545\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f70bccdefe51462c9e186ed52d593856 : SUCCESS in 57m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/94d77cd99d994a7b80f36cfba1fc9465 : SUCCESS in 20m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a712d4ad3ee944baa4b4d9bd207c7885 : SUCCESS in 1h 05m 40s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/429733e57dea4fb18dc853aa1a84f2e9 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e498adcbd9034953bcc342e264c77dba : SUCCESS in 2m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5bf213c8e8604e85a57316397cb67940 : SUCCESS in 4m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/409540fc69cd4d399b371c5dbe150714 : SUCCESS in 17m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/858129ad20bf42a7846107cb87b18a3a : SUCCESS in 57m 20s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"3c4c2da83585ce071458ee51824aba572fce3f27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 21:51:05.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dacdf328e2144e7a4b95987561ae4bf\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b784c3a4e6944f19bee0b7f2d936137f : SUCCESS in 21m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6aed2131bbdd456b8284ff641564e0c6 : SUCCESS in 58m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/17e63794b6564c1e84a0133dae4d8210 : SUCCESS in 19m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/17b1955270964346b882fbfd9d3079e9 : SUCCESS in 1h 25m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3966568d2b4c4623b0d2af6bdd4bf12f : SUCCESS in 59m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/883e0d8d00ab44799287699e036c84c7 : SUCCESS in 2m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/585d884e2163473892446f290bcb92ab : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4132a34c7bba4665bc0bc23bc420043c : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac685583c00643cea425a6c341e35f01 : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/663bb5e5bc6d48469ac9cbb6c3f9dff3 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae406376eea46d2b6b0933d58cf608b : SUCCESS in 4m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/426540a25ab34320a2076eff7841dc36 : SUCCESS in 18m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/acff4c70f2d9486a945f47ef80666d97 : SUCCESS in 56m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e3760334030a40b19aa378ceb870840f : SUCCESS in 1h 15m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/70682e0ae3794b919f748525981fd6e1 : SUCCESS in 26m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/abcd6abacff947ffb5934a1be2f105ed : SUCCESS in 53m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3cbba5fe7ba44842a00d31ebc2e8d67b : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":13},{"id":"07f3963da6d79d587a81fb679b92bba5dd06badb","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-08-10 14:37:27.000000000","message":"Patch Set 13:\n\n(13 comments)","accounts_in_message":[],"_revision_number":13},{"id":"8cd541d85d73c24c1c607d97c4e7871d91fed8fc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-10 20:17:15.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"2e8d0806951bbd00245ba16ebc9a3ada3712ba46","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-10 20:18:17.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"83cb87972c03b8a2599326ff3f04434143fbfe2c","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-08-11 09:10:33.000000000","message":"Patch Set 13:\n\n(13 comments)","accounts_in_message":[],"_revision_number":13},{"id":"54c111127b7f3336787a96141fa86db37fec6920","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-11 09:21:40.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"4cd00777916871b1dc196ee6d85098be879fd843","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-08-16 07:37:34.000000000","message":"Patch Set 13:\n\n(4 comments)","accounts_in_message":[],"_revision_number":13},{"id":"d74999f49611746c1e9ce476104fa53f70cc8cb6","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-08-16 09:32:30.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"52443150aeb64c9574129318bc6f15b09926805d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-17 15:20:18.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"0f3c5c45b7ceb953943d9b323bdf5f5babdf6191","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-17 15:29:27.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"d1b5e1a35b8a1f1fb533fcc1cea597a87a710bcb","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-17 15:32:44.000000000","message":"Patch Set 15:\n\n(6 comments)","accounts_in_message":[],"_revision_number":15},{"id":"0fcb7c240ac45dde6a29447a2d56b97c4099a33c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-17 16:58:54.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fad686df10d84ee18d2bc74f48894e11\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0d3c0d8eb9a14209abbd3b691135d2c0 : SUCCESS in 20m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/84afccd417a44273b7e41100e6728347 : SUCCESS in 1h 08m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8557b222ce244a3182916b87ed76a5b6 : SUCCESS in 18m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/88deb7872ec34795a86fdac1e0922a71 : SUCCESS in 1h 05m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3260497353ba423eb78c5bc4643d3995 : SUCCESS in 1h 23m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99879e555d194c06a78521acea73c3bc : SUCCESS in 2m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e975d8dd40d4ca6b77b6344413113d6 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46bc3f30ca4041b5a8e55e8545da5aae : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/441cc632897245ed994dfcee08e102a1 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db3aef9a0c394ebf810d9f392ff63d95 : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adaa53e8d8f64323b4d40492773759a8 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/199a76bb69924047b06185a0caddd3d5 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2acbe0aea2b54e2781814a9353ced45c : SUCCESS in 3m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6fa194ef40bc461aad7d72b69c14b6ea : SUCCESS in 21m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5e01ee77fbb54073864c977606ba2d9e : SUCCESS in 1h 03m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/685fd780503648b99165082e458ecda4 : SUCCESS in 1h 24m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/190871c2d2d84d138618cd1bf7df1af6 : SUCCESS in 27m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1fa1c2eb41774cd59960a73b6df4a92d : SUCCESS in 52m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ea1dab50970d405b818f5f3207e22359 : SUCCESS in 1h 02m 51s","accounts_in_message":[],"_revision_number":15},{"id":"9cd50d9c19a8d98a28767ab21197a73d9fa605ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-17 17:41:33.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15ea37b129fa494083bc7839e42afa6a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/171c9c4dba504468a3e081c31ebb0d2b : SUCCESS in 1h 08m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3ffcdba632704af2a91a454588d5f76f : SUCCESS in 1h 06m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ca5c457bdb9349c791eeda6a206f34ff : SUCCESS in 2h 10m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2232d2b5f56d46eb9c019fe49623e836 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d8768ca93364bf5a2e9414a8915e03b : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/521a50b4a3374ba1aa9461205b4dc432 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8e4ceeea3c364bdb954e7eac656b8922 : SUCCESS in 5m 08s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/894fc56fe2ef4a82b85b16519e15b71c : SUCCESS in 1h 02m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c3e6bb550e284e2f9276f67129702010 : SUCCESS in 1h 09m 17s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f91dbd509bf5ea78e441604ab9ef20d7956ef9c1","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-18 08:15:48.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"4cea33f79da2b1e7c95c8b660dd5a8c4bb4080e4","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-18 08:27:04.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"2d559aa6846fc26392f240fce558f1e9cce308b8","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-18 08:54:29.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"841f0df79ec40fb1158142d8ddb4005861d746ed","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-18 08:59:18.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d23e3baf4985f0db27e463790de6b52c1e475e4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-18 09:27:01.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"a2ec0c9a24a286c5da5b30fadd92c0a131389984","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-18 09:27:50.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"fc75620150f531a2a3020a1d4e5319afe1620e7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 11:00:38.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/479d3f16ba384d90907522c942b2ca8e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0c926df1362a43b694b1876f570c4b8d : SUCCESS in 23m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/841cd5e82a454282b8ed5b3e02c8bdd0 : SUCCESS in 49m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/aa051f70fc9448f383b7896f5a576aab : SUCCESS in 23m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/348a2b5839fb4c4cb6af224fb99b6855 : SUCCESS in 1h 00m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fcc4af7d505d4a9cbbc310c3314ebb68 : SUCCESS in 1h 27m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/446c482fe95a406f8fc5369dcf9210b4 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f723f22858614df9861656d8a951b891 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41903f867e8742c688c6404507e570f7 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9070bcb4e914db3ae72a3e245281d38 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/025d697d3edd4cb7ae05f20e999fca39 : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6178028ccb784f17a3eaf0cf7089ac8b : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5f0c17979b649e4aa97b0125c317e44 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71fb1ce9400646e887b6f39d9eebd1d5 : SUCCESS in 2m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b63821351c8b4716a814b8c8d7169bf4 : SUCCESS in 16m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4002115d8f484be4bc75b71f190bb2e5 : SUCCESS in 1h 05m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/01259510fa7946a7a861efebb282c380 : SUCCESS in 1h 27m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/208dd56669344da1ad807823a254e6c4 : SUCCESS in 16m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d98ec7a0dc794a03bb9353a5cef8b04e : SUCCESS in 1h 00m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cd79701d9e0c44ad9b10181ee3c66b53 : SUCCESS in 1h 17m 59s","accounts_in_message":[],"_revision_number":16},{"id":"021d148db614e423360621cfc543126638897ab1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 12:00:28.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c0f070c2c324406942094ee1b539664\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4fc360ba3df34f288a383718a259ee0b : SUCCESS in 1h 22m 20s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/95000c98f0d04c1d965a0df881fdb835 : SUCCESS in 1h 16m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fd882047c24a48fdbd5f6e4f500c4c03 : SUCCESS in 2h 31m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/081f2a41f20e47f2badb286177e7ad5c : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d371111cdbd4ec2ad3780f34193ec59 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fd179ab8b45e435580e75fd30517404a : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/986edea4dec84b7d8ce03ecfc0e4382b : SUCCESS in 5m 02s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c9a06a7463ef44bab8c46439e776afbd : SUCCESS in 1h 11m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2d0db5b91c4940c0b171280dc7eca639 : SUCCESS in 1h 09m 45s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"c82948c049a86e0a7f16c7313252d114c960736a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-20 13:21:59.000000000","message":"Patch Set 16: Review-Priority+1","accounts_in_message":[],"_revision_number":16},{"id":"f1e1b871ad16035150fe5f2595b9022bcc72c1ca","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-20 15:02:42.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"5f26bd00eff8a9e5f1a110616c7be1ee085f4842","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-22 08:29:41.000000000","message":"Patch Set 16: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"8552f6c0553dc61c4262726cc24dc1e11b9c953b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-22 20:25:26.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"19b54a009f64d82782b2782821404021ca4c9787","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-26 10:35:18.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"c030f3ff18c5a24f02b2cac2cc652c74d890f949","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-26 13:03:32.000000000","message":"Patch Set 16: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"69e0a995d1ba6dcf615b12ae46ae9957b500c3f4","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-09-27 07:15:24.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"bbcc2502af1295e5826face2fa45f460f31f88c0","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-10-24 17:58:22.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"009100750283706fcce0542f93fbe1c2e3a44fe9","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-10-30 01:24:59.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"beb65ab0dbbee0838d2a2a0ac82c04738d5eb332","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-10-30 08:14:02.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"e0f3a5c5910b05cb8fb3631cccf9b067374c9f50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 08:14:25.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"5bb48c27b8b4691f4496c4cdc1824a4dbf727655","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 09:51:05.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cc8ae69ceec49a8ba5b5503a1b530b0\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/987dc03db4e2416c833311a30d1e035f : SUCCESS in 18m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/da59bbcb95f1470ab88dda20c1e76b53 : SUCCESS in 23m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bc635a8c3971426b95c3c8e27f9e3127 : SUCCESS in 1h 10m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3617c57f746f4a0ebe5a9dd47f6e68e1 : SUCCESS in 1h 09m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/548cdaf882594f9cb312214decba30b0 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01ab382553124ec7aa92c3100c779a17 : SUCCESS in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35b8e7ca761b4ff98ce28520f77859e7 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f782706da874ee49bed587ae249868f : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b98e0099f8754c36854d875f52aff7ab : SUCCESS in 3m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/71f09611fa424fbabf42cb147c124320 : SUCCESS in 22m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d15c6596a3d34148adc27d407bc2484c : SUCCESS in 28m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4a71a3e3ca2f4324aa5c496fe93c44bc : SUCCESS in 1h 07m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/029d183c89714d6985366c6efdbf9502 : SUCCESS in 1h 28m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/353de7e56cf048cc90e8c8d2b9a46068 : SUCCESS in 18m 22s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ceaefd04d6cd4d1cbe675d9f97b6f829 : SUCCESS in 23m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ea8989aed4ba4b6b9467cf6757d5799c : SUCCESS in 48m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/94124f515c43469e935489c69c8c592e : SUCCESS in 1h 16m 59s","accounts_in_message":[],"_revision_number":16},{"id":"c23b41fb0f54ee728e322c9c3b86ba42f29b181d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 09:51:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"edce5c94c049466ea9ac3461b0527af521d86045","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 09:52:08.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4efdbfe1ce984f718380f54b6fcb2460\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d3371e7fae48baad3e0916b530502b : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcb827710c204b558f771bc8ac46a005 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"3213dccb8c318134695bcc9c9ba480448964314f","revisions":{"7c37ea747f82302defbeb091eda3462450d6138a":{"kind":"REWORK","_number":1,"created":"2023-06-29 23:30:01.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/1"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:58.000000000","tz":120},"subject":"review letsencrypt","message":"review letsencrypt\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c37ea747f82302defbeb091eda3462450d6138a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c37ea747f82302defbeb091eda3462450d6138a"}]},"branch":"refs/heads/master"},"50d7f24020d560d5821b7dd2562f450d9cab5556":{"kind":"REWORK","_number":2,"created":"2023-06-30 07:30:17.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/2"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-30 07:30:15.000000000","tz":120},"subject":"Rework letsencrypt","message":"Rework letsencrypt\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50d7f24020d560d5821b7dd2562f450d9cab5556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50d7f24020d560d5821b7dd2562f450d9cab5556"}]},"branch":"refs/heads/master"},"5a356a5ff01238287877e0a07a75f6c597549a07":{"kind":"REWORK","_number":3,"created":"2023-07-13 23:24:11.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/3"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-07-13 23:24:09.000000000","tz":120},"subject":"Rework letsencrypt","message":"Rework letsencrypt\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a356a5ff01238287877e0a07a75f6c597549a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a356a5ff01238287877e0a07a75f6c597549a07"}]},"branch":"refs/heads/master"},"bce067e83f85d5b49fed0da6c15bfdb0677abb92":{"kind":"REWORK","_number":4,"created":"2023-07-14 00:29:55.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/4"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-07-14 00:29:47.000000000","tz":120},"subject":"Rework letsencrypt","message":"Rework letsencrypt\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bce067e83f85d5b49fed0da6c15bfdb0677abb92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bce067e83f85d5b49fed0da6c15bfdb0677abb92"}]},"branch":"refs/heads/master"},"89df6c4d79c74ca4c172d77eb742ff516ee8ecb4":{"kind":"REWORK","_number":5,"created":"2023-07-15 10:39:01.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/5"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-07-15 10:20:04.000000000","tz":120},"subject":"Rework letsencrypt","message":"Rework letsencrypt\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89df6c4d79c74ca4c172d77eb742ff516ee8ecb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89df6c4d79c74ca4c172d77eb742ff516ee8ecb4"}]},"branch":"refs/heads/master"},"489a6fe26ff27ba7ae7cd44e10ff9e8618c31324":{"kind":"REWORK","_number":6,"created":"2023-07-16 17:06:12.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/6"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-07-16 17:06:05.000000000","tz":120},"subject":"Rework letsencrypt","message":"Rework letsencrypt\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/489a6fe26ff27ba7ae7cd44e10ff9e8618c31324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/489a6fe26ff27ba7ae7cd44e10ff9e8618c31324"}]},"branch":"refs/heads/master"},"154e04bec35d33b639b91af8d824018f9766097f":{"kind":"REWORK","_number":7,"created":"2023-07-17 10:22:18.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/7"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-07-17 10:22:09.000000000","tz":120},"subject":"Rework letsencrypt","message":"Rework letsencrypt\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/154e04bec35d33b639b91af8d824018f9766097f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/154e04bec35d33b639b91af8d824018f9766097f"}]},"branch":"refs/heads/master"},"ed8088eec52a91924804e63ae97024a393c2e02c":{"kind":"REWORK","_number":8,"created":"2023-07-17 15:42:27.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/8"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-07-17 15:42:25.000000000","tz":120},"subject":"Rework letsencrypt","message":"Rework letsencrypt\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed8088eec52a91924804e63ae97024a393c2e02c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed8088eec52a91924804e63ae97024a393c2e02c"}]},"branch":"refs/heads/master"},"4c3be47bb566dfed66a30439f8f61bbb072a7977":{"kind":"REWORK","_number":9,"created":"2023-07-17 17:15:52.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/9"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-07-17 17:15:49.000000000","tz":120},"subject":"Rework letsencrypt","message":"Rework letsencrypt\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c3be47bb566dfed66a30439f8f61bbb072a7977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c3be47bb566dfed66a30439f8f61bbb072a7977"}]},"branch":"refs/heads/master"},"79468aefb6de02ebc46b16714ee171c294c20bc6":{"kind":"REWORK","_number":10,"created":"2023-07-17 19:31:36.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/10"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-07-17 19:31:15.000000000","tz":120},"subject":"Rework letsencrypt","message":"Rework letsencrypt\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79468aefb6de02ebc46b16714ee171c294c20bc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79468aefb6de02ebc46b16714ee171c294c20bc6"}]},"branch":"refs/heads/master"},"bb074cc7a81a33a98e75f5d1e971a691fe0973a6":{"kind":"REWORK","_number":11,"created":"2023-08-09 17:12:27.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/11"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-08-09 17:12:24.000000000","tz":120},"subject":"Rework letsencrypt","message":"Rework letsencrypt\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb074cc7a81a33a98e75f5d1e971a691fe0973a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb074cc7a81a33a98e75f5d1e971a691fe0973a6"}]},"branch":"refs/heads/master"},"b7fe086ee81101c5ac1c836d1e70585fc9a9a979":{"kind":"REWORK","_number":12,"created":"2023-08-09 20:18:29.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/12"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-08-09 20:18:20.000000000","tz":120},"subject":"Rework letsencrypt","message":"Rework letsencrypt\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7fe086ee81101c5ac1c836d1e70585fc9a9a979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7fe086ee81101c5ac1c836d1e70585fc9a9a979"}]},"branch":"refs/heads/master"},"d20e4e74611db9fa4d7155d2dda9e8a410ab7e62":{"kind":"REWORK","_number":13,"created":"2023-08-09 20:24:13.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/13"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-08-09 20:24:05.000000000","tz":120},"subject":"Rework letsencrypt","message":"Rework letsencrypt\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d20e4e74611db9fa4d7155d2dda9e8a410ab7e62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d20e4e74611db9fa4d7155d2dda9e8a410ab7e62"}]},"branch":"refs/heads/master"},"fdd4ac37671276761b39ef4acff1fd12927b63f6":{"kind":"REWORK","_number":14,"created":"2023-09-17 15:20:18.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/14"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-17 15:20:04.000000000","tz":120},"subject":"Rework letsencrypt","message":"Rework letsencrypt\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fdd4ac37671276761b39ef4acff1fd12927b63f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fdd4ac37671276761b39ef4acff1fd12927b63f6"}]},"branch":"refs/heads/master"},"bf5e435defee2c4ee96bef43566f0b8b5552b709":{"kind":"REWORK","_number":15,"created":"2023-09-17 15:29:27.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/15"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-17 15:29:13.000000000","tz":120},"subject":"Rework letsencrypt","message":"Rework letsencrypt\n\nThis patch is adding \"API layer\" to letsencrypt\nimages which is represented by set of scripts inside.\n\nThis scripts are called by kolla-ansible orchestration.\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf5e435defee2c4ee96bef43566f0b8b5552b709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf5e435defee2c4ee96bef43566f0b8b5552b709"}]},"branch":"refs/heads/master"},"3213dccb8c318134695bcc9c9ba480448964314f":{"kind":"REWORK","_number":16,"created":"2023-09-18 09:27:01.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/47/887347/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/887347/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/887347/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/887347/16"}}},"commit":{"parents":[{"commit":"00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7","subject":"Merge \"base: Debian 12 uses new APT sources file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00cebc3cde7aa35dfdca04dfe490c9dd7c3c40b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-29 23:29:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-18 09:26:22.000000000","tz":120},"subject":"Rework letsencrypt","message":"Rework letsencrypt\n\nThis patch is adding \"API layer\" to letsencrypt\nimages which is represented by set of scripts inside.\n\nThis scripts are called by kolla-ansible orchestration.\n\nChange-Id: I61b70fb4e12ba03b96e79004e735d2ead0f52319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3213dccb8c318134695bcc9c9ba480448964314f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3213dccb8c318134695bcc9c9ba480448964314f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
