)]}'
{"id":"openstack%2Fkolla~887642","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I42e7c4bfd6865388603a0fccaf29f2850a3a481e","project":"openstack/kolla","branch":"stable/yoga","topic":"bug/2023668","attention_set":{},"removed_from_attention_set":{"29873":{"account":{"_account_id":29873,"name":"Joris Vleminckx","email":"joris.vleminckx@gmail.com"},"last_update":"2023-07-13 08:16:52.000000000","reason":"Change was abandoned"},"34579":{"account":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"last_update":"2023-07-13 08:16:52.000000000","reason":"Change was abandoned"},"34405":{"account":{"_account_id":34405,"name":"Boris Lukashev","email":"rageltman@sempervictus.com","username":"sempervictus"},"last_update":"2023-07-13 08:16:52.000000000","reason":"Change was abandoned"},"34437":{"account":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"last_update":"2023-07-13 08:16:52.000000000","reason":"Change was abandoned"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-07-13 08:07:09.000000000","reason":"Reviewer/Cc was removed"},"36151":{"account":{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},"last_update":"2023-07-13 08:16:52.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-07-13 08:16:52.000000000","reason":"Change was abandoned"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-07-13 08:16:52.000000000","reason":"Change was abandoned"},"34585":{"account":{"_account_id":34585,"name":"Franco Mariotti","email":"fmariotti@whitestack.com","username":"francomariotti25"},"last_update":"2023-07-13 08:16:52.000000000","reason":"Change was abandoned"},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2023-07-13 08:16:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I42e7c4bfd6865388603a0fccaf29f2850a3a481e","subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","status":"ABANDONED","created":"2023-07-04 19:55:38.000000000","updated":"2023-07-13 08:16:52.000000000","total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"ca04945ad44e9dce21789ee2fab29bcccbea1cab","_number":887642,"virtual_id_number":887642,"owner":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":29873,"name":"Joris Vleminckx","email":"joris.vleminckx@gmail.com"},{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"_account_id":34405,"name":"Boris Lukashev","email":"rageltman@sempervictus.com","username":"sempervictus"},{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":34585,"name":"Franco Mariotti","email":"fmariotti@whitestack.com","username":"francomariotti25"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29873,"name":"Joris Vleminckx","email":"joris.vleminckx@gmail.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34405,"name":"Boris Lukashev","email":"rageltman@sempervictus.com","username":"sempervictus"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34585,"name":"Franco Mariotti","email":"fmariotti@whitestack.com","username":"francomariotti25"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29873,"name":"Joris Vleminckx","email":"joris.vleminckx@gmail.com"},{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"_account_id":34405,"name":"Boris Lukashev","email":"rageltman@sempervictus.com","username":"sempervictus"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":34585,"name":"Franco Mariotti","email":"fmariotti@whitestack.com","username":"francomariotti25"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29873,"name":"Joris Vleminckx","email":"joris.vleminckx@gmail.com"},"all":[{"value":1,"date":"2023-07-13 08:08:25.000000000","_account_id":29873,"name":"Joris Vleminckx","email":"joris.vleminckx@gmail.com"},{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"_account_id":34405,"name":"Boris Lukashev","email":"rageltman@sempervictus.com","username":"sempervictus"},{"value":1,"date":"2023-07-13 08:08:25.000000000","_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":34585,"name":"Franco Mariotti","email":"fmariotti@whitestack.com","username":"francomariotti25"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29873,"name":"Joris Vleminckx","email":"joris.vleminckx@gmail.com"},{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"_account_id":34405,"name":"Boris Lukashev","email":"rageltman@sempervictus.com","username":"sempervictus"},{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":34585,"name":"Franco Mariotti","email":"fmariotti@whitestack.com","username":"francomariotti25"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29873,"name":"Joris Vleminckx","email":"joris.vleminckx@gmail.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":34405,"name":"Boris Lukashev","email":"rageltman@sempervictus.com","username":"sempervictus"},{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"_account_id":34585,"name":"Franco Mariotti","email":"fmariotti@whitestack.com","username":"francomariotti25"},{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-04 20:10:49.000000000","updated_by":{"_account_id":34585,"name":"Franco Mariotti","email":"fmariotti@whitestack.com","username":"francomariotti25"},"reviewer":{"_account_id":34585,"name":"Franco Mariotti","email":"fmariotti@whitestack.com","username":"francomariotti25"},"state":"REVIEWER"},{"updated":"2023-07-04 20:16:51.000000000","updated_by":{"_account_id":29873,"name":"Joris Vleminckx","email":"joris.vleminckx@gmail.com"},"reviewer":{"_account_id":29873,"name":"Joris Vleminckx","email":"joris.vleminckx@gmail.com"},"state":"REVIEWER"},{"updated":"2023-07-04 20:21:36.000000000","updated_by":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"reviewer":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"state":"REVIEWER"},{"updated":"2023-07-04 22:29:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-05 12:31:05.000000000","updated_by":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2023-07-05 12:31:05.000000000","updated_by":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-07-05 12:31:46.000000000","updated_by":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-07-05 12:31:46.000000000","updated_by":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-07-05 12:40:21.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2023-07-10 08:39:23.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-07-10 11:13:03.000000000","updated_by":{"_account_id":34405,"name":"Boris Lukashev","email":"rageltman@sempervictus.com","username":"sempervictus"},"reviewer":{"_account_id":34405,"name":"Boris Lukashev","email":"rageltman@sempervictus.com","username":"sempervictus"},"state":"REVIEWER"},{"updated":"2023-07-11 10:40:11.000000000","updated_by":{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},"reviewer":{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},"state":"REVIEWER"},{"updated":"2023-07-13 07:51:42.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2023-07-13 08:07:09.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"}],"messages":[{"id":"a120d7d0a4ea4d4cd39f8b2a6832b3f97f852ee0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"date":"2023-07-04 19:55:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1168ef4e853c5a725ffc828fc95fc6f74220013","author":{"_account_id":34585,"name":"Franco Mariotti","email":"fmariotti@whitestack.com","username":"francomariotti25"},"date":"2023-07-04 20:10:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"20b0639e41ae386a65f6ce2890c4493a8fcfc7eb","author":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"date":"2023-07-04 20:16:22.000000000","message":"Patch Set 1: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d113414f13e5688b4c6cd5a82d3b42b10d8999c","author":{"_account_id":29873,"name":"Joris Vleminckx","email":"joris.vleminckx@gmail.com"},"date":"2023-07-04 20:16:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"930c0a16e5df39731b3dbe92886134ec5b6e5b43","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2023-07-04 20:21:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"718edb58360fcaafeb0db62bed7f6f547ebca964","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"date":"2023-07-04 21:06:55.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9b772453a6da7c524d3c92e54cd3837698aa6679","author":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"date":"2023-07-04 21:07:25.000000000","message":"Patch Set 2: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3bb05cc27fc55d8d7711681c7085c334353c04ef","author":{"_account_id":29873,"name":"Joris Vleminckx","email":"joris.vleminckx@gmail.com"},"date":"2023-07-04 22:02:40.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4953d65f7e59c772f1599c92f8135c8d346ee30b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 22:29:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb5d3f7a5cfe4266a84b24086f8e5673\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/640d29f96f634f03ae12e1e11e81b24d : SUCCESS in 33m 51s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4e48790fa16c45db870e29361e39ecd8 : SUCCESS in 39m 54s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/288700034d4b4164b4abe0934ae93f08 : SUCCESS in 32m 44s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/64a9e6ef187243beaa7d80c736568793 : SUCCESS in 1h 09m 50s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/67aab4bb55cb4b4b840f18d7fa09485e : SUCCESS in 1h 16m 19s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/fc64e677d86c42c2bd4d7f8c5f178ffd : SUCCESS in 1h 03m 02s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/714ff34361324d19887d511a3b0d973a : SUCCESS in 1h 19m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/afdc74a72e34487a95e5f65d6f0aa17e : SUCCESS in 20m 15s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dc7af955dac54217944e661f76968e73 : SUCCESS in 15m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0aa03fecdc0e419b9e67a3500bb59067 : SUCCESS in 58m 18s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f939dfe7f0e949af83e05c4b6b03523b : SUCCESS in 1h 07m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57eb7604f261415bac0576a8a928144a : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c0e47170df64069a3192ac58e33bbcf : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd1eaf70176f403db9e0094b8f6c71f9 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b17a9f4d63b9425cb5193fcb6c4ea61e : SUCCESS in 4m 56s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/241ecb42157d4fbabedec87b982247d8 : SUCCESS in 25m 18s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/6534c7df6056481ab727d774a9fae197 : SUCCESS in 1h 06m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/214929c88d8a418197857f142e351635 : SUCCESS in 25m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c85a126febbc4a989b436633eb7bf0b6 : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":2},{"id":"03854c099fb581ced12608b2352a853a53b84c1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 23:04:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af1da0caef144761990d1f2f5040ea9d\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e9e2ef5f2be9441097b1ed14dc74ee82 : FAILURE in 24m 23s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c07d3f93044b4df484384e05beb2aadb : FAILURE in 19m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/98825f7066284da997384d4f3cb1545e : SUCCESS in 17m 37s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9bfa776ad5824990967ac2987a2c2cb4 : SUCCESS in 1h 54m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c44518de7504db6bef49fc30ef93857 : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9983933ba0094be2a7912416dc29139c : SUCCESS in 3m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cfb8dba2118b439b82d689187300228c : FAILURE in 18m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7f6a4fa90e9339cfa42fcb9e64e2f3a4696e6c19","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-07-05 12:40:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9601c8cf8528074868c507e2318e40768c0c4380","author":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"date":"2023-07-05 13:09:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0f04aa00cd6ddb828a2bea1355f9c300e2664fb4","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2023-07-06 18:42:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6c8fab7301bacc27d948c81045a8bcfef7185848","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-07-10 08:39:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"83508f70659b7260e69a1b3e1798d5b250fb204d","author":{"_account_id":34405,"name":"Boris Lukashev","email":"rageltman@sempervictus.com","username":"sempervictus"},"date":"2023-07-10 11:13:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f1a645533ee48767a30b63dbd2faaa403a0c7eda","author":{"_account_id":34405,"name":"Boris Lukashev","email":"rageltman@sempervictus.com","username":"sempervictus"},"date":"2023-07-10 11:14:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e3f55fccd5c53d053b21cb811275872144c04f8c","author":{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},"date":"2023-07-11 10:40:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6733d354f7843e5af44252c2c8cf7ec754ff91cd","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 07:51:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c1d9ef65254bd1a00a76b857a4ae622ba5b766ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 08:07:09.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"06f463b7cfe6fe28fd1a0de34f07ec45dadecd6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 08:07:38.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":4},{"id":"34c18d018189fdb85ba4b9b2a7686003ecc68e1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 08:08:25.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":5},{"id":"ca04945ad44e9dce21789ee2fab29bcccbea1cab","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 08:16:52.000000000","message":"Abandoned\n\nabandoned in favor I7e289bd5ce52319604717e0b319d0642a8e598db https://review.opendev.org/c/openstack/kolla/+/886674 (for better tracking cherry-picks)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a2d1901d694255e7f6e7a529e638a6bdb63a2100","revisions":{"592017183eed8d63cb4e8a30b360be62e1576cb3":{"kind":"REWORK","_number":1,"created":"2023-07-04 19:55:38.000000000","uploader":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"ref":"refs/changes/42/887642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/887642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/887642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/887642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/887642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/887642/1"}}},"commit":{"parents":[{"commit":"482ebb999756bd2ed3fade8380a9a7fd4d3c1c0e","subject":"Update support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/482ebb999756bd2ed3fade8380a9a7fd4d3c1c0e"}]}],"author":{"name":"Alejandro Garcia","email":"agarcia@whitestack.com","date":"2023-07-04 19:51:30.000000000","tz":-240},"committer":{"name":"agarciaws","email":"agarcia@whitestack.com","date":"2023-07-04 19:55:15.000000000","tz":-240},"subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","message":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu\n\nRMQ team is now maintaining erlang-24, erlang-25 ppas, see [1].\n\nThis allows to solve compatibility issues between erlang and\nrabbitmq versions.\n\n[1]: https://github.com/rabbitmq/erlang-debian-package/discussions/33\n\nCloses-Bug: #2023668\nChange-Id: I42e7c4bfd6865388603a0fccaf29f2850a3a481e\nSigned-off-by: agarciaws \u003cagarcia@whitestack.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/592017183eed8d63cb4e8a30b360be62e1576cb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/592017183eed8d63cb4e8a30b360be62e1576cb3"}]},"branch":"refs/heads/stable/yoga"},"1b87820e65588ce592ff890331bdb0ac71c62c7c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-07-04 21:06:55.000000000","uploader":{"_account_id":34437,"name":"Alejandro Garcia","email":"agarcia@whitestack.com","username":"agarciaws"},"ref":"refs/changes/42/887642/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/887642/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/887642/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/887642/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/887642/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/887642/2"}}},"commit":{"parents":[{"commit":"482ebb999756bd2ed3fade8380a9a7fd4d3c1c0e","subject":"Update support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/482ebb999756bd2ed3fade8380a9a7fd4d3c1c0e"}]}],"author":{"name":"Alejandro Garcia","email":"agarcia@whitestack.com","date":"2023-07-04 19:51:30.000000000","tz":-240},"committer":{"name":"agarciaws","email":"agarcia@whitestack.com","date":"2023-07-04 21:04:18.000000000","tz":-240},"subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","message":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu\n\nRMQ team is now maintaining erlang-24, erlang-25 ppas, see [1].\n\nThis allows solving compatibility issues between erlang and\nrabbitmq versions.\n\n[1]: https://github.com/rabbitmq/erlang-debian-package/discussions/33\n\nCloses-Bug: #2023668\nChange-Id: I42e7c4bfd6865388603a0fccaf29f2850a3a481e\nSigned-off-by: agarciaws \u003cagarcia@whitestack.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b87820e65588ce592ff890331bdb0ac71c62c7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b87820e65588ce592ff890331bdb0ac71c62c7c"}]},"branch":"refs/heads/stable/yoga"},"4523f272702c8f45f0e6b70765247e3895db9a21":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-07-13 08:07:09.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/42/887642/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/887642/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/887642/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/887642/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/887642/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/887642/3"}}},"commit":{"parents":[{"commit":"482ebb999756bd2ed3fade8380a9a7fd4d3c1c0e","subject":"Update support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/482ebb999756bd2ed3fade8380a9a7fd4d3c1c0e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 07:53:13.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-07-13 08:07:07.000000000","tz":120},"subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","message":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu\n\nRMQ team is now maintaining erlang-24, erlang-25 ppas, see [1].\n\n[1]: https://github.com/rabbitmq/erlang-debian-package/discussions/33\n\nCloses-Bug: #2023668\nChange-Id: I42e7c4bfd6865388603a0fccaf29f2850a3a481e\nSigned-off-by: agarciaws \u003cagarcia@whitestack.com\u003e\n(cherry picked from commit 0e881148d0036f27b95c5a27459e527fb3b0476d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4523f272702c8f45f0e6b70765247e3895db9a21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4523f272702c8f45f0e6b70765247e3895db9a21"}]},"branch":"refs/heads/stable/yoga"},"722ada394b69fd8a68f70133b90ecfc92c8c6fee":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-07-13 08:07:38.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/42/887642/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/887642/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/887642/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/887642/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/887642/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/887642/4"}}},"commit":{"parents":[{"commit":"482ebb999756bd2ed3fade8380a9a7fd4d3c1c0e","subject":"Update support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/482ebb999756bd2ed3fade8380a9a7fd4d3c1c0e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 07:53:13.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-07-13 08:07:27.000000000","tz":120},"subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","message":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu\n\nRMQ team is now maintaining erlang-24, erlang-25 ppas, see [1].\n\n[1]: https://github.com/rabbitmq/erlang-debian-package/discussions/33\n\nCloses-Bug: #2023668\nChange-Id: I42e7c4bfd6865388603a0fccaf29f2850a3a481e\n(cherry picked from commit 0e881148d0036f27b95c5a27459e527fb3b0476d)\nSigned-off-by: agarciaws \u003cagarcia@whitestack.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/722ada394b69fd8a68f70133b90ecfc92c8c6fee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/722ada394b69fd8a68f70133b90ecfc92c8c6fee"}]},"branch":"refs/heads/stable/yoga"},"a2d1901d694255e7f6e7a529e638a6bdb63a2100":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-07-13 08:08:25.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/42/887642/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/887642/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/887642/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/887642/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/887642/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/887642/5"}}},"commit":{"parents":[{"commit":"482ebb999756bd2ed3fade8380a9a7fd4d3c1c0e","subject":"Update support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/482ebb999756bd2ed3fade8380a9a7fd4d3c1c0e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 07:53:13.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-07-13 08:08:14.000000000","tz":120},"subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","message":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu\n\nRMQ team is now maintaining erlang-24, erlang-25 ppas, see [1].\n\n[1]: https://github.com/rabbitmq/erlang-debian-package/discussions/33\n\nCloses-Bug: #2023668\nChange-Id: I42e7c4bfd6865388603a0fccaf29f2850a3a481e\n(cherry picked from commit 0e881148d0036f27b95c5a27459e527fb3b0476d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2d1901d694255e7f6e7a529e638a6bdb63a2100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2d1901d694255e7f6e7a529e638a6bdb63a2100"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
