)]}'
{"id":"openstack%2Fkolla~888335","triplet_id":"openstack%2Fkolla~stable%2Fxena~I7e289bd5ce52319604717e0b319d0642a8e598db","project":"openstack/kolla","branch":"stable/xena","topic":"bug-2023668","attention_set":{},"removed_from_attention_set":{"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-07-13 12:09:53.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-07-17 10:54:45.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"hashtags":[],"change_id":"I7e289bd5ce52319604717e0b319d0642a8e598db","subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","status":"MERGED","created":"2023-07-13 08:51:06.000000000","updated":"2023-07-17 12:32:45.000000000","submitted":"2023-07-17 12:31:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888335-bug-2023668","cherry_pick_of_change":886574,"cherry_pick_of_patch_set":4,"meta_rev_id":"efdbc24fc3f37c5d205a71acc9c6c4c3a8f77a83","_number":888335,"virtual_id_number":888335,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-17 12:31:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":1,"date":"2023-07-13 12:24:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-07-13 12:09:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2023-07-17 10:54:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-07-17 10:54:45.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-07-17 10:54:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-13 08:51:06.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-07-13 09:12:33.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-07-13 10:16:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-13 10:29:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-13 12:24:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"aaa4ac7582fffac4c3ec8dbca6e9cf5363809c51","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 08:51:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/base/apt_preferences.debian\n* docker/base/apt_preferences.ubuntu\n* kolla/template/repos.yaml","accounts_in_message":[],"_revision_number":1},{"id":"b7de56537a5aaa2507a0f0fa9dd048f177285e76","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 09:00:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"852df81c7ace31b444674e00231277bcfd35d4a3","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 09:00:38.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"613d20687d514448c9c6e2b6bfa64f9e6449c026","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 10:16:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6be60333ccb45c992ad1facb6f11965\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/88234ad2143d4624be72e67064ed0b53 : FAILURE in 37m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0cbf0953edbf4f8eac1950306751bf13 : SUCCESS in 28m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6320fff2768e40ac9e3bdb03f9ad8828 : SUCCESS in 1h 07m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/903c75c4101646daa1df80a93d818c98 : SUCCESS in 2m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4158011c5b542868c37c456e3fcc426 : SUCCESS in 3m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/77ffdf8b41b9455abd21a2437ffcae79 : FAILURE in 1h 14m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"21801f0c8505fc18e52d9b5a86b03203285f55e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 10:29:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40cce8c4eebf441e836fc18fc67abb68\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/995d6cef95f548788737dcb9f0c34980 : SUCCESS in 35m 20s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/83206fd6539d411a9bd160b2498a249c : SUCCESS in 33m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d0e1e9089f3b479184464668cd98de50 : SUCCESS in 1h 16m 27s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/24122dc8ed37495cb91f4ff88def0b12 : SUCCESS in 1h 21m 11s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/68d1701e0aab4f1d96ce6d07d3d64b94 : SUCCESS in 26m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0cfc0f82c8c54b4f866bd3c74a8d7c88 : SUCCESS in 30m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1ce5e226a8f8418286f860815a922473 : SUCCESS in 1h 15m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38098270678e49b5ade3becc60fc49e3 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8faf278af9064ffdb4d43b4193da6e48 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/556e840e42904ee3a05e4b526b9b2399 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffbab827148e4a6d9209c284052d4d7c : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11ab05fb9edc422dbe2d601cea8a3cfb : SUCCESS in 5m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ec5666902b824a4786a97590d04acba8 : SUCCESS in 23m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc79ad9a0d60405f84305d7a2b1f1f15 : SUCCESS in 28m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/303be5ef3b5f4af380334ae098aff9ff : SUCCESS in 48m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/09b98d3467924e8db3e4e37c8816fe03 : SUCCESS in 50m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fbec393269a241e293a5f9520e2f762d : SUCCESS in 1h 09m 57s","accounts_in_message":[],"_revision_number":2},{"id":"79942f720dfdcb25a905514a8edf6a12f7b43715","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 10:31:52.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"4f519396a95a6399167d5554220d57c030e4bc9d","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-07-13 11:45:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e453a28e8149e28bc61828f9f5e9035b35098e07","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 11:54:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"75bbeeec5d0eff84f6c1da5cf1fe07caab7193f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-13 11:55:19.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"130f2f92ef700def928c5a2f36274c5ffb7262e3","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-07-13 12:09:53.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"61e70cbec95545d1ab5734e0d2057e64b824d3d3","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-07-13 12:10:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6831f88adeabd8fb2348c36d07a24855cafa169e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-07-13 12:24:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b28e8d9a98edd4e1acfc0030012252d5e2c2cdd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 13:11:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24109eda3fb34b29b66e3284dcb1363c\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/16d1128bb7b44668963b9b8a02589ba9 : SUCCESS in 34m 43s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/929c20d46cfe49dfb057d74c3ec8634e : SUCCESS in 36m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0108a6eeb5d44447ac5ad39100871a2f : SUCCESS in 1h 10m 16s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/909b75e6a490484c8497327dda11b668 : SUCCESS in 1h 05m 57s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/80d27aebc68940ce86b7eca15faeb0b4 : SUCCESS in 34m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/90ceead1b28a461ca4a6ef852e6c5bca : SUCCESS in 22m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e7498d8872f04a82b5c4ac8ca019d075 : SUCCESS in 44m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fdf3ec2ed9145afbd1814b93b3874ba : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c63a52d9be74c29922538cb3798052f : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8a4a91497d54203936534d81c268c0e : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8abed148c4b458dbed0279fb40746ad : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc8b24757444374aa242b2eb8d4de48 : SUCCESS in 6m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/879fc6d3cee649849dafabb2ebb42c0c : SUCCESS in 22m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac6df19ec0fd4132a79a2f57fd5a7442 : SUCCESS in 23m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f8a0631484246e9944e01167285ec08 : SUCCESS in 57m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/56d7a97d012b4e0cb1945583ea5b819a : SUCCESS in 56m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e7aaf6bf9afb4d50bceb715090013d9a : SUCCESS in 1h 09m 19s","accounts_in_message":[],"_revision_number":3},{"id":"2ddfda02f7796cbf4e53cb1697360e732c616dca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 13:11:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1642038d0ad349d4a9e98df3a947f23d\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/92b13004a9c949f98b90a20b664978e1 : FAILURE in 32m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/492b957fb4714699b6d3727323a7efd3 : SUCCESS in 25m 13s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c51b34da32f64942a08c702fa8df723f : SUCCESS in 1h 03m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f78e5d6340d4ebeaca85a94439cd29f : SUCCESS in 2m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/acd58a3852be49a6b87e6d3a40509c58 : SUCCESS in 2m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/caa086c896ec48928e0450f0405a23f0 : FAILURE in 1h 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8adfca1609a262726203d85a418d5b7a90ef84f0","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-07-17 10:54:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bf21212031153a4dd457952261bd62cbd9f38126","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 10:55:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a1465a01dcc18a460edd95fcf02be6fd55b96799","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:31:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f6098d3473e4c95a26774174ac037d4\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8be30fdd14dc4b7f9746d611924bc832 : SUCCESS in 25m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/22d83a2504ab4b6baf78af6e9e7db93f : SUCCESS in 1h 23m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cce7048b55c94c59817efb6efcd9e977 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d43f7a261cc946b2b1f70831634de664 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1e19f75950443e5ae76deba47ed30d8 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f7ed12e91c4b6882156f0dc8efa51a : SUCCESS in 5m 31s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/583cb1f393e6460fa91072c74b57967b : SUCCESS in 23m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b9ba390e3d5941249e18088d6c97d3c0 : SUCCESS in 1h 04m 11s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/18b38d8f066f4de38927b3ba069876bc : SUCCESS in 1h 18m 36s","accounts_in_message":[],"_revision_number":3},{"id":"9bda35d4032ff3bd793e72fe0377a4ccecee9a09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:31:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"efdbc24fc3f37c5d205a71acc9c6c4c3a8f77a83","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:32:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4406a50a814b483782dbefbbb3e56c7c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ffaa5b19b3d4c7e8c4628424f908d4f : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dc38e719758f607c4cb2c9ba7d242c73e55e6238","revisions":{"00141f5ecddd4c02a4441581d760d47df708fe90":{"kind":"REWORK","_number":1,"created":"2023-07-13 08:51:06.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/35/888335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/888335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/888335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/888335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/888335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/888335/1"}}},"commit":{"parents":[{"commit":"fb41d37463c53526fedfd8bb99d558c23e3a2031","subject":"Fix cinder-volume: add missing nvme-cli rpm package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb41d37463c53526fedfd8bb99d558c23e3a2031"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 07:53:13.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-07-13 08:51:06.000000000","tz":0},"subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","message":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu\n\nRMQ team is now maintaining erlang-24, erlang-25 ppas, see [1].\n\n[1]: https://github.com/rabbitmq/erlang-debian-package/discussions/33\n\nChange-Id: I7e289bd5ce52319604717e0b319d0642a8e598db\n(cherry picked from commit 0e881148d0036f27b95c5a27459e527fb3b0476d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00141f5ecddd4c02a4441581d760d47df708fe90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00141f5ecddd4c02a4441581d760d47df708fe90"}]},"branch":"refs/heads/stable/xena"},"08773f986fb302fd4b2f553224887bad9d9a068a":{"kind":"REWORK","_number":2,"created":"2023-07-13 09:00:07.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/35/888335/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/888335/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/888335/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/888335/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/888335/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/888335/2"}}},"commit":{"parents":[{"commit":"fb41d37463c53526fedfd8bb99d558c23e3a2031","subject":"Fix cinder-volume: add missing nvme-cli rpm package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb41d37463c53526fedfd8bb99d558c23e3a2031"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 07:53:13.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-07-13 09:00:05.000000000","tz":120},"subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","message":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu\n\nRMQ team is now maintaining erlang-24, erlang-25 ppas, see [1].\n\n[1]: https://github.com/rabbitmq/erlang-debian-package/discussions/33\n\nChange-Id: I7e289bd5ce52319604717e0b319d0642a8e598db\n(cherry picked from commit 0e881148d0036f27b95c5a27459e527fb3b0476d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08773f986fb302fd4b2f553224887bad9d9a068a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08773f986fb302fd4b2f553224887bad9d9a068a"}]},"branch":"refs/heads/stable/xena"},"dc38e719758f607c4cb2c9ba7d242c73e55e6238":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-07-13 11:55:19.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/35/888335/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/888335/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/888335/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/888335/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/888335/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/888335/3"}}},"commit":{"parents":[{"commit":"fb41d37463c53526fedfd8bb99d558c23e3a2031","subject":"Fix cinder-volume: add missing nvme-cli rpm package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb41d37463c53526fedfd8bb99d558c23e3a2031"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 07:53:13.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-07-13 11:55:03.000000000","tz":120},"subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","message":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu\n\nRMQ team is now maintaining erlang-24, erlang-25 ppas, see [1].\n\n[1]: https://github.com/rabbitmq/erlang-debian-package/discussions/33\n\nCloses-Bug: #2023668\nChange-Id: I7e289bd5ce52319604717e0b319d0642a8e598db\n(cherry picked from commit 0e881148d0036f27b95c5a27459e527fb3b0476d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc38e719758f607c4cb2c9ba7d242c73e55e6238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc38e719758f607c4cb2c9ba7d242c73e55e6238"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
