)]}'
{"id":"openstack%2Fkolla~889725","triplet_id":"openstack%2Fkolla~master~I7c01ba1127f3f0bbfddf373fa66cd2627e627415","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-08-01 13:50:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7c01ba1127f3f0bbfddf373fa66cd2627e627415","subject":"mariadb: Add log for mariadb-upgrade","status":"MERGED","created":"2023-07-26 15:20:18.000000000","updated":"2023-08-01 13:51:37.000000000","submitted":"2023-08-01 13:50:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889725","meta_rev_id":"0793007cd3563cb369abc7784c686c754e2874cc","_number":889725,"virtual_id_number":889725,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-01 13:50:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2023-07-28 09:07:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2023-08-01 12:20:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-07-26 17:47:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-07-28 01:42:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2023-07-31 08:56:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-07-27 09:36:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":1,"date":"2023-08-01 12:20:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-26 17:33:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-26 17:47:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-07-27 09:36:08.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-07-28 01:42:28.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-07-28 09:07:32.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2023-07-31 08:56:35.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-08-01 12:20:19.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"08fea26c95a47e67fe6165a9a12f0b2468abfa16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-07-26 15:20:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"770c08cdb623f17f1ef64f60b6d1736400dfd10b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-07-26 17:20:17.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"80033ac689aa9a34018e6f01997c8ad4298f5a27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 17:33:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d364d3b63a4d46f5a0e8d31e68fb1210\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9f003501bfa6470da53102e33b102489 : SUCCESS in 19m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/82b484b5ba3a4e9d975cc9a0706c0b1d : SUCCESS in 1h 04m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c229791549b14fa98f57aa315e3ffa24 : SUCCESS in 17m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c5f420aa39a84420b477401ed2406a38 : SUCCESS in 58m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b09e08010d0d4a638b9d638ac1ad3431 : SUCCESS in 1h 05m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/255d947e5cd24c579c5cae8bdc0d7225 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cb00628d39247de9298d92e795b592e : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bca458e3b714010aa28023973552ad5 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90f3d2d7393f45ac894343a06f6f230c : SUCCESS in 4m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb3026c4a12847fbbc2ef044e0e53dd6 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1455938bdada45588ee3537a84cd1d55 : SUCCESS in 4m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2aa8ad5e73124ae4b66f469b10b8fbc0 : SUCCESS in 18m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1cccfe6412884ae5b42df6db8398eec8 : SUCCESS in 1h 07m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/166b74d6f5154c07a2d5eeb9f8c37170 : SUCCESS in 1h 10m 22s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a5cb34e5b48a4260aea763c597368d05 : SUCCESS in 1h 18m 13s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a821065efa614fc08383aff30a104069 : SUCCESS in 2h 00m 40s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/3417c81808bf472a8b32664d5f60468a : SUCCESS in 28m 04s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/813e8517eeb447019ff61cfcbe136d53 : SUCCESS in 19m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d18dc7b579b0458bbb47320cfc5739cd : SUCCESS in 59m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c147c1c782b8442ba11c40de09379f1a : SUCCESS in 1h 04m 20s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ed23d512735e4ce4b3977958257671db : SUCCESS in 1h 33m 48s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/26584b5f08404edf80dadefdca887b95 : FAILURE in 39m 40s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/edb3f91ec3bd4485a12a51841d5c60e0 : FAILURE in 25m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe476ffa94b7fcfdcbe3cc4d548642cca0f9bcd6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-07-26 17:47:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0647efff5bb77ee222c023878c20dc52065cef70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 18:07:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/404a2e453fb945c59c6ed47974321475\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fa251039d91145f992e40e5a99e05590 : SUCCESS in 1h 20m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/848257296e08444fa42fb40140d9080b : SUCCESS in 1h 27m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8e5baeca7d52482f95b927dba55b601c : SUCCESS in 2h 32m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3d833dad84548f08ce92c1b26365f60 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d0eec52677b498797d92ca49baaba12 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6cb340d47280464a847c09f609ded7cb : SUCCESS in 6m 14s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/93a1b2f00cd24956a035048c883c0d3e : SUCCESS in 1h 20m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fae6b7d68ad441228eb8046ef8532383 : FAILURE in 1h 35m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0eaa65be98008ffa4c92702399af36af83ca2bc","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-07-27 09:36:08.000000000","message":"Patch Set 1: Code-Review+2\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"23ac0ddd44b0c38df00c04c9f892982cde66671a","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-07-28 01:42:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"48571aaeeefa409e959877c5f02179ef334fc6d7","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2023-07-28 09:07:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b722700f3c3d0f2095feee56566b4bcfa52eb6d3","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-07-31 08:56:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e3061d21680717ca0bf9a9db358ff2a6d826df75","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-07-31 09:39:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc09fd101d22f5db807440882bade290c8d3b604","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-01 11:41:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc70e82a6b9798a46c0d47bf5b9918b51d787e06","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-01 12:20:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"594a7feae6c01d57a0fbf001f91f47545188fa02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 12:20:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cdcdcf8b1cda429f0a9291163f88c4425ebd8de3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 13:50:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d49f81e3f84648cca8af5d268743598d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b34998a8cf9f45eeb2af0477e5d6b0d8 : SUCCESS in 28m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1faaa1e3effb4e51b4e92cfeefe702cd : SUCCESS in 1h 18m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fb35f9337a314c578e842325584ce5ea : SUCCESS in 1h 25m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a00257373bc9499ebe78935c3337280f : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/849ef1e88cd6453cada949e20ef3dccd : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47d05840ef43441c84ff440faa9106ea : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3285b5cf42944416997d464c0a39a027 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d993a61da1d45cfbec233e9e76aa415 : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8e85d3ccb15842bf8254e709a2a6d0a8 : SUCCESS in 15m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e17052f264354b2fa8b1a5d306225d3d : SUCCESS in 52m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/80a86bae66c14798b1f8bae81b30e90c : SUCCESS in 1h 22m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/94c8f6c6ced24269a7814594127a39e9 : SUCCESS in 25m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f95ddb2999914c99984542a43f0ed913 : SUCCESS in 53m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e63d5d04ba4d412ebc8a6b07c1c3c00c : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":1},{"id":"dd28c67d6c4aea9ab57e2ef9748d0fb616f940a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 13:50:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0793007cd3563cb369abc7784c686c754e2874cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 13:51:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6103b724b5da449bafd7de8103001066\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee3eb7fcaf514aa893111af25a1ffe52 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1841cc852da0adaada3bf29398c9e3f6a48ccbec","revisions":{"1841cc852da0adaada3bf29398c9e3f6a48ccbec":{"kind":"REWORK","_number":1,"created":"2023-07-26 15:20:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/25/889725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/889725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/889725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/889725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/889725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/889725/1"}}},"commit":{"parents":[{"commit":"1c3b7cec6731d555caa59a571ef25603182d0cad","subject":"Merge \"Fix dockerhub secret name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c3b7cec6731d555caa59a571ef25603182d0cad"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-07-26 15:19:57.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-07-26 15:20:16.000000000","tz":120},"subject":"mariadb: Add log for mariadb-upgrade","message":"mariadb: Add log for mariadb-upgrade\n\nChange-Id: I7c01ba1127f3f0bbfddf373fa66cd2627e627415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1841cc852da0adaada3bf29398c9e3f6a48ccbec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1841cc852da0adaada3bf29398c9e3f6a48ccbec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
