)]}'
{"id":"openstack%2Fkolla~890464","triplet_id":"openstack%2Fkolla~master~I4da94a1725f563d8a957a064573323708fdc87df","project":"openstack/kolla","branch":"master","topic":"update-etcd-v3.4","attention_set":{},"removed_from_attention_set":{"25733":{"account":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"last_update":"2023-11-17 18:49:38.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-08-26 19:31:44.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I4da94a1725f563d8a957a064573323708fdc87df","subject":"etcd: update to v3.4","status":"MERGED","created":"2023-08-03 22:18:38.000000000","updated":"2023-11-17 18:50:41.000000000","submitted":"2023-11-17 18:49:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890464-update-etcd-v3.4","meta_rev_id":"26642765112b4d7914916a52cec22adc1ee5363e","_number":890464,"virtual_id_number":890464,"owner":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-17 18:49:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-11-17 11:36:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},{"value":2,"date":"2023-11-17 13:37:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-26 19:31:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"date":"2023-08-08 15:34:09.000000000","_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},{"value":1,"date":"2023-11-17 13:37:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-11-17 11:36:31.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},{"value":1,"date":"2023-11-15 11:52:57.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-04 00:09:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-04 07:27:08.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-11-15 11:52:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-11-17 07:01:01.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-11-17 11:36:31.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"ac1d23f38d89232e4ec3430d22f2430361d497f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2023-08-03 22:18:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b44acd8a42f7a027e8105d0e30bf506171d9c6d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 00:09:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/722e717f9d8944ac89c68f552bf98d0b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a68da6b5b928484a96d1c2ffb520690d : SUCCESS in 20m 51s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0142ac5be869430caea5db64a4124979 : SUCCESS in 1h 07m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d9a106da90eb4935866d2fdf76d662d8 : SUCCESS in 18m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0095107261054ffa84ff6b0ab562421a : SUCCESS in 1h 00m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/98870f19b5df4d028f3638aef382be8a : SUCCESS in 58m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/204f6a43db6441319f221724a6451dab : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9f1621fc3b14bc6acc7b1eb3874ccec : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d82840c98dd4a7b8ee9557ee72896ac : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a228430275f24a7a88e31d6bee34857f : SUCCESS in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b156d810cdc24aababb86a20c3739f76 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/386acbef06514ceabec3669b21de248e : SUCCESS in 5m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a32c41e67b474a1796d55614f076ec32 : SUCCESS in 14m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7c6282525bbd4a2bbb5df696702eebd1 : SUCCESS in 1h 00m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4fa5d97660d4416499a49cd69128e658 : SUCCESS in 1h 16m 32s\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/d0008bb342a643ac8fa057dfff38fcac : FAILURE in 56m 52s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/716efaa9634540e5b3ed25a83eebdf82 : SUCCESS in 1h 10m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f5f70621d26045a68e9dd7fb221e190b : SUCCESS in 1h 19m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b8ad3ffb7abd4aa1852dd2355c27ac7e : SUCCESS in 55m 25s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/e1e70ca98d67429db0fa51c695cbaf6a : FAILURE in 47m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8cb64e9c9a048f6701e1157380f6ff22e71619f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 00:18:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6627b706a08942a59226bcc4fe7474e6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/475a36441f7a4fd68578c3201e492960 : SUCCESS in 14m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eb00de10197c4516b7b371e923ad8c52 : SUCCESS in 1h 58m 04s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/47fd6b17b664489e82b2917dec408d38 : SUCCESS in 57m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8dcca3705b8e4e94b8323906170a24a7 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8962dcb94be543f18984235b040ad7ee : SUCCESS in 2m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eca2060933bc432c94b70eaa5f3a27e2 : SUCCESS in 5m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/445aef76e03c4a8bb42b3e953ca99ffa : SUCCESS in 15m 40s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/062f13f35f9a408e87a270be81dc423a : SUCCESS in 1h 36m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd09fcb2e27e74963dd91a3cb2598d9515cb84f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2023-08-07 11:43:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2450caba8b2a061eb33639aead5637e85a1e51dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 13:21:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9782495bd02e4aa5803e947f7b25a499\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9fff04df4f184aa58dba8fa721587763 : SUCCESS in 26m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9eca4030130f40a485017d0fe07eb832 : SUCCESS in 1h 09m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0769bef78f1a48adaeccf08b665ed7d8 : SUCCESS in 22m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7ad79e9c17454e5f8bf44ac80b58048b : SUCCESS in 59m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bf104cce12904a2e91b7b4bf56ae192a : SUCCESS in 1h 08m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07b5968d635b408f91a3319b11265de5 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80c74c4da5154bdc8d462776dc9ae53e : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6edaded54e0d484d82ace2e3873a3644 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5fb526289374ad88ddee6f0c125fbe4 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3733ca94f33b409bade538da29f44529 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd3e4c920a6b4a38a0c717caf8f85a90 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd0cd85077a047868205f221053e052f : SUCCESS in 3m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f92e98f2e0cf499aba057f056e2cf2e4 : SUCCESS in 17m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4955362b9c5b42059363c5a2f34d6986 : SUCCESS in 54m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2aa44e2f571d405594202b13672d3040 : SUCCESS in 1h 25m 31s\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/cf0edc9f296d467a856792855d5c3f13 : FAILURE in 1h 00m 05s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8c3a040dd41549738d381f0270c85d39 : SUCCESS in 22m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/09ab32cd183343018d84d8a00d107c6b : SUCCESS in 1h 10m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/91c7250275b74f5da0709b0d93911a75 : SUCCESS in 1h 13m 46s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/3f9043188aa54fe196cf647d8a27fb79 : FAILURE in 1h 09m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ea07cc6c090953698e8ad0f41d493796efb4ff1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 14:11:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0391ad3e6ab346eb95d0d3dd1f807813\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a938cefaf4f8487bae4807272d89568e : SUCCESS in 1h 18m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/64470188375f4b55a1e56e697dead162 : SUCCESS in 23m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b63f53bff56846e8b84857bbd98d0a51 : SUCCESS in 2h 26m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/005b593eae4b4773a930618ac545c138 : SUCCESS in 2m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/429cc17b63d54062a2284b5ff549f3ce : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89cf98e9948943a4b32b8bfc4aa44ad9 : SUCCESS in 2m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/60cecf38c2134ce3b71bf403f91e334a : SUCCESS in 1h 38m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e8b73be342334bc2858bdf509280127e : SUCCESS in 16m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a5e145a1794c77673d5968b4d18de10d640dd5d8","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2023-08-07 16:07:39.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6a6f23c86e1a799082023c762f2d5e7b8fad2058","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2023-08-08 15:34:09.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0ac9a0993c7d445f9c1b92319bef8b33efc33eeb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-26 19:31:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"50d1c2fedb106b645aeb5e273b6a892eced7c0fa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-15 11:52:57.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"22a28c7ef83c5f55d91bce2f0331945ae63441f9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-17 07:01:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"faa7ae8acfb0a006b7976f92f20f7a13e3202f15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 08:54:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ca1a28165e84985bdf7c017bd8e4835\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1d590a77b30c4ef991ee1350e657690e : SUCCESS in 22m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c65e75acd6e84acfbfa8b5d2c5019824 : SUCCESS in 1h 13m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dd5e62a78fdb4b8f982eecedee235379 : SUCCESS in 18m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cad19df68d86498f89a7ea2552d12350 : SUCCESS in 27m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ef68ad8b5909435780e335d510ddf581 : SUCCESS in 1h 06m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c2bfdadb42d34254b8b0977e2640bd1a : SUCCESS in 1h 47m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7da179bbab154f79a5d6b99137919fc6 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a96ffd5ede314a4fbb7c266db2e8323e : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8d6431500954670881da74c7306878d : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f5056467b5542e3ad0616de1060d474 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28cb2657d95947438b89918917a9955e : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88c2f93f8e7440428fe12603a67dca94 : SUCCESS in 3m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f9fa285e181b49bdb286c57666d4c018 : SUCCESS in 18m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a5678cd8e3584fc481bd893c6f7a2bc0 : SUCCESS in 27m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5e4b05de164d4d03b2d70c1a9be61504 : SUCCESS in 1h 04m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3fb69c256ac84794be26e3a1fb83885d : SUCCESS in 1h 35m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ef7a5a99953f403ca798c8c184c49dca : SUCCESS in 18m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9f49d05eb5f042caacb75cb1faaba5d1 : SUCCESS in 24m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6d491af2d2db40d8be30309fec7e5a78 : SUCCESS in 44m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/196dbddcc7e2437ab2e4bc3cb46b21f7 : SUCCESS in 1h 25m 32s","accounts_in_message":[],"_revision_number":2},{"id":"8e5531e5cccb2d62e4438222254ec178d64a2b4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 08:55:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/616666b7875e42b6ba20856d46ba8e67\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6ea259d8a1d147d0aed4aa8b7c6c3b67 : SUCCESS in 23m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f84de0545db04433a0f8fbd3b72b26ed : SUCCESS in 18m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c831da82c7ed4690bb56f2d310d32e9d : SUCCESS in 1h 52m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/23d5f273b99243489abcce06c8035a69 : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cb515fabed0a4aa98e2540863274595b : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e3156e5e75b44f6dad4c1d895fb3a3c6 : SUCCESS in 2m 55s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1eabad9e093c476c9425928e992cce28 : SUCCESS in 1h 20m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/50b5c280d593430a9892e66c42f23128 : SUCCESS in 18m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f904b919ad36e6af18e15755f1c5536899f3ad55","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-17 11:36:31.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"45aebe2c98615f5d3fb6fc882fc181be2480d77f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-17 13:37:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d7e048d7da0d98dcaeecd603c5c4a2480672521b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 13:37:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c5d6650a3795a775cf282c7b366e7bb614fadeb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 18:49:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b693826e45b450988b48ddf2554bbce\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d505b477859749109c5fa320d2897afc : SUCCESS in 16m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9455b3d76455414db47009d530600e7c : SUCCESS in 28m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/930cb9d37bcd458b987dea5b006e1ed1 : SUCCESS in 52m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a91aaad6aa47425c9e1e01f212125c26 : SUCCESS in 1h 30m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b050b7daa28b495699a1f386485d0f14 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46d952e0d7b849a584cc1266a368eb74 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2252327cd664a1e8c2d979e595786e7 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f10146eaf6e4581b9c35eeb81160450 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3040ad486cdb4d1cae905efcadc88e82 : SUCCESS in 3m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ca0778dddc914006ba1247efa2e7ac00 : SUCCESS in 23m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/81af4bc14862467b8c62507ef3e75706 : SUCCESS in 30m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f48f1b29dfa7455485269e0f8fa95fa8 : SUCCESS in 1h 03m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d027f50def03424a94f201559c6a5526 : SUCCESS in 1h 24m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/abecfd961e644a8daff6c7f341facaa0 : SUCCESS in 34m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/39566cbdd33049e38e824592997697c3 : SUCCESS in 31m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/648216ab7ae74b01aa28411297b3cce1 : SUCCESS in 1h 08m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cba004d9e8bd427b8caa302cc5ee4926 : SUCCESS in 1h 33m 34s","accounts_in_message":[],"_revision_number":2},{"id":"8e18dc24f7828ea84e3d1c3aec5289a42bc076f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 18:49:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"26642765112b4d7914916a52cec22adc1ee5363e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 18:50:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05862b1720b04a338d4d60fb1307a3e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/619c55ce7acb48fabc2668587d0be16a : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d33693251bcb41f7a199c661d5ef391a : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c3905e4b4c83b5e983be6f3b98c690cfc84c7d8c","revisions":{"41ea84fa71b6002f607d066e732c55cb43459827":{"kind":"REWORK","_number":1,"created":"2023-08-03 22:18:38.000000000","uploader":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"ref":"refs/changes/64/890464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/890464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/890464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/890464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/890464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/890464/1"}}},"commit":{"parents":[{"commit":"bd68acb60fe850534541347d7d44091927c5f416","subject":"Merge \"Update stable branch management doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd68acb60fe850534541347d7d44091927c5f416"}]}],"author":{"name":"Jan Gutter","email":"github@jangutter.com","date":"2023-08-03 22:15:47.000000000","tz":60},"committer":{"name":"Jan Gutter","email":"github@jangutter.com","date":"2023-08-03 22:18:37.000000000","tz":60},"subject":"[WIP] etcd: update to v3.4","message":"[WIP] etcd: update to v3.4\n\n* Updates etcd to v3.4\n\nTODO: Proper release notes are missing\nChange-Id: I4da94a1725f563d8a957a064573323708fdc87df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41ea84fa71b6002f607d066e732c55cb43459827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41ea84fa71b6002f607d066e732c55cb43459827"}]},"branch":"refs/heads/master"},"c3905e4b4c83b5e983be6f3b98c690cfc84c7d8c":{"kind":"REWORK","_number":2,"created":"2023-08-07 11:43:19.000000000","uploader":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"ref":"refs/changes/64/890464/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/890464/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/890464/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/890464/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/890464/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/890464/2"}}},"commit":{"parents":[{"commit":"bd68acb60fe850534541347d7d44091927c5f416","subject":"Merge \"Update stable branch management doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd68acb60fe850534541347d7d44091927c5f416"}]}],"author":{"name":"Jan Gutter","email":"github@jangutter.com","date":"2023-08-03 22:15:47.000000000","tz":60},"committer":{"name":"Jan Gutter","email":"github@jangutter.com","date":"2023-08-07 11:42:43.000000000","tz":60},"subject":"etcd: update to v3.4","message":"etcd: update to v3.4\n\n* Updates etcd to v3.4\n\nChange-Id: I4da94a1725f563d8a957a064573323708fdc87df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3905e4b4c83b5e983be6f3b98c690cfc84c7d8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3905e4b4c83b5e983be6f3b98c690cfc84c7d8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
