)]}'
{"id":"openstack%2Fkolla~890512","triplet_id":"openstack%2Fkolla~master~Ibbf75fa73b8397ea3686b42cec6cad6d8f6d3858","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-11-17 19:01:29.000000000","reason":"Change was submitted"},"23303":{"account":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"last_update":"2023-11-17 19:01:29.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-08-04 15:34:37.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"Ibbf75fa73b8397ea3686b42cec6cad6d8f6d3858","subject":"Adding db sync script for venus","status":"MERGED","created":"2023-08-04 08:52:28.000000000","updated":"2023-11-17 19:02:50.000000000","submitted":"2023-11-17 19:01:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"890512","meta_rev_id":"dd3e9d70445afb5031f97561826112f84c663d02","_number":890512,"virtual_id_number":890512,"owner":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-17 19:01:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-11-07 08:36:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-11-07 10:17:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-11-17 13:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-11-17 13:37:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-11-07 10:17:57.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-04 08:59:50.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-04 10:18:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-24 13:21:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-11-17 13:37:57.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"f511573cdebb7f6f10d1492257d9c4acf9a00f03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"date":"2023-08-04 08:52:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14f2503f768e4bb910befedac52732d16a796361","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-04 09:08:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"65636973d15369e2aff95767a1bf9f0621f8d6a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 10:18:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d79df7376a140949e6576879a050bfd\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0665ca68b15948ca825a16fe84ed1828 : SUCCESS in 18m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/afbbd30fd04741caa13926e883531f59 : SUCCESS in 1h 03m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/078b40c7b220427f87ff7fb9e8b64105 : SUCCESS in 14m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5916386a554f46f997932bd0d448c37a : SUCCESS in 1h 02m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/50d233dfbaf64b9c862b3b9dfab0f418 : SUCCESS in 1h 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff345dffa9524533b5de7e379997af30 : SUCCESS in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/201849b5bf9745cf902b2c30414b5584 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a13e746c9e1d428c935c43e4ba1ec81c : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/237c8d180fdd46adb23025c00d89a730 : SUCCESS in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6990b9166f0540b080d7fcc008908978 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4cacee6e2b84dd5b7ae9115441b4d36 : SUCCESS in 4m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d07dff472e2d436ab4e6fca0760fc6ec : SUCCESS in 16m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b4425d55163f40c083300f4685c5362c : SUCCESS in 1h 06m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f20c8a617cde42a6810186905dc6af07 : SUCCESS in 1h 19m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/70f75cdb10e242a5b0ac872133533822 : SUCCESS in 18m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4c558f2e249247219dc025d4f2d7a5fc : SUCCESS in 52m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f1fedc9b19ce4945b311420f3bbb84e8 : FAILURE in 1h 09m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e80756d146e9882b3463468bf94d58d4969a6ca5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"date":"2023-08-04 10:25:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"57f5ed4fe6248dffb37511be51e50c4d77b0ed45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 11:45:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e04a5306b86c48968130c7d904891d88\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6ef0f54a5f2e4231bc8b692f0368df46 : SUCCESS in 21m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a1004ffa91e2494fb1e7f84d64168164 : SUCCESS in 52m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/67348ceeb54048cf966f68d6dc666b23 : SUCCESS in 16m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ef727eb8613d4432a8a049914d60a2ab : SUCCESS in 52m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/640f828b0c1548a8a62de0ade607f906 : SUCCESS in 58m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df1720f7e9f74cf79030f704790781cb : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/047ea6a3360240fb98449f039b7bda0e : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/135a50ee58ba47188d7e370747f0af73 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/675d8082e98c4110b0a4bad5f604eb05 : SUCCESS in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50a2e7907f0b4ca3ac511f1c0cb194f0 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1734c7d734ee45c080c65626c1eb6e96 : SUCCESS in 5m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bd8c726814b349ecaafc5ea3b25264ea : SUCCESS in 17m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b2fe215e89fd4194a6d4250fffbbc915 : SUCCESS in 1h 06m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e4bcb0d9bc6c418bae8b8f331c87de5c : SUCCESS in 1h 14m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8462943a0ebf429e98f0d8d961fe35d5 : SUCCESS in 28m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2e9fdb8f1a4648baaf272db9b4c67e18 : SUCCESS in 52m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9ec823ea3a7e4417a9b66fc8f01aa698 : SUCCESS in 1h 08m 20s","accounts_in_message":[],"_revision_number":2},{"id":"fe4794ac2923f1273cad2648fc5a6c0f6edc677f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 12:40:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4967d701b274eef9353dcb64f172ba9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e46e16fe520c41ee977b9a31c0b4988a : SUCCESS in 19m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cc883be2626a4e98b65733628cb708a1 : SUCCESS in 22m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/598f7366460e4236b72012a67ac50e49 : SUCCESS in 2h 08m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81f60c16802e41e6ab8add56aab14174 : SUCCESS in 2m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4362a2e950274807ade992f80e31bdb8 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ec839faf21364fe9b4e315217b34b3e4 : SUCCESS in 2m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e9c9e8e27daf4692b1f215287500cb1d : SUCCESS in 1h 06m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/210c2a72e23246cbb5fea298d71ffa07 : SUCCESS in 17m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c824e86b005119f1f02f2ca516dae1907efba80","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-04 15:34:37.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"75243db8dc4e165d1829da0a06f749665e688206","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-07 08:14:20.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"a36da044af6f05c1177b97f948acd126f304aba4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-24 13:21:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0ba1c142b890012a1124d3df8b78a66add600104","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-24 15:37:26.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"fd7031169cc07895ed4d1a5157a43b6332a997c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-07 08:36:03.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"694cd8957330d2e7c6232bd3f43befab59d527f2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-07 08:36:17.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ce83328f011f16182d97494c8c7a2ba0c5c1a85f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 09:52:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21235ff445e34c5abbb7f4df70dd0999\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c5ee60a8cfe641a3952852507a5ef70a : FAILURE in 19m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3735b0b4da2943aa83be56163646067e : SUCCESS in 18m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c5f72f0e8a1d40d98f04915350507794 : SUCCESS in 1h 13m 51s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e1cecfcd08024605bbb56c032e91cffb : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/351c3f4869b54076890505db7ae6c9f5 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/edfbe0d6d432416c84b97740bc3dbde8 : SUCCESS in 5m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4b83f907cde144b2863396e5305d3ceb : SUCCESS in 17m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e467f722e915452c984ca65968ea0ef4 : SUCCESS in 1h 09m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6ce654b951637db06a0e79e9dbcfaf7223ac7a5d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-07 10:17:57.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"84b9c8cce234d0e25039d2a2c142e36d4046f2b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 10:22:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b06a26e5783646c891eff11ece6dbd19\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/917702c4d1d34541a35fd8abc94a37ee : FAILURE in 23m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/26dd6e0aaf7c4928a21ab5d73628af61 : SUCCESS in 53m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ba08515af809470dbc22d0b480263961 : SUCCESS in 16m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e37e86590f894bffa3e90654cf642bb9 : SUCCESS in 21m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/089e1fa15f20458dada903225d21cf7b : SUCCESS in 47m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ff5455b72c694ec889cb46152ce63ec2 : SUCCESS in 1h 05m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeff0c0d2ca34b95b477236eec2a9af1 : SUCCESS in 2m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d03532cbc76a44059e279cdee88d64f8 : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5bdde089c70e485eb3a891382bdb794a : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ef905433d84473aad2361730e349498 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd7e1bf76254b75af5cc8af0e7b68cf : SUCCESS in 4m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bacea1384cb94c9682669b89aa3906f0 : SUCCESS in 21m 20s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/57d7f1a0135d432ebdfd91636eacf3ad : SUCCESS in 33m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/74aaa69af07649d293d010ea2ce88f60 : SUCCESS in 1h 07m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bf3c86800f654663ae5581c371d19ea9 : SUCCESS in 1h 40m 10s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/68cd3c4cb3564aca88bd736e56ebd058 : SUCCESS in 22m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b1d193bb91034926bd876aca2bdc6d19 : SUCCESS in 29m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ab3ddaf22b9542a5b98706728ffe963b : SUCCESS in 48m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c8a44a9b93654912be612d387e00380f : SUCCESS in 1h 28m 20s","accounts_in_message":[],"_revision_number":3},{"id":"ad77196f8b8cc5f24951e5b9413c163ff4e0836a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-17 13:37:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fdb7f27feaad6fd9d81dad684bc474dc57d2adc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 13:38:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4659f74936fa01641849307ff4d131385dc167c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 19:01:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f7aabe096c64f288476beed6fed21bd\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/45b14ff8d9c946cb8744d2cbe79cf58a : SUCCESS in 19m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7281e3a6369548b88f30782eb2e0c906 : SUCCESS in 25m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3e512102e2cc46ef9b221b26df0e354e : SUCCESS in 52m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a7fd5d2f599f43f0a897aeebabe27d9e : SUCCESS in 1h 36m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5612805b6a94c4db9a1825758e744f3 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be686b7977d74b3496c60206b859e17b : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f09e9298fd04272b611cfa413b948fc : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92ed3a8355eb488d931403272c429618 : SUCCESS in 5m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a86763388e3047eb8bb0cbea009c57cc : SUCCESS in 22m 41s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0d725fe0ac834b8eb5acb544fd348978 : SUCCESS in 54m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bb8a366528f7422489986fdf5b10e282 : SUCCESS in 1h 02m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4d7b048c8444406f91712fdfa55ee05e : SUCCESS in 1h 31m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/86048ea529c2456da36b3c8914d50df3 : SUCCESS in 34m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8ebc6dfc97604bb6af9c7bc54e79f7df : SUCCESS in 36m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/081f2f9c33384ef182cd4a9ccc3860e7 : SUCCESS in 1h 04m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/11f314d977784bfdacf9dc8bb4a61e19 : SUCCESS in 1h 39m 50s","accounts_in_message":[],"_revision_number":3},{"id":"afd2da1e769ffc13d3a16413e1866a1bd2c6df7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 19:01:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"dd3e9d70445afb5031f97561826112f84c663d02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 19:02:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed76b47819504f9d82d4f8c556631780\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e7b6e8733f64d739fdef4165cf8af83 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"772de9662caa5e2041d557322a9b6a7344b3e0f0","revisions":{"ccf76e1ddeeb3c98512b8267a828b28e31575ab3":{"kind":"REWORK","_number":1,"created":"2023-08-04 08:52:28.000000000","uploader":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"ref":"refs/changes/12/890512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/890512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/890512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/890512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/890512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/890512/1"}}},"commit":{"parents":[{"commit":"bd68acb60fe850534541347d7d44091927c5f416","subject":"Merge \"Update stable branch management doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd68acb60fe850534541347d7d44091927c5f416"}]}],"author":{"name":"howardlee","email":"lihongweibj@inspur.com","date":"2023-08-04 08:46:52.000000000","tz":480},"committer":{"name":"howardlee","email":"lihongweibj@inspur.com","date":"2023-08-04 08:46:52.000000000","tz":480},"subject":"Adding db sync script for venus","message":"Adding db sync script for venus\n\nThe script for database synchronization is missing in the venus-api image,\nwhich makes it impossible to create the required database tables during\nvenus deployment.\n\nCloses-Bug: #2029915\nChange-Id: Ibbf75fa73b8397ea3686b42cec6cad6d8f6d3858\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccf76e1ddeeb3c98512b8267a828b28e31575ab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccf76e1ddeeb3c98512b8267a828b28e31575ab3"}]},"branch":"refs/heads/master"},"1e5e44ab486101d576020ded2b08bd0de0c130b4":{"kind":"REWORK","_number":2,"created":"2023-08-04 10:25:47.000000000","uploader":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"ref":"refs/changes/12/890512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/890512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/890512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/890512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/890512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/890512/2"}}},"commit":{"parents":[{"commit":"bd68acb60fe850534541347d7d44091927c5f416","subject":"Merge \"Update stable branch management doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd68acb60fe850534541347d7d44091927c5f416"}]}],"author":{"name":"howardlee","email":"lihongweibj@inspur.com","date":"2023-08-04 08:46:52.000000000","tz":480},"committer":{"name":"howardlee","email":"lihongweibj@inspur.com","date":"2023-08-04 10:22:05.000000000","tz":480},"subject":"Adding db sync script for venus","message":"Adding db sync script for venus\n\nThe script for database synchronization is missing in the venus-api image,\nwhich makes it impossible to create the required database tables during\nvenus deployment.\n\nCloses-Bug: #2029915\nChange-Id: Ibbf75fa73b8397ea3686b42cec6cad6d8f6d3858\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e5e44ab486101d576020ded2b08bd0de0c130b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e5e44ab486101d576020ded2b08bd0de0c130b4"}]},"branch":"refs/heads/master"},"772de9662caa5e2041d557322a9b6a7344b3e0f0":{"kind":"REWORK","_number":3,"created":"2023-11-07 08:36:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/890512/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/890512/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/890512/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/890512/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/890512/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/890512/3"}}},"commit":{"parents":[{"commit":"bd68acb60fe850534541347d7d44091927c5f416","subject":"Merge \"Update stable branch management doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd68acb60fe850534541347d7d44091927c5f416"}]}],"author":{"name":"howardlee","email":"lihongweibj@inspur.com","date":"2023-08-04 08:46:52.000000000","tz":480},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-07 08:36:02.000000000","tz":0},"subject":"Adding db sync script for venus","message":"Adding db sync script for venus\n\nThe script for database synchronization is missing in the venus-api image,\nwhich makes it impossible to create the required database tables during\nvenus deployment.\n\nCloses-Bug: #2029915\nChange-Id: Ibbf75fa73b8397ea3686b42cec6cad6d8f6d3858\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/772de9662caa5e2041d557322a9b6a7344b3e0f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/772de9662caa5e2041d557322a9b6a7344b3e0f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
