)]}'
{"id":"openstack%2Fkolla~890538","triplet_id":"openstack%2Fkolla~stable%2F2023.1~I3ca4e10508c26b752412789502ceb917ecb4dbeb","project":"openstack/kolla","branch":"stable/2023.1","topic":"disable-gnocchi-build","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-08-08 07:12:54.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-05 11:43:55.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-09-08 11:17:42.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-09-08 11:17:42.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-09-08 11:17:42.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-09-05 11:43:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3ca4e10508c26b752412789502ceb917ecb4dbeb","subject":"Revert \"Disable build of Gnocchi containers\"","status":"MERGED","created":"2023-08-04 16:24:12.000000000","updated":"2023-09-08 11:18:46.000000000","submitted":"2023-09-08 11:17:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890538-disable-gnocchi-build","meta_rev_id":"863f390e1543169802a51e7959b47ea126f84c5a","_number":890538,"virtual_id_number":890538,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-08 11:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-09-08 09:56:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-09-08 09:16:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-09-05 13:18:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-09-08 09:57:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-09-08 09:57:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-04 17:48:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-04 17:52:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-08-04 17:52:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-08-04 17:54:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-08-04 17:54:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-08-05 09:01:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-09-05 13:18:22.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"65a7c567efe22d6e2a3e5e58d134ca4a7bfedc58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-04 16:24:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"baaf3d2c7187d542f1e521feefb0144c2eab2a39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 17:48:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdd8f757af174288a96479c0bb49b6a2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/93e52dda77fb45dd9b93e1cef9aba4f2 : SUCCESS in 23m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8ba691142bdf423e97070dc94d5f1e98 : SUCCESS in 1h 03m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8a58a7c226104c09afce7d259547190b : SUCCESS in 18m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ee03c63fc28d45fc9ede3329c1ab5f8d : SUCCESS in 1h 09m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b988cf1eaa86406d985ba4393e358183 : SUCCESS in 1h 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4172de9c6d44b899be23289b37d5d60 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27554adbfb0f4d44bf2f8c5af4b5b2bc : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c69c87f6c2c4673b84041deb803e531 : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a31d80a473474c7497a7494244336314 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3f279b48ac04af09f5810f14b80e701 : SUCCESS in 5m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ae41b0a484bd466888e383f1efc3f457 : SUCCESS in 25m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/40c19d33bc6649b5943a79cb79aaa128 : SUCCESS in 1h 07m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a351a26cfb68498293bffab962369aee : SUCCESS in 1h 17m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/da82ec3413ea41c8a77a9a8006e8345d : SUCCESS in 31m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9e54b0e6330c4c438f80f0a61b0ebc9c : SUCCESS in 59m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1f6dece3c00744b1a86876daa6cda696 : SUCCESS in 1h 08m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b9e2e26abdeba7402236ec3a17a8ff0bcde677d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 18:11:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e6e52eb43694f4d80c5faddcd833ba6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8072030922a544baa8a18bdab6ba4e6a : SUCCESS in 28m 30s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0bacee8f13124365bf5527666737a777 : SUCCESS in 26m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6c4e676741fd4db49bc926f1509ab5a8 : SUCCESS in 1h 20m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ec1fd10cd1b47afb7bb72a5ae33370d : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e8e3ef33e3fe4f74a3eba23f1bbb4ce6 : SUCCESS in 2m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/dd6a478e90f046b09766bd58b2663a07 : SUCCESS in 1h 42m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6597a46879254023a52547de4b39fa13 : FAILURE in 16m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"18c2191258d6c8fe3c026266380bc64b16bab5ff","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-08-05 09:01:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e8d2beaac9e34e472483df16965b2d39706f9992","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-08-05 09:05:15.000000000","message":"Patch Set 1: Review-Priority-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49c36bd4c138e47d0441c5c6dda619823ff970d1","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-08-05 09:05:23.000000000","message":"Patch Set 1: Workflow-1 -Review-Priority","accounts_in_message":[],"_revision_number":1},{"id":"9091b131b352c5b633681fec53fc25d57fb076d5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-08-05 09:05:35.000000000","message":"Patch Set 1: Code-Review-1 -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"0e56e9258262698dffb8fcc7f2c7ea4e3eb21b11","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-08-05 16:06:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2518f2240b2e7b1d80a77439bb7589935e4017c1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-07 07:51:02.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49931d22a4c99992c329f6c01f25ac87a4e00327","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-05 11:43:55.000000000","message":"Abandoned\n\nneed backport https://review.opendev.org/c/openstack/kolla/+/893722","accounts_in_message":[],"_revision_number":1},{"id":"2dcb11b39c9adc61596dfebe15fb7e9e4d0664d2","tag":"autogenerated:gerrit:restore","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-05 12:00:25.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"fd09426ff82e7533299abaa38aca644fb2da8e11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-05 12:04:31.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4a6dd1044ea4fc1ec629f38ebab2b3f4f0e9e460","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-05 12:04:56.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b5ce8c7a18d1606f8e29772faef322f3ed224a69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 12:47:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74199c78108443cda205aa94685d9261\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b42396c84536434bb9ab9d02c0b1199f : SUCCESS in 36m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/062b405688ee45e5b18311209914186b : SUCCESS in 38m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/50fe1f7cd00246ac9d04b2fb0cf167b8 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/606111fa6c3f49c1853a16cecf2be928 : SUCCESS in 3m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/69851238845b441f8e24bea517832561 : FAILURE in 31m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4de557d3a2eb4841a68b92fbcc8ac20d : SUCCESS in 36m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"39a2d46460a9bb90a19f7f809572d8616bb143d4","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-05 13:18:22.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1df8c2f25f8865079b426da711cb765380700569","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-05 13:21:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6dcf9e6c41fcf242beda33f4ca75d47e2c4d11dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 13:24:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/44a9b0dce9774d678440e4221413d7d9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0a6292fcc06a4c1abe3ef3cc4c41fc2b : SUCCESS in 23m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1b0f45bdd10c41a4b5a4b383f564d824 : SUCCESS in 1h 05m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1bc5c694cd14413a9ae5791e6a9d9259 : SUCCESS in 21m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03c5a5b05df94812aa8dde606fe747ec : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0cbc59c1d0e48cdaea918c94c9541cf : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df667484716743109bea33bc6b697b50 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b9373e34b8f4326a239f5dfa52c6835 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42d8c60dafbc4202aa2c0b27a606bd15 : SUCCESS in 5m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/796bd03a1c114d40bcc929652fd722af : FAILURE in 37m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f41aaab4e3b4466286b626b5fe240c86 : SUCCESS in 1h 14m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8a78005f9f254dfc8dbbadeeda763a06 : FAILURE in 8m 49s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9fadde85488446a2ab3b22213a5aabe0 : FAILURE in 18m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/39b2c356bd80468f8fe1a2334d3cd735 : SUCCESS in 50m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/27fbd3ea6b2f43139f1bf4cf2c438d27 : SUCCESS in 1h 06m 16s","accounts_in_message":[],"_revision_number":3},{"id":"f0f6ee4d6e940b14010dfd135196939d26767282","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-05 13:25:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dbb009ea96cd1f390853ecb979a9a53e01cea36d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-05 13:34:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c5713be4292dfed811ce6c76a2204b9285dd621e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-05 13:34:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cc12786f10c876ede8565ac156b0790d1f989cc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 14:49:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a861f085834b470fad8e08dc3b45f49f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a9916f28d74d46938001bfe6995b6e1d : SUCCESS in 17m 48s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/34270e23945d4861a5d7c1488577d5b0 : SUCCESS in 57m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1e9a7df402cd4700a13871a9ff059ed2 : SUCCESS in 17m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3394dabc79a34f0f9f4ee893c4712675 : SUCCESS in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2f4dd0b537d4839beb25c0e7b63c4a9 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eaf986f43fa24f49872910226ca41774 : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c7014725e34424dbb1a1d0876655457 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cce96703fd3b4ba9ad272667cfcec834 : SUCCESS in 4m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c0d31b79743b48ba8d605dfd996365ed : SUCCESS in 18m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6f731dc3e056455c9b4680ac91720fbe : SUCCESS in 1h 05m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f2a8458c9cc6466e8ca331e0d6c0320c : SUCCESS in 1h 07m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/18bfa84af0794365a4c08593e8459da1 : SUCCESS in 21m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d4e28f963bda43d08c7439e1b073c0f8 : FAILURE in 40m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/853ebdd4cce74806ad1ed63e2f65df46 : SUCCESS in 1h 04m 37s","accounts_in_message":[],"_revision_number":3},{"id":"cbe0a9c893616032de8a9a9a7338e0aba8da1363","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 14:55:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1f049b1090e4bd68d15760ead58312e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b4b1ab4929724e8d969179242cfc5729 : SUCCESS in 1h 10m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/efb5043e34bf4fa48f63d9a81751bf8f : SUCCESS in 1h 18m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3dc204cf3d934c8aab83cad90d4bdf2c : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3785af99448149b8b33a7393a8a77e1f : SUCCESS in 5m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6e237d6a520242f1b1e6e133f9ae116e : SUCCESS in 26m 16s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9e2dd1302f0a43e1b2bffcc5c5b95d05 : SUCCESS in 20m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f5984d6ee8da687375295d4ece0cda5a205068ff","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-05 15:50:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"149c244cf6e02d1a004c819c1e43ab87ef9c2478","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-05 16:20:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"653b0f4c7aea4296f0cdcd234010fd5dac5f090c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-05 16:51:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"59cbb6024e204ec7d27a6c59f26be4b3676aa488","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 17:16:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cafc4de89f24ed0a9e224d49609e01c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ebee86299d964114a00a991602c84d2b : SUCCESS in 30m 54s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/56c79245d75d470db77e62827c43058c : SUCCESS in 59m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5104b158db1f4516a20e149f5e160a09 : SUCCESS in 21m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a1b2cb04851492984372312802413e2 : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70134c2fc38146aeb895df534fc30ccc : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a5fa6e710324614ae143b67586c046a : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8730e5cb185647ae8a62dd199e7d314b : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a06e2a4072a430397dc622af6fa1aba : SUCCESS in 4m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b980b225b9234c248d93f4975b1e4362 : SUCCESS in 22m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c745a1b441cc4a7c9cd9b8d2f52c610a : SUCCESS in 1h 04m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d7f7f6b5058c4cfa935188df6e51969c : SUCCESS in 1h 21m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/62dabb6e227043479b261341c19d4502 : SUCCESS in 24m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2b4c7484067c476c89d3b1fa08c7d9f2 : SUCCESS in 1h 03m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/37177eb2bf654837873546e641898f60 : SUCCESS in 1h 14m 36s","accounts_in_message":[],"_revision_number":3},{"id":"f742ae933490f5526fbf2c61ed0c8699b7a81bb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 17:16:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72cf23e7968d4f749b212d14281e355b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ac7131fb092645d6b78c38e1d6c4fddb : SUCCESS in 22m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f4ab69e041ce452d808b4eedf304e8ed : SUCCESS in 20m 55s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1c06e7fefa64635bccb81ce87578709 : SUCCESS in 2m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/85d69a33897f44d7bb4375a61939a07c : SUCCESS in 2m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/05a9dda18d2949cb8a557798d3c68b30 : SUCCESS in 1h 24m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3e12657c71ec4afdb3ec935f5d5212c3 : SUCCESS in 18m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b85e017a8a784481b133bfd437a78435ba4172b1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-08 09:16:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5e08463ac688320935481978ff37c4187a2bc81d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-08 09:56:00.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7a2f2176f7d52e646f0e9d141be2d87109d9eabd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-08 09:57:17.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"65bf6c9a0a12ea893f1ce686f8611f15cf47cffb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 09:57:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a754f7b2ab0dbf3113723e3e37711d9e6bafa9cb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-08 09:57:49.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8ab9988909239cb7fdd7899459801be0ee56b20b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:17:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c38b3904c46e4dde8e8e30f274e864dc\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4969ee5082ba4abf81f8051b13313562 : SUCCESS in 17m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cabd612522346d9a8786426456f4f5c : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/313f9c20ef904ae79f4b9def2c335f46 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab677aa19a4f4ed78153041fbc03d039 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c67bf639c1bb4de2a73ed39ebd82372e : SUCCESS in 4m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c950d1cb3a254c0f9e97ee9dc733ddf9 : SUCCESS in 18m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c02e7ece6a2d45aba312fa84ee82c6b6 : SUCCESS in 1h 12m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/38d789aaae8b489fa4855825912d483b : SUCCESS in 1h 15m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/25f4ccaf746a4cfe9c1083030868dae6 : SUCCESS in 17m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/140431239d864d1aa166f7a04bb37092 : SUCCESS in 1h 00m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3698cfdbf29d46ef930782ad79e5fb03 : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":3},{"id":"2f3460d76ac8bb5a1ea1a677975b42283c45bfa6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:17:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"863f390e1543169802a51e7959b47ea126f84c5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:18:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78349ed7113a4697b3cb261f512f4659\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5df6b4e60fe41a1b0729b15b9dd401c : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2da6ceab7891372f8324dccf9982d80c1c2cc0aa","revisions":{"89c6722087421fbacf3026a01397337b7fb46f82":{"kind":"REWORK","_number":1,"created":"2023-08-04 16:24:12.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/38/890538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/890538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/890538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/890538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/890538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/890538/1"}}},"commit":{"parents":[{"commit":"91a4079d84cec7ea384b301b5ba0cb79db495503","subject":"Use erlang-25 from copr on aarch64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91a4079d84cec7ea384b301b5ba0cb79db495503"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-08-02 07:50:46.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-08-04 08:28:07.000000000","tz":180},"subject":"Revert \"Disable build of Gnocchi containers\"","message":"Revert \"Disable build of Gnocchi containers\"\n\nThis reverts commit 5505cd000ce7f12c272c5b62074c9cc6bf1bb7a7.\nReason: Closed #1304 [1] as completed.\nAlso build from master until new tag releases since #1304 fixed\nin master [2] only.\n\n1. https://github.com/gnocchixyz/gnocchi/issues/1304\n2. https://github.com/gnocchixyz/gnocchi/commit/b52f7414d595e8d765fde50992f2776a77b98de6\n\nChange-Id: I3ca4e10508c26b752412789502ceb917ecb4dbeb\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 4e5e508282dfd40a7d7831217e109a055073bd7c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89c6722087421fbacf3026a01397337b7fb46f82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89c6722087421fbacf3026a01397337b7fb46f82"}]},"branch":"refs/heads/stable/2023.1"},"fe0dd8287595ac868bd0f905c2d7c35d180f1a88":{"kind":"REWORK","_number":2,"created":"2023-09-05 12:04:31.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/38/890538/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/890538/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/890538/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/890538/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/890538/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/890538/2"}}},"commit":{"parents":[{"commit":"91a4079d84cec7ea384b301b5ba0cb79db495503","subject":"Use erlang-25 from copr on aarch64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91a4079d84cec7ea384b301b5ba0cb79db495503"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-08-02 07:50:46.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-09-05 12:04:22.000000000","tz":0},"subject":"Revert \"Disable build of Gnocchi containers\"","message":"Revert \"Disable build of Gnocchi containers\"\n\nThis reverts commit 5505cd000ce7f12c272c5b62074c9cc6bf1bb7a7.\nReason: Closed #1304 [1] as completed.\nBranch stable/4.5 is taged 4.51 with Fix [2] already.\n\n1. https://github.com/gnocchixyz/gnocchi/issues/1304\n2. https://github.com/gnocchixyz/gnocchi/commit/c4f142d946f65493d5e4fb86c2c12f478853845f\n\nChange-Id: I3ca4e10508c26b752412789502ceb917ecb4dbeb\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 4e5e508282dfd40a7d7831217e109a055073bd7c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe0dd8287595ac868bd0f905c2d7c35d180f1a88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe0dd8287595ac868bd0f905c2d7c35d180f1a88"}]},"branch":"refs/heads/stable/2023.1"},"2da6ceab7891372f8324dccf9982d80c1c2cc0aa":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-09-05 12:04:56.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/38/890538/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/890538/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/890538/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/890538/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/890538/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/890538/3"}}},"commit":{"parents":[{"commit":"91a4079d84cec7ea384b301b5ba0cb79db495503","subject":"Use erlang-25 from copr on aarch64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91a4079d84cec7ea384b301b5ba0cb79db495503"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-08-02 07:50:46.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-09-05 12:04:55.000000000","tz":0},"subject":"Revert \"Disable build of Gnocchi containers\"","message":"Revert \"Disable build of Gnocchi containers\"\n\nThis reverts commit 5505cd000ce7f12c272c5b62074c9cc6bf1bb7a7.\nReason: Closed #1304 [1] as completed.\nBranch stable/4.5 is taged 4.5.1 with Fix [2] already.\n\n1. https://github.com/gnocchixyz/gnocchi/issues/1304\n2. https://github.com/gnocchixyz/gnocchi/commit/c4f142d946f65493d5e4fb86c2c12f478853845f\n\nChange-Id: I3ca4e10508c26b752412789502ceb917ecb4dbeb\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 4e5e508282dfd40a7d7831217e109a055073bd7c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2da6ceab7891372f8324dccf9982d80c1c2cc0aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2da6ceab7891372f8324dccf9982d80c1c2cc0aa"}]},"branch":"refs/heads/stable/2023.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
