)]}'
{"id":"openstack%2Fkolla~890541","triplet_id":"openstack%2Fkolla~stable%2Fxena~I3ca4e10508c26b752412789502ceb917ecb4dbeb","project":"openstack/kolla","branch":"stable/xena","topic":"disable-gnocchi-build","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-08-08 07:13:06.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-05 11:44:12.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-09-08 14:41:28.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-09-08 14:41:28.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-09-08 13:26:57.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-09-08 14:41:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3ca4e10508c26b752412789502ceb917ecb4dbeb","subject":"Revert \"Disable build of Gnocchi containers\"","status":"MERGED","created":"2023-08-04 16:28:53.000000000","updated":"2023-09-08 14:42:38.000000000","submitted":"2023-09-08 14:41:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890541-disable-gnocchi-build","meta_rev_id":"85d4f28f0a612eb9de0ea1b15dac405d7d7dd722","_number":890541,"virtual_id_number":890541,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-08 14:41:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-09-08 10:03:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-09-05 13:19:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-08 13:26:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-08 13:26:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-04 17:53:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-08-04 17:53:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-08-04 17:54:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-08-04 17:54:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-08-04 17:59:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-04 19:40:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-05 09:00:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-09-05 13:19:47.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"1ef9e1fd22e588960ab003dbd0e3d65ed098ff3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-04 16:28:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8f5177ada224b470f7c63ec51b3a1e03696df6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 17:59:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a957ec1893c74597add92a178326e57e\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/740cdd064442474299faf38b6e9c2239 : FAILURE in 31m 53s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/950fe52511ea4bb2a52daa6116e72d1f : SUCCESS in 30m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/82eaf65e6c7c4f23a9777ce04c83bb84 : SUCCESS in 1h 28m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e7e53785c8d40e6ae17e061199611f8 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42d26eb553b74566bb06d37d29eddb79 : SUCCESS in 4m 33s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/dd0316b2b2d14ad781a12ee98df4570c : FAILURE in 30m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e19725fc594af8eca54932e176f43be00424c12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 19:40:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb86280a58964495b0743efade0bb738\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0be1d20c29b14b678581e9e86eed0d94 : TIMED_OUT in 3h 01m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e5f4d6b39b594522bde77d9333a3b998 : SUCCESS in 24m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7555f8d41a7e40d9a6e359fdf12a1365 : SUCCESS in 1h 06m 07s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5f9a7c82371c45bba319b27b865a6eb4 : SUCCESS in 1h 15m 23s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/23e78d8384b8409a892f5f6c05cce496 : SUCCESS in 15m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/45e63b63b77e43d38bcf7aae73afdb3f : SUCCESS in 18m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/61ecefb4e61c46a09a2fa5e499446cf2 : SUCCESS in 56m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dac1dee419d34930a8d8b513845dc44c : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c37cae264784e5b90ff31206992a3c5 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/514fea3fdafe4175b50cba6635a22a03 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc85379beaef4dd68da4530672b987a2 : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/949f40f8046b437e919d8f75b3a39afe : SUCCESS in 5m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3077ad11a4f44a118896f5147910a6fb : SUCCESS in 22m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d901e84c92b9447695821f07501e2081 : SUCCESS in 28m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5c3181d404f0406ea965704350a292b5 : SUCCESS in 57m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/82cb0453156b4d1998a1861b5ec2a229 : SUCCESS in 53m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6414ea29dd004eaeaf4bfc67cb3b3161 : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5a4efcb1d90dd2ac968177c77c5f93da68ecb9c1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-08-05 09:00:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"818c88f50d855a16e1f1822c3899679a301b142f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-08-05 09:06:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2aefc6e7582add896d3f4699062a2ba42be1844","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-08-05 16:07:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b78e0b2374e3e5172728a99cace36daf4004c56","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-07 07:50:47.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"96854998c834c4cb3afb41d6f5ca308b39bf81c3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-05 11:44:12.000000000","message":"Abandoned\n\nneed backport https://review.opendev.org/c/openstack/kolla/+/893722","accounts_in_message":[],"_revision_number":1},{"id":"615faeb39b3a1fbf4fb0c15282c4049f2d547660","tag":"autogenerated:gerrit:restore","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-05 12:10:52.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"169ac9ce7686dfc18359301a052dfdc2af9181dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-05 12:11:41.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bcd6ad1633bd8cbc1067505946255323688cad5a","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-05 13:19:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1407a6b8f9c941b9999b24c81582636e04a5b0af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 13:33:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/603582078c504174be062d99cfde6a0a\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9dc4c74b73b44cfb999c3ae77378498a : SUCCESS in 32m 45s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/75acbcc579ab49ad8d393ef3ad16403c : SUCCESS in 43m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d8a77450a58f4ca5b6e216c61396cb45 : SUCCESS in 1h 16m 40s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c4ab7c860bb84c9c87932e03d253f3e1 : FAILURE in 1h 01m 27s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3f86883691b64b258a6ab3a58a3a4ab7 : SUCCESS in 26m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6e35515889854b39bc9036da028d7fa2 : FAILURE in 21m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2adf8ff2d6fd48e4a01eaf78206eb17b : SUCCESS in 55m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3b69fefae414606a09833a6efc9779c : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3350130787547a2ade8c9efd20ec9d4 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49d894d2d9ca4155819fa177a532d288 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05222cd2538c4537b57a78973627b112 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd15f5d4b42d42079ded0792eeb51fe9 : SUCCESS in 5m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9baec7ed0a5141e2906d4701c91d0052 : SUCCESS in 18m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b81f585c618246a8a0f8ebd1557ef6f5 : SUCCESS in 21m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/816357d6e4c14325a8114f3c94c09a0c : SUCCESS in 1h 06m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6877cfc10f124a92a2b0d00bb3f7e0e6 : SUCCESS in 49m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/580346eeb29b4ce9b8c8085a0cae8511 : FAILURE in 6m 58s","accounts_in_message":[],"_revision_number":2},{"id":"882981e93cdb2030be4354f16d4027aba59cb369","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-05 13:36:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e2aacf2c355fa0eb503e3e0ed9af780137ffd931","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 14:37:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f236ba1b9b7400e965077f811b236ce\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ab7e9df54fa14e35adb1f4f317268538 : FAILURE in 2h 23m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/532191f51cfd40d0aba9c19ff071b843 : SUCCESS in 2h 16m 09s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8a9b88bf1b024fc8b293a8ebf32e4458 : SUCCESS in 2h 22m 08s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1bd8a38b90284a59bcc334655afc6b5c : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c894f09dbd84ebbb0664cc6ba5a8062 : SUCCESS in 4m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8eabe974b3d84e03be153b541cb5822e : FAILURE in 32m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"852a3eb1ab6a6608f7f49e670cec25f6938ffc29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 14:50:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/edb52ecc29f14c11ac9c038145f03283\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f88257c3611949c89871726b9a74cb8b : SUCCESS in 26m 58s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/208dcf7ef7ab4e91a391b047f559081e : SUCCESS in 25m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3d7e40d6d1344adb95e98f635ac15736 : SUCCESS in 1h 07m 32s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/be9c23fde7cb4051a611bc0202e99a8d : FAILURE in 51m 51s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7ea2d45652f741d0acad29c736adfdce : SUCCESS in 17m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1a1d39d7b64f4d72b5b4669397d7c57d : SUCCESS in 19m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9f8084ee29cb437298b92bc22cd589f0 : SUCCESS in 59m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20a667f10eb74519afea11e9d8f9da05 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29db045daf394ddf90234efa69dfa4f6 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02313de68ad742a2ab9f68896cd55e49 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/132d914b2ce44c34b859e2236000f61b : SUCCESS in 3m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29cbd08f9d364076b0f236f746923542 : SUCCESS in 5m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/73062f2afe264e268348aa8036a2993b : SUCCESS in 21m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f83fd1515b04fb39396c95a72d82706 : SUCCESS in 16m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b190b84ced540c790714060950f8143 : SUCCESS in 57m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6dc5d7d8c3f44dc98a753b811996f621 : SUCCESS in 47m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e227d7dd485a41219d824c7a82654f52 : FAILURE in 8m 32s","accounts_in_message":[],"_revision_number":2},{"id":"808871f5ad308be0e2f6a5dd765a8bb6e26788dc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-05 15:51:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ac6c779d1595b402b65fd0601acd2810027800e9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-05 17:46:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1177ae50650a7f2f792744ebe05c9b56dd0fe797","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 18:07:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d8e672675f14a6cb211a027babf6e52\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5dab48acdfeb4a3bba2059d2ecffe95d : FAILURE in 2h 04m 15s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d7f1993687dc41cc9f6de5e8175568e1 : SUCCESS in 2h 13m 50s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a096384fe3604ba1a28bac43d08d9af5 : SUCCESS in 1h 17m 15s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ee08be4cd8249708f7256c6e0dd078d : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ceae3a9e455047f0948ac574693b310e : SUCCESS in 3m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ffb36710dd6f4a9cbf12a1052b60d7a0 : FAILURE in 1h 46m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"234af963107e41afd66029cc4c652461a4ff4c08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 18:54:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/661faa15f964488f86ab0d431a158d52\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e4431e4825054ebe86fd93fd17bddb26 : TIMED_OUT in 3h 02m 31s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/98cf7730285f4c838117116b328ab6fa : SUCCESS in 36m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3d3d0b1fb023492fb10a01ba840ff06f : SUCCESS in 1h 17m 38s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1179c83ca57c4a34b8b88e4abd6f47b6 : FAILURE in 1h 03m 02s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/443e1eef20234572b1a2d9721486e7ca : SUCCESS in 17m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bd8507bb3ff44d95bd9a199171172d14 : SUCCESS in 22m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3c34eeefdcd74eb19bdb4dd651ff710c : SUCCESS in 1h 13m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7cb8b1c8337413e875586b8e9f5359f : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c779a7102e04d13b06d72ceebf133f9 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37ffa7c81b3e41d7b8af2de0ee7977eb : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cef2b4a3bf94ec0a2439565f72a7ebd : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24fec835e5a5445aa7a7cae5c535fe75 : SUCCESS in 5m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dbfe9a387b1d4c819137a009bf67c2d7 : SUCCESS in 20m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/172f5479c3bb47c69cf02e62696dca44 : SUCCESS in 17m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/074744e44efc4ceaad37ecab087758b5 : SUCCESS in 47m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d831314d1a93434c95881a25a10fdda5 : SUCCESS in 58m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/82d32fc5f6404c6ab733409d9a8f8747 : FAILURE in 7m 23s","accounts_in_message":[],"_revision_number":2},{"id":"89ed2705fb9072417545768f630b82c04cf0ac23","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-06 10:42:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0192ec50552f43bac2341486ab8f9f42edcff2f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 12:09:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80dabadd6bc148d4a51b7e7216b771d6\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/cac4744bc1f9459dbd5d23df645cbf07 : SUCCESS in 34m 43s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bc77946f9a1643b295eb013b74f5d88f : SUCCESS in 38m 26s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d1035060e9424116a0b49705da52d3bb : SUCCESS in 1h 21m 40s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2d667c985fe74992abf1cebd52efe01e : FAILURE in 57m 37s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/56b2aa34107d4167b46f445baf18f14c : SUCCESS in 18m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2ff2ac7e2ff041f4ab0c403d30ee7133 : SUCCESS in 20m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/37ebd0bfa99c4187b9b4b46d75f4f71c : SUCCESS in 1h 11m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5a4408450e0472c9b6aee2c37655cad : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c11b2ae6aaec40a6b23f9ab4fc851264 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d71d6fbef596478b920ccc17b078c9ab : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bece0d56ba4454ca5ac855d6acb29da : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e72d3777379948abbec7544d2435269d : SUCCESS in 4m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bcceb8c4e9724b21928beabbaf4b34bc : SUCCESS in 24m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff8a8436bc2a449b906586c3a403aed3 : SUCCESS in 19m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b2131f74ea654833aaac011ad0ec8d00 : SUCCESS in 44m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c134578ce9094fd1b685d16051ebb50e : SUCCESS in 51m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"90aaf13650fd3f86a56e0a77920478005dfdb968","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 13:30:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fc59f33aeb94e06ae9a3ad34157b30b\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ad630f0b9e5e44dcb642a7fb542b1cee : FAILURE in 29m 54s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fe5243b43baf403aa33e5333223e977f : SUCCESS in 30m 11s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4eafbaeea3044703920b1656a94b23be : SUCCESS in 2h 08m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc51b46093a94f02badcef02ef379aae : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d239860aff1540afb11bc83f4ba03d83 : SUCCESS in 4m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e3867fcbb2524452a4f5b0e14706b37b : FAILURE in 23m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"22430e5f44f33357785b808dc8ef56b4a4935972","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-08 10:03:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1c335e216f23c048f2aec5f81e38bdc66604d421","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-09-08 13:26:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8a20cdc7bf005f6277fcb002d64d82d10a182bc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 13:27:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aeff574f74cd105d2d7994ac5cff4aa648e47dfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 14:41:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52e00ef2ed1842cea803a9f191d6685b\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/42b3a8bde9104040ac037ff2f07e07a5 : SUCCESS in 39m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f142c7bdfb5247f59404e7180d216adc : SUCCESS in 1h 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52b95f3030404c1e894dfe018767ce95 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90128de70f1e436d9295722cca797b6e : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81622b27c8c241ffbc91f040a05c8fe5 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39fbd39acfc04aaabcd5f6ae1f62e9c3 : SUCCESS in 4m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/58fbe9a231ce492486b9e03cbaa0c797 : SUCCESS in 21m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c4e191d8eb944b68bc6f6de059a5b15d : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":2},{"id":"4a9a60b4e34c45d4493f87cb692b03b46ca3f7ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 14:41:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"85d4f28f0a612eb9de0ea1b15dac405d7d7dd722","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 14:42:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/501de7871f214ae1872397da5878ed63\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fe754c3bd10436bb05af317a643a5b5 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9171cbfab0321e6a0cb6338fa52e35df67f7128","revisions":{"28a4ead6d6473bff6aade5c400b8ea17cef112be":{"kind":"REWORK","_number":1,"created":"2023-08-04 16:28:53.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/41/890541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/890541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/890541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/890541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/890541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/890541/1"}}},"commit":{"parents":[{"commit":"dc38e719758f607c4cb2c9ba7d242c73e55e6238","subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc38e719758f607c4cb2c9ba7d242c73e55e6238"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-08-02 07:50:46.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-08-04 08:46:30.000000000","tz":180},"subject":"Revert \"Disable build of Gnocchi containers\"","message":"Revert \"Disable build of Gnocchi containers\"\n\nThis reverts commit 5505cd000ce7f12c272c5b62074c9cc6bf1bb7a7.\nReason: Closed #1304 [1] as completed.\nAlso build from master until new tag releases since #1304 fixed\nin master [2] only.\n\n1. https://github.com/gnocchixyz/gnocchi/issues/1304\n2. https://github.com/gnocchixyz/gnocchi/commit/b52f7414d595e8d765fde50992f2776a77b98de6\n\nChange-Id: I3ca4e10508c26b752412789502ceb917ecb4dbeb\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 4e5e508282dfd40a7d7831217e109a055073bd7c)\n(cherry picked from commit 789fbe3665c71d3f83d1d6acf1c1fd8edf39589b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28a4ead6d6473bff6aade5c400b8ea17cef112be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28a4ead6d6473bff6aade5c400b8ea17cef112be"}]},"branch":"refs/heads/stable/xena"},"d9171cbfab0321e6a0cb6338fa52e35df67f7128":{"kind":"REWORK","_number":2,"created":"2023-09-05 12:11:41.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/41/890541/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/890541/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/890541/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/890541/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/890541/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/890541/2"}}},"commit":{"parents":[{"commit":"dc38e719758f607c4cb2c9ba7d242c73e55e6238","subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc38e719758f607c4cb2c9ba7d242c73e55e6238"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-08-02 07:50:46.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-09-05 12:11:38.000000000","tz":0},"subject":"Revert \"Disable build of Gnocchi containers\"","message":"Revert \"Disable build of Gnocchi containers\"\n\nThis reverts commit 5505cd000ce7f12c272c5b62074c9cc6bf1bb7a7.\nReason: Closed #1304 [1] as completed.\nBranch stable/4.4 is taged 4.4.4 with Fix [2] already.\n\n1. https://github.com/gnocchixyz/gnocchi/issues/1304\n2. https://github.com/gnocchixyz/gnocchi/commit/1336d930c18ce16c67d7542bf4c1e99358486a4d\n\nChange-Id: I3ca4e10508c26b752412789502ceb917ecb4dbeb\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 4e5e508282dfd40a7d7831217e109a055073bd7c)\n(cherry picked from commit 789fbe3665c71d3f83d1d6acf1c1fd8edf39589b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9171cbfab0321e6a0cb6338fa52e35df67f7128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9171cbfab0321e6a0cb6338fa52e35df67f7128"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
