)]}'
{"id":"openstack%2Fkolla~891258","triplet_id":"openstack%2Fkolla~master~I2e4a638e057d0e99e5182d823ae58617c158d786","project":"openstack/kolla","branch":"master","topic":"setup","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-10-08 02:02:29.000000000","reason":"Change was abandoned"},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2023-10-07 01:26:50.000000000","reason":"\u003cGERRIT_ACCOUNT_32029\u003e replied on the change","reason_account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}}},"hashtags":[],"change_id":"I2e4a638e057d0e99e5182d823ae58617c158d786","subject":"Add py10 support in testing runtime","status":"ABANDONED","created":"2023-08-14 02:15:18.000000000","updated":"2023-10-08 02:02:29.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"41a2d7bd10814fb40e21de94caef347b75e26b47","_number":891258,"virtual_id_number":891258,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2023-09-26 12:29:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"all":[{"value":2,"date":"2023-09-26 10:28:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-09-26 13:09:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-08-16 09:19:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":1,"date":"2023-08-14 05:56:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-09-26 10:28:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-14 03:38:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-14 05:56:17.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-08-14 09:16:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2023-08-16 09:19:29.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"},{"updated":"2023-09-26 10:28:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-09-26 13:09:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"dce0cba96fbd96a8a0376803fb077bb46e5cfe51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-08-14 02:15:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d47ea5142f45ea59f123a94ebe17eed5672ba15c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 03:38:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e54c620fd01c4aed8e8ac183ea186f3e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ec29239eb7bf47ccb9942cb346fddb89 : SUCCESS in 2m 01s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cbb5a65c1863449db584e0304a96f73a : SUCCESS in 24m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/31e501c4e6db4f348796ca3f66e886b3 : SUCCESS in 1h 03m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b28f41c6890948e6a33c6bde123142d7 : SUCCESS in 21m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/57ce27d1f4b34dc89f8bb44fdd30aba6 : SUCCESS in 56m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1605f5fda5064432934b71359906a35b : SUCCESS in 1h 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db4afcb459534f6a9a494c781ea2d387 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58565930ed99438c9b0134570f01abad : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd9405bed2fc45a1a93c1a0fa99618fb : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b2ffe03740e48b9ab6d77ada2fc7e5d : SUCCESS in 3m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/923a6fe050a94005a9f347f409d99571 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd9068005d2a43e78908ef05945be498 : SUCCESS in 4m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/08b23fcb243e4d0bb1767ad2a42f697a : SUCCESS in 16m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7adb97eca102411b8920e0fac41ad032 : SUCCESS in 56m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f980da3e0a2d4a46b82f1185100b75bb : SUCCESS in 1h 16m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/369e74e30a2b4aeb9197b91e6092f7cd : SUCCESS in 33m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b5f6fb5134fe4b6e9d4a2684d1dfa1c7 : SUCCESS in 1h 05m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7870375905f74147b44799ec84ab7768 : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e9018a2338de0593f7992803fd7a94afb652d085","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 03:54:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21e4d617435e455e92830feb5bb7b127\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/20129976735d4c518a3c3a05fbbb980d : SUCCESS in 52m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bc700eb9aece4a449dbc5ea07b3a11ef : SUCCESS in 48m 59s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/626fbac80d7940e7899458c6f15f8c46 : SUCCESS in 1h 36m 21s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f572b5805f949e29d3f40019fb3cc75 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/34f423381f1a418aa16b2b33edeba44e : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/51e5f7f2772a4d638d21d60b896f42cf : SUCCESS in 5m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2f579bc802194ca9aa433f38886ce4fb : SUCCESS in 1h 17m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ef9b568d98cd4b3ea8d4269ba8d93e7e : SUCCESS in 1h 17m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"509415077a0a68f4fd5c806a454e676c6ca4015b","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-08-14 05:56:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b74494c65fe523eab5607d5a4a4ca7b50a175a43","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-14 09:16:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"933dfa23aa23a0ae46df749224da5641efc97ab1","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2023-08-16 09:19:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"18ee161913310aa4797e578623bdbf2c071dae16","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-26 10:28:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e00091fe76b139c5cfe917497d914e0e2bb8351f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:29:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c11f5c123d8a6021dce5b55680f868cb75559109","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 12:29:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f9d73c4ca7c4d0dba94fd064230dc83\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2de79d5f42f94ac5b5c4f163d6c2d20b : SUCCESS in 1m 57s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6fb81e42694d43c7a8cb39cfff8d942c : FAILURE in 15m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/06915059b4074f41bc37c183078ead0a : FAILURE in 30m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/478d00fb4de04404b1b08d0c8254edc8 : FAILURE in 13m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/00bd488503f243ae92e9a3adf2fd424c : FAILURE in 19m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bd580196da04c2ab95c7b20699054d3 : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be7f679d6b714a8fbcedb59c5a71e038 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec9713cfc1d945a99240e01da34e8d81 : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b572816d53794a39b664d57235a98966 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cd1273d131e4bff933f93d3a6f0743b : SUCCESS in 4m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/78f4bdbbe5a84c73a8b591872d022c44 : SUCCESS in 19m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/78eea0adced64874a83533f1b44e38a7 : SUCCESS in 1h 03m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3eeff7d2fc034b76853e3ae411c6e2b5 : SUCCESS in 1h 11m 37s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/11f0810d7dd34f0784799fbc346af88b : SUCCESS in 1h 14m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0619e322a25040c189bca28e5e52c489 : FAILURE in 15m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6a63c3a50cea4cb285fa260743973446 : FAILURE in 22m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/314b455c31e24882a91a96f9a5e160f2 : FAILURE in 16m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/40b6b247eee545b98fba63988235adcf : FAILURE in 14m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b2f1b0b10684aed234b7d61edd58cf998fafa1a0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-26 13:09:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f5691c2638f37a2dcbffc3a2bb603c8518718487","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-10-07 01:26:50.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"41a2d7bd10814fb40e21de94caef347b75e26b47","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-10-08 02:02:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2c029deb6a0aceab767f943eeab447773a57937","revisions":{"e2c029deb6a0aceab767f943eeab447773a57937":{"kind":"REWORK","_number":1,"created":"2023-08-14 02:15:18.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/58/891258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/891258/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/891258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/891258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/891258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/891258/1"}}},"commit":{"parents":[{"commit":"7f12d216dc4de2c8d32291c3d6223185ecf2b510","subject":"Merge \"Revert \"Disable build of Gnocchi containers\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f12d216dc4de2c8d32291c3d6223185ecf2b510"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2023-08-14 02:14:13.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2023-08-14 02:14:13.000000000","tz":480},"subject":"Add py10 support in testing runtime","message":"Add py10 support in testing runtime\n\nIn 2023.2 cycle testing runtime, project test in py38 py39 and py10\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2023.2.html\n\nChange-Id: I2e4a638e057d0e99e5182d823ae58617c158d786\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2c029deb6a0aceab767f943eeab447773a57937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2c029deb6a0aceab767f943eeab447773a57937"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Verified\u003dMIN"],"failing_atoms":["label:Verified\u003dMAX"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
