)]}'
{"id":"openstack%2Fkolla~891617","triplet_id":"openstack%2Fkolla~master~I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6","project":"openstack/kolla","branch":"master","topic":"frr-ovn-bgp-agent","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-12-02 09:18:44.000000000","reason":"Change was abandoned"},"35638":{"account":{"_account_id":35638,"deleted":true},"last_update":"2024-12-02 09:18:44.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-06-18 09:39:59.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-11-27 15:26:37.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"36250":{"account":{"_account_id":36250,"name":"Przemysław Kuczyński","display_name":"przemekk","email":"przemyslaw.kuczynski@ochk.pl","username":"przemekk"},"last_update":"2024-12-02 09:18:44.000000000","reason":"Change was abandoned"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-09-03 08:55:32.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"36670":{"account":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"last_update":"2024-12-02 09:18:44.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6","subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","status":"ABANDONED","created":"2023-08-16 17:55:24.000000000","updated":"2024-12-04 18:34:33.000000000","total_comment_count":133,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"9e81ce9b8845a9656c78bb5c43b29ce0e0b15369","_number":891617,"virtual_id_number":891617,"owner":{"_account_id":35638,"deleted":true},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35638},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-09-09 10:21:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35638},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":35638},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":35638,"deleted":true},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2024-09-09 07:53:06.000000000","_account_id":35638},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":-1,"date":"2024-11-27 15:26:37.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35638},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":35638,"deleted":true},{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"}],"CC":[{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},{"_account_id":35875,"name":"Jörn Kaster","display_name":"joek-office","email":"joern.kaster@epg.com","username":"joek-office"},{"_account_id":36250,"name":"Przemysław Kuczyński","display_name":"przemekk","email":"przemyslaw.kuczynski@ochk.pl","username":"przemekk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-16 18:42:11.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-16 19:20:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-17 06:56:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-08-29 13:49:40.000000000","updated_by":{"_account_id":35875,"name":"Jörn Kaster","display_name":"joek-office","email":"joern.kaster@epg.com","username":"joek-office"},"reviewer":{"_account_id":35875,"name":"Jörn Kaster","display_name":"joek-office","email":"joern.kaster@epg.com","username":"joek-office"},"state":"CC"},{"updated":"2024-01-11 16:09:58.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-01-11 17:59:01.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-02-08 10:43:18.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-02-09 14:39:40.000000000","updated_by":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"reviewer":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"state":"CC"},{"updated":"2024-06-14 14:53:54.000000000","updated_by":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"reviewer":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"state":"CC"},{"updated":"2024-06-14 14:54:58.000000000","updated_by":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"reviewer":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"state":"REVIEWER"},{"updated":"2024-06-16 22:00:19.000000000","updated_by":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"reviewer":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"state":"CC"},{"updated":"2024-09-19 15:33:33.000000000","updated_by":{"_account_id":36250,"name":"Przemysław Kuczyński","display_name":"przemekk","email":"przemyslaw.kuczynski@ochk.pl","username":"przemekk"},"reviewer":{"_account_id":36250,"name":"Przemysław Kuczyński","display_name":"przemekk","email":"przemyslaw.kuczynski@ochk.pl","username":"przemekk"},"state":"CC"}],"messages":[{"id":"7c5bcacf1729887e4c060e2ec91db7bd44d24a6d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35638,"deleted":true},"date":"2023-08-16 17:55:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e6adac00393c253f3bb8e7552725e547bbcc4b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 19:20:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/97c374e05e224d069ec03a42b2318e6c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/10263b611e6a46ea9c796b7a17969dd0 : SUCCESS in 19m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/92d7827cdf234f0cbfa411a3f3aadb5a : FAILURE in 36m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c3a44520ea804ca3b2b6295008b25e36 : FAILURE in 15m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0db1d01cc5aa4f19a0f7d51e23c21073 : FAILURE in 34m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0c660001bc2b4d328505cdf734920399 : FAILURE in 58m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15a02363b76846858fdfbde87b0b0727 : FAILURE in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d49b423ea554243977fa3b1fb6cb51d : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43b01f7bc8384823b8d3ff6766965d34 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89e2a2ccea914e48babb606d8eb534a6 : SUCCESS in 3m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d499ee6e1274181b304b201a793345e : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b788e20fba2c4603a71144d258a5c51b : SUCCESS in 5m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b3cc36d3fab84ca19f387512ffd4b214 : SUCCESS in 17m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a6eb17aa8a404620ba8c3d8fc9649941 : FAILURE in 39m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2a629de12ae74c968f81e10692503ef3 : FAILURE in 58m 34s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c55ebe40cf564f019c38004ed74f835d : FAILURE in 57m 33s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/925713a4cda1441da9194b388f7b98eb : FAILURE in 1h 16m 30s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/8684b969539240ceac235a2ff1f723cd : FAILURE in 48m 59s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/d36a00cf64814b86a58c1c2a7c2f7717 : FAILURE in 30m 27s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/c4beec3c477242ac8298e199f1493d95 : FAILURE in 38m 32s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/2bb749e1fc964db6a395a4099e3b1bec : SUCCESS in 25m 51s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/4b848d424a2349978a04e73bfb79088e : FAILURE in 32m 51s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0012be8dbd464b929cf2e3ad11347dad : FAILURE in 32m 19s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/999099b756fb4904b8f8d745ce10abbc : FAILURE in 39m 53s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/64213f1ef524484aab65706a353b979c : SUCCESS in 32m 27s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2ed59810a3454bc39cdad56fac3dfc17 : FAILURE in 30m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b28afb9b938a41be94cbf7cd1106cba6 : FAILURE in 36m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3fb7ea202f024d2382642c9b3180dd75 : FAILURE in 54m 50s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/77e9c8dc7d8041e89c1d568877c8d99c : FAILURE in 46m 52s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/321816cea6094e0b8beeb7d6b8a9f3f1 : FAILURE in 1h 17m 11s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/8003ed2b728a42f388e6720902629d20 : FAILURE in 42m 17s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/587082e969da4359ae22741f93640949 : FAILURE in 36m 26s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/30a9ccf43c8c4d95984f63bb12e5fbc5 : FAILURE in 52m 38s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/35e99a338fca49569d07d5fd7518a9ce : SUCCESS in 27m 29s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/46130549fc2f4c3f81df5978208a3e62 : FAILURE in 36m 59s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/55d8ed1f199747d6a2325b67fb9ce8a2 : FAILURE in 40m 00s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ba3fd03c07354c01959b4c36c9b5c8ed : FAILURE in 16m 23s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8fa006cc688e48ccbf3659364c3016a6 : SUCCESS in 31m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0bd82dae0c5f82f9b53e7b946e4b75b0e13a2216","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 19:37:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de93b5dc25bb4169a0d7de9b13daeab0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b1078100a65049148a98584abae0bb44 : FAILURE in 1h 36m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0628cefcb8624f6fbca11b51ff696655 : FAILURE in 1h 37m 16s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/11c6b401e30e4cddb1faee1982dd23b7 : FAILURE in 37m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe77d88d75de4a29ac660e006525c2da : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f6b22e6a22c841b2a46a06d23a2cd9d5 : SUCCESS in 2m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30aab6e1764f4f898580fb5e374f86c1 : SUCCESS in 5m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c29f3da8d08c416ba777c56e287fcf0f : FAILURE in 17m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d2c2dedc1a214c29ac6e23831bf17c2c : FAILURE in 1h 40m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c0d571633a79d805d3e3ffbd8a3725816023924","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-17 06:56:34.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"167b86876b9ea4c1849e9603ed231da2c7d29059","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-17 06:56:42.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d6d85c23984ae93a6788d2162ea9365e376fcb61","author":{"_account_id":35638,"deleted":true},"date":"2023-08-24 15:30:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37551992e10a3241d84f03990fdd2016ec885f0d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 23:25:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a8f420f7acfe0987d3103a37a0a15ab9bb48019","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 23:29:46.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bcb38e3b9eb858556d0c07705641353ff4beff9e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-11 16:05:59.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"62785d9c59edf64bb71fce8e4ee645b916ce7a1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 16:06:16.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 891617,2","accounts_in_message":[],"_revision_number":2},{"id":"6f756572294b08e144ad0a25e6ce68d03b75f5bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 16:06:19.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 891617,2","accounts_in_message":[],"_revision_number":2},{"id":"2183f35983458aa0db60902476d16db844dd678d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-11 16:06:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36b2ab433056e29d3748da62b652a221420fd6ca","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-11 16:06:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nThe following files contain Git conflicts:\n* kolla/template/repos.yaml\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"bb8bb2e87c41f1851d52d399ecd7bd965872c261","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-11 16:08:29.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"dab05b1c16dc2e94a44ab4c6d4b309bd55ed8bee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-11 16:09:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3dbdb8259364bd96cc368b6f1af29ebf29658bb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 17:50:40.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6f885e934bc4df09ebe06c82cbc84a1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/bf7c6d59c1d74cecad8411d5314a23b0 : FAILURE in 21m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/085eb1a1c07b4bdfb9218de8ccfd65d3 : FAILURE in 1h 39m 49s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/08cf88dbf19b4e17ab68695eb12cb4f5 : FAILURE in 34m 34s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb72eaba94a94f7c8e34cfd398268808 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0687b5af91594b708112e7f723a3f19e : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a45f4daa399d4ad79c01e0f6dc3f0745 : SUCCESS in 6m 02s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2bb4c7ceec1f4d7da4613ebc1a6e9fbf : FAILURE in 23m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a9fb72fb922143dfabe693c325d95ab4 : FAILURE in 1h 36m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f9d9d670f61900c7d794ab84b98a4915e8a2d4a3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-11 17:52:43.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"3ccf587b738ccc8e63b1e90aa243e5d070a48a4b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-11 17:55:13.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"50dae77a890620a2ad4816026264f7d8cc9c1410","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-11 17:56:38.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"6aa416187e4ca5f08d057f0f90cc3c53119d26dd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-11 17:59:01.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ba8363804c29a0a8c6d149bec4d584a9a767ce13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 19:28:28.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0b0153760ad4bcf8e9004ccf45e8db1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e3420cf5da324cf389891052738303a9 : FAILURE in 1h 31m 22s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/485a7dcc34d745b0bf42c47571798952 : SUCCESS in 16m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4bdbba4016554040b2011156ebf76380 : FAILURE in 1h 11m 22s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8235331465b4b958cd236a0532aedd6 : SUCCESS in 2m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e5b50ef7ee2b40a78addff82f12fecad : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6c5ffe9daa7e461787bb8104a3f9a26b : SUCCESS in 2m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5d48f579ef8a4150ac4c41212c0b3a1f : FAILURE in 1h 29m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/cd3825a12dd0426cb896e2d0c14438d0 : SUCCESS in 15m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"355cf93932477f306ff8efb1b6a05b4794c50aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 20:08:43.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ce1217dd24d4621891cd255e2ebe5a5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7fd4346da17f4640a705e477718dae1c : SUCCESS in 24m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/17b8f3fb6692416ea5316e3bcac5fbe3 : FAILURE in 38m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6f4236b169db44abab09028fb82188e2 : SUCCESS in 21m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e44abb6a8818497ba6a3d9e7eeabc738 : SUCCESS in 29m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/76143dc725c14ba29864c6499fffb5e9 : FAILURE in 38m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/22d65bc0c385460c947d08aecc0dcc0d : FAILURE in 1h 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd22680384a451db8ca52dadc55eaee : FAILURE in 2m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46417893d7af47dda5d3fbc537eb8550 : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85c7be032007414d8d7d289099a9e3ec : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64fb97f354b64dfc8081a8d90cfa6cab : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3698ce26830406a993554c87f377145 : SUCCESS in 4m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/de00d93e42014c44a76a6d87e40d7caa : SUCCESS in 24m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/160d22be82434fe4ab045716bff9ca95 : SUCCESS in 34m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/39738d31d6644c94b8ad9061092c7044 : FAILURE in 39m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b5478a44dec64831bc04a0e08acfb6df : FAILURE in 1h 14m 28s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/d98ee11755b74864b4495e180c51978b : FAILURE in 58m 22s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/59872b84a0e44fdbbaaf4e64206305ca : FAILURE in 2h 06m 35s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/58ed2ec565934b68bbfe2092c7d6d7c0 : FAILURE in 45m 01s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/39d1e55893c646ef8849fdf8867752af : FAILURE in 53m 06s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/744e6a656991468393452074dc0adaad : SUCCESS in 27m 15s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/e97df388acce4d8a9b92d505f28bd07a : FAILURE in 43m 41s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/3ae61f66a5d24ab5a3facdbde22867f9 : FAILURE in 42m 04s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/0872a18750b546f7bef1fe9935207580 : FAILURE in 49m 00s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/19f7a4a725904b4c89015eec4c5c31ae : SUCCESS in 31m 43s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5a0d905f54fb4dcaa28966b674bd2e55 : SUCCESS in 15m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7a199e9b98ee4b6080b9d1eb9407f2ec : SUCCESS in 1h 06m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8bd4bde36075421eae6cf612fd206af9 : FAILURE in 40m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/de9823f4e2894857974432fcbe292721 : FAILURE in 1h 03m 37s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/14237c759701495b9f37a8d11297fdf4 : FAILURE in 51m 52s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/db00941c26514a06bcb42543f3be4f5b : FAILURE in 2h 00m 07s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/025c02b2547846cd8035b373aecdaef6 : FAILURE in 33m 00s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/2af7b6dd21174a049a8e212fd4c7df61 : FAILURE in 41m 25s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/e40dfdcfb706464295e98bae8e74c9f3 : SUCCESS in 26m 02s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/a1d0864efe64479d825ae3c25825e762 : FAILURE in 37m 29s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/9d11863bec9f44b6a9d0251cb5c8b444 : FAILURE in 27m 11s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/2a5aeb71b77b41e4a0fb02475b41528d : FAILURE in 35m 55s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e15405bd961944b2b6c80e10a0b447a0 : SUCCESS in 28m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ca40c11dd53a76305b44f3655cc57b7446011efb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-05 18:37:27.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"a6aec69217fa3b93311138a63d6f331a75bc84ca","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-05 18:38:20.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"f2335ea4e290f86569344c727b1b7cba51992f76","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-05 18:38:44.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"e86a1828ebfe6c0fcb5739274d39a446239f84b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 20:17:18.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12ce7629742d46159c31b6a989d3a099\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4ea743cdc3e942bb863a394a2299bc0a : FAILURE in 1h 37m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2aca226126db4c77a7e2d973925a4173 : SUCCESS in 18m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7d8a0680a8ce4b5faeda64fe1553075b : FAILURE in 18m 45s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab888ae3f15e46fcab6a688e61debaaf : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f0202613b3a443419375667cbb1ab9a2 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a7b3e962ba34413889df480a8c00f880 : SUCCESS in 3m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b5149c9545cd4afca8dd2cbe8a059b48 : FAILURE in 1h 34m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c1ef0ab20b834226b8944fc78ee63109 : SUCCESS in 22m 56s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6fab60d318f0dca298c534688ae34933be95224d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 21:37:06.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5ae013dfb9049f4a60676b359ef6181\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b23058158ba54a4f8c43ebdf6e4bcce2 : SUCCESS in 28m 53s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bde2f00ac98c44ecbf11b848aa2dbbed : FAILURE in 49m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a29c4f1fce4b44b191e8823ce8ceb33d : SUCCESS in 20m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5ff8a0e4a5b44afa9140c0f679b61ac0 : FAILURE in 5m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3aee0226041a4078a7cc68439ce1d5b6 : SUCCESS in 1h 02m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3c2d32fc610f433b90920cd0e307f760 : FAILURE in 14m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f8e722a8aff4fd9b9a29fb0093f0b92 : FAILURE in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/541b7a46f2bf409cbafba104eaf3d12e : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ffac9057f6d4dacb35141eec4f07327 : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bea8f278e79f41779f488c520d35041b : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c00e3f17a2a4c8697a48e064c68df10 : SUCCESS in 4m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1588d8f4abaf4e2b881692aa915acf10 : SUCCESS in 26m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1cfb44f31a0041c0b76c995d7880e76c : SUCCESS in 25m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1539fdfa01d341309078b73026d417ba : SUCCESS in 1h 02m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7dc3df880eba4b80bdcd76b2d1150d8b : SUCCESS in 1h 23m 23s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/ec13b8d098c84d86b461c8926e1a9ac0 : FAILURE in 1h 18m 05s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/83997255683a4946b41a23c3ab593a19 : SUCCESS in 2h 50m 00s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/74eafe508b1a488b954ce7c363252ce0 : SUCCESS in 1h 25m 24s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/469794aff6544e509fb4b8d10b58a933 : SUCCESS in 1h 17m 42s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/32a9e78df22b43589460facbb984b401 : SUCCESS in 32m 38s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/8ce1e66e9ffa4dc4b67345126ecce18c : SUCCESS in 1h 11m 07s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/1704b622f60448bb8ef9459846c7f33e : SUCCESS in 1h 13m 55s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9127b2c6b11a4152aa8cf4f2e6042ba5 : FAILURE in 59m 20s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/fac42f06dae245db9b65359c88d1cdf4 : SUCCESS in 34m 02s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4802bd176df24293bd7dae7d9cf567f7 : SUCCESS in 18m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f80447ce6c4342e9b17a19c40acf55a9 : SUCCESS in 50m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/83a9e06a8c7c4de5a285c8f857eb718e : SUCCESS in 44m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b5e9fef430b8485397e90f6e5e646d34 : SUCCESS in 1h 11m 02s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/abe030584fa44d2db46ba7d852b80a72 : SUCCESS in 1h 24m 45s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5166d48edbeb47eeb6e9b75c4d200d48 : SUCCESS in 2h 05m 39s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/f1db279a656b4f239e50ad67008b9591 : SUCCESS in 1h 12m 37s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/d2fe2f8525874be081ce85e5e9489bff : FAILURE in 9m 49s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/b6d497244e0142bdb7bc2a463121db60 : SUCCESS in 26m 26s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/5c16395cf94b4fa59e407477b83fb93e : FAILURE in 8m 40s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7018ea57c94243aea2704b51c4f14f0a : SUCCESS in 59m 46s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/7cb77b06495a4a1b87034723f9f95053 : FAILURE in 55m 21s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e395840aa6594901bdf5f8cb32b110f6 : SUCCESS in 23m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6da2bf8207ab00f9c95649bdbf4b470ff3215e00","author":{"_account_id":35638,"deleted":true},"date":"2024-02-06 14:09:24.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ba2ea2e3b5b6e968ea303480a9940769b3882c10","author":{"_account_id":35638,"deleted":true},"date":"2024-02-06 14:11:14.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2a010ae0d0b84134705d1f815aec6d02d9f7aeaf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-06 16:47:07.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"6b5f604c7af3ad386b0b4f6404a411933d43f2d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 18:56:59.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c379157206644f70bbe943dd591ec85e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b311e07d23ab411eb8211d9230851c04 : FAILURE in 1h 21m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cbe4519fe630488890e0794b4ee11d20 : SUCCESS in 25m 13s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7fe7ade1e02b4421bd230a5ba8d6b2b0 : SUCCESS in 2h 06m 34s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37c91758e2d9424cb5f190fe9483b668 : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3791b69ac96c489681b927b9a061fb64 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aab11cc59fc9489985d5855e0cf94b05 : SUCCESS in 3m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e909310c7cdc4fbaad12f0de74a2c981 : FAILURE in 1h 22m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/cadb68c1412d4255a3b877c8f59dbb75 : SUCCESS in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8ce45c708978c3e52af55e8ecd623abd277dfab7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 19:58:27.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3dc56ddce3c47b2b41e382c3ec8ff30\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b029e2ebf63347fea4d3b9c67f40476c : SUCCESS in 25m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/95c0b70c2a974ac49f2f6b032ced4050 : FAILURE in 39m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ae4d33545583402d84682176583371cc : SUCCESS in 18m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/19ab3e4456b246dca3c12c88c16dbd17 : SUCCESS in 26m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/52f0dc48ad6f48c39fddbf6cc2b3b6a7 : SUCCESS in 1h 07m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/afe2f6b905cc4629bfaa8e9748885a63 : SUCCESS in 1h 46m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e28c24a3c7ff4f44b96cbc51a37d7985 : FAILURE in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/803e611ef0734a90b2668a3e2b94d0ff : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5717d99c32c349c4bca1966f9a6a3801 : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/608664e235f943059352f4468b9f52f6 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2153a4af4973461ba6786350b1353ba8 : SUCCESS in 5m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b32eab134d3f40e4b8e362eb8522c18d : SUCCESS in 21m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f6319d6e4ef34ecc8343703b7f5d57ad : SUCCESS in 28m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/915265807bfd403bbd4410b87da5d7c5 : SUCCESS in 1h 05m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0a219e2beb5846d28ca17a081d1ae169 : SUCCESS in 1h 32m 59s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/e7de66e0be594d49ad13071c4819ff40 : SUCCESS in 1h 41m 59s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fcb92a20a1704ca28cfdcea2500cc3f4 : FAILURE in 1h 52m 42s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/f1361521695c46b7ba963692f1ae3877 : SUCCESS in 1h 35m 00s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/346724927bdf47a59a48c7fc7b5ab014 : SUCCESS in 1h 26m 24s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/b2afd33fbddd4d40bfbaa36c14df46ca : SUCCESS in 31m 36s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/5b6b07dc8f9d41c2962a37f51236fb94 : SUCCESS in 59m 11s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/187cb37095164645a6ba2d8c0056136d : SUCCESS in 1h 03m 47s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/e6ef9ecf820c4defbfbe15113641684d : FAILURE in 1h 01m 21s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9c51f812a67c41fbbee183d76c108417 : SUCCESS in 28m 28s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d11cd8da46cf4e9fbe83ca413429318f : TIMED_OUT in 3h 01m 38s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f3cc16e12d474521896471833a9e4fb3 : SUCCESS in 1h 01m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/915cdcc426674a8d9539ca347327dafd : SUCCESS in 1h 00m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/86317c44467c4d6b8b07c215d05bde7e : SUCCESS in 1h 12m 55s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/8335b46569754217a2d860d5932437b5 : FAILURE in 1h 08m 46s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d0f6a28305b44b1db118ba9c24154282 : FAILURE in 1h 00m 34s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/e05c94aac0f84216a8a57c82d6ff6a7e : SUCCESS in 1h 09m 01s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/5355a4f9ff30483fa4300a89030cafa8 : SUCCESS in 1h 04m 22s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/543d2ea47f38490c9944fcbb9164faee : SUCCESS in 23m 31s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/7f38f37604924360b69fc73b9b7e961a : SUCCESS in 51m 40s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/bd1a425231d845c499e5412359819a3c : SUCCESS in 55m 22s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/dbcb33acdfa644138e9f0a4e54206a98 : FAILURE in 52m 39s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/47f8920732ef4a38bf702d64eb990307 : SUCCESS in 24m 42s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1522067dfe260decf6744bd0ba7ae33e10ae4771","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-08 10:42:07.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"4b5a3f9b69151e515e44033b3705d5e8bc047a95","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-08 10:42:10.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":11},{"id":"69a44b77c3e26adfa9a457fadabe5a76001b2887","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-08 10:43:14.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"19d299cea67042f80a3555649f535deb9ba4c10a","author":{"_account_id":35638,"deleted":true},"date":"2024-02-08 12:19:47.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d7e073a0b280bb1194c43fa073639350161f8672","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-08 12:21:02.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"cec5668db2366c294d6686f9bfdb8ac712976b24","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-08 12:21:28.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c586ac149f356300853df42dd060800161ae6d67","author":{"_account_id":35638,"deleted":true},"date":"2024-02-08 12:30:01.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ed858e42b1f72717b6d206117d6c8a703166edb0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-08 12:34:13.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"4d6e820e5fda166942f5eb90b0caa79c8b267b19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 14:21:20.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0752b4c714df4828a8f85b9b42b31111\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/388e9f31a982433d92fa4e71a9fe37dc : FAILURE in 25m 58s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3758702557154ca09602d7ac0bc82bda : FAILURE in 50m 52s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f0844c882db74a12a0be0fcb0004d358 : FAILURE in 21m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/23209c3fba584fdb90b0ed52b0c0152b : FAILURE in 28m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/257d479548e24ea687464f86a4dc5fda : SUCCESS in 59m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/53db4a3eb8e94a0a816c37d1f7158cb7 : SUCCESS in 1h 33m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c598d7638a494f28a2689a978c3e49de : FAILURE in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2097f4f895cf4de581a98917c8fd7b38 : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6da303b7ac7141beaa822f8d8eab4323 : SUCCESS in 3m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bd9231484064022add821ea7d9bb2a3 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e88b014127f448b9bacdc051f1d4703 : SUCCESS in 4m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b796ae764d334c5e8e2653d0e389c511 : FAILURE in 26m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/07e7bb63e64b4b6db3e35ebb0ff3b0af : FAILURE in 29m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/826369b6f2984d719c82da14c28dfd68 : SUCCESS in 1h 10m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/89bb241cd8eb4f6c8f0f7ef017187ff5 : SUCCESS in 1h 48m 07s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/7ee457ab78084b228e35b7eb430ef032 : SUCCESS in 1h 42m 02s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4bb0d92d77a44e25a7e01a02f47d7e36 : FAILURE in 1h 19m 22s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/ced77c54788e47ecae00e31df509a6bf : SUCCESS in 1h 25m 13s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/6c6be956e13c4ba69b1029a8ae8613a4 : SUCCESS in 1h 28m 58s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/5554aaaba8384da4ad6e6da2752c3613 : FAILURE in 27m 36s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/7210bbd8ebc9428aa579376ac68db740 : SUCCESS in 1h 10m 00s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/e5899ee9708240848599c8b2c1b73ccb : SUCCESS in 1h 39m 36s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/b356d5f06ac44b579b583017aafb27d2 : FAILURE in 19m 13s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e3a2450c09084d5c927bd3912c53357d : SUCCESS in 33m 36s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4a8d5affae9946e8a8920883187fbfe5 : FAILURE in 28m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/132242741d304cf7afed7d61769c9586 : FAILURE in 1h 15m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/128b4839d05a4325bde1b0cf81c7b52a : SUCCESS in 1h 03m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b33dc5c73fc64e04b4b9fbe0d7dc3cf2 : SUCCESS in 1h 30m 34s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b26560fb28524a4da3f6e6c88c1cf97b : SUCCESS in 1h 44m 12s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/42c59d0fd3924d7ba28a2fd6619b9994 : FAILURE in 1h 55m 35s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/63a20bafb80f43b88e39788080575eb0 : FAILURE in 51m 59s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/cf934849805941d7b672737b5d87c199 : FAILURE in 5m 11s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/b2e092850aef449a94db67714fee400f : SUCCESS in 25m 57s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/06c2e72d438041faaa15bce16870b1b7 : SUCCESS in 45m 50s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/f9646be274764daaa8618977e3f88f54 : SUCCESS in 1h 04m 46s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/14693ced1350409bb86ed3ff8f2bd33e : FAILURE in 16m 29s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9198ce2b791940709e6c475a612741f4 : SUCCESS in 25m 44s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"abae20fa64a01bc69d8696c173197ac0951df188","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 14:33:49.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dba3b861b3c4e1da03933f92b8bde64\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3caaebd3ecd547d19eea7a3772ad1d88 : FAILURE in 32m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ec4d4a80559c46f7a1dff302c7870070 : FAILURE in 30m 15s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/328a333e5084424e8f06bef6e3daba22 : SUCCESS in 2h 01m 04s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/82b61f265d9b45858bb2af3ea69a3e24 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fd1ea06581784c37b53ca5401e75ebe2 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b1b984355c2c430380194b15467716a2 : SUCCESS in 6m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ba881c6f673c4e309fea8f56ffdf8316 : FAILURE in 32m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b367b50891cf4e5d90b5c7cdb7b9c2ba : FAILURE in 2h 10m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"fe7185d0f7b02869ac5df1b6c5b96c078d4fbd38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-08 15:03:43.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"408090a2756d928a0b08a192fb90ba76f34d620b","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-08 15:21:07.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"51be711871149d40af6b0cd35b40ccd2f6d1ccab","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-08 15:23:28.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"282d674296a2a4399f4a67c532d42500b5e8f533","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 15:38:00.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9421ae5dbe38498e8139a84472abce50\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/94321f1e80494516b089f2f66065dc4b : FAILURE in 10m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8b0372af4eec4b0fb37254c5de8ddb4c : FAILURE in 26m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cb6bdc71d6ab45efa765b1b1293787b7 : FAILURE in 24m 00s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1dbe9a73b98949fb89b5d2454ba3ade0 : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/206b122290364d1f8dd1b309ffcca1ab : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a0c70921f8b94db3b7b3c91b566ccd31 : SUCCESS in 3m 04s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/940a015bed484c089a47c84d209c5d0a : FAILURE in 12m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/dd546ce5eec64c31930a9dfc3390d279 : FAILURE in 25m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8e424db1d9787275498209b70374ff5158bd6194","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-08 15:53:54.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"00dd048c3e831bcf8bcba38894c01b94951ed483","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 17:22:46.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40da17c7e03a45efb562c1739da95ab8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7646d14a4a9941a7bd95589eebfe4803 : FAILURE in 8m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9bd026feb451450584794cc79a818a09 : FAILURE in 1h 26m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/64c0112386f246d7aa78b628e1735c90 : FAILURE in 42m 15s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67ddaf9b78ab4756873b2f3b419d641d : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5ef4a606309349f8bcafc567f17129ab : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f2ae5ae100ce4e94afb8f73bd4d2061a : SUCCESS in 3m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9a7b571138d3407cacfaf68edfc7bb9a : FAILURE in 6m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/61bbdf7c26f14debb877bcc152ddc2bd : FAILURE in 1h 16m 18s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a6c2fff2797ba1123a16037878558805607d524e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 17:37:34.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2dadac43f17472996d343ce75df2ff4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ca107cecb4794051a1ef37d7092ac4bc : FAILURE in 24m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4d93dc2c1b0749f38fab010b4bd6dc55 : FAILURE in 8m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/97b0c47503a24266887be0c1ebb60e4f : FAILURE in 18m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/27ca252ad2f445ae87a162319ca01a7f : FAILURE in 34m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c06c81a156c24778874212d28d9bbb09 : FAILURE in 17m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/51db753a15e346088bb056dda18a6a0a : FAILURE in 30m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c061660fb81c47a29c072cee81ba226c : FAILURE in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aaaaf96644ae40beaa041763fc435b8e : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84f72dcae33b48f2be9e1effbeeed5da : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/476be121d12e41b0972f4d83ce6984a6 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bb3050b64c849b2b71f8cc344142e8d : SUCCESS in 4m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d47ad21041b24443b551aff659d66ae4 : FAILURE in 25m 51s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3cc925dfba414585b870d9eb114ac8c5 : FAILURE in 32m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5c758c3232804d11b98e4c7699fb6e8c : FAILURE in 21m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cdfa7b03e02a46dea229bd1802262e45 : FAILURE in 20m 27s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1b0cdce1e5414b03a0849a53a3ec672d : FAILURE in 21m 48s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7e2e57f91d1a4ea89a5a605ddab69910 : FAILURE in 25m 45s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/054b2f0a80b34c35976193b19f30e4d4 : FAILURE in 25m 39s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/b1997a21202b42d5bf1db4d23f13a755 : FAILURE in 27m 19s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/5e3a3bf76bc24e8bb4c8a2419617eca5 : SUCCESS in 28m 53s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/418091d6386647adb5b2620a97783bc5 : FAILURE in 22m 29s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/9ae748c554bf49d89769c321143ada34 : FAILURE in 21m 34s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/321e50d8999c4af492db0b9260bb6167 : FAILURE in 22m 10s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8c153b08548d4f0e8a87cbd97ddd461f : SUCCESS in 37m 27s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0a6fcf8df4364717a95dc6f187ef4917 : FAILURE in 21m 35s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a856be56afac4af9869365bd43bcd2ef : FAILURE in 1h 31m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b5b31bdc3e164e70a3381a27c7b7de5f : FAILURE in 20m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3ce5b51ee1c643c996779aa3c0e66688 : FAILURE in 19m 20s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/c2a76df23d8f4bf1b72b1e4387a91f7d : FAILURE in 22m 40s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/cfcffe47e1c14c4fabf702ca9e5080b6 : FAILURE in 17m 15s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/afaadca3754347b6aaf507e144cfdad3 : FAILURE in 17m 01s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/9f7922de9d6f4c6dbf36975ced8f0e82 : FAILURE in 4m 44s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/43d5699445084a198dd1a0781c008606 : SUCCESS in 23m 31s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/5bf2f29ecf7145b6ab3267a19d14d3a8 : FAILURE in 21m 45s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/181470ae8a9c4baea21d020f5901a751 : FAILURE in 17m 55s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/49a26c6ce195486ea5cac9048fc8de3c : FAILURE in 17m 41s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4aeb5942c2a043f98f99c70696abf819 : SUCCESS in 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b5cf34c69ecef7dcaf97a8bacbbdae277aa022ce","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-09 08:51:28.000000000","message":"Patch Set 14: Code-Review-1\n\n(18 comments)","accounts_in_message":[],"_revision_number":14},{"id":"36112f35bfed4151ed87fd2ce2d4df82ae8fa5ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-09 15:38:43.000000000","message":"Patch Set 15: Published edit on patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"446ce3a0beeb8ce1a71ce51cff569b2b63d6e8e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-09 15:39:42.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"305c15f6f1b15439d9d1856940a5e8c0cd70af5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-09 15:42:14.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"0c0a4792214054656e4eb30e7ba6d98592c645b9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-09 15:45:29.000000000","message":"Patch Set 17:\n\n(6 comments)","accounts_in_message":[],"_revision_number":17},{"id":"1276b1334afc8a0f5a91b30f90e208338e498f3b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-09 15:49:09.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"ecc3122fb608e8b81049fa446f53341249d64e4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-09 15:51:03.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"aa325970410955373815f00b1e8dc6afb7d85d7c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-09 15:51:19.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"688a446c61122c4be9bd85a9583aa3a5bf6cb948","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-09 15:52:13.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":18},{"id":"86674068f220659425ef7313efb2964114c9d0d6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-09 15:58:05.000000000","message":"Uploaded patch set 19: Commit message was updated.","accounts_in_message":[],"_revision_number":19},{"id":"6bf5b1a5e3a3789c9395530df8cd6b0f1af1e9b5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-09 16:09:49.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"74b36b15b7933f91ca11ae515eec55226b604376","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-09 16:11:25.000000000","message":"Patch Set 20:\n\n(3 comments)","accounts_in_message":[],"_revision_number":20},{"id":"2c230ea06488ef10cc794edaef7de0417cdd5cb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 16:42:08.000000000","message":"Patch Set 20:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f53cd4f0cc494d2eb562be957124573c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a2b7d96f811a447a9a236ee103a17cd4 : FAILURE in 30m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f52569a33fb5431ebcf11aea58daba76 : FAILURE in 24m 35s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f996d36bf24c4aca8fef3626de9197d0 : FAILURE in 23m 24s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/640ecac2da7c4571bf5217d3c8639689 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c546acc9dda54a0f9b9799d7291eea8b : SUCCESS in 2m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/39b073dbdc4e4108afcb8f146ac0283c : SUCCESS in 4m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/49dfc57af7f64b9d8754fd749f53bf2c : FAILURE in 30m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/de384387d5404156973e2a7f159e9d88 : FAILURE in 25m 22s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"04ea0cd4cf2359f90333a46fa92e990e1a5e676e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 17:09:04.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7310f36a80c04198997b18cec207b259\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d6070558a8d64efe90181d226979f9d7 : FAILURE in 27m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/42b74478031741f5878e3dafdff15295 : FAILURE in 21m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f09c84f2f8934a28ba72fcaa23344ded : FAILURE in 21m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3c57ee2f056948f1b9b5559aee2b1653 : FAILURE in 44m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e3001ef400834b31888c63817bac00c2 : FAILURE in 17m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/13f0457fb2b742eca0b9449c2de8b7a9 : FAILURE in 23m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/577f4c1069734036821648619d52754e : FAILURE in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/575035a1b4294840ae4c6b639f7f87bf : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84be17239f974986b37c2c48b0ddad10 : SUCCESS in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/092047a6a0c74c65a36c35435a1b111a : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec718de7ce2b4f5788ac30c6525437c0 : SUCCESS in 4m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8114ea849855462886080d45eaf9ae87 : FAILURE in 25m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/88ea6ceef4a84b04a1c39a6279df5b76 : FAILURE in 27m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c81b56545204423988717d9c2aec961a : FAILURE in 23m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e9fdeac10cc5409b802dde4e926698f2 : FAILURE in 20m 58s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/755b1a31283642a99363be11fe1061cf : FAILURE in 26m 25s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5b7af829148b4370963176f2bc003e30 : FAILURE in 19m 50s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/d44ec13d98fa4b0eb82d1a30e2ecd00a : FAILURE in 21m 29s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/56a6894f671948bb8a9bd97aa7044862 : FAILURE in 26m 36s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/39ba9d59a14e4f048e34d9b32180e5c8 : SUCCESS in 31m 13s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/23738afba0df48409f778c93cc05e61d : FAILURE in 23m 13s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/fbe8bf4d60bb4f0eb91f86fde0ccb75d : FAILURE in 21m 53s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9fc6f27a733e48cebe0bb7033f098945 : FAILURE in 25m 18s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/28e47484cd1a4fd59a6e5a0892e35f64 : SUCCESS in 36m 24s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/174f520815634126889837cab26c4647 : FAILURE in 25m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/07043bb8ba424d4c98651113ba17db11 : FAILURE in 57m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/93dc55a1c02c40ed8b64cca835bb5d7d : FAILURE in 19m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a059eb1c66ca457284f71b290d1cc19c : FAILURE in 19m 50s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ba9cad142bd84b63b175675231e0f0a8 : FAILURE in 20m 45s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/80d4fc43e5ec430d928d42f879f1e37e : FAILURE in 21m 57s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/51657c70bfa34c9b8640840a88158dc7 : FAILURE in 20m 06s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/f5be6166197343b5b3ba3eb118d3a1a4 : FAILURE in 4m 40s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/3c4c0904f5d64fdca41789a7f7af6e65 : SUCCESS in 26m 56s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/81a199238f1c4713a57cd6cebceeab10 : FAILURE in 15m 12s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b8b3f50fa0d5494c8226aa2fcf9035bc : FAILURE in 17m 07s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/f5a9ea4d0fe6455a8723946a70a45fe4 : FAILURE in 21m 34s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a21c7c202b8e4ac99323e7e2ef9fcce6 : SUCCESS in 32m 17s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"be59163e21c386c4d3e935dceff921a84132196a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-12 13:41:38.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"6ae4e9753e926a199ef8db8c0cf89cce9ac79bd2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-12 13:49:22.000000000","message":"Patch Set 21: Published edit on patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"eb8e0ee1386ac4bac2fda49d50f5bf2e8dfb2a25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 14:21:28.000000000","message":"Patch Set 21:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/270e07ed1ea34b03a53dc7698ab51e57\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b1bba7d505834a4098c6c3eb84313c94 : FAILURE in 24m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e04d420314254d52bca27d09b05f62b2 : FAILURE in 24m 26s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5279565ef9fb489d85e8cb68da878edd : FAILURE in 19m 10s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d57ce6fd91f041bcb99a284a6e3e97d6 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2e5fde9ff9f94503a697674b143edcc9 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fb44352938eb4385bff95ef5bba26825 : SUCCESS in 2m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d02d1ea34c1e4f6b8509e35ace5dc36b : FAILURE in 27m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/15130474aefb4311abd7458c9a6aa207 : FAILURE in 22m 10s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"0f37c16c34152e670a89006ea1a9a2b99bfe07bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 14:42:43.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f0d5d0c69c546ada68028964601e8ad\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1bbd2d7e68a644fcaae8ff0833aada24 : FAILURE in 25m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a2e9df783453418bbb38a2b812a59a6a : FAILURE in 19m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/297440a809494426bc5d8028227fcc0d : FAILURE in 20m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a6c362167298401680c06d9a59c0a29d : FAILURE in 28m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/91a41c1e74064260af76b2614d0f099c : FAILURE in 19m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ed4de19783144e10b526f30250985278 : FAILURE in 21m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9250c0cd1cb14ba99fdc8fba8b8affe7 : FAILURE in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a344eaef27ac485e8516719cd9c1d56e : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be973db742bc42d98c8a13692a62b810 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a69ca781974942d3834bf9a8e0fb8f3c : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047c05e08a21473f99624c0c4ace852e : SUCCESS in 4m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/51206f674a0c4ade8ae63c5b2328c800 : FAILURE in 32m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cd1527eb5b024b6ca2f77b65757d27cf : FAILURE in 34m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/56d04876bca74f9d839ca958807926c1 : FAILURE in 18m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2e76176142c2407ab52d5dfb2293e07a : FAILURE in 18m 41s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/18471c306ec34f55bbae4f7d33c8dbb4 : FAILURE in 21m 51s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/873878346150450b81ecdc4c2fd56077 : FAILURE in 19m 52s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/951b898bf4df42218aae0dfce26a10e9 : FAILURE in 20m 13s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/556ed1bb691c43c4b80f2e2593571888 : FAILURE in 21m 08s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/8fc4a74043ce47aa914af56a656caab9 : SUCCESS in 30m 30s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/9f6e3f609aeb4c4887dbe703327822c5 : FAILURE in 24m 01s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/397364e06bcc4bcfba9a36a04f97169b : FAILURE in 21m 05s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/2627531d0f5a48599f52655f9e62bd78 : FAILURE in 22m 47s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0a1a2b53d17a4a418d3bd94a4c487771 : SUCCESS in 34m 12s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/24b6fb13a09848b2aea9943bb0d0e3b4 : FAILURE in 24m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e5f276282d0d4d118c68b83e22a15a91 : FAILURE in 51m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c36d59eab282456a84217913505d3fcd : FAILURE in 19m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6080cf2493124110b3f8cf241274d7e0 : FAILURE in 17m 41s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/762d5678b6234636b08d1d4babc27adb : FAILURE in 16m 28s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1d696aad42144772aaaca347a1306468 : FAILURE in 20m 20s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/8ea13dc53fb74178892c1d3b884dff8e : FAILURE in 15m 02s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/a821cd0400644f6185ae4c10b6cc90c0 : FAILURE in 6m 01s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/c7f6d804633146dab794928f945ee149 : SUCCESS in 22m 18s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/7cf8c805226348898c5afde4db65c3ee : FAILURE in 16m 41s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/5000808c2aca452db296e01ba2f0d5ee : FAILURE in 17m 10s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9c1c87a07e734226864abf7c9ba01c94 : FAILURE in 17m 40s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/40cc8065b2504b7e9caf9f97361290da : SUCCESS in 29m 49s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"f22b120000dedc492b30b2f128072de968be15fe","author":{"_account_id":35638,"deleted":true},"date":"2024-02-14 17:04:38.000000000","message":"Patch Set 21: Backport-Candidate-1 Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"e28478e6ba8e62de5f1064d1f7f4a0485aa5a446","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-15 07:23:55.000000000","message":"Patch Set 22: Published edit on patch set 21.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":22},{"id":"2f85cd8aced5df7a8edf9ca724206d63e6ba9c42","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-15 07:24:23.000000000","message":"Patch Set 23: Published edit on patch set 22.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":23},{"id":"1663c1681d423c0a55484791d8a54f5418c629db","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-15 07:25:00.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"3c064d268755eb3a9ff087928b8121c005ca67b9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-15 07:28:34.000000000","message":"Patch Set 24: Published edit on patch set 23.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":24},{"id":"f7707d0b4955caf54376aaaa8d5c2b5f8c533faa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 09:15:48.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65adf33737d64170916d1752b8a60916\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1b1cb17f68ef42ffbf6be266336d5e23 : FAILURE in 1h 06m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b509db564f6f42c8ae8ce7128eaa1a64 : FAILURE in 13m 22s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6eddeb8c7dcf488aab492462dc483e63 : SUCCESS in 1h 45m 36s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b9d6947bcea84813b5e6cb46850e69c4 : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/62a7a2446568437abbb04c9d8ab7ea19 : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2291fa5479e643d4b9097f24e37632fe : SUCCESS in 2m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b47694c3f3ad489b8e4c5c240fd8c09e : FAILURE in 1h 00m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/527bb1bb878b4b6abe6f1a285b0c905f : FAILURE in 13m 46s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"15cc1cb9eaae098c4e8d76f6ff852165bf68718a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 09:25:53.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0eb151db762b44db95717bd52213f38f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4e0775b2489944e0872e668e36beb4da : FAILURE in 24m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/08d4f1e45de7415ea5766a69da705858 : SUCCESS in 1h 06m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4a392e0d80e749708002f2cc93998cf5 : FAILURE in 20m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/330fae99658c4749b7fecfa1186586a6 : FAILURE in 26m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/537634bbaeeb4e77b6c40045986c6dd2 : SUCCESS in 57m 14s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a2d8f28a506f4c6e8333d3f1b99215cb : SUCCESS in 1h 18m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f79f6b0b15964207aba39cb9d3332470 : FAILURE in 2m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22c66aefa41047899ff2255cb7a3a58e : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2ae96989ed143248bf0f454724be4f6 : SUCCESS in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/744cb1af67964add94a4485dac176f2b : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8a2a601da304c06876d55d115acc315 : SUCCESS in 4m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4da95eb059a3455da2c10620bb51edf3 : FAILURE in 22m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6e62b73c64ed42e998fed296d155e6ca : FAILURE in 28m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0d7a76e68ece4649b1913248bd037bf2 : SUCCESS in 1h 05m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b46a1e48e1eb4f5b992b040006dfbb8f : SUCCESS in 1h 32m 59s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8bc4ea2b0b8f4bc2b0c7ef3113575788 : FAILURE in 29m 47s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9052b2b711ff4e63a631dbb4992f3b88 : FAILURE in 1h 46m 53s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/fcf99309491445a08bf5a28df825ae92 : SUCCESS in 1h 19m 17s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/4f1ebb0fe5d948f2a71bbea1116ed9a2 : SUCCESS in 1h 30m 24s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/42c24b4405ef4b7a8be39378119a43e6 : SUCCESS in 22m 45s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/0b7e9a5ae6284a6989f5a5b650b9306e : SUCCESS in 1h 07m 43s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/9078362cb1a94fdaa2751447467f7619 : SUCCESS in 1h 15m 30s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c8d0ecb16e5a48b192acf13cc8c2a804 : FAILURE in 20m 20s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/dbdc803595e54eafba38c18e76388189 : SUCCESS in 33m 42s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a8a34253b182492692965ec18d354397 : FAILURE in 18m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f73c06e890cf4e4ab4608bc9d6805e40 : FAILURE in 50m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6d9c1d2335234fea8d59f9d162c95861 : SUCCESS in 58m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5c2b1d9758484c7ebff5c23d98bdd390 : SUCCESS in 1h 11m 31s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ce8561b047014da5a6c2ca52fbf11f14 : FAILURE in 27m 54s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2d3e116fa79e4445a45525e167ebef35 : FAILURE in 1h 15m 21s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/fb926916290c46f8902e1605db1a5fe4 : SUCCESS in 52m 56s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/c2c3df83eca34733859870b1cb4a15b0 : FAILURE in 5m 52s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/a352fb249de846a7b7028c560f6cd5d4 : SUCCESS in 19m 59s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/f989e62a80cd40219826616ca37c4d7d : FAILURE in 30m 38s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/36f802fd0057424ea05fdd74bd2c1d89 : SUCCESS in 49m 53s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/c5bdb81b5b2049d39c85ac1d25a8a848 : FAILURE in 14m 03s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b73f92eb7491486fb289db48907ea2a3 : SUCCESS in 26m 59s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"aaa07efda96ecdae038478a468eb4b26ce2d8e3c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-15 10:41:58.000000000","message":"Patch Set 25: Published edit on patch set 24.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"a09890f688d2940c48c2615d2114e6ebb390b045","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-15 10:43:21.000000000","message":"Patch Set 26: Published edit on patch set 25.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":26},{"id":"9db437371f4a5138c720d0e5546fad56971c0d6b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-15 10:43:44.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"e5c0f1c539ba3efaa29ab2d04d362ee2ac6f4d53","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-15 11:31:45.000000000","message":"Patch Set 27: Published edit on patch set 26.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":27},{"id":"5955c23a7a895dc062f4df8884f8b4c702c97577","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 12:46:08.000000000","message":"Patch Set 27:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbf4ea56a83f4cada2506e1ae9c1546d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/460fde758ed6422ca85407a32dff65ff : FAILURE in 1h 12m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/aff6b7d107e64be68814e6d292e7666e : FAILURE in 25m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/81e32494edf347b39fe72bb976b068fa : FAILURE in 1h 06m 27s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/066c97c8ccae426ebdbaa04dd0a0e7f8 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/38943d665e9d4b3e963feb1c0b0a85b1 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b76222acb7374855a51acb19b414c67d : SUCCESS in 3m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0c3a087f983c4b3682613a52b40033be : FAILURE in 1h 11m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9d97f5c93afe4eca9b4a2069be51605b : FAILURE in 23m 43s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"629bfe586009aac5889103cbbbbdad95e0f009e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 13:16:43.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d10517db40a42f3ace44be286e852e7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/51f6e2394af54886bbea1caa600e3c95 : FAILURE in 33m 53s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/09addfb2328349b297d3aad4b5630ea3 : SUCCESS in 1h 11m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4dd15754342a41beacfad53341911c33 : FAILURE in 17m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5734ff10d6ed43c3afad5f1497a7e028 : FAILURE in 31m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/019e9fe08caf438286e976e022327b62 : SUCCESS in 53m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/293b2684d4ce4866bbc21a12cad37b9e : SUCCESS in 1h 34m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/008ab00958f04e8899f264ef68719d00 : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aba11ee46c7a415f8a4638536d20335c : SUCCESS in 2m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78ee4b2d7c4941c2bd32f54b3a82d8aa : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08a5d75e3d144fa49d9134887313f9bd : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9284c5b4448433c9ea99cf8af24cc83 : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2c68663becc24d4e850f11d4a6c4b32a : FAILURE in 23m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b99755db8e0d4a60ad10eaaf72635bfc : FAILURE in 29m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/572fe767c54e4b1f829da673a4d93fab : SUCCESS in 1h 16m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/de41cb8f8a9b4c47a0ce3ce8e45f8b8f : SUCCESS in 1h 17m 51s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/5fbb407b8ebf496da3785460a13b0a9f : FAILURE in 32m 16s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/78dd554bcacd4acca10b77e85f6244d8 : FAILURE in 33m 30s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/d38615e63d1146cc8959ecf3d981ba58 : SUCCESS in 1h 22m 04s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/755bc9cd522c4920b32603745f73d5e0 : FAILURE in 43m 13s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/988d064b829240c2bbeab0a23c244540 : SUCCESS in 29m 19s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/b402c8361c6047e4be47afb0f05ee6e5 : SUCCESS in 59m 08s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a8e0ce7d69dd4b3c923cf80cd63204db : SUCCESS in 1h 16m 11s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/89cd442ace9b4a2d91cb1137484d4f53 : FAILURE in 19m 29s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f804b97a4c6f4ba1828482bd6574c6e0 : SUCCESS in 34m 06s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a3a6e26df0604875b390cac3a0f61adc : FAILURE in 21m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ef61aa3f4db94344832f295ab0770162 : FAILURE in 1h 03m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a4b68c5049414ce69e48a5a0712b300e : SUCCESS in 57m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c0a0443fe134446f832406816062e018 : SUCCESS in 1h 35m 00s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/295377f5a68f4315b9a76c7e5ee3d4f4 : FAILURE in 25m 13s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bec3657a6ab541b490e2d07baf843d7b : FAILURE in 1h 33m 50s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/7ed972306f3e4b4590b7fc9978c4253a : SUCCESS in 1h 01m 10s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/53f548feec6b44ea836b5341a2e00241 : FAILURE in 3m 50s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/e9d94c94c1f9480782f0c86c4dcd5904 : SUCCESS in 25m 02s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/53bef8c10da1429384374015615ba258 : SUCCESS in 59m 19s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d8c46a49382945a58833bb4a88a9fd22 : SUCCESS in 56m 30s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/42a4e328904e41b4b3d3a76ed54378fa : FAILURE in 14m 34s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2c5b077d265f4160bb07e2f6444fbea8 : SUCCESS in 27m 12s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"30580d50b7daffd24e3483e46da57d40e35c39fa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35638,"deleted":true},"date":"2024-02-16 10:54:09.000000000","message":"Uploaded patch set 28.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"fc67c6ee5c4247515f41ceef8a78515c10fff167","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 12:02:51.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23a852a0b7d24103adcd8c9082b74640\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/59afa77b816447e584b58ad0a9096049 : FAILURE in 21m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fe42bb651fd849d5a3b97aa784200646 : SUCCESS in 17m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ff56344fd09f471e88f0f6540213d9ed : SUCCESS in 1h 07m 27s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/02b9feaa7ffa496eb2ae780a7989a715 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cb25fd5742194821820b0d240fc49ab5 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9014ce364517490e83a9d0c595b30f4e : SUCCESS in 4m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/986714ca211a44c683812a9cad9ca5ca : FAILURE in 21m 40s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3dbf6b57a2df488a934285803300891b : SUCCESS in 56m 13s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"ddc8adf7a611bec8ec2f4a32b09992d38517eac4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 13:12:07.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebdca9fb659f4c468cccc9607c22c438\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/754a6f0118544d8782f96a209a914b63 : SUCCESS in 26m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4e35a19d266543a49f8bbe37ed1838d4 : SUCCESS in 1h 05m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/075fa943acbc45a3a5674858c0978fb1 : SUCCESS in 17m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f2e0db6ca96441dc94421234a82488f1 : SUCCESS in 28m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/14ea57ff3f33430fb99110c8e7ccfd36 : SUCCESS in 1h 09m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d6c84bfb095041ca869bd88962699790 : SUCCESS in 1h 29m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6155c34ef1cc4141b58740b74e22422f : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e22b2cb1d30449f8334de2535a27dea : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/117ba87b0c3043b9acf315fa7a14d217 : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5525f4df460c45598c10aeaac481c84b : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a4483bbc7dd422cb7f2ffe87e3f15d6 : SUCCESS in 4m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/84a151877eea4660a8c6078ad179bae5 : SUCCESS in 22m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/40ae7b0f34f44fa8bbda744546c0a356 : SUCCESS in 29m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f339fc7bcd1b4806af8d8e48eab6d057 : SUCCESS in 1h 16m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b2409fbce8e9462a9b56ac9171f05823 : SUCCESS in 1h 16m 33s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/eac7b4b516104b2fae4c0c820ecb13b5 : FAILURE in 29m 36s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bb5503952b4b442bb508b0c7b1d811d1 : FAILURE in 1h 43m 05s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/70bd6840e5ca45caa3d76a3d3daef1c7 : SUCCESS in 1h 16m 36s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/5778ccba419c429188bd06f371cd8d44 : SUCCESS in 1h 18m 27s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/7694fe28e50e469d849f1d7d02a9ebcb : SUCCESS in 28m 16s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/086c92c999964f3590d40a59cdfa1819 : SUCCESS in 1h 07m 24s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/5054f5508d444d529a750349c9e06341 : SUCCESS in 1h 13m 53s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9d4810d2903141da9b682e510907cf7d : SUCCESS in 1h 32m 24s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/16de82856af64caeacdd557d3af9ea67 : SUCCESS in 36m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0e4a3809350a4c198f9a82a20b4c8f14 : SUCCESS in 27m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b960c4949e494c63b25f3ea0eda657b4 : SUCCESS in 55m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d51b99ee371648dab200c4926d327c9c : SUCCESS in 59m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dd3c0ecd05434d48a764a1eed9324a9e : SUCCESS in 1h 21m 36s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/218e825c5f1d418a9778fdf876b49228 : FAILURE in 30m 03s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/01b2f209802b483ba3ff706a2e3f7d87 : FAILURE in 1h 29m 40s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/01b87728ee704802ab4a32bdc9f6d1be : SUCCESS in 1h 05m 30s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/6dee200a8784475488e07d6e41ce106e : FAILURE in 4m 43s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/f5546f49a06e4459aeb245e9958645d8 : SUCCESS in 25m 25s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/ea02a90b2ae648dda983339b002c3271 : SUCCESS in 50m 17s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/0ad94942b20e4ea28bb7cfb4fbcd4978 : SUCCESS in 1h 05m 41s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/f849298931aa4009beec9ce17f77d510 : FAILURE in 38m 19s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f605b614c69b41aaba41cc48a9396fa0 : SUCCESS in 28m 38s","accounts_in_message":[],"_revision_number":28},{"id":"b606d66e9c4634845b7f5afa2b0b4b074bcb3a07","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-19 15:30:04.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"1801a7099f8ce97387c5a42d1c4683f11519aabf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-19 15:52:58.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"4d82a0465db03fe87e1f649c6ebaefef37151847","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-19 16:16:59.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"c7d7233369ff9d63eab5d366275f035dfa2a69d4","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-23 11:17:03.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"c827f515760b08f91d3ae3603c0a0834ee5bbed8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-26 08:23:38.000000000","message":"Patch Set 28: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"e38a52559bdf05d5f0b4017a00c8be4cd5aed826","author":{"_account_id":35638,"deleted":true},"date":"2024-03-04 21:53:27.000000000","message":"Patch Set 28: Backport-Candidate-1 Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"5c8be928fc420cb412e90974d49ed87af029cdb5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35638,"deleted":true},"date":"2024-03-17 12:38:20.000000000","message":"Uploaded patch set 29.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":29},{"id":"1ddab124b32ad43adebdeb3a67b3d4668133bd03","author":{"_account_id":35638,"deleted":true},"date":"2024-03-17 12:38:58.000000000","message":"Patch Set 29: Backport-Candidate-1\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":29},{"id":"4df9f2b93647fd34d10276cd6e513c79fc08ed4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-17 13:49:29.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adbe683df0b14eb1ab2d6ac64ddcb696\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d80fc2d54ffe42c3a87dfcac2d292726 : FAILURE in 19m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3520049407bb493889e14151beccd0af : SUCCESS in 16m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4047391faec7439da308be5fa74c336b : SUCCESS in 1h 09m 51s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d47da26b16f5486197a4e542e0e7c7a0 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/415c02bb10aa4f5d890be738399ee727 : SUCCESS in 2m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/01bf5e6d2e964365b18d6cadc284d44d : SUCCESS in 4m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/eff313c5edf24f77863783f192488991 : FAILURE in 1h 01m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/05528f59209c45f9b88b0cea3ea88e2b : SUCCESS in 20m 14s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"d0e48d098e6fdf9f017adb6a7f6c744352c33c7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-17 15:46:54.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90372010def748e6a116511c351f68c2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/35aa5b6e43b04e779297b91193a2f475 : SUCCESS in 18m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0fdf2c0838eb4b72a1b08594aab63b5b : SUCCESS in 52m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e09529cfe0314a918a8ea1c98f85b104 : SUCCESS in 13m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b3938d02ec324e35b3603f7cf009ebcb : SUCCESS in 29m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ce432f96c2c34f71aee45fa24f258854 : SUCCESS in 47m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8f543a56581b41308a1530c7dec5202a : SUCCESS in 1h 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/988e70924cd446c893ea7366c5f3bddd : SUCCESS in 2m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc86e0a8fdb24506a46ed3afcd501696 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a861c7fd45d4ed0932051261395c01b : SUCCESS in 2m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d7609775d394ff99ef543883d200de8 : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b0b8211ef5242c4a1ffd8d828a7289a : SUCCESS in 3m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3e1d2be3031a4fcbbc96a80223d38970 : SUCCESS in 20m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/02e7b24c324d4fb68fb46e012788d3a3 : SUCCESS in 22m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5764a9f1e0f9450b8aa51a7951ee26c2 : SUCCESS in 55m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cea0af58d9ab4e2ca978708ff0df26e6 : SUCCESS in 1h 36m 39s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/fb9dbe9e229541148b88f4a4039ff047 : SUCCESS in 1h 35m 55s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bf7769cd65ea473e9df60fc3f8de8a3f : SUCCESS in 2h 27m 15s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/182535d9571749a6ab3548d5c22ecaf1 : SUCCESS in 1h 21m 21s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/ff33bb79e071481aa227baf887b9d6b1 : SUCCESS in 1h 29m 31s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/3dc3e3ffbabe4d2d9cb372006c38a18f : SUCCESS in 28m 47s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/49e2214388ad456e932f699a3a777c85 : SUCCESS in 1h 05m 39s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/c08a0d957e524092afe389a4575ed0e1 : FAILURE in 1h 01m 01s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/8f5fc234e5514cb0b290b5eceeeae297 : SUCCESS in 1h 30m 59s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9cd9ef38160442098c023b625f4055ad : SUCCESS in 33m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/194d8aaa60454063a664034f12ad2687 : SUCCESS in 17m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/69de33a9ac5e4e52b6c7c9a0e0006141 : TIMED_OUT in 3h 01m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/11423b9a274e466b9add2827b53f6941 : SUCCESS in 54m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bbd3dc73775140309b754e293868fbbb : SUCCESS in 1h 20m 43s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/fc91ed9d23f3490c803b44edbb77b3d0 : SUCCESS in 1h 31m 59s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c245ae8417fa455592713a42ec334a64 : SUCCESS in 1h 49m 43s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/4fbbfa2e3f4d4e25b0d1bf4152db1a38 : SUCCESS in 52m 58s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/189f9682aa704a6f85a2b6a43a0e7f8d : FAILURE in 4m 03s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/53159c256a6c4b70902649d7dd149941 : SUCCESS in 22m 41s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/462ca2e32ae24aa1bb5d2cf977a80890 : SUCCESS in 38m 31s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d6c7531a8a8148b8a865d281aa8a13a2 : FAILURE in 41m 37s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9e8ae185947f4fab9dc5f119aeb601ba : FAILURE in 35m 28s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d8a934a167d24648b09014012cfd31e4 : SUCCESS in 27m 20s","accounts_in_message":[],"_revision_number":29},{"id":"a8c25155d220511b929594f29ea0baff5785956b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-28 09:31:30.000000000","message":"Patch Set 30: Published edit on patch set 29.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"4a3b3aef1d8d8a41a3103c97f94288daa856144b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 11:19:44.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8677f67bb5b47f0897b71f4b86afcbf\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f52fbd2707bc48749efe7b90b53a17dd : FAILURE in 8m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/73d680d0a3274e51b811631b6f5cf71a : SUCCESS in 16m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9ea71cf873e3449b8d9b42be1f13ce22 : SUCCESS in 1h 46m 15s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d1d57050232e40efb7ebfc9950805b63 : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ea6338b6c72347f6aa196d92912a634b : SUCCESS in 4m 08s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/df98cbd6ceb74795ab80539df9e90b6f : FAILURE in 21m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9c16632765c24a8d936dbf789129ef32 : SUCCESS in 1h 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"fa1801facfb238a6d1fe7084e5d592e9e5421118","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-28 11:20:15.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"f7b12ed3878c83bdafc9f7960d0b8085c71f4ca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 12:12:27.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/99515d7938d94631b303440eb2f2990d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/580e3b6c3bae4fec9db3660963e73c20 : FAILURE in 15m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/10ba44e0c661475c96e3fed7ff49b25e : FAILURE in 12m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/971c9917969d47fa98744c86aeb89350 : SUCCESS in 20m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b68d1f95fb604ecda3833f5ed6ac6402 : SUCCESS in 33m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/97632851af56480fade6520a02d75b7d : SUCCESS in 56m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/92f59659d69d4e5ca63a9b5190890f75 : FAILURE in 1h 38m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7211c5a0f4c40558a37bdd385fe234b : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08924193ff404751878e9b2a4df80cbd : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3ac7485fd394f759b2551dc8c226b93 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ed62bf425ac4a719c0876e267a9d86d : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39d69a126ff5436e98d6e7b735fee7c4 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da310418d24e488c8d0bebb258efd7b2 : SUCCESS in 5m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/744ef5d651d84219a6584ad8234a49db : SUCCESS in 22m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8be0b391efaf4a72ad5e0c9b23112893 : SUCCESS in 31m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6789bc8468194b8b80944002bcd77185 : SUCCESS in 1h 04m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c7512a82b90a4e299aa580aa13f6a653 : SUCCESS in 1h 34m 36s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8e996a1853bf46a8a68b18648bbdb5f8 : SUCCESS in 2h 08m 03s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9d4d5d03b62447deb3a91edb23bfb278 : SUCCESS in 2h 35m 58s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/8959db2da7ca47d28f0361ed69e775b1 : SUCCESS in 1h 26m 59s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/12ee424f34a544bebcfd5a6ef9bcd4e2 : SUCCESS in 1h 19m 03s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/7b86bed544b74fd29a06b59cb5cb8c7e : SUCCESS in 32m 16s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/851684f97b18425ba04884ce2d72e21b : SUCCESS in 1h 08m 00s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/1e8aa920fe7c4b74b7fd270fa4a33556 : FAILURE in 1h 01m 05s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/a38993ab5a514c399ef836fb76025a10 : SUCCESS in 1h 24m 59s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1207dd4b876d451887c77772484afafa : SUCCESS in 33m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ebff70eb26ae4234aaca1d39c2afddcf : SUCCESS in 20m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b99457971a154ceaa002cbcfd5f00035 : SUCCESS in 1h 00m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3604bad3223a47a2b7b622ec91dfc535 : SUCCESS in 50m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/315959235a6647f98bd7cf523fec48a3 : SUCCESS in 1h 32m 34s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/a40795c04b0c40f1b1fad4d98fad9988 : SUCCESS in 1h 25m 12s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/cfbe153379a44daeac2cc66dacb94b31 : FAILURE in 2h 30m 05s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/93c86a350bd243759985cf2f86135e5d : SUCCESS in 1h 05m 51s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/72a2be5b6cca44a89d868bcc5ddbeaed : FAILURE in 5m 31s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/3c6b997f1e3c45f9af725ee59ff76e53 : SUCCESS in 21m 34s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/82be2e33d6a94864807a8553c3a1acec : SUCCESS in 49m 24s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c8d31fd58df140df84c05776ce3c9837 : FAILURE in 49m 22s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a797f13899d84034ab15ca24f2121c1c : SUCCESS in 1h 05m 51s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7e99e1fc9826428e87450e94b7127bd4 : SUCCESS in 32m 43s","accounts_in_message":[],"_revision_number":30},{"id":"8f31a9442c7f024f9fb5a52c0b2e380eadf66bb7","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-02 09:29:11.000000000","message":"Patch Set 30:\n\n(3 comments)","accounts_in_message":[],"_revision_number":30},{"id":"38473cc31837ff8fbaa0d7212d6b75a46c7155a7","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-02 09:37:35.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"18184afa9444118df7aa336e32ae36be18d85a22","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 07:00:31.000000000","message":"Patch Set 30:\n\n(2 comments)","accounts_in_message":[],"_revision_number":30},{"id":"dc761c8e77b1a345311b753d279db3541e24f86c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 07:00:49.000000000","message":"Patch Set 31: Patch Set 30 was rebased\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"add649848ba9856dba053fab3b97f1d729b1d74c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 08:29:58.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca18b488a6ce441ba5b0a1e32fc7a136\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/72c6373acd6e48aebafadb185d96ad2c : FAILURE in 59m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ce97fb1935574df0afa0f23cd24c6998 : SUCCESS in 17m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b9f58b0ecc53406e85087a55116f5255 : SUCCESS in 1h 27m 19s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/188f2b98ee0e4103b6630f8219148f4d : SUCCESS in 2m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/81eb2078c3064d3d831acbb1c9f60e53 : SUCCESS in 4m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/516c8c74f3ff4b8a9705814466ab3856 : FAILURE in 19m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b0f3621ad3d34dfdb3e2bd34121a2f7e : FAILURE in 53m 08s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"cf6c1f11cb925e93422338e59270b09124f9e85e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 08:32:19.000000000","message":"Patch Set 32: Published edit on patch set 31.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":32},{"id":"e28f9e4130922d458431e98823d55487fcb56e73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 09:44:31.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd20789b18354843a4df49dc370af52c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6d7ca6b7649b492a8a86fa9e0bbfde4f : FAILURE in 22m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2e172489d3ef4cc98f05116d0b5ca96b : FAILURE in 1h 01m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6d26e5d290f24e3fbb94946568a97c33 : SUCCESS in 1h 11m 23s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af401ad93300453b8499ebce9be7a32f : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/973a984299ba4e28a42bc192c2d39e49 : SUCCESS in 4m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/fbcbd708dc41400a9299faefee047edf : FAILURE in 19m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/178914840aa041c587919e9d90db9618 : FAILURE in 20m 06s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"6ea3337112590841443bbfba04b4fe5036c99019","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 11:35:46.000000000","message":"Patch Set 32: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d4d9134e04d4ae0ac5ff65c4173557c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/439355ebca1346998b1015e00a7ef615 : FAILURE in 28m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/49a434130709402da32530b8c951a535 : SUCCESS in 1h 07m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/238535f6e0fe42aba4582de9d7690fd7 : FAILURE in 23m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f57338de24e44091bb793a5895a11e73 : FAILURE in 33m 21s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/28f13f29efca438db502383d327b7e45 : SUCCESS in 1h 07m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fe203411dab04c6297006a07e7f4f61d : SUCCESS in 1h 26m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/488ba6b72fc0408cb9b143367ace547f : FAILURE in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1245b4d89ab4c029c346a52adf23284 : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e7c9154dde4041f0bc1522c029744e47 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fa4c4a62499f4f8e82c5c1930ffe4680 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa1ae11b25ec4ab0b032f9d4198d0091 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa3f02f7a5147cbacef58d5158ebb15 : SUCCESS in 5m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d566f1d6f62d40b39616a0436a1648f0 : FAILURE in 27m 52s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/98c22683409e4df6b5ee838f74659f54 : FAILURE in 34m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/812e334e17dd4e058bf131f5fe6018ac : SUCCESS in 1h 07m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/75ecd701a4f144bba87e32c1647114bf : SUCCESS in 1h 41m 48s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/84ac7356b8914e5f89fdd567545d7461 : SUCCESS in 1h 48m 42s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8e1d478933ed4a039ba2e3c22f1d8d4e : SUCCESS in 2h 30m 56s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/db1d1d4b84874f33a7579a283a3c2f68 : SUCCESS in 1h 03m 56s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/0c091ccb22c245da88c1fe69ee0c106a : SUCCESS in 1h 19m 05s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/956343c6d59644a0935e8c853e487db8 : SUCCESS in 31m 24s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/d3035b4740ac462391f0e68c299cc02c : SUCCESS in 1h 06m 01s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ae36541130134b608c56f8db3496536c : FAILURE in 1h 04m 13s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/ebeeb4bad54f4384b88a50d639a7bc46 : SUCCESS in 1h 40m 10s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ab905f0efb364c3390e97e691ef56f49 : SUCCESS in 32m 30s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/18f920fd58994b40beb0b7fb17d68db5 : TIMED_OUT in 3h 01m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/38705f324b664e558ac64f00c7f3a0ed : FAILURE in 47m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/32ed069276244a11ac388e50eeec9c5b : SUCCESS in 48m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/83f0f2f2189f43a8bda26a2ff9e62bc5 : SUCCESS in 1h 08m 58s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/1fc7b34ca1a8491387f01e154c384607 : SUCCESS in 1h 26m 05s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f8026447a7024dc9b0c23f7d944eb2b4 : SUCCESS in 2h 24m 25s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/86496d518c7f4f72a860c0372546f7c8 : SUCCESS in 1h 00m 40s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/6f2486bc66ed44b28b0c313e053330ba : SUCCESS in 1h 08m 41s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/add810aa30b743a1aa7053cf2709a9af : SUCCESS in 21m 59s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/f471b6296d3246b6938d9aae703b0f92 : SUCCESS in 49m 19s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/4bfeab82df4647a19ae46b782d0c2900 : FAILURE in 47m 59s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/fd71650d8c194c59a5adbd32e1448279 : SUCCESS in 59m 23s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/372170f47ca347e29a372dc07489d534 : SUCCESS in 27m 34s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"9236d2fc7a99046217d3c500eeeb6d08b491c864","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 18:40:05.000000000","message":"Patch Set 33: Patch Set 32 was rebased\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":33},{"id":"bd35a435c31c4bab8e3e567bc9002ff240fe3bff","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 18:40:15.000000000","message":"Patch Set 33:\n\n(3 comments)","accounts_in_message":[],"_revision_number":33},{"id":"88e929792158663d099e0d7570306b904c2fcc61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 18:40:38.000000000","message":"Patch Set 34: Published edit on patch set 33.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":34},{"id":"33ac9998ee893eab29b81836d0f039f705ba7254","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 20:27:50.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b0b95d7e51c407f818d07b2ce757929\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4fbc510c11474d5cad387a6f4d07c3c1 : SUCCESS in 1h 15m 42s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/519ddf2c036b40c88a00ad8a19c15163 : SUCCESS in 17m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f725b97b6e21468e8fb313aa9ab741a1 : SUCCESS in 1h 45m 28s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/320be36030ef4abeb4a3ce4af7bea743 : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9e51bd0dee5a4f5cbbf1de6e327f0a68 : SUCCESS in 4m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3af98db850d5447c8f8d077ce24b01d9 : SUCCESS in 27m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c399c24eaf6945ea920f177fd350a8d1 : SUCCESS in 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"001bec5be3d77550685d6faf91394a639e852140","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 21:29:59.000000000","message":"Patch Set 34: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ec9bc4e3e7f457bab8c9b8255b22e72\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0272b9db5799438bb587995177e8bb17 : SUCCESS in 22m 27s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/24aa424f77224d1fbbb33705eed35fe5 : SUCCESS in 1h 06m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b0cf7a1aff40426ba6eecca409427f60 : SUCCESS in 19m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bf0f18b263c74948872067283704bf0b : SUCCESS in 32m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2714d1b7d9b640b6930d1acb694064ce : SUCCESS in 1h 00m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/af2a7d486f1642518f0733dfa1058107 : SUCCESS in 1h 22m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30ef4bc561ea4ff18f5f647e02205000 : FAILURE in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6bf7da3b6b942ab812f4bd44a94d810 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/150c0b71fea24c7b87096ba5b4bd0cec : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c40615dbd9a5409f8429cd904e816d9d : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ea4955ea5e041ea9bd5738bea01fdb9 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27789765d8d044b484f8e0b0d8f0dba1 : SUCCESS in 4m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/312ec692a8984928a68fed1f5402655e : SUCCESS in 24m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/68339c97c2f34b95ae17181dec7a3603 : SUCCESS in 28m 38s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bbbc49942baf43d39f8a2ac56396ab5d : SUCCESS in 1h 06m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7a4b9cf4f24a478c805da44f44913739 : SUCCESS in 1h 40m 13s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/865f1882a43647b4a1e2d160086feafc : SUCCESS in 1h 34m 02s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5afa4427a67e4262b11aa11ff4bb06b3 : SUCCESS in 2h 36m 49s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/73338eb667d34c1cbfed5d8b1b425877 : SUCCESS in 1h 25m 34s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/7e21ab88f4e14c2db0e8aab72dea67ba : SUCCESS in 1h 29m 09s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/52c16ec1e74a463d9ea3190a298d5e47 : SUCCESS in 28m 20s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/d6c2e823dd314b12ba80a58dc14bfaec : SUCCESS in 1h 04m 49s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0a31498175d146bb831b97721e2190cb : FAILURE in 1h 00m 53s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/42928c1ed1a64686b12a681e833ec8b3 : SUCCESS in 1h 32m 49s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8c80ac53ba344933a9a5d43d2e7c5190 : SUCCESS in 34m 21s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a81607f7be2b46b9b71a7ec288df6a31 : SUCCESS in 19m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7e373e6f39ec4fa39977bd43cde2a846 : SUCCESS in 42m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a3674ca1b34548b4893540b3302895d0 : SUCCESS in 52m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a6abfcb5495246f894620a0384c55038 : SUCCESS in 1h 15m 44s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/7638c0d99d3f4e70b0f99dfa70d6653e : SUCCESS in 1h 32m 38s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/484206caab7e4f848de8a1932c045f45 : FAILURE in 47m 35s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/89ac6e46a84b447aad897da8812cf6be : TIMED_OUT in 2h 01m 17s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/06b596181c60433799054ee22dc45a8a : SUCCESS in 1h 06m 54s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/5bb2fbc6e1a047eda3116074cb725d9f : FAILURE in 23m 41s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/26f04a534ac0427eb9008c7a21fb61d1 : SUCCESS in 40m 49s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2b5ee733314f4bbc8587de32fbabe45a : FAILURE in 53m 17s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/eae11e22e19b48a1958deb15d17e0566 : SUCCESS in 1h 18m 06s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/082504412243433f83fbd8aa16246487 : SUCCESS in 22m 26s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"8ea60ead6b2c3e72a7f6fddd8c8240b08773a292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-17 05:55:44.000000000","message":"Patch Set 35: Published edit on patch set 34.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":35},{"id":"22e9dd4ecda0cfdbcb0c2f0ce039df03351928d8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-17 06:00:54.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"ffd216a12189654be36b97fc5b51b8f810e85344","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 08:29:13.000000000","message":"Patch Set 35:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e418f19029cb4bf5b85bfad0fe8a0141\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9a2988cc371a4dbbae7ddcb8a401d006 : SUCCESS in 1h 59m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e4734adb8abb4672939f2524ccb21b74 : SUCCESS in 2h 31m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/853acb45e9544de19de8c5f06ed89ec1 : FAILURE in 2h 10m 36s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec5ffd4ae905422582681ca5f281ce64 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3a5f0d5da1414ccb874d2b5f0822b9f1 : SUCCESS in 2m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4e3854eaa2e944149f7490bf6eaab6fd : SUCCESS in 2h 05m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b81465cefd1245d9b25c59b5148c9470 : SUCCESS in 16m 14s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"e9a55b9d4634473ef276784cab05c1c47cecb4b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 08:32:53.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20b7e88be5b847cfb1f89e7a2da929e9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/87785c88108547cf9028dbf1893118cf : SUCCESS in 25m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e2f7f69202164bfda3f6a7bd65e75a1f : SUCCESS in 57m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8b81cab4308347869666d9c5ad99bd8a : SUCCESS in 16m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/30f776e1831b46a5a79ebfbcc7564bea : SUCCESS in 23m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1dae32399a604f23a9097c52e5884592 : SUCCESS in 1h 06m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/16356da0e79a4ef69277200c543cfa9c : SUCCESS in 1h 36m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08a8be8166694941b02bd8ab357b6419 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e798f5f6167d45158bd8afc047028c63 : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4946dfb220ac48868022e51d70c78a77 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/084eaf9d628645749f7a586dc338b2cb : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e464de90d48b411fbb583e198934a81f : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7301fe3968d24fbab5487b1a6838fe6a : SUCCESS in 4m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/67ba60104d6c4402b4f5f83fcfc765e2 : SUCCESS in 22m 49s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a74404cb94a2464b842a33c18daad4f1 : SUCCESS in 34m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a414679af61643cdad3379683e8c505a : SUCCESS in 1h 10m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7fd3fc363ae848b6a025594b2f47bb0b : SUCCESS in 1h 33m 06s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/9846cd43636f452b8a1c8ab3a469e17c : SUCCESS in 1h 57m 02s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e235e3caeffa4945b72f0188940cb081 : SUCCESS in 2h 25m 07s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/bf751430184846f5ba83c576e22f18c6 : SUCCESS in 1h 15m 09s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/5ae6370f4af842ada896b80d3f2e1205 : SUCCESS in 1h 31m 31s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/416b30378584409fbb6d9ad66c83c063 : SUCCESS in 26m 42s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/f2bb21a1fc8845179a5740834f6e4382 : SUCCESS in 1h 03m 19s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ff5acf09cfc549e99d2afde132f8146d : FAILURE in 56m 06s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c97cc0bdf6604229bfe490ca1c41f073 : SUCCESS in 1h 19m 18s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/cebea78b6a09400cb1c9978ee7785a26 : SUCCESS in 28m 46s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c03cccdc7da54841ab13c19ec5429229 : SUCCESS in 21m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/76610bf15d3d4aaa9d69dab026ae07bf : SUCCESS in 51m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/15496afaa0ad44ab9eb91a18b23a234b : SUCCESS in 52m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/14af6b8477a2419c9795c4b56f1675ca : SUCCESS in 1h 16m 11s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/8a9aaf9db19d42ffade4de6b805fc7dd : SUCCESS in 1h 29m 55s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bfef554d5af64d49a3f311b7f3b43f65 : SUCCESS in 2h 03m 17s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/f5396bdebd9a4baa98de979461e413d1 : SUCCESS in 58m 15s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/58c72f0e1b5c4cc4b03970c9864dacd6 : SUCCESS in 1h 03m 32s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/91807d16bb114d8791de6eb2cadafba2 : SUCCESS in 23m 25s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/78e948aa47df4056b78f41766fd26dbc : SUCCESS in 50m 54s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/145a84429e9743359d475f92b5725169 : FAILURE in 51m 33s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/0d060dc74727420b9e17e09bcebf3251 : SUCCESS in 1h 04m 44s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/bdc61f7208224b2aa39cf529ded19a67 : SUCCESS in 26m 05s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"db64525f5179136955ddbab980c0155be47e4759","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-19 10:44:29.000000000","message":"Patch Set 35:\n\n(9 comments)","accounts_in_message":[],"_revision_number":35},{"id":"29da9f4db4bcdc42e04ab19cf23cc44ccce59361","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-24 13:03:07.000000000","message":"Patch Set 35: Review-Priority+1","accounts_in_message":[],"_revision_number":35},{"id":"d51ccfbe15921d2b956b77a59285f32fa030f02c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35638,"deleted":true},"date":"2024-04-26 11:17:18.000000000","message":"Uploaded patch set 36.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":36},{"id":"1a171cbd61940ebc9fb0e321dd4966522dbbab37","author":{"_account_id":35638,"deleted":true},"date":"2024-04-26 11:18:07.000000000","message":"Patch Set 36: Backport-Candidate-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":36},{"id":"e7545390ac26f11f28f0ffa21ce1f54f728add11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 13:38:59.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7037c4daa5a640fbb52373ca1ce7f8d3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/955687f24cac4ae79278146091f5ca09 : SUCCESS in 26m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/afc9ef17b41e4e0497f5a733dc610a52 : SUCCESS in 2h 04m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8cb550f27c514d129fc7d5ffdf96c822 : SUCCESS in 1h 11m 07s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d3cbc7917874421b04b9bd261373c0a : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f40f1d7c197b4248a80f07f4975cf728 : SUCCESS in 2m 40s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f9f150482da14c9bb83417e3f1c96050 : SUCCESS in 2h 19m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/825db2e24a9044c690bd962197fcfe0a : SUCCESS in 17m 39s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"5dc0de9cf32a574ac80c87009ae3b63f13cc723f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 13:58:43.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78f8fd0b738e40ffa9f749eb9019cbc7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f93cf3bb951c4994a6f43debbb082591 : SUCCESS in 27m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/cf5df8e718d84dbbbc86ded020059304 : SUCCESS in 58m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3009bb1feefc4750943d2839230c5917 : SUCCESS in 21m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3bcdf1e8e11443478a010d36de2898e3 : SUCCESS in 27m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f74b4d6b10794b85952f7b0936f0fb05 : SUCCESS in 52m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/69dd6b0afea645ffb7009d8e7096a633 : SUCCESS in 1h 32m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e33efc3237246dcbad5d38f7da8d8bf : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9545dd306946475db1c473c499f0fa66 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be2ea222d4a04a039d943385663567d9 : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/abc0ab8c183243d1aa612918394410a8 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbfd6b23fbf641699ceee8d244cd78b0 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ca83c2346984a7a864c23c669d0bf17 : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7de084300c6948768b0dd4c41a551abd : SUCCESS in 25m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/25867a2360a546adb8ef75bdb6877856 : SUCCESS in 32m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1eec6227fa0f4ecebefd9adabe1496cd : SUCCESS in 59m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6dcb654af75e4f7cbb3ac0282cef12e4 : SUCCESS in 1h 31m 29s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/4b16145297db4d46b36b175c316eb497 : SUCCESS in 1h 49m 58s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e99d3a4486be44ddb6f44ae7c11ff46d : SUCCESS in 2h 34m 40s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/b6df6689ec4543748e06be62639deb46 : SUCCESS in 1h 23m 23s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/7c9ada87d9cf4ab7b5907da7b36b18af : SUCCESS in 1h 37m 31s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/0eafa0e99fc94c1fa0c731872bb36dc8 : SUCCESS in 29m 16s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/5e3d5145c58945d9b1867e00031afc0c : SUCCESS in 1h 05m 46s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/7cb9ef291a49454b9cbb06d5d4fe7d2c : FAILURE in 1h 00m 20s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/8290074b4f2743cf8de980a5be0666d3 : SUCCESS in 1h 32m 44s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c00ac3cf117a4493a45e8e06a08c2e5d : SUCCESS in 36m 00s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/63604f90f138422fbe2ea1a049b5cb97 : SUCCESS in 19m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/053c5298f2a8447ab61a16bb1422d452 : SUCCESS in 49m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e1e32bd72e37459aaa364ede2ba92e3d : SUCCESS in 50m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4e2f2ac2caf24db0af53bfb14251ade3 : SUCCESS in 1h 18m 29s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4060679651ee47e8a68d4b113727ef58 : SUCCESS in 1h 29m 22s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1d06acaced174a75853f20717a5e2d50 : SUCCESS in 2h 33m 36s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/34bcb0098ed84d80a52e818d3dcc0ffd : SUCCESS in 1h 06m 29s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/aafa780a797a4d179b29597d03346d3c : SUCCESS in 1h 03m 49s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/a5d9d46ea0684641b7dc7bc488c6f352 : SUCCESS in 26m 30s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/ccf439793c3e4ee08dc70551bb81dddb : SUCCESS in 47m 08s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/381d93e5272d4c7caf337b547427ea09 : FAILURE in 45m 50s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a84f3667f87a4c09a7c77704e2da039a : SUCCESS in 1h 01m 33s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f4fd8422603546e1960f572d496f7097 : SUCCESS in 30m 29s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"18e58efc8309f29b1ccb235069a572f7f0693043","author":{"_account_id":35638,"deleted":true},"date":"2024-04-29 09:49:40.000000000","message":"Patch Set 36: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"c2b4e8c09be109cadd16af01d89a154dd49d2f59","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-29 09:54:17.000000000","message":"Patch Set 36: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"e285f0ba086ac85056c88f88c85fefc2f46e047a","author":{"_account_id":35638,"deleted":true},"date":"2024-04-29 09:55:10.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"1a3be033c03e1b01e900dac49b6b613fcabac8b6","author":{"_account_id":35638,"deleted":true},"date":"2024-04-29 09:55:42.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"1281df9811ed78aab5314ebaea737acb9e18c26c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35638,"deleted":true},"date":"2024-04-29 11:07:59.000000000","message":"Uploaded patch set 37.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":37},{"id":"7feccbd790a810b3ad8f00a87daefdb74670e608","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 13:31:08.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70229bb0ccc148d5861f7dced773320d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a6877ff27b5a4346a43f099b7670e0a0 : SUCCESS in 19m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3d4cdf071a1544efbc004bf034d35422 : SUCCESS in 2h 14m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8d5838ad437941728fc166b9c1fa2fdf : SUCCESS in 1h 15m 42s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d96618d658e8412cb6a6527b9ba24132 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b3ee32756d2744ce99685d40625ecfc3 : SUCCESS in 4m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3683fbc1e4e04f1b9f10ca40227093dd : SUCCESS in 2h 19m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5c12e549945d429db44175791fc798ba : SUCCESS in 21m 21s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"e6e542397c41ee8094cdad611ca94eecacc52b89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 13:45:05.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/084c7b2f49f74c9db5bb835073ca7448\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ed1311c7269c4e5eba623de6bd6b8266 : SUCCESS in 27m 50s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5e54723bdacd4dc08722a7c6ad988f1a : SUCCESS in 54m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/25831c5807394f6089095b17d375dec2 : SUCCESS in 18m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e20e89493615471d913131fd17178cee : SUCCESS in 21m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/934fc94dbc484a2389d56ae9094b8009 : SUCCESS in 1h 01m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4db28a3ba4194040bc76f00cdd3fcae3 : SUCCESS in 1h 38m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bab94e3dfba943fc9fb6f846c06c1889 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1ad6669a97841ee978d8a0e44136d8b : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8842c9015a34690a2a9a0a5ebd76bb9 : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6cc00125709423f9f8c8c84139c107e : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa2ef5ceb23b401cab7f946d78d25cdd : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d57d9df741d5473a8429390b85d049c0 : SUCCESS in 4m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c95c885641a84ddb85fb879cb7a3d99e : SUCCESS in 23m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1de74335c1c346d0b8fe5b7968a381f6 : SUCCESS in 36m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/029426a8dc9f447494a531cfa07fad98 : SUCCESS in 1h 08m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0d65a77175a14001be2973d3b81b1542 : SUCCESS in 1h 38m 25s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/5f5d23b6327c4bf986f5d2f8967a03c0 : SUCCESS in 1h 59m 21s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c937b5bcd92f459d9f1e2e544f3fb982 : SUCCESS in 2h 33m 13s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/393f6ce1063d43e4a5c2e3e08a9cb1d6 : SUCCESS in 1h 25m 42s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/ced9c22f026c49cfb318b25fa8ac9739 : SUCCESS in 1h 09m 04s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/bf0cdefba47d45bebf47cfb833a06114 : SUCCESS in 26m 22s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/280d55077f71424fa96bd3b94dcdf1de : SUCCESS in 1h 05m 20s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ebe6b7de8fd941b7b3b5eca48c871b30 : FAILURE in 1h 09m 27s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/644429d0161f4c76b900958a40e997f0 : SUCCESS in 1h 16m 59s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/078533e4cc26490d899a309bb2c42343 : SUCCESS in 43m 33s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f07f752118944d6281d36f2f54215a9d : SUCCESS in 27m 28s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f3c912a358b4406fa94d85882610f163 : SUCCESS in 42m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4f82a28990f04e3a89fffee2a973e4df : SUCCESS in 49m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7f1795cc93324b21be678f458d83f270 : SUCCESS in 1h 19m 15s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/06b557e292b64127b712c47ce3055729 : SUCCESS in 1h 38m 00s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d14e8fb283204899bee24b8a888b9a61 : SUCCESS in 2h 27m 13s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/7edc56d9ce8741d2a54814acc13258fd : SUCCESS in 56m 14s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/37e650def0c141de902dc63842f0a36e : SUCCESS in 56m 32s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/cf98b2e53de74732b2c4b85f546bf588 : SUCCESS in 21m 38s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/0bcd541203bb4f2187e8af7fc60c4c66 : SUCCESS in 42m 25s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/aa2e483e0e2944e5b6f47539661a034a : FAILURE in 52m 47s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/4a8688031a5d45029cc0067f7ee08ce2 : SUCCESS in 1h 07m 12s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/db1ec27bd818438aa04e1b074ede4968 : SUCCESS in 27m 54s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"0e303890d80b815ed6aaf9c6812233fe2481b996","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-05-03 14:06:42.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"f06f72d1156bcce6f391f1fffaea3e5ec28531f6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-06 07:07:23.000000000","message":"Patch Set 37: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"ea47e2743da9846d9166c6ccc6b02588dd68be1d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-06 07:07:29.000000000","message":"Patch Set 37: Review-Priority-1","accounts_in_message":[],"_revision_number":37},{"id":"c0a02981f518cd3b6c225c48cc23349a884a54b9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-16 05:24:59.000000000","message":"Patch Set 37: -Review-Priority","accounts_in_message":[],"_revision_number":37},{"id":"d826380c38d64e0a60f22703ce0eb35f0db82586","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-05-27 15:18:16.000000000","message":"Patch Set 37: Code-Review+1","accounts_in_message":[],"_revision_number":37},{"id":"fd6d65e760b2f2080bd849e977b03a8e53ac05f1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-05-28 11:46:22.000000000","message":"Patch Set 37:\n\n(2 comments)","accounts_in_message":[],"_revision_number":37},{"id":"bce6dcc97ec115d87c02981b5f2ca194a7160c75","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-06-05 15:58:04.000000000","message":"Patch Set 37: -Code-Review Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"d65ec1a9dadb38c43b1a1c16a60fda7ebef45e8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35638,"deleted":true},"date":"2024-06-07 10:31:52.000000000","message":"Uploaded patch set 38.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"72c1f378a5bc9f48da51c07ceca913f46ecf1a5f","author":{"_account_id":35638,"deleted":true},"date":"2024-06-07 10:33:17.000000000","message":"Patch Set 38: Backport-Candidate-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":38},{"id":"9100880d40d7d4702c270a2597c49b107a33ee69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 11:45:33.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/01e181d46d39446eb0e2b5f6543fdbbf\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5df2c134dd704bf7b11ca75dfb8ebdda : FAILURE in 39m 54s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/84b9047fc89648ebb9cad9fa0e7929ae : FAILURE in 19m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/01d235a67d6b4ec8a8cfeef85b43241b : FAILURE in 24m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1e608f0a0d0a4c8194f50ff4ce92560b : FAILURE in 30m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f1cc953ce4fe4d40b62f389dbae27b82 : FAILURE in 26m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3e22ef30b6074da586a8b0f407b6f99b : FAILURE in 31m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/116b1eb2697d42a79a0f13e2b3eca4e2 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c38ff46d7d34a24958a3de48e7209b0 : SUCCESS in 2m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e034d8832c5b4def9652855aebd96065 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c0f90789e647402f8ff906c1c804d82f : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26ef4cb243ad4e92be44a53ce84fb479 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c8d5fa2ec6f4b37b22b2c3547834530 : SUCCESS in 4m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/13ed7b19082c4fea8f48a4abc663ee81 : FAILURE in 24m 10s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0d8d77b7aad84f088f2cda4b7b117489 : FAILURE in 32m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fe2ade0348334b7fa8838355f7e4372f : FAILURE in 23m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/43c4d7567d6d43c7abaa94014da15c83 : FAILURE in 17m 34s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1fc5ea6daf574a5e937a117968608dd3 : FAILURE in 18m 46s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f2d1ca6c3951432ba69ddf3f075c1056 : FAILURE in 22m 13s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a445c9ec4a67451faa8d155a2fd66883 : FAILURE in 19m 42s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/6b75c41db6e94e30b37cb7275ac1f17c : FAILURE in 23m 41s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/0c4e8169aaf942ae99b8e063ff5f2d15 : FAILURE in 17m 52s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/b8971cea2f9a4b21b4d00e22e549c5be : FAILURE in 19m 47s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/7d59356422844892b3b4a8c8310e37ea : FAILURE in 17m 07s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/3b0d90e189dd47f9a766780e0d3403c3 : FAILURE in 19m 33s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/fef61cf88ebd4aaeaf5c6476555a2352 : FAILURE in 20m 49s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a5a6facef30b4e2886866d9aed81260d : FAILURE in 23m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/eb868111df2d45b3aa8348813ddabdc4 : FAILURE in 1h 08m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6f9d8177b7374d4ead8a700e46a0333f : FAILURE in 27m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a7ec9b06ca4f428f89ec830f925f831f : FAILURE in 24m 23s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/acaa268f61174040be11d44a3987c9a9 : FAILURE in 26m 39s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f2fb464d157d4b238c151ddcc1a68a1c : FAILURE in 26m 35s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/fbb5449eb1534fde8293c0bd42f7f3e5 : FAILURE in 24m 09s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/36a91c8a752847bd9d2c2ccb3b5a4045 : FAILURE in 28m 46s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/9531f23979e249cd81aeea6048854211 : FAILURE in 25m 48s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/166e374122514d9f87cee4bdd483a0f0 : FAILURE in 26m 13s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/5d63dfbec08c415d918bb8f323da6f80 : FAILURE in 24m 33s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/3f33ec2f574c407ba4cbc524a5b0bcb9 : FAILURE in 28m 29s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/70951cf6a3fd43999ea5233891070b88 : FAILURE in 26m 59s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"de35602a112a0af9cb07a407607306918ed37596","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 13:41:41.000000000","message":"Patch Set 38:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c1a6cb4c99d490b95f2e9fc8d9bc0a0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ce19e1ed372547acbb2d96aa47bb7546 : FAILURE in 22m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6b0626663560400a8ac639b1149893ef : FAILURE in 25m 15s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/96066bf5bde64e2b94105ba78652c05f : FAILURE in 28m 44s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f050267f19174a5c93f144660bf39f94 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/87f6ac40971142389173b9b49991774b : SUCCESS in 4m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/64d42e9efb0b4afcb2454a40c15e16b8 : TIMED_OUT in 3h 07m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/aa371b9d5f124d51864caeb04543b258 : FAILURE in 24m 49s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"ed0f626cb8e2a34f1a3bdf17a30dd4b11cb35e7f","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-06-11 08:09:41.000000000","message":"Patch Set 38: Backport-Candidate-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":38},{"id":"3fb4f342d6f28469cf31ecf88efb43dfea5c94f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 10:10:43.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c91495c5877845d89e2e7766a2382a4f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fad3dc6cff4345808201f98bec06002b : SUCCESS in 30m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/161cb8198c1342a6b741b39de54fae61 : SUCCESS in 24m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2485c1a3b5cc4f34b4927402b0d5aa38 : SUCCESS in 1h 06m 07s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5bcba7d3591e4decb3a178198c37bc3c : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/05f76b99720e4f9895b4f15b2f386381 : SUCCESS in 4m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/440725ee7ad54c22a1dc0ddddfe4d4a1 : SUCCESS in 1h 04m 53s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c206b2f534e64b40b6ebf674e39f6244 : SUCCESS in 1h 59m 02s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"ee28905ac66f14f44571f5e99671e77abbaa75bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 11:01:07.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c22a8585ff54f02b8cdaef5f895eac0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3f6b2a5483404ea69e9a764ef774b821 : SUCCESS in 27m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5a1b8cf99d9b4211938c863152bcfe31 : SUCCESS in 1h 11m 56s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/39315c9b01f24a30915b54f9e3cc627f : SUCCESS in 20m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/eb077c1744f54719bcd0d85847529a29 : SUCCESS in 30m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ac5b4ab6a76a4f06b8f821d78af1588f : SUCCESS in 1h 01m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7e49a0cae19c4c94a8b7da93b547120a : SUCCESS in 1h 31m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ead27f937e764b4f842482aa1c8cb171 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a185cf7b5294c98885a470b7d4f2dbd : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a5cf1faabac34638b0a09159e117a821 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/415e748521ee437e80562868622481e7 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f13c47917f54e9f8adc79cb01808727 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b560bab6a2384995b6e23d9cbd579d6d : SUCCESS in 4m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/796ceb471af84e9796a8716a535d9a5e : SUCCESS in 31m 04s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/23d16f43451b41ddac1070a5f0506e0e : SUCCESS in 37m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4007f36ebc144712a7018be7d1ed9075 : SUCCESS in 1h 08m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/48e8bb6967b64dd3927b87ccb355b638 : SUCCESS in 1h 43m 29s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/0bdc6a1c453a40d0803e0583e1073745 : SUCCESS in 1h 53m 31s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/479d8fb117ba4e83a5da37d974e3340b : SUCCESS in 2h 30m 56s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/ef728e658e49437588b9f5150fd138cd : SUCCESS in 1h 21m 27s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/7bb8268d162d4d849c3d1c6c8f97f5ad : SUCCESS in 1h 25m 54s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/a100cb01bde24a12a3b013afd743f017 : SUCCESS in 31m 20s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/786b48ebb6f5458e8d3aec3645517a3b : SUCCESS in 1h 02m 50s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/6f9522c60df94f009c487f1c3299c11e : SUCCESS in 1h 15m 00s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/f8fb099ab6ad4b37a8c30ef57ec2c4af : SUCCESS in 1h 28m 05s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8838781c1c4e459f9fccba87ec50d114 : SUCCESS in 34m 39s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7060d81fc148477681ab28e9a9cf3c1e : SUCCESS in 20m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e728045626fc4fada01a48054faac876 : SUCCESS in 57m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/465ddce3fc0146d9830ae323e8cc11ee : SUCCESS in 54m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9a1ba7bde1c54dd694fe064d522e587b : SUCCESS in 1h 23m 55s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/d125344ba7c34abc815ee117cb10b8aa : SUCCESS in 1h 28m 33s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3198acb0ba4e4cf996d498c466be27e1 : SUCCESS in 2h 18m 47s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/1c4913a3de674aa9a8725b884f4e3f45 : TIMED_OUT in 2h 01m 13s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/9f77dec978cd4ab2b24ba739c4161365 : SUCCESS in 1h 07m 03s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/6148c025375e4a09a0273801baef73ba : FAILURE in 22m 02s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/61a0e337a5af42d59cf11ad54ac1657c : SUCCESS in 51m 00s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/1f6c6cedfaf94129a244772a1e463b9d : SUCCESS in 59m 57s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/bb8d2150521a4adcae4fc7ccfc9522bf : SUCCESS in 1h 10m 01s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5fb02581f4714d9e82541fedb7511d63 : SUCCESS in 26m 43s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"a2cf9c637d38283af9acda1b16e6fa275aa8cc28","author":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"date":"2024-06-14 14:54:58.000000000","message":"Patch Set 38: Code-Review+1","accounts_in_message":[],"_revision_number":38},{"id":"a6b837629a2a9b630b051f4392cb7939ff7abd9f","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-06-17 14:59:12.000000000","message":"Patch Set 38: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":38},{"id":"0a45cdc9731d2cd8ade4740e8981320ca3f36830","author":{"_account_id":35638,"deleted":true},"date":"2024-06-17 17:24:53.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"0782701723ddbd61a8b2b35e31b78cb14d4f5138","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-06-18 09:39:59.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"76c5795290ea1cb478fcb2c4553a81f6161fa528","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35638,"deleted":true},"date":"2024-06-19 10:43:10.000000000","message":"Uploaded patch set 39.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":39},{"id":"2cee8e9b718e7a3d87666a3650d54b65e291427e","author":{"_account_id":35638,"deleted":true},"date":"2024-06-19 10:43:58.000000000","message":"Patch Set 39: Backport-Candidate-1 Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":39},{"id":"e339c5a9c29a5bc5fde218a79fad388324d41f23","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-06-19 12:32:29.000000000","message":"Patch Set 39: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"b64622ea5608b7ac10eae15bd8cf906848140c11","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-06-19 12:32:49.000000000","message":"Patch Set 39: Code-Review+1","accounts_in_message":[],"_revision_number":39},{"id":"55664973424a5c9e7a2582bbb8c5606fc752f1f6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-19 12:35:31.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"368b03aa7826c273df5e1941cf7e49b1f98d07ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 13:17:22.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99532dc71d404b3fbdefbb1f2a48d8d1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e383357b2262497e9263110d92d7f0cf : SUCCESS in 24m 51s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/566781463a0445c1afbc6eaff87c8447 : SUCCESS in 1h 08m 56s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a295ade8eb2a4db1b054664c6d3b0b08 : SUCCESS in 20m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a2a471a67d9a4d77a81ccf5cd2046348 : SUCCESS in 27m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4524c5821f274945818618a5d7a909cf : SUCCESS in 1h 03m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ed780723b6cf4243a2080c22a460ef92 : SUCCESS in 1h 22m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65622d295e384bef9087fae3c93b4fa6 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c63593bb7881400c8f37f6af9f31224c : SUCCESS in 2m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fcbaa0248d9a47128b486b0823dc1473 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/434281e1800245e89f53f035a552c01b : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/265560ab02284a338b1548f1c73935f6 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7852b08dfc444cb4be2129551b0ea924 : SUCCESS in 4m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3f4f7a4845514ea0ac799537bb9e0bfa : SUCCESS in 29m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/14c525fb34004766865af52d074a1ef5 : SUCCESS in 36m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2f500c9496454540acbb91d1d3e4731a : SUCCESS in 1h 10m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/91875d724ad5452f96a9bdc0e6c54956 : SUCCESS in 1h 37m 21s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/b8a3c6bfa3aa4398bc2d9536ca20e867 : SUCCESS in 1h 51m 22s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2237ef518dc94a57b9357f88819b9a97 : SUCCESS in 2h 28m 38s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/b5366813fcbc4e89b560070e67b1b4b6 : SUCCESS in 1h 19m 19s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/8f8d2b50ea1b4add84452dfdf537bed2 : SUCCESS in 1h 20m 51s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/56700a198c494902a9054a8419b5f5f2 : SUCCESS in 31m 18s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/60e2ba391cd74bffb410b74f3c7d1090 : SUCCESS in 55m 23s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/e5ac6b917b9b4f83be6ff381003b7cbc : SUCCESS in 1h 09m 40s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/b102895b57fa45ab94e9670cacbcda7b : SUCCESS in 1h 33m 05s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/50594f1ff10a4588ba0db41f75ea1aab : SUCCESS in 35m 35s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d03106f0587943468bcdb11ab9e30ff1 : SUCCESS in 19m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3f536147cebf4eb5ac771f296da4023c : SUCCESS in 1h 08m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4e3260d0d94042bf8903c0ac6f4454bb : SUCCESS in 49m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a9afb122c9ae4b27aed4bcc2a39d7832 : SUCCESS in 1h 16m 54s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/2e31f9fe3f044969a38ad045ef67810e : SUCCESS in 1h 22m 48s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f7054cb9043c49569cf2d12e688ce331 : SUCCESS in 2h 13m 33s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ae37bf87f49e4ec6b20257da83265a6b : SUCCESS in 1h 04m 29s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/1c722607947b49778e0d846b377d80d9 : SUCCESS in 1h 06m 19s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/db95a0f334ab4316979c86969046513b : SUCCESS in 21m 30s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/a89df23fb71e482482bd0fcd6ad1ca78 : SUCCESS in 44m 23s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/494937af5968427488eb0e381d79d585 : SUCCESS in 1h 00m 53s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e960eddd5b0a4e42a5c7db6c8c04b744 : SUCCESS in 1h 05m 37s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/cdf84c241d5b477ba224dd6d99975ca5 : SUCCESS in 27m 37s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"00b52f3b4d060b1a1aa2f6e196ebe0bfff0d70d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 13:58:21.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8b57a75e38f494f92e44dafe8e7ce5a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2044605ab1f14176bc21a774726d9bdf : TIMED_OUT in 3h 06m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d446553984f8488c93a2ae2cddb89c29 : SUCCESS in 30m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/021bcc0872974efdb321fc768d7e64e3 : SUCCESS in 2h 00m 03s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b57d9e06a55a4e8faba7abf44788e677 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cf96110a7ae245c7913c69f3a34bcb1b : SUCCESS in 5m 02s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4564ef9fe1b44bef9ea882402bf355de : TIMED_OUT in 3h 09m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a686bb3272ed43108fb555f4018af2cd : TIMED_OUT in 3h 08m 55s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"1fecfc69a92c339e226660495e90d187e1f05330","author":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"date":"2024-06-28 17:07:26.000000000","message":"Patch Set 39: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"af3b402be6cfe47b51c65dbf0f4563bec9734ac6","author":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"date":"2024-06-28 19:08:23.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"eee4a5a4ff9263b5dcc982018bc5514afe705bf1","author":{"_account_id":35638,"deleted":true},"date":"2024-07-04 07:39:38.000000000","message":"Patch Set 39: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"a5414e445351f20f7fad41645bace1dca0f942d0","author":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"date":"2024-07-04 14:21:50.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"1ac93643d9254cbef382668fb3223298207aad2d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-04 14:31:19.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"6ca5b1cfb0754dc06866d773136100ae947809c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35638,"deleted":true},"date":"2024-07-12 13:18:30.000000000","message":"Uploaded patch set 40.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":40},{"id":"ff964eaaa79152406f84cb53fb36680b6c8eaf68","author":{"_account_id":35638,"deleted":true},"date":"2024-07-12 13:22:27.000000000","message":"Patch Set 40: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"ebfe221fcd78e104325f6959edb709a66b21e9d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 17:10:27.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d240cab75f73425aa737898ced5d8ee8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7406f766584a47b38e5257224c796c00 : FAILURE in 21m 19s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/906e9c44600746fead41f643b6666b04 : TIMED_OUT in 3h 03m 23s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9926f029d83840e09dc1835bc2cc98c1 : SUCCESS in 1h 54m 54s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8e38d9e78724bf4a54633e239e0a70b : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7860508e98234f70ba387a096b4ce3b2 : SUCCESS in 3m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/37f7e5372ee640c681d3326a662d1320 : FAILURE in 23m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9945d1b6f8e248edb87e3fa055b8a6fb : SUCCESS in 2h 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"88f4c011e7cf203f2f293e57272d6b4cca749934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 17:21:48.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c4d952ccb2b44498e18b75d5f8487e6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/eea651f5e24f467c8b45e0d819bc509b : SUCCESS in 30m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f592dbabce9f46f29035fd6dd86b74e5 : SUCCESS in 1h 09m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ca432ca5d9d840f9877c8fbd01a37e1b : SUCCESS in 22m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/338bb0c6c42944caa439e8f56b8636a3 : SUCCESS in 34m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/869888b77217498681e442eb9f53eb9b : SUCCESS in 1h 06m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/85b59ff84fa744158f51e5f56393f989 : SUCCESS in 1h 31m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6323a0c608a4dd88271277dbaaf12f8 : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d96a1ee241d414b98d7ca888ac8e79a : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c19df82252744bf937f82fcd25daeab : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44d3a8cc8a3948b2a394710eaf48b98b : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8595e4a8f584e6aaaf5247bf37c9d03 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed1cb122d4bf43af875f33fff725a851 : SUCCESS in 4m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d17974682d204e38964ba6b4954e4541 : SUCCESS in 26m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c7625c452193496483f7ad8c620b4021 : SUCCESS in 33m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a2ee657697a748268b1e49450e69994f : SUCCESS in 1h 08m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/23b8f9bdaaf34669936240397fdae74b : SUCCESS in 1h 39m 58s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/d5dc779e68ca469c940fc8d1a71af2e2 : SUCCESS in 1h 44m 16s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d9d050bc9e024767b344090319dfdf8a : SUCCESS in 2h 42m 53s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/7648413faacf4b15b0f3edc19d7cd0e1 : SUCCESS in 1h 20m 17s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/ba81ac843a0b4bbab142cefe61ccb573 : RETRY_LIMIT Host unreachable in 1h 09m 53s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/a654310d2f2b4f1b8ff80c7c663d09cb : SUCCESS in 26m 38s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/ce1f18d6ed4645499e0056f5728518cc : RETRY_LIMIT Host unreachable in 39m 21s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/7a66ec13966c4509bae2fd29795a77d2 : FAILURE in 59m 08s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/ad610b2035b44e7898a3b7c0392ba448 : SUCCESS in 1h 20m 29s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9151fc0d39fb4a73a5555b582c548044 : SUCCESS in 36m 18s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/894282a22f0c495fa9930c9f01460c72 : SUCCESS in 28m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/44698f957a76411995078874ae73a9a5 : SUCCESS in 56m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c850bfc3b63f49b5b569e9cb5b596c79 : SUCCESS in 55m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/94bb7650c7b34af6a61ae0e78d8a5c7c : SUCCESS in 1h 21m 04s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/85fa499ff4f940f69d232ef34beb15af : SUCCESS in 1h 27m 33s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1d9564e9e86940ada2892cca12415079 : SUCCESS in 2h 14m 09s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/8e4ef4811c8542eaa7ffe1063f895bdd : SUCCESS in 1h 01m 18s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/f06963458902489a95b3134b1fdf10a1 : RETRY_LIMIT Host unreachable in 51m 15s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/407c42a5687b4300b171547dff0afc0a : SUCCESS in 24m 40s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/9985c02316d647c5a52a5d5919f4aec0 : RETRY_LIMIT Host unreachable in 28m 11s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ced68b56af4246ddbb83fd97b0b686f4 : FAILURE in 51m 02s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a1b5f5cd1a5041e1ab6c50fd7ed0eaee : SUCCESS in 1h 13m 05s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ab18ec43301b42b29f64cf1f1b4c48d2 : SUCCESS in 27m 17s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"29bac5531da6a2251b39c7a62b294f90cf6ce87d","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-07-18 15:54:39.000000000","message":"Patch Set 40: Backport-Candidate-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":40},{"id":"0f1331df1b540165f19c8c52cae5ee03c9c9240c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35638,"deleted":true},"date":"2024-07-25 11:25:29.000000000","message":"Uploaded patch set 41.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":41},{"id":"2e9443682b3e86b3530a750cab266e19d479678f","author":{"_account_id":35638,"deleted":true},"date":"2024-07-25 11:25:33.000000000","message":"Patch Set 40: Backport-Candidate-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":40},{"id":"6d38f156efce2d1ba5c950f6f9dbcf6252d60d9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 14:12:49.000000000","message":"Patch Set 41:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98faa19eaa1e45c9830fdf371e829f02\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/01162cb4063c46759bb0fd638c14cc95 : SUCCESS in 37m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6c00eb1ab4be4ad792471c2a5f42d961 : SUCCESS in 33m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e3e1b5b722cd49c39f64189f191468aa : FAILURE in 1h 18m 16s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7cca8ab5160a4c50941e74c0dd67a55e : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c1975f7cb0204d948cb33729865e10cf : SUCCESS in 2m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9a5f01f43a54493a9ea2cc230afc2963 : SUCCESS in 2h 44m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/79739213709645ec9093c57fbb44283b : FAILURE in 29m 49s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"50791239444366124a27909fc09978529568276b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 14:27:27.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94f8c93ded2a4d5a8bab7191fcae08a7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/45d941434f5748228a3e86d33f3842c4 : SUCCESS in 1m 55s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/328a0578550c4adcb48db0a41071e9b2 : SUCCESS in 24m 27s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/14050c57f62f468aa11a48d98742f00f : SUCCESS in 1h 16m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/288a8893495d4742b6b37ec11d5dfd11 : SUCCESS in 18m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7212dc41a1dd4456924a2f6d88173538 : SUCCESS in 29m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/95c729bfdfe346adb7bf8d563bad49f7 : SUCCESS in 56m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/09c5ed66df6b427188a7257f56a4f504 : SUCCESS in 1h 36m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d14a5f1fd214bf7a88a719bd75ce86e : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae5ac7fb57874f1eb056bf9d117c22a5 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3dacc7bed5bd425299b935c82a17fe62 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d4e4f4bef52448e847eea48e486561f : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e48722e67e14fc88d2b91646b2bb358 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0b3668f359a45848f533ae02c3c23a1 : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e94073092a844c33b6a9e7d8c4dc6185 : SUCCESS in 4m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d778a31d7b684931aead3b7ea4ad2590 : SUCCESS in 24m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4e9b502c4c7442d59325492e7d887cb2 : SUCCESS in 42m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7a20dc16fc664861972c91d57c40a80b : SUCCESS in 1h 21m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f1002755fa9d4be087cf018c9ca0d5c2 : SUCCESS in 1h 36m 17s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/3ce013fa4110414ab603729dcb34cb78 : SUCCESS in 1h 53m 03s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0c04556b201243139f23098faf40391c : SUCCESS in 2h 54m 50s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/f76b5687350541bea0b56dd2f2b6f2ee : SUCCESS in 1h 21m 51s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/76be259c068a41feac6b563d82d546eb : FAILURE in 1h 15m 54s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/027f5e90532d4e5c889a9c20353e473f : SUCCESS in 29m 13s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/cc86424437084e8f8ccc779c6db4606a : SUCCESS in 59m 11s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/8041c416e1c341259d6c8cec5d9f644b : FAILURE in 59m 20s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/d62593e74586459aa897fc96e7af794e : SUCCESS in 1h 37m 28s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/71469d5e35384e538b7f0f58165301c3 : SUCCESS in 38m 19s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6e85e4dd48a1498aa83ffdeb1d7e5e2a : SUCCESS in 17m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ba8109a1e9ce45b1b02fd730c81a12bd : SUCCESS in 50m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/55ba7370478c464ab790f2549829f4e0 : SUCCESS in 51m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c1660b10e915451e8dde5ad0803e6e87 : SUCCESS in 1h 19m 20s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/abfa191436fa497b8447838c9052e984 : SUCCESS in 1h 29m 47s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/09ee75bea6654448994e2004e6d0a480 : SUCCESS in 2h 28m 51s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/277ea28ec53a4508b4383da3fcbd5afc : SUCCESS in 1h 01m 00s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/8a2f33f02f234354ba55ce2776108750 : FAILURE in 54m 18s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/5932130689264da2b94bf4aa7f353584 : SUCCESS in 24m 11s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/5212493ecf32493fa93a91dcb3b8bab8 : SUCCESS in 44m 34s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/a005272d9ab1457a9be89d0fe9312a01 : FAILURE in 42m 27s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b35f709205f74eafbcd9d36666519b3a : SUCCESS in 1h 06m 24s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/aee1d46f06ab425296c86d69697e664e : SUCCESS in 26m 05s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"c08174d5741bd6b1d7b6fe50a113193e4cbc30b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35638,"deleted":true},"real_author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-27 11:42:49.000000000","message":"Patch Set 42: Patch Set 41 was rebased on behalf of \u003cGERRIT_ACCOUNT_35638\u003e\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":35638,"deleted":true}],"_revision_number":42},{"id":"d49876846fbc5f199c6b476be58378c71bf701ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 13:52:06.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc063d3cd2824771a3bdce18bf514f89\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4d7a18d7f0d64680843b7b7874325ddf : SUCCESS in 1m 45s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9440a0320b6841e6bb1fc2eb7473bf11 : SUCCESS in 36m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b5267d4f2e404223967fa1f960f340ea : SUCCESS in 1h 07m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5cb5cc917f0643d4a55338a9d654f9ca : SUCCESS in 22m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7df71204e1954e56b0838a78c57ef2b0 : SUCCESS in 24m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5bbf7c214b15411bbb18d0c6931a0c11 : SUCCESS in 56m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a57b4ce3d7294bc2866fa14a7934317b : SUCCESS in 1h 23m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59daf62be7994a988d557bfc480add6d : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e090e7952304424943cff36a619aac6 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4119b3e898b04f11b792e6bd20dc022e : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f50cf751461431096415cc6bcb1f01c : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/998f71a75dc84434988f00e40532347f : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/193a7824408e4de196d37176b20e70e4 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/226a2ae727f4403c8b94072da4171b03 : SUCCESS in 3m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a890836c349549c4a1dce2c00fb838d6 : SUCCESS in 31m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e3e5f69de9ec41129109073e247d0a68 : SUCCESS in 34m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8fc2b59bec3843c59233e0eecd840c96 : SUCCESS in 1h 10m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5ba78c426fed47288e6bb9e88f4af7bf : SUCCESS in 1h 25m 09s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8b21cb27c13b4b68a380b5c84a9a4786 : SUCCESS in 1h 49m 22s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1a709ce2ad4646a9ba18122acd44e994 : FAILURE in 1h 35m 06s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/25fad3162bce4aa3aa6c475d1cecf0f4 : SUCCESS in 1h 10m 36s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/403e6449f16448c9bc1f47117d4874ab : SUCCESS in 1h 26m 47s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/54e623b431aa4648abbf36f0ec55715e : SUCCESS in 26m 15s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/6f8f243297d74f38994d2f29e0f570ed : SUCCESS in 53m 43s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/5d1e4cf5ec794feebf7b496082fc8c15 : SUCCESS in 1h 11m 11s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/421035d662914e658945225390ec230c : SUCCESS in 1h 21m 53s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ee353cb0738b45e0a679bbf10f69ff37 : SUCCESS in 34m 30s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/14ba73df4d7f4e5e87d957ab81fa83f1 : SUCCESS in 30m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2c4323a2ec344d46b6384b8325964c86 : SUCCESS in 59m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7a2f6fb33ec6472da1d5dcd86f54da6f : SUCCESS in 52m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/408cdc227ffb4ee4b3aa84a80388ac56 : SUCCESS in 1h 16m 19s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/7ce339d6369740f8805a1410ce9c5dfc : SUCCESS in 1h 32m 43s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6ec0bbe1b72649cdbf61838caba3b43b : SUCCESS in 2h 02m 09s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/479b91c507724ce79f3a10e03f9cb463 : SUCCESS in 1h 02m 45s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/0e47041ec7324749a4cc4e9c3bd33563 : SUCCESS in 1h 05m 14s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/eb7e687b389046f88f98a42d99dfe534 : SUCCESS in 22m 27s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/ef79e6f1b6fc4c4fab0d8f70583cab07 : SUCCESS in 56m 26s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/be246f563ca94c4b9bc058ab9b3a393e : SUCCESS in 1h 02m 41s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/f1a329b94f7b4024ae3d0a9a8b879a2c : SUCCESS in 1h 09m 41s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9bb36e86acd8403da957c2ef25cd6568 : SUCCESS in 26m 09s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"b4c982fb504cfe57f4e4ff794386664ebf9da664","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-28 11:51:39.000000000","message":"Uploaded patch set 43.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":43},{"id":"1ac93ba337ccb73cd0d0bcce028752e10738361f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 14:32:19.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2337f5f10ba34014a1fe00588b00bed0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a8959a34ddd045cba5901c5694f884fa : SUCCESS in 2m 44s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/dcc7da63dd7a49a4b440c1a07f1ffc8f : SUCCESS in 24m 56s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c5ce62c3a60548448657d414ffe33d27 : SUCCESS in 1h 08m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9cdb5593f52d47faac85b2b076c018e3 : SUCCESS in 17m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3493c8798fc8482a917f86cdbbfc7e42 : SUCCESS in 29m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/64fb9ef3a08a47a498580238786160e0 : SUCCESS in 1h 04m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2227ca7ea95845c18f9670ee42b480e1 : SUCCESS in 1h 31m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b29a2d5756d4945b58fd1e768f14e4c : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f3d108bc7bb4fce81d263f937f0d69f : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b1777847e1ad40d290c3d9831f5c945a : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/148316f97b5e4bfb85666770ba775cfb : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2239766316794c28b9000349ad7b321a : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e5379252eb642b183b41fac12e8c52b : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2516a131cd0345c39d5a98b8ef5120b1 : SUCCESS in 4m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/db577530a7ea46c28da7dd98c2262df3 : SUCCESS in 25m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e19d4bd85d144baf943a925bdf2efe5c : SUCCESS in 38m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b5dc9cc1045c462885d95b913d6ee461 : SUCCESS in 1h 01m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1f3cfb06060c4709b69a0657708723db : SUCCESS in 1h 17m 23s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/7c59ce32285c444b9ad2e767aada191f : FAILURE in 9m 58s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d747cdac6ae548cf9c3547632abc71bf : SUCCESS in 2h 34m 34s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/3bbf5a66d15d4abcbf2afe6b69786d67 : SUCCESS in 1h 07m 54s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/d8e4e95afc2a465e9c1eb4a59831f88b : FAILURE in 10m 31s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/5c879d95039d47989a599c220914f4ab : SUCCESS in 24m 34s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/d0f6a6f9108b4ccbb2cca6372800f3bf : SUCCESS in 54m 56s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/20bbf52a94644c39bf20b007a0e658e0 : SUCCESS in 1h 09m 10s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/49643c8c36a94b90906df9ea3e23b14c : SUCCESS in 1h 24m 06s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c110e69f9004441184abe36cf248ebf0 : SUCCESS in 30m 29s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/129a3ba96e6b42079e7cba1274e0e2b5 : SUCCESS in 17m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/eb7edc15ca6b48e79ccc40af4713b6ca : SUCCESS in 1h 04m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/26ba20618e5344ae99584c45814b4562 : SUCCESS in 54m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6be80f48c0004f95834be463e0de1be2 : SUCCESS in 1h 22m 06s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/c666c2b57afe44c083e988faadcb084e : FAILURE in 7m 45s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/39f39a0ab61c4fdab00321b918a83672 : SUCCESS in 2h 31m 04s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/38b9941b82bf48d7a47910f6bbb8aafa : SUCCESS in 1h 05m 33s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/ccf47b95ceff4e1fbb027e7f2766e320 : SUCCESS in 1h 00m 44s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/8d5b9f68ea3d4df1827145081d1ef15a : SUCCESS in 26m 37s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/95127519bd264903a6ca5dd4d2570b17 : SUCCESS in 47m 57s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7bf93e200df44381bf5ddb4b9d6c9643 : SUCCESS in 58m 10s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/bf2dfa2576aa470c84126cecad90067d : SUCCESS in 1h 04m 52s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4cb5e3fd101145f58eb1ff51ba590c4e : SUCCESS in 24m 44s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"08ae56ac2ea4248a8db946c46b01754126e3ec45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 14:34:30.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db1c3cf319164acdb66d6730bab86db9\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/35b97418009b4d67a25463d09ddeea3b : SUCCESS in 2h 29m 23s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e597ab19479b4025b5b2fab89066ffc8 : SUCCESS in 1h 35m 34s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ad9b39004054290bd59236b668f52b4 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0578635eb0cb47898a528e413b35c557 : SUCCESS in 3m 49s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"b5759152d5b93b76ca4f9a1b45ead1539ebf393c","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-30 13:24:50.000000000","message":"Patch Set 43: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":43},{"id":"c1ff244a3f98f5ce36b091a10987fac98cdff796","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-03 08:55:21.000000000","message":"Uploaded patch set 44.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":44},{"id":"5a1456c1538a0f7e31081d6024ad40c7baf5f51e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-03 08:55:32.000000000","message":"Patch Set 43:\n\n(1 comment)","accounts_in_message":[],"_revision_number":43},{"id":"1821caf9c2a6b3756c8ee5166f27e3b720b69030","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 11:09:49.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59713c471ed04059b23c14953806656e\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ccde00b0889b40b69ba5334ce4f54020 : SUCCESS in 2h 02m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6557c69feaf44b58a32471adca70f1f7 : SUCCESS in 1h 36m 16s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/97e7fb9b9afb475dbcd34f75bd9b5b29 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/257fb43e41a74020a5aca9b18e5ebe84 : SUCCESS in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"6eda4acab1af61c2e446326e19a9d4bcaa3009ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 12:03:11.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78a21915f8cf4ca5a30ca31947e96a18\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d018799b9cd5425d88022b260a8fba78 : SUCCESS in 4m 32s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c98b4f65fc244fc4b2ebb7be67ba20b6 : SUCCESS in 24m 27s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e22621dec3f44213914d353f7de81cb6 : SUCCESS in 1h 09m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/190ef481210640da92d032e2774f92f5 : SUCCESS in 17m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/29ef3e55a21c4e7285a870ee7e844328 : SUCCESS in 33m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/424faa624e9d412eb24116c7d3862cbd : SUCCESS in 57m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/83c85a5aeb2d4e09a7030b3ba6c3e9f4 : SUCCESS in 1h 30m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb95ed47fb0a4b6983ff029ac0f57737 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b32dafb0279e48e5a3fd931fd188eb79 : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a7d10c6677845eabcf8ee76fb42e102 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc3d2fff975149028b6557459910de7a : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/243d5da591ec4fac90dfd4b8cdb582ef : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c1e3c46d964a31a4704ac2133021e8 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e624da233a947029c37e7b56fe3d006 : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/11479ea862614f81a3f9daadf0552335 : SUCCESS in 31m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4c534d23f45742e7a5334c3571ed8271 : SUCCESS in 35m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0cda5eaf1c8e4807af4d8e382475e029 : SUCCESS in 1h 05m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/54734aba60d2451e86bb9e9110aba484 : SUCCESS in 1h 30m 21s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/905c4d0c9b584fe8b6f33c6c67b34a11 : FAILURE in 1h 30m 35s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/20064985fa7e43a79ee12c3a00cbd49f : SUCCESS in 2h 22m 39s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/9523e6ec5f3e4170b88adfe02b445967 : SUCCESS in 1h 10m 02s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/8145793119b640959f4446d7870d6818 : SUCCESS in 1h 21m 11s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/ba459d88e7854bf9b8eeac03cfc4b730 : SUCCESS in 30m 43s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/dfaa8314ab6e4f5d8376a66f6895a0a0 : SUCCESS in 1h 01m 05s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/fab8b9c04fdd444f8e0c242dedb8d864 : SUCCESS in 1h 07m 16s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/b071397d93544024bb6719a3512e7704 : SUCCESS in 1h 18m 44s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1f751712a0c2422fa6dc82ef7aba1694 : SUCCESS in 34m 52s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b75936e4ea6e40518e08942eb412ddd1 : SUCCESS in 23m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f837f52d372644cb99b23866641b8b03 : SUCCESS in 1h 01m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a37e81a5658749a69a1fc00e7be88001 : SUCCESS in 1h 02m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f8509aab1c46474cab9c000e13ecbc9a : SUCCESS in 1h 06m 19s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e29f9955222745c8b2c3809bc05a1223 : SUCCESS in 1h 25m 35s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d9fa5ca5fe3c416f90d59294015056f2 : SUCCESS in 2h 18m 27s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/b8014a14462f40269c01d1385c8b0618 : SUCCESS in 58m 03s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/5b01d12e9c3643018b1a6704246c85aa : SUCCESS in 58m 28s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/5e40b1036e064e1089594e638e2ee2ce : SUCCESS in 20m 40s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/348aeddf5c10453f9e72d7d02d9ce7d7 : SUCCESS in 48m 11s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e4fcaf41fb2a40158eb275de55aa57d0 : SUCCESS in 1h 07m 28s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/6be2b9236e2345b5b04a2975ff9d2d8f : SUCCESS in 1h 15m 11s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7690d81a2daf4c69ad209c3e9544fa8f : SUCCESS in 22m 52s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"dbb6158d81423fbe285a9e177d013d524ab2ecd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35638,"deleted":true},"date":"2024-09-09 07:53:06.000000000","message":"Uploaded patch set 45.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":45},{"id":"3c636f2b492b9856b6fe4c48bc9ab7bbe6a8a57f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 10:21:23.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ace8c64b481b4aa08ff3e22af03cf0a6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d188fd8ef97e44a7a1bdc380b500650c : SUCCESS in 1m 54s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/aa984422a5594fceb6b9b235f79fad71 : SUCCESS in 23m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/955c1c6afb51412db36783e1f273c143 : SUCCESS in 58m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/73280aaa482346e38ef7dc4aa7a1c23a : SUCCESS in 16m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2774f4594f6d40bdb665c1950946eeab : SUCCESS in 27m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3ed5a2f594054b9d8968b9e8d5d3ae42 : SUCCESS in 1h 02m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/692203bbac0e495793b94157c81e49d8 : SUCCESS in 1h 25m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e273577c96141aabe56126fcb4aa198 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b32805468384a6a93787506d1ea627f : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7f193d7572f4b2cae559ac9ffbb825c : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/67607f38afe34afaa1bcab33b25ddb1a : SUCCESS in 3m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f97e65ba876475a89c3ddda62f272ba : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23af653d447d4b5292134ff599b46776 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52ada049a9ce4f418ecd88dfe823a956 : SUCCESS in 3m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/aa3bf35b7ff54537b72787dcb2ac8bae : SUCCESS in 28m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a5e55d23b6094ac9af9d1b7ac7ac1b1a : SUCCESS in 37m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b00bb17f8b7b4a5eada20286c12416ad : SUCCESS in 1h 06m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2481754ca2b54adaaf96a72715c9b5ad : SUCCESS in 1h 25m 45s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/bf6fea44e3b848e68d56adc489398091 : FAILURE in 1h 33m 36s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bfbfd92b21d14f99bc217ead06f70b82 : SUCCESS in 2h 18m 12s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/5fbde591d6ba45c1829c1986224667ff : SUCCESS in 1h 08m 26s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/3b9312c72d9646a7853135afe6e6671a : SUCCESS in 1h 11m 48s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/39cf182627c1467aa344ca1ad44291c8 : SUCCESS in 25m 03s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/382d204d35e54e5caee01c404366e0c0 : SUCCESS in 59m 10s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d45c150cbcdb4f00ac6bfd57f8c63be6 : SUCCESS in 1h 02m 44s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/a5bd8f2a4f2e4820b086e8a6014b3bbb : SUCCESS in 1h 15m 31s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ed68889bcbab47a48c4929cb421436bb : SUCCESS in 33m 55s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/123077412d5a4eaebf92579c1a1ebf85 : SUCCESS in 21m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a7ddcaa6f4894ab2a2dcb28a9b34ed37 : SUCCESS in 53m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5b4ec114672e4cea9c2c69a5d7d626ea : SUCCESS in 51m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b39fd5bd662c4cc385dcebc76f39853e : SUCCESS in 1h 04m 55s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/0915050ad8f248b196d07a680fd01ed1 : SUCCESS in 1h 34m 29s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/59870b5761ac459781c7e1a4fcbcc402 : SUCCESS in 1h 57m 59s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/42749efe6c8e4280b4fb5e0ea47ca058 : SUCCESS in 56m 01s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/faec4746cf334d998321e8ef44360865 : SUCCESS in 1h 04m 58s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/c12f2ab0c85842969ff82cf687feed2d : SUCCESS in 20m 03s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/a9318446590a45c3bc9d2c19ad77fb68 : SUCCESS in 45m 25s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/42ecb51f835540ffa9ccec2a4c02f45b : SUCCESS in 51m 03s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/acb594223cb54c8294e2d5d5bc74f6bf : SUCCESS in 1h 00m 57s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ffe79b80e9c14a37883e70e78ba65a0b : SUCCESS in 22m 09s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"d902fb6c72fdef21dafdc80c2a15b0d437c31f89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 11:01:40.000000000","message":"Patch Set 45:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9fa1944913f473a8e369b4129797ed7\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/630c83db31124c5c8ef45db23fce38a7 : TIMED_OUT in 3h 06m 56s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/39e2908a0d2744909b792fb83399d2f6 : SUCCESS in 2h 00m 58s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52226f5c2b57497295cf49d81f5a3c7b : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/610eb99cc4eb4c1e8fbd01d463f382ee : SUCCESS in 3m 56s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"762eec52448a76986f977bf69c36b7914c5233fe","author":{"_account_id":36250,"name":"Przemysław Kuczyński","display_name":"przemekk","email":"przemyslaw.kuczynski@ochk.pl","username":"przemekk"},"date":"2024-09-19 15:33:33.000000000","message":"Patch Set 45:\n\n(1 comment)","accounts_in_message":[],"_revision_number":45},{"id":"543a3e2c451429618df55a2d3f3b8d6c4bb4aaa2","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-27 15:26:37.000000000","message":"Patch Set 45: Backport-Candidate-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":45},{"id":"066e2a9a8c968cb7766330f68e8b09b3409f4780","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35638,"deleted":true},"date":"2024-12-02 09:18:44.000000000","message":"Abandoned\n\nI will abandon the change as I\u0027m currently missing the time to complete this. Others are welcome to continue this work.","accounts_in_message":[],"_revision_number":45},{"id":"9e81ce9b8845a9656c78bb5c43b29ce0e0b15369","author":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"date":"2024-12-04 18:34:33.000000000","message":"Patch Set 45:\n\n(1 comment)","accounts_in_message":[],"_revision_number":45}],"current_revision_number":45,"current_revision":"8c73b8863759ddcd8b70ca874b9645d65ef434a4","revisions":{"fd1e4512a01aef8286f9e5d46f3ec88cf54b9fb9":{"kind":"REWORK","_number":1,"created":"2023-08-16 17:55:24.000000000","uploader":{"_account_id":35638,"deleted":true},"ref":"refs/changes/17/891617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/1"}}},"commit":{"parents":[{"commit":"6ba4582e2838ed4bd152e487eeb95d9ab7995676","subject":"[release] Use UCA Bobcat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ba4582e2838ed4bd152e487eeb95d9ab7995676"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"subject":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support","message":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin, that is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are nessecary to fully implement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd1e4512a01aef8286f9e5d46f3ec88cf54b9fb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd1e4512a01aef8286f9e5d46f3ec88cf54b9fb9"}]},"branch":"refs/heads/master"},"0317c08519664a354b5aca37133a0a606df34ef6":{"kind":"REWORK","_number":2,"created":"2024-01-11 16:05:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/2"}}},"commit":{"parents":[{"commit":"6ba4582e2838ed4bd152e487eeb95d9ab7995676","subject":"[release] Use UCA Bobcat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ba4582e2838ed4bd152e487eeb95d9ab7995676"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-11 16:05:59.000000000","tz":0},"subject":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support","message":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin, that is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are nessecary to fully implement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0317c08519664a354b5aca37133a0a606df34ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0317c08519664a354b5aca37133a0a606df34ef6"}]},"branch":"refs/heads/master"},"02a5875f79615913605b1590210655d5a2021675":{"kind":"REWORK","_number":3,"created":"2024-01-11 16:06:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/3"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-11 16:06:52.000000000","tz":0},"subject":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support","message":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin, that is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are nessecary to fully implement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02a5875f79615913605b1590210655d5a2021675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02a5875f79615913605b1590210655d5a2021675"}]},"branch":"refs/heads/master","description":"Rebase"},"f3a78a775bd59653047425e8c779d5319f33f1d1":{"kind":"REWORK","_number":4,"created":"2024-01-11 16:08:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/4"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-11 16:08:29.000000000","tz":0},"subject":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support","message":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin, that is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are nessecary to fully implement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3a78a775bd59653047425e8c779d5319f33f1d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3a78a775bd59653047425e8c779d5319f33f1d1"}]},"branch":"refs/heads/master"},"2fb0ae2696a011db8bfaedd3e5a7ae0db232ac68":{"kind":"REWORK","_number":5,"created":"2024-01-11 17:52:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/5"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-11 17:52:43.000000000","tz":0},"subject":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support","message":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin, that is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are nessecary to fully implement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fb0ae2696a011db8bfaedd3e5a7ae0db232ac68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fb0ae2696a011db8bfaedd3e5a7ae0db232ac68"}]},"branch":"refs/heads/master"},"8c2a7c09f5a56914e17805c001dce45d1343a381":{"kind":"REWORK","_number":6,"created":"2024-01-11 17:55:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/6"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-11 17:55:13.000000000","tz":0},"subject":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support","message":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin, that is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are nessecary to fully implement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c2a7c09f5a56914e17805c001dce45d1343a381"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c2a7c09f5a56914e17805c001dce45d1343a381"}]},"branch":"refs/heads/master"},"9044d7a9fcc6b4b59990eb86ee00a52448f469d2":{"kind":"REWORK","_number":7,"created":"2024-01-11 17:56:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/7"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-11 17:56:38.000000000","tz":0},"subject":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support","message":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin, that is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are nessecary to fully implement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9044d7a9fcc6b4b59990eb86ee00a52448f469d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9044d7a9fcc6b4b59990eb86ee00a52448f469d2"}]},"branch":"refs/heads/master"},"9bae059636c06e37adaac8f7375f1f8e5ee45c45":{"kind":"REWORK","_number":8,"created":"2024-02-05 18:37:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/8"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-05 18:37:27.000000000","tz":0},"subject":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support","message":"WIP: Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard support\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin, that is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are nessecary to fully implement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/901695\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bae059636c06e37adaac8f7375f1f8e5ee45c45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bae059636c06e37adaac8f7375f1f8e5ee45c45"}]},"branch":"refs/heads/master"},"54a1af986fbc99dae7b029b77bda2e753ddfc397":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2024-02-05 18:38:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/9"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-05 18:38:20.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin,\nthat is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are necessary to fully\nimplement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/901695\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54a1af986fbc99dae7b029b77bda2e753ddfc397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54a1af986fbc99dae7b029b77bda2e753ddfc397"}]},"branch":"refs/heads/master","description":"Edit commit message"},"08b1b3d3b9bdd9a31d6a86e0cc47f9ff1e99e2e0":{"kind":"REWORK","_number":10,"created":"2024-02-06 16:47:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/10"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-06 16:47:07.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin,\nthat is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are necessary to fully\nimplement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/901695\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08b1b3d3b9bdd9a31d6a86e0cc47f9ff1e99e2e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08b1b3d3b9bdd9a31d6a86e0cc47f9ff1e99e2e0"}]},"branch":"refs/heads/master"},"96ba24eb47833563480ac721b10450ebf94d59c5":{"kind":"REWORK","_number":11,"created":"2024-02-08 10:42:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/11"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-08 10:42:07.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin,\nthat is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are necessary to fully\nimplement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/901695\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96ba24eb47833563480ac721b10450ebf94d59c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96ba24eb47833563480ac721b10450ebf94d59c5"}]},"branch":"refs/heads/master"},"553c0d77c0e40665fabe8bf009ffc064afa44722":{"kind":"REWORK","_number":12,"created":"2024-02-08 12:21:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/12"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-08 12:21:02.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin,\nthat is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are necessary to fully\nimplement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/901695\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/553c0d77c0e40665fabe8bf009ffc064afa44722"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/553c0d77c0e40665fabe8bf009ffc064afa44722"}]},"branch":"refs/heads/master"},"df5d2ec6980191cc2098b6d8c1a471e106340ced":{"kind":"REWORK","_number":13,"created":"2024-02-08 15:03:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/13"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-08 15:03:41.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin,\nthat is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are necessary to fully\nimplement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/901695\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df5d2ec6980191cc2098b6d8c1a471e106340ced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df5d2ec6980191cc2098b6d8c1a471e106340ced"}]},"branch":"refs/heads/master"},"df829a0ccd2d1e15e6abc359da753f7cdbcfee75":{"kind":"REWORK","_number":14,"created":"2024-02-08 15:53:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/14"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-08 15:53:54.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin,\nthat is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are necessary to fully\nimplement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/901695\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df829a0ccd2d1e15e6abc359da753f7cdbcfee75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df829a0ccd2d1e15e6abc359da753f7cdbcfee75"}]},"branch":"refs/heads/master"},"90344b0dbd2f9903624c8644802c92dffadceaf3":{"kind":"REWORK","_number":15,"created":"2024-02-09 15:38:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/15"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-09 15:38:43.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin,\nthat is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are necessary to fully\nimplement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/901695\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90344b0dbd2f9903624c8644802c92dffadceaf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90344b0dbd2f9903624c8644802c92dffadceaf3"}]},"branch":"refs/heads/master"},"365fd7af45d99cc8c05c097bc7bfbb23bd44841a":{"kind":"REWORK","_number":16,"created":"2024-02-09 15:39:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/16"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-09 15:39:42.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin,\nthat is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are necessary to fully\nimplement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/901695\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/365fd7af45d99cc8c05c097bc7bfbb23bd44841a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/365fd7af45d99cc8c05c097bc7bfbb23bd44841a"}]},"branch":"refs/heads/master"},"2031836591344b5d0a0aa7e279baa20ba3480179":{"kind":"REWORK","_number":17,"created":"2024-02-09 15:42:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/17"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-09 15:42:14.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin,\nthat is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are necessary to fully\nimplement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/901695\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2031836591344b5d0a0aa7e279baa20ba3480179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2031836591344b5d0a0aa7e279baa20ba3480179"}]},"branch":"refs/heads/master"},"f9cc2590f03651983e1f43f3d7f31f16fdf0249d":{"kind":"REWORK","_number":18,"created":"2024-02-09 15:51:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/18"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-09 15:51:03.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin,\nthat is also required for certain scenarios with ovn-bgp-agent.\n\nThis change is part of a set of changes in several projects that are necessary to fully\nimplement ovn-bgp-agent.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/901695\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9cc2590f03651983e1f43f3d7f31f16fdf0249d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9cc2590f03651983e1f43f3d7f31f16fdf0249d"}]},"branch":"refs/heads/master"},"27cfeaf514ea0f22b3c27d5abcba229ece8f1cfd":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2024-02-09 15:58:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/19"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-09 15:58:03.000000000","tz":60},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27cfeaf514ea0f22b3c27d5abcba229ece8f1cfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27cfeaf514ea0f22b3c27d5abcba229ece8f1cfd"}]},"branch":"refs/heads/master"},"0e211037897920e9b3a3632d3d47c165564efbac":{"kind":"REWORK","_number":20,"created":"2024-02-09 16:09:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/20"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-09 16:09:49.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e211037897920e9b3a3632d3d47c165564efbac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e211037897920e9b3a3632d3d47c165564efbac"}]},"branch":"refs/heads/master"},"b2580d760074f83278374fadd0a88ca13a4c7c8d":{"kind":"REWORK","_number":21,"created":"2024-02-12 13:49:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/21"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-12 13:49:22.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2580d760074f83278374fadd0a88ca13a4c7c8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2580d760074f83278374fadd0a88ca13a4c7c8d"}]},"branch":"refs/heads/master"},"14feb010c403ac252c678486c376e6c20779ae17":{"kind":"REWORK","_number":22,"created":"2024-02-15 07:23:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/22"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-15 07:23:36.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14feb010c403ac252c678486c376e6c20779ae17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14feb010c403ac252c678486c376e6c20779ae17"}]},"branch":"refs/heads/master"},"817f812ba0d607def1591ff55ff442399c248654":{"kind":"REWORK","_number":23,"created":"2024-02-15 07:24:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/23"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-15 07:24:23.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/817f812ba0d607def1591ff55ff442399c248654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/817f812ba0d607def1591ff55ff442399c248654"}]},"branch":"refs/heads/master"},"fa83fb83ef570f692bf3077b65ce7325748a4c09":{"kind":"REWORK","_number":24,"created":"2024-02-15 07:28:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/24"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-15 07:27:57.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa83fb83ef570f692bf3077b65ce7325748a4c09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa83fb83ef570f692bf3077b65ce7325748a4c09"}]},"branch":"refs/heads/master"},"047537c65d4bbc7509d8e14860f299e3b03fe305":{"kind":"REWORK","_number":25,"created":"2024-02-15 10:41:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/25"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-15 10:41:58.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/047537c65d4bbc7509d8e14860f299e3b03fe305"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/047537c65d4bbc7509d8e14860f299e3b03fe305"}]},"branch":"refs/heads/master"},"8adae9a2f8823b70fe0dd08333d7e4461aab5d6c":{"kind":"REWORK","_number":26,"created":"2024-02-15 10:43:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/26"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-15 10:43:20.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8adae9a2f8823b70fe0dd08333d7e4461aab5d6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8adae9a2f8823b70fe0dd08333d7e4461aab5d6c"}]},"branch":"refs/heads/master"},"4edbdbcc697f6e9d98c4938c5bc11801150ce52a":{"kind":"REWORK","_number":27,"created":"2024-02-15 11:31:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/27"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2023-08-16 17:53:44.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-15 11:31:45.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4edbdbcc697f6e9d98c4938c5bc11801150ce52a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4edbdbcc697f6e9d98c4938c5bc11801150ce52a"}]},"branch":"refs/heads/master"},"da85e194c0477dbf2bfe3c1d9e3d119a9be6fc33":{"kind":"REWORK","_number":28,"created":"2024-02-16 10:54:09.000000000","uploader":{"_account_id":35638,"deleted":true},"ref":"refs/changes/17/891617/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/28"}}},"commit":{"parents":[{"commit":"5f604316bf44d2a36c0724d57458bcab39ddb8b1","subject":"docs: update monthly stable releases loop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f604316bf44d2a36c0724d57458bcab39ddb8b1"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-02-16 10:53:33.000000000","tz":60},"committer":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-02-16 10:53:33.000000000","tz":60},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/da85e194c0477dbf2bfe3c1d9e3d119a9be6fc33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/da85e194c0477dbf2bfe3c1d9e3d119a9be6fc33"}]},"branch":"refs/heads/master"},"64913524961fc0a7109e314fc2c74c63ce81673c":{"kind":"REWORK","_number":29,"created":"2024-03-17 12:38:20.000000000","uploader":{"_account_id":35638,"deleted":true},"ref":"refs/changes/17/891617/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/29"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-03-17 12:37:30.000000000","tz":60},"committer":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-03-17 12:37:30.000000000","tz":60},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64913524961fc0a7109e314fc2c74c63ce81673c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64913524961fc0a7109e314fc2c74c63ce81673c"}]},"branch":"refs/heads/master"},"57e871e5c398169ccdecb207bede6e3b42a98cbc":{"kind":"REWORK","_number":30,"created":"2024-03-28 09:31:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/30"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-03-17 12:37:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-28 09:31:27.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57e871e5c398169ccdecb207bede6e3b42a98cbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57e871e5c398169ccdecb207bede6e3b42a98cbc"}]},"branch":"refs/heads/master"},"465c10c6d2a9682be4f48baa6054e5843c8675a0":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2024-04-15 07:00:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/31"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-03-17 12:37:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 07:00:49.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/465c10c6d2a9682be4f48baa6054e5843c8675a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/465c10c6d2a9682be4f48baa6054e5843c8675a0"}]},"branch":"refs/heads/master","description":"Rebase"},"86e5ab6d1dbf942d38a5a6c95bc0fdbd37c046a9":{"kind":"REWORK","_number":32,"created":"2024-04-15 08:32:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/32"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-03-17 12:37:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 08:32:19.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86e5ab6d1dbf942d38a5a6c95bc0fdbd37c046a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86e5ab6d1dbf942d38a5a6c95bc0fdbd37c046a9"}]},"branch":"refs/heads/master"},"4ad2190ad62f577ffd8347f8679bec8b00841602":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2024-04-16 18:40:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/33"}}},"commit":{"parents":[{"commit":"c9e46b16e8e4c8c5a2e0284c7fc42fd6d80be8a5","subject":"Merge \"[release] Use Caracal sources by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9e46b16e8e4c8c5a2e0284c7fc42fd6d80be8a5"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-03-17 12:37:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 18:40:05.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ad2190ad62f577ffd8347f8679bec8b00841602"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ad2190ad62f577ffd8347f8679bec8b00841602"}]},"branch":"refs/heads/master","description":"Rebase"},"5f222aed320958c2a690ccea01406d0ae691038e":{"kind":"REWORK","_number":34,"created":"2024-04-16 18:40:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/34"}}},"commit":{"parents":[{"commit":"c9e46b16e8e4c8c5a2e0284c7fc42fd6d80be8a5","subject":"Merge \"[release] Use Caracal sources by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9e46b16e8e4c8c5a2e0284c7fc42fd6d80be8a5"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-03-17 12:37:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 18:40:38.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f222aed320958c2a690ccea01406d0ae691038e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f222aed320958c2a690ccea01406d0ae691038e"}]},"branch":"refs/heads/master"},"331a452ef67aaa9ab2d81783afe68dbeef77dc10":{"kind":"REWORK","_number":35,"created":"2024-04-17 05:55:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/891617/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/35"}}},"commit":{"parents":[{"commit":"c9e46b16e8e4c8c5a2e0284c7fc42fd6d80be8a5","subject":"Merge \"[release] Use Caracal sources by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9e46b16e8e4c8c5a2e0284c7fc42fd6d80be8a5"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-03-17 12:37:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-17 05:55:44.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/331a452ef67aaa9ab2d81783afe68dbeef77dc10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/331a452ef67aaa9ab2d81783afe68dbeef77dc10"}]},"branch":"refs/heads/master"},"b314e354dba9d9875fba3679800330a80d440306":{"kind":"REWORK","_number":36,"created":"2024-04-26 11:17:18.000000000","uploader":{"_account_id":35638,"deleted":true},"ref":"refs/changes/17/891617/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/36"}}},"commit":{"parents":[{"commit":"e47e24c9463b6f3dce177f7436376c2b3413c4de","subject":"Drop Vitrage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e47e24c9463b6f3dce177f7436376c2b3413c4de"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-04-26 11:16:38.000000000","tz":120},"committer":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-04-26 11:16:38.000000000","tz":120},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b314e354dba9d9875fba3679800330a80d440306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b314e354dba9d9875fba3679800330a80d440306"}]},"branch":"refs/heads/master"},"68ef25d416b1e6327bccb588ac81daccb8fd2576":{"kind":"REWORK","_number":37,"created":"2024-04-29 11:07:59.000000000","uploader":{"_account_id":35638,"deleted":true},"ref":"refs/changes/17/891617/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/37"}}},"commit":{"parents":[{"commit":"e47e24c9463b6f3dce177f7436376c2b3413c4de","subject":"Drop Vitrage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e47e24c9463b6f3dce177f7436376c2b3413c4de"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-04-29 09:44:09.000000000","tz":120},"committer":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-04-29 11:07:36.000000000","tz":120},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers through BGP on OVN\nenvironments.\n\nThis change extends kolla to be also able to build images for ovn-bgp-agent and FRR, as this is\na requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the BGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68ef25d416b1e6327bccb588ac81daccb8fd2576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68ef25d416b1e6327bccb588ac81daccb8fd2576"}]},"branch":"refs/heads/master"},"a273e3805c412b59238d5be76b7166dc7352162f":{"kind":"REWORK","_number":38,"created":"2024-06-07 10:31:52.000000000","uploader":{"_account_id":35638,"deleted":true},"ref":"refs/changes/17/891617/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/38"}}},"commit":{"parents":[{"commit":"e47e24c9463b6f3dce177f7436376c2b3413c4de","subject":"Drop Vitrage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e47e24c9463b6f3dce177f7436376c2b3413c4de"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-06-07 10:29:47.000000000","tz":120},"committer":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-06-07 10:31:12.000000000","tz":120},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers\nthrough BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for\novn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the\nBGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a273e3805c412b59238d5be76b7166dc7352162f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a273e3805c412b59238d5be76b7166dc7352162f"}]},"branch":"refs/heads/master"},"6fa300f66797c253186ac310a8f7fe09f6d38c96":{"kind":"REWORK","_number":39,"created":"2024-06-19 10:43:10.000000000","uploader":{"_account_id":35638,"deleted":true},"ref":"refs/changes/17/891617/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/39"}}},"commit":{"parents":[{"commit":"e47e24c9463b6f3dce177f7436376c2b3413c4de","subject":"Drop Vitrage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e47e24c9463b6f3dce177f7436376c2b3413c4de"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-06-19 10:42:42.000000000","tz":120},"committer":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-06-19 10:42:42.000000000","tz":120},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers\nthrough BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for\novn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the\nBGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fa300f66797c253186ac310a8f7fe09f6d38c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fa300f66797c253186ac310a8f7fe09f6d38c96"}]},"branch":"refs/heads/master"},"9235bbc2da9c296efc520552a5f046705e292437":{"kind":"REWORK","_number":40,"created":"2024-07-12 13:18:30.000000000","uploader":{"_account_id":35638,"deleted":true},"ref":"refs/changes/17/891617/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/40"}}},"commit":{"parents":[{"commit":"e47e24c9463b6f3dce177f7436376c2b3413c4de","subject":"Drop Vitrage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e47e24c9463b6f3dce177f7436376c2b3413c4de"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-07-12 11:27:29.000000000","tz":120},"committer":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-07-12 11:31:21.000000000","tz":120},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers\nthrough BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for\novn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the\nBGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9235bbc2da9c296efc520552a5f046705e292437"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9235bbc2da9c296efc520552a5f046705e292437"}]},"branch":"refs/heads/master"},"1510d4e7778b758ab64b1769fa26a46ac9509648":{"kind":"REWORK","_number":41,"created":"2024-07-25 11:25:29.000000000","uploader":{"_account_id":35638,"deleted":true},"ref":"refs/changes/17/891617/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/41"}}},"commit":{"parents":[{"commit":"e47e24c9463b6f3dce177f7436376c2b3413c4de","subject":"Drop Vitrage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e47e24c9463b6f3dce177f7436376c2b3413c4de"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-07-25 11:23:21.000000000","tz":120},"committer":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-07-25 11:23:21.000000000","tz":120},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers\nthrough BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for\novn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the\nBGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1510d4e7778b758ab64b1769fa26a46ac9509648"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1510d4e7778b758ab64b1769fa26a46ac9509648"}]},"branch":"refs/heads/master"},"cd0fb2ab6ade9d715dc04932c66dd6b2c50cb3b0":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2024-08-27 11:42:49.000000000","uploader":{"_account_id":35638,"deleted":true},"real_uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/17/891617/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/42"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-07-25 11:23:21.000000000","tz":120},"committer":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-08-27 11:42:49.000000000","tz":0},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers\nthrough BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for\novn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the\nBGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd0fb2ab6ade9d715dc04932c66dd6b2c50cb3b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd0fb2ab6ade9d715dc04932c66dd6b2c50cb3b0"}]},"branch":"refs/heads/master","description":"Rebase"},"c585e554bec20fec705e101e3fb70fac5a379019":{"kind":"REWORK","_number":43,"created":"2024-08-28 11:51:39.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/17/891617/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/43"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-07-25 11:23:21.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-08-28 11:51:21.000000000","tz":120},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers\nthrough BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for\novn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the\nBGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c585e554bec20fec705e101e3fb70fac5a379019"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c585e554bec20fec705e101e3fb70fac5a379019"}]},"branch":"refs/heads/master"},"787e75d5185ec3b93703b077b2be256a7d609433":{"kind":"REWORK","_number":44,"created":"2024-09-03 08:55:21.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/17/891617/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/44"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-07-25 11:23:21.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-09-03 08:55:15.000000000","tz":120},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers\nthrough BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for\novn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the\nBGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/787e75d5185ec3b93703b077b2be256a7d609433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/787e75d5185ec3b93703b077b2be256a7d609433"}]},"branch":"refs/heads/master"},"8c73b8863759ddcd8b70ca874b9645d65ef434a4":{"kind":"REWORK","_number":45,"created":"2024-09-09 07:53:06.000000000","uploader":{"_account_id":35638,"deleted":true},"ref":"refs/changes/17/891617/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/891617/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/891617/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/891617/45"}}},"commit":{"parents":[{"commit":"e47e24c9463b6f3dce177f7436376c2b3413c4de","subject":"Drop Vitrage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e47e24c9463b6f3dce177f7436376c2b3413c4de"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-09-09 07:52:39.000000000","tz":120},"committer":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@epg.com","date":"2024-09-09 07:52:39.000000000","tz":120},"subject":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard","message":"Add ovn-bgp-agent / FRR / Horizon BGPVPN dashboard\n\novn-bgp-agent [1] is a new component that exposes VMs/containers\nthrough BGP on OVN environments.\n\nThis change extends kolla to be also able to build images for\novn-bgp-agent and FRR, as this is a requirement for ovn-bgp-agent.\nFurthermore, it also extends the horizon integration to enable the\nBGPVPN dashboard plugin.\n\n[1] https://opendev.org/openstack/ovn-bgp-agent\n\nChange-Id: I788fbb76af3cd2fc4bd1cafb6004ba09c2362db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c73b8863759ddcd8b70ca874b9645d65ef434a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c73b8863759ddcd8b70ca874b9645d65ef434a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
