)]}'
{"id":"openstack%2Fkolla~891655","triplet_id":"openstack%2Fkolla~master~Ib8cd2356c6e4cdf7b2108a7fdedcb5a4b02768ca","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-22 14:22:23.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-09-22 14:22:23.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-09-22 11:50:57.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-09-22 08:35:35.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}},"hashtags":[],"change_id":"Ib8cd2356c6e4cdf7b2108a7fdedcb5a4b02768ca","subject":"Drop openvswitch-netcontrold","status":"MERGED","created":"2023-08-17 07:08:55.000000000","updated":"2023-09-22 14:23:35.000000000","submitted":"2023-09-22 14:22:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891655","meta_rev_id":"8d8a59b52ad5ea34ec157137383b4a0ae26d9814","_number":891655,"virtual_id_number":891655,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-22 14:22:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-09-22 09:15:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-09-21 12:04:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-09-22 08:35:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2023-09-22 09:38:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-22 09:21:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2023-09-22 09:42:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-22 09:21:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-09-21 12:04:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-17 07:32:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-17 08:32:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-17 08:38:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-17 12:56:14.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2023-08-21 10:14:14.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2023-09-22 08:35:35.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-09-22 09:15:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-09-22 09:21:52.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-09-22 09:38:39.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-09-22 09:42:48.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"05b1fc453f55254924dcfc7ed313209611661f08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-17 07:08:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba25f9ed937c78e5497df992735cc5b65a81c944","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-17 07:32:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"454519f95563aac07065cdd249fb0dfe135d95df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 08:32:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cabf8db498145b7a88020ec8094a05f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f7538254cf4a4e28979ed742b7597f8d : SUCCESS in 1h 17m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6e30205142a64ef3a006a0e8deda879f : SUCCESS in 18m 32s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8aa066446ec841d6bd18bfceb358143f : SUCCESS in 1h 13m 48s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37853eba11a5464c8c3af3ee8037c04c : SUCCESS in 2m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9db2b5af2daf475ba912a3245858b1c9 : SUCCESS in 2m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a7c619f556a14b3ba312dda401912b8b : SUCCESS in 5m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/08ba4523759845249b1f76d06c12cdc7 : SUCCESS in 18m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5e336b5b998c4911b055f20d3170fb73 : SUCCESS in 1h 18m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c8fce281f3bdb32e2b4ca0971ac7740e4f1781f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 08:38:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36090cc220d84c3c9159a21d4a44bbb7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9606efc5182145a39c559dd4586f763d : SUCCESS in 20m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/86aae30c2c2241a3bd34e3fe16fc313d : SUCCESS in 1h 03m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fff02ad7cafc44628118c7fe5f16d205 : SUCCESS in 19m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/47e401751d9f414e84b31a01601b174d : SUCCESS in 1h 03m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/74b7a548e9114f82ac6b87b2807a6884 : SUCCESS in 1h 09m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a4af4cbfb84472d8868a46d3dbe2223 : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fd853f8000c459ab6919734aa19d1b0 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/598a9f5457f648abb2d663205119c931 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93f6859d1eb44d45ab522c6684061a00 : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7896963ff22540e6b91a90a9a4d5b752 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be64f2a8717a4a3a857458c2c341ee62 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65526fca45004cc298321362ae460d9d : SUCCESS in 3m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/136dfac95b4d4c5fa33297bfef256a20 : SUCCESS in 14m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/84808ba7df7c445e992a025df4cb9a8f : SUCCESS in 1h 03m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/101491b919bb447481fb446fdccbc0a1 : SUCCESS in 1h 04m 01s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ea8f16e1174b48a6afc3b107925dec64 : FAILURE in 7m 29s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/ea0d8d8551a04db99644f5f0016968b2 : SUCCESS in 1h 17m 45s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/afd3fe434c074a6c9355787e945c3efb : SUCCESS in 33m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d511cb68f7624e21af71194a78ca922d : SUCCESS in 1h 03m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ef6889b276354f03b968fe092bdbbb3b : SUCCESS in 1h 11m 24s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d958465c16974451a129e54839a4e24a : SUCCESS in 1h 09m 39s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ec7f4bfe7b0a4406b24421f13ba41330 : SUCCESS in 1h 29m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8d9125dd8e6a72b19bf6a29f93d31e169397d04","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-08-17 12:56:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f6290d13e297e71dbe164ee874bbade1df9b803","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-08-21 10:14:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cc8d7bf6dc9834a08d5556e09a66a90bb6b1870","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-20 13:27:49.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"2fcd23db429e7caf106eb9c885e8d0f8e78badc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-21 10:35:10.000000000","message":"Patch Set 2: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f65f3b95c313f168d1f3a9b2cff2dbbb104835d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-21 12:04:35.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":3},{"id":"09ee9fe0ac9f813b1f7558e87da0812e400adad4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-21 12:04:47.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d80d41aa4d93bfa8804b12b2e755e47a8d1cbb71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 14:00:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/593ca4a5fcc24ff2b8fcdaef8574d0c7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4ab36cbc15d846e09721cc4fa6c529d6 : SUCCESS in 21m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3274172aec0d436da6f1931d7c466418 : SUCCESS in 1h 00m 04s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6c0799e653df402788acb48db6f25a63 : SUCCESS in 19m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4047bd081fd1469ea57154ee174ca38d : SUCCESS in 57m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1088581188354445b3987813f86a3f51 : SUCCESS in 1h 24m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/444edf5f5929440483ba568db333daa5 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d865382c48d4f7aa629773aef729a36 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/431b4573b889460f9910b0f9516e1ff8 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26492f0ac9e04301b2eb47a318b9410c : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2fdb702260f24916968e2fb65ec472c8 : SUCCESS in 3m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a7f14f24dec449886b39b588f8fb67a : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0272d4c292e42478732de0bc4f17476 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ffdc07c9149476f966f0d5dc29d9654 : SUCCESS in 3m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/64c969c745ba41deaf40ea1ccaec74d1 : SUCCESS in 15m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b82450fb7b9f4abcbeeb928d57bc9060 : SUCCESS in 1h 07m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0cc1dbedb0744326b716dc1d1ee2e018 : SUCCESS in 1h 29m 53s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/5fdea59e8a1646a79adeeea4aca0c003 : SUCCESS in 1h 12m 41s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/87cdfc9258274479a610a3b6903a70f9 : SUCCESS in 1h 48m 55s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5c4bd4886e7d486f9cec06e21379ee0d : SUCCESS in 30m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/060334b389754c9990ab73297ce3ece2 : SUCCESS in 55m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/942d51063b07447c80e59ca59f385102 : SUCCESS in 1h 26m 53s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d7334ee872c14cfda5b321181eeb9e33 : SUCCESS in 1h 07m 55s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/491d26e3bc544f3cb0dc7490e94dffe3 : SUCCESS in 1h 20m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b1cc915df2569edb6e87cd86350369618d840b4f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-21 17:15:13.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b0f5f798d556607c6118da0490484c6611bc51ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 17:23:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b5e8c06fcc54a4a8f7003769522896a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1e219110cd3e47d68c56a4750edfe342 : SUCCESS in 1h 24m 59s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b65c0ad6026f4011bb0f8bfc87ea1c5d : SUCCESS in 1h 28m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4dc84863e8954e679446301af6d06bdd : TIMED_OUT in 3h 02m 21s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32c9dc7cb8ac412b85f0728959764335 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba73d9727be54dc4a3b2aad7dfcc92bc : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6ad7495b623c4863b1d0766330c486d2 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c97f6c22cb674c0d9165782ce8abe136 : SUCCESS in 7m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/460178ae2a39497781862de41cc7df19 : SUCCESS in 1h 19m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c66919e494e24301b6be62fe8ea88d65 : SUCCESS in 1h 25m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"45ffd75e86cec4ec302b523e3d6cd92348667a49","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-22 08:35:35.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"27ebcbf4967b2696abf709fd599c0eae1b28651b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-22 09:15:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d88dbd559eccd74b0985ef3b6c9c5277ad474830","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-09-22 09:21:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2241136315d9e2623af02f812e78d0f9d124401e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 09:22:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4282ea6860ec3f457ee0fd46d7b6eae9bd1dcab1","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-22 09:38:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7922760ca9ed5400df902e5ac5f282880ad78a46","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-09-22 09:42:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0bfbf02335a3fff670faa16922ff8caebdaa1cef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 11:00:02.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34e3fca53ac44650b477f96d2e319ba8\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5da1bc3d19884995917114352396063b : SUCCESS in 21m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7e469a0c01b549ab981bcbea2727600c : SUCCESS in 1h 11m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cd3e960851f742d88cc7fa735ac0dabe : SUCCESS in 1h 31m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a35f506d82fd4f2ca3b3dfc64d257adc : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37e088059613419cb4bda81a570621e8 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b70dff662a8496597fd65bac869145e : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e357f84a24e7407da6546e2e850e5de1 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d45a0362287442739fedaeb9d8f74eab : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3d63cdee59f40b49868755013508ccc : SUCCESS in 4m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1d51574052854cafaaadb164517bc6d5 : SUCCESS in 15m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/45a11cef45ea4ae08f471e6607a8c5b1 : FAILURE in 50m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4a1b1c2ae7864e09a3762751281d3c80 : SUCCESS in 1h 17m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bfc82c9c6bd043fda6b76cfcf6b6bd7f : SUCCESS in 18m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8278fb1333a44c52aaee972bd6461aec : SUCCESS in 59m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2f80f010a96b4a59b8bfdde3789ce03f : SUCCESS in 1h 20m 22s","accounts_in_message":[],"_revision_number":3},{"id":"3cd4ca9f3340ae077194f8bde32c6213444131bb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-22 11:03:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b642f8be5ff3815b7e28188744736ffcc5107a6b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-22 11:25:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5b92ce3469a473eac9e8d1b612abac6d27ca43d6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-22 11:50:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"41a2f3a15b9f5b8c9e4bb9c4075d9afb05987bb6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-22 11:52:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ea316927fb95cb81eb8fb0124021cf450bad8cbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 12:45:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9087297c3ad4408ae0b0cce616acbc4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/725f58dfbdc34bb6927d359182fbc40c : SUCCESS in 20m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6c6438ad842847eaa16bd3a2825a1bde : SUCCESS in 1h 12m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b830a143b2894a0e8dc96d5755f6b424 : SUCCESS in 20m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cc20d4b3461b4288b58c8994b774b5d1 : SUCCESS in 1h 06m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e4428d4b84cb438c89877d87fe116de5 : SUCCESS in 1h 13m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d69d6160bc2440ba999986f4f02097f0 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6be153296eb14aa3ba437cb9434631ff : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/451e57f6ae2e419fb602188fb1a39cde : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d589ab1c722240bbbbedd8ff2d80e1cf : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67a15f0aca9743029b9243efbbad62f2 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6222d455d924332ae21c6e6c05147b5 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4d8badc9d3f42559e43b2d27276d301 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d75445a9fa8b4c3285bddf8180000ae4 : SUCCESS in 3m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/07b2ff7bb37a4868bf07fd9a0e6d217d : SUCCESS in 19m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c787a0501ef04c3fae7bbbf4ec80ae24 : SUCCESS in 1h 10m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c2cda1f868414da4a86a745839aff2e7 : SUCCESS in 1h 24m 51s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ff9d0d52c8a14fbdafd0ba8cc96030a2 : SUCCESS in 59m 25s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/eb469912a58f4ceab266f7cc313e5809 : SUCCESS in 1h 34m 32s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2cdab4efd2ab492fa8d2d06309558e99 : SUCCESS in 21m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ee8cc2eace064265a5e2dc0e9733fd35 : SUCCESS in 53m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1aba9549f19f4904914565d59bbdeb2d : SUCCESS in 1h 07m 42s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/6e95264bcff94b02903b263ddd85107c : SUCCESS in 53m 31s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/d4b7b26cb0d743e1b872b15b42a1c47c : FAILURE in 47m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eabebb7f6ceef3346a6fe02f5f7fad8e9c6865c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 12:45:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5500a745d062590467b8ce61b455793cc7ec6ed3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 14:07:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd18b9d5f7aa4234855dbf35180a7669\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/66b7558dcaa94f64842f013aad723f3e : SUCCESS in 1h 28m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e9ec4617057148f092abff0b147114f0 : SUCCESS in 1h 36m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e4c1f4ac8f6742c9bd9a8611e4dd6216 : SUCCESS in 3h 01m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/16549743838146a2bebbe56471ff0742 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3ba7424619741f79071e6a4cdad217b : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/41ea2a1bc32043f09d6dcec216ded2db : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/21134b5caeb04c2dbaa3c503b91d256f : SUCCESS in 4m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/41bf841c38c64d2abd949741959c4f81 : SUCCESS in 1h 23m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/907014b3a260486b984791c5c3c2469e : SUCCESS in 1h 26m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5a47abb47f189289048ad095d97c1187682a92aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 14:22:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d30b6c73333244fa89ce27bd51a5953e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d481bdbefb30435ea900d14dd4ddcb49 : SUCCESS in 18m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/49c2355409434bac8c43c8263a66e4ec : SUCCESS in 23m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/10be829f53bd475cb4e50838fd60ca56 : SUCCESS in 1h 05m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1a053512d3d4414fa2ef60caf51c962b : SUCCESS in 1h 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/205ab87bb5ed4e3fa29b52d2aa4008dc : SUCCESS in 2m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4500f5e9af21440d9689abab878bcb4f : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bd0411040934c6392a7f268fe3ea749 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3284e4c1cb249a3b97761c3f94086dc : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f631a7e655ce494b98f6c693ceb47cc9 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f18c9937726d4fcb8b4e3a71b1d0e8aa : SUCCESS in 4m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/453ae6a5b5a74af380ff6d70eae295e9 : SUCCESS in 16m 49s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8af40c49f5e74ab78354e45f469c5a79 : SUCCESS in 1h 25m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/616f3452ecea47b5b3e4c0841c96e349 : SUCCESS in 1h 09m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dff103e1abcd44aa9ac8381da0865f30 : SUCCESS in 1h 34m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3d8f272f1b6a4ca59e58ec1d766ae880 : SUCCESS in 26m 18s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5a5799a2901d420abf8031414c4d2901 : SUCCESS in 35m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d6950f1dd9384be0a0fb2d3b40d6937d : SUCCESS in 56m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/539d4db9d44e4bc29881294c76f56f64 : SUCCESS in 1h 10m 30s","accounts_in_message":[],"_revision_number":3},{"id":"65ddc455da26066dff61532c2aef7c07999edb71","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 14:22:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8d8a59b52ad5ea34ec157137383b4a0ae26d9814","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 14:23:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa1e960e71a54209b881d2bd320488bb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdb9428f1f8541b48bdc8f6faa5611e3 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85bc956aeb4e469186263e03ba0e7855 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dbc9b1193e4361bbab3742970ea268b5a419c911","revisions":{"4e0493e1fc3f1c0a66c69951831907f2a1308035":{"kind":"REWORK","_number":1,"created":"2023-08-17 07:08:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/891655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/891655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/891655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/891655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/891655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/891655/1"}}},"commit":{"parents":[{"commit":"7f12d216dc4de2c8d32291c3d6223185ecf2b510","subject":"Merge \"Revert \"Disable build of Gnocchi containers\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f12d216dc4de2c8d32291c3d6223185ecf2b510"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-17 07:08:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-17 07:08:53.000000000","tz":120},"subject":"Drop openvswitch-netcontrold","message":"Drop openvswitch-netcontrold\n\nChange-Id: Ib8cd2356c6e4cdf7b2108a7fdedcb5a4b02768ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e0493e1fc3f1c0a66c69951831907f2a1308035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e0493e1fc3f1c0a66c69951831907f2a1308035"}]},"branch":"refs/heads/master"},"3d23d92bdc08e95114260c950fe6006f20077bc8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-21 10:35:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/891655/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/891655/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/891655/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/891655/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/891655/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/891655/2"}}},"commit":{"parents":[{"commit":"7f12d216dc4de2c8d32291c3d6223185ecf2b510","subject":"Merge \"Revert \"Disable build of Gnocchi containers\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f12d216dc4de2c8d32291c3d6223185ecf2b510"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-17 07:08:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-21 10:35:10.000000000","tz":0},"subject":"Drop openvswitch-netcontrold","message":"Drop openvswitch-netcontrold\n\nCloses-Bug: #2027668\n\nChange-Id: Ib8cd2356c6e4cdf7b2108a7fdedcb5a4b02768ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d23d92bdc08e95114260c950fe6006f20077bc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d23d92bdc08e95114260c950fe6006f20077bc8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"dbc9b1193e4361bbab3742970ea268b5a419c911":{"kind":"REWORK","_number":3,"created":"2023-09-21 12:04:35.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/55/891655/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/891655/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/891655/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/891655/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/891655/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/891655/3"}}},"commit":{"parents":[{"commit":"7f12d216dc4de2c8d32291c3d6223185ecf2b510","subject":"Merge \"Revert \"Disable build of Gnocchi containers\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f12d216dc4de2c8d32291c3d6223185ecf2b510"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-17 07:08:30.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-09-21 12:04:35.000000000","tz":0},"subject":"Drop openvswitch-netcontrold","message":"Drop openvswitch-netcontrold\n\nCloses-Bug: #2027668\n\nChange-Id: Ib8cd2356c6e4cdf7b2108a7fdedcb5a4b02768ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbc9b1193e4361bbab3742970ea268b5a419c911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbc9b1193e4361bbab3742970ea268b5a419c911"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
