)]}'
{"id":"openstack%2Fkolla~892058","triplet_id":"openstack%2Fkolla~master~Iab9a4e674e685356c2250b988b6e8e6608808405","project":"openstack/kolla","branch":"master","topic":"enable_platform","attention_set":{},"removed_from_attention_set":{"32523":{"account":{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},"last_update":"2023-10-09 22:15:42.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iab9a4e674e685356c2250b988b6e8e6608808405","subject":"Add Platform Variable When Building Images","status":"ABANDONED","created":"2023-08-19 22:03:55.000000000","updated":"2023-10-09 22:15:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7a391eb2f7c8a6ed2f6597a1d310934e5bbfea51","_number":892058,"virtual_id_number":892058,"owner":{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-08-19 23:30:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-08-21 10:09:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-19 23:20:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-19 23:30:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-21 10:09:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"d2f437e4325402f231b870ec50a24809d3a7a493","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},"date":"2023-08-19 22:03:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae4eea56386e3e0d6f613782315dff4cc9ef8438","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},"date":"2023-08-19 22:05:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ea454db15c8ce1955edcb9ef2199d549143ff9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-19 23:20:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/552e2dc0226345b9bdde12ef82996d6a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3fc571346de14a12b72bd622a1f8a04d : SUCCESS in 15m 19s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8f766619d81d40ecb46d91fec96679a1 : SUCCESS in 1h 00m 00s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/309b87bbb55f4e018f7bba46f09cec44 : SUCCESS in 1h 12m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/57fe413869a94cc7a467a1170c8dfe4c : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aff36f23a4f8484ba68558648cee0bec : SUCCESS in 2m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/afdf089601dd418187c74cfa9de51aba : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/671187721e8b4e319f495c37fd73fb01 : SUCCESS in 5m 14s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8067f076ee56485b967ab0c320cad52f : SUCCESS in 16m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fc9c27957a4741d190f5b5786910e6b2 : SUCCESS in 54m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7390000e7a3347584b8b7c287e436a0a0bfc7acb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-19 23:30:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/643297f3b0684a6d90d4923cd7926883\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/24e563a073084fa9936d307707a14e53 : SUCCESS in 17m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ed125c1d1d8b423ebb789464e932fb77 : SUCCESS in 56m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fbac3decb6584b58b67bc612ebc93e13 : SUCCESS in 12m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7239322d6fa243879b08c73059a0d262 : SUCCESS in 53m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c6e5998027c1417099daeaf65878056e : SUCCESS in 1h 02m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3912027ba15d4d7da6b3127da7492f04 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e13cfcdc18e940078cd870250bc69678 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0ceb788112e460c92beb56a2f98cb38 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f82973b28fa44da9bded60cc04f8111 : SUCCESS in 2m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62d8013b2b0e476c8be19801e4208891 : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef2121f7d1c34b1eb8935c3cc812ca6c : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfbd76f3279e451c918d818aaa980626 : SUCCESS in 3m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fdd9ac2e63bc47feb4ad6a7de75a7f85 : SUCCESS in 19m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/509b2f003eba494497035c852baef780 : SUCCESS in 57m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7861aacdd5fa4cf7aacc2e311769027f : SUCCESS in 1h 20m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d849cb07d8cd4675bd49af4e481fa880 : SUCCESS in 16m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4e4f2914dcf94010aa59f3c6013c360b : SUCCESS in 41m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1bcf790582304217a9b81a724deceabd : SUCCESS in 1h 03m 32s","accounts_in_message":[],"_revision_number":2},{"id":"157720a362451f72448d3967e7a5366499755d17","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-21 10:09:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7a391eb2f7c8a6ed2f6597a1d310934e5bbfea51","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},"date":"2023-10-09 22:15:42.000000000","message":"Abandoned\n\nDocker Python SDK doesn\u0027t use builds which cause multi platform build to not work.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"987eb7b97c8f61b20ca050da6156ae29c44ff182","revisions":{"7997af82cdbe6681673fe76d084bc3100a00b38d":{"kind":"REWORK","_number":1,"created":"2023-08-19 22:03:55.000000000","uploader":{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},"ref":"refs/changes/58/892058/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/892058/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/892058/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/892058/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/892058/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/892058/1"}}},"commit":{"parents":[{"commit":"f000360c2d699e5b2bd9bee00e78055deb1540b4","subject":"ubuntu: mark collectd and telegraf as buildable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f000360c2d699e5b2bd9bee00e78055deb1540b4"}]}],"author":{"name":"Gene Kuo","email":"igene@igene.tw","date":"2023-08-19 21:57:46.000000000","tz":540},"committer":{"name":"Gene Kuo","email":"igene@igene.tw","date":"2023-08-19 22:03:52.000000000","tz":540},"subject":"Add Platform Variable When Building Images","message":"Add Platform Variable When Building Images\n\nThis commit adds platform variable to image build task. This may be\nuseful in environment where we need to mix different processor\narchitecture with a single image repo. When deploying, docker will be\nable to pull the image with corresponding architecture.\n\n[1] https://docs.docker.com/build/building/multi-platform/\n\nChange-Id: Iab9a4e674e685356c2250b988b6e8e6608808405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7997af82cdbe6681673fe76d084bc3100a00b38d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7997af82cdbe6681673fe76d084bc3100a00b38d"}]},"branch":"refs/heads/master"},"987eb7b97c8f61b20ca050da6156ae29c44ff182":{"kind":"REWORK","_number":2,"created":"2023-08-19 22:05:23.000000000","uploader":{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},"ref":"refs/changes/58/892058/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/892058/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/892058/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/892058/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/892058/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/892058/2"}}},"commit":{"parents":[{"commit":"f000360c2d699e5b2bd9bee00e78055deb1540b4","subject":"ubuntu: mark collectd and telegraf as buildable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f000360c2d699e5b2bd9bee00e78055deb1540b4"}]}],"author":{"name":"Gene Kuo","email":"igene@igene.tw","date":"2023-08-19 21:57:46.000000000","tz":540},"committer":{"name":"Gene Kuo","email":"igene@igene.tw","date":"2023-08-19 22:05:15.000000000","tz":540},"subject":"Add Platform Variable When Building Images","message":"Add Platform Variable When Building Images\n\nThis commit adds platform variable to image build task. This may be\nuseful in environment where we need to mix different processor\narchitecture with a single image repo. When deploying, docker will be\nable to pull the image with corresponding architecture.\n\n[1] https://docs.docker.com/build/building/multi-platform/\n\nChange-Id: Iab9a4e674e685356c2250b988b6e8e6608808405\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/987eb7b97c8f61b20ca050da6156ae29c44ff182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/987eb7b97c8f61b20ca050da6156ae29c44ff182"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
