)]}'
{"id":"openstack%2Fkolla~892586","triplet_id":"openstack%2Fkolla~stable%2Fxena~I976781ed0bdc47760163e25c480ca8fe93f66517","project":"openstack/kolla","branch":"stable/xena","topic":"bug/2028048-stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-08-28 15:09:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I976781ed0bdc47760163e25c480ca8fe93f66517","subject":"Pin elasticsearch-curator\u003d\u003d5.8.4","status":"MERGED","created":"2023-08-24 10:46:18.000000000","updated":"2023-08-28 15:10:52.000000000","submitted":"2023-08-28 15:09:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892586-bug/2028048-stable/xena","cherry_pick_of_change":891721,"cherry_pick_of_patch_set":3,"meta_rev_id":"88a69c22333e10d81ec3a5eb5727862aa70cf0f9","_number":892586,"virtual_id_number":892586,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-28 15:09:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-08-24 10:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-24 11:15:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-08-28 13:38:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-08-28 13:38:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":33466,"name":"Lukasz Zalewski","email":"lzal3wski@gmail.com","username":"lzalewski"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-24 10:46:18.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":33466,"name":"Lukasz Zalewski","email":"lzal3wski@gmail.com","username":"lzalewski"},"state":"CC"},{"updated":"2023-08-24 10:48:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-08-24 11:15:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-24 12:06:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 13:38:24.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"5f7662e218312472fc5ac40e44f593746bdff8c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-08-24 10:46:18.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"d7ae603704d218774b3740eeccea40ac5d5c2cb1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-24 10:48:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b00a9593668e28ef8e274afa45f0ffe6fc1f2844","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-24 11:15:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"647dbed4b639d79627ce15762216625b0289fe91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 12:06:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6345e99d4a9d452fbcf2b32e49fabab2\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6833dbaffdca4264894f88045614c5b6 : FAILURE in 29m 47s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d4a94f4420564f7c94d3967c719daa16 : FAILURE in 33m 58s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/238230abb038455e80ed229f2141bf6e : SUCCESS in 1h 13m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4bbe14ce58ce43bcb5e3c704cda810ab : SUCCESS in 1h 15m 24s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a827cea6039a47b7b0c1352b5684eb17 : SUCCESS in 18m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/14509f7b800542378d2957f4770a8de3 : SUCCESS in 16m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/91c71ba0e3b04e0088270468d1a67c3a : SUCCESS in 1h 10m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef1a43b3c25143c3b4539ecec4a3d9b6 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fdebc870c1744b2937e3593bb2758a8 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89741fd2499f431aa0743d8c629b7d99 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa3604050d324e4bb9738174ef1bae5b : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05c53334fc114822a1840afe97e8c143 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fc3ea2c36514dc69f1917d5af95222c : SUCCESS in 3m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e66f5dff3e054c20aaffd36f68290b0a : SUCCESS in 21m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/16c5fe27515a4fd38e6b1e07135fd19a : SUCCESS in 24m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/532dec810f804d4eb21fd022933dd3e5 : SUCCESS in 59m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f17704fcdfef4c65b7045fd279c2b5ae : SUCCESS in 45m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a7223fdb8ff14de9ac03ae9a9fdf70cf : SUCCESS in 1h 01m 09s","accounts_in_message":[],"_revision_number":1},{"id":"eddae39d1ce14ffe202c51570c87898ee6cae352","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 13:01:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31a4e5e5f5be422b8f66d37e2593d523\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9022f29367024b548624a5c82cb053f6 : FAILURE in 26m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3dabb5404dc945678f29e0e35beee1f9 : SUCCESS in 20m 25s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/467ab3e5253c4b6394e4c0eff0285220 : SUCCESS in 1h 16m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/105d3bca71ac49f7a81aa2187a148f81 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6de526da36254d65ba7bab75f8cc8b5b : SUCCESS in 2m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/37371865e90f4c3aa4aa0da7bb5b8d87 : FAILURE in 2h 13m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37d524f8b12d33a2a70d419170d59fb9ccdf1e15","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-25 12:50:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a243db03938c4a2f137585d8dc79d65cbf7f90ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 14:55:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83ead66e42a8467b945180c73b6ea599\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6913bb22afe84f8b9d14ed5ac9ee61b8 : FAILURE in 2h 01m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8ef63f2dead94b9f8da494967ed8064a : SUCCESS in 1h 43m 40s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5acb164187b8483e98ee46ad5e35c80b : FAILURE in 48m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63287ae3a793444ba9c6dbd93416910a : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5890c88bc24e4e768968eb467bec07c0 : SUCCESS in 4m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e56f9848cc92429b85a121040047893c : FAILURE in 2h 02m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e9d514488fbe7a9070c29a89bb30b0f0548f9e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 15:53:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b8b36f6f06647bbb78e96cdbd291d23\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9b5415a75cbf47e2b05c142998211b21 : TIMED_OUT in 3h 01m 30s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f606013fe24a482ab0715b1dbbd3689a : SUCCESS in 32m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/457a638342ad48ccac83463f387743a0 : SUCCESS in 1h 16m 12s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7e0672f6c3e64413afe73e47e77a358d : SUCCESS in 1h 16m 57s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f5c9746cc7514d8c8655b5fb53b5a446 : SUCCESS in 21m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d9fa1dd97dc94f70a86c80d8154dbe3f : SUCCESS in 18m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/de6c2913993f4ca397cdc01cca5b5ffc : SUCCESS in 59m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab41650caf6941ff8a6cced9993e7528 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47714514cdfd4299b02f0ec3eba74dff : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/873c150f31804b5d863668776320fcc0 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c45194f8de4e46c9b673ff464e714afa : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9560b648cdca442684def723358d724d : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/281ab3216a564a438c4f886a75d0b8c5 : SUCCESS in 3m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/43c95c31f0d94d91b6b439ff816853c1 : SUCCESS in 20m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1fdb5898ef744067b3f9bcc77b294dd5 : SUCCESS in 20m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/88b8051c5f5d465f8796b719d19808ae : SUCCESS in 53m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/47aff260d73f4663a660977ed6932f6e : SUCCESS in 51m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3e251404f97449a187cbeb200c2519de : SUCCESS in 1h 08m 03s","accounts_in_message":[],"_revision_number":1},{"id":"9f0e6bd27faa11f5fa0648c1d8f870af34a957c4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-28 13:38:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0446da560c7a4577fc45b59066292c1ee73124ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 13:38:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c72c3ad6d1628be1059ea9ad1ebac14109847618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 15:09:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95eed53716394972b8d728d8a745a902\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f8fd98fd965546c4b9e0e91352a3deac : SUCCESS in 36m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/055bd2c339a2499fa8d55516f2ee44e6 : SUCCESS in 1h 24m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd53645407204e7e9aad5d83e770e365 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0328ff9eb6cd43a1a89a1e9f2206d9e5 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e94986fb0aba4635b92ce3ed84af759e : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3dd943107d64d148bfd8fdae6538b5e : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b875013dd8d14d639d637b69251d0987 : SUCCESS in 3m 32s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f46c47d7269485fbde4c58b7389c43c : SUCCESS in 19m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fa94e59f2907458ba3efbe66af64ad35 : SUCCESS in 56m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/380f2ac9f6104edda8727660be1b2846 : SUCCESS in 1h 01m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d6fa6b00dc81f0e98d76f0886bbdf24a5c616bcc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 15:09:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"88a69c22333e10d81ec3a5eb5727862aa70cf0f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 15:10:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7501c666ba0446338c1e89cffe9aff7f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc4db39913bc41149b883532634af7f4 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e9d5ffb28ba4b08af265a3b33baf2da : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b156ad0de5affeadeac7e0dffb5d43cf799b217","revisions":{"6b156ad0de5affeadeac7e0dffb5d43cf799b217":{"kind":"REWORK","_number":1,"created":"2023-08-24 10:46:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/892586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/892586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/892586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/892586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/892586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/892586/1"}}},"commit":{"parents":[{"commit":"dc38e719758f607c4cb2c9ba7d242c73e55e6238","subject":"rabbitmq: use erlang-25 ppa on Debian/Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc38e719758f607c4cb2c9ba7d242c73e55e6238"}]}],"author":{"name":"Lukasz Zalewski","email":"lzal3wski@gmail.com","date":"2023-08-17 13:09:44.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-24 10:46:18.000000000","tz":0},"subject":"Pin elasticsearch-curator\u003d\u003d5.8.4","message":"Pin elasticsearch-curator\u003d\u003d5.8.4\n\nelasticsearch-curator 8 is out and is incompatible with Elasticsearch 7 release.\nWe pin curator version to release 5.8.4 which should support any Elasticsearch\n7.x releases [1].\nAs elasticsearch has been dropped from Zed onwards, this only applies to Yoga\nand below.\n\n1. https://www.elastic.co/guide/en/elasticsearch/client/curator/5.8/version-compatibility.html\n\nCloses-Bug: #2028048\nChange-Id: I976781ed0bdc47760163e25c480ca8fe93f66517\n(cherry picked from commit aadc290613217bacc14f81b17a7d26d2d3d716d2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b156ad0de5affeadeac7e0dffb5d43cf799b217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b156ad0de5affeadeac7e0dffb5d43cf799b217"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
