)]}'
{"id":"openstack%2Fkolla~892876","triplet_id":"openstack%2Fkolla~stable%2F2023.1~I67f5f12700d7b55f26bff81e9b54559303da6d83","project":"openstack/kolla","branch":"stable/2023.1","topic":"bugfix/kolla_container_restarts-stable/2023.1","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-09-06 08:19:09.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-06 07:00:58.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I67f5f12700d7b55f26bff81e9b54559303da6d83","subject":"Add comprehensive checks for container restarts","status":"MERGED","created":"2023-08-30 07:13:49.000000000","updated":"2023-09-06 08:20:29.000000000","submitted":"2023-09-06 08:19:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892876-bugfix/kolla_container_restarts-stable/2023.1","cherry_pick_of_change":873972,"cherry_pick_of_patch_set":7,"meta_rev_id":"528646aa81fa99b9c442c34724c0f645b7468e9c","_number":892876,"virtual_id_number":892876,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-06 08:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-08-30 08:24:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-30 09:52:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-08-30 08:39:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-08-30 08:39:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-30 07:13:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"state":"CC"},{"updated":"2023-08-30 08:33:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-30 08:39:41.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-08-30 08:47:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 09:52:58.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-09-06 06:33:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"980e6eed77f19ac236f2af2cc054c882c3823772","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-30 07:13:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"87ce2fdacce85e4fa62f1784cd4817c8debdde62","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-30 08:24:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f982ba664127ac0e62922f2bbfa1a4efb4570a72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 08:33:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/339bcf5356b8484aa789e92330c79a45\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1133f486da0c4c0fb6c88ae0fa5523ee : SUCCESS in 23m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/665ada6b834944da9fc0b3135c6d423e : SUCCESS in 19m 26s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/05d884ece6a9457ab8e0273caef24da2 : SUCCESS in 1h 17m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/59457d3d05dc42bda7a0fca20a4b4988 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/13834df32edb46e5b848e3f6a97e5b05 : SUCCESS in 5m 18s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/97eba781260541df878a2a159c9dc082 : SUCCESS in 21m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fa2d9d90e9284fb49481a31ec24ecc84 : SUCCESS in 59m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b6c560bb89ad4213a56ef53e99c9adeec150c1d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-30 08:39:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c307475861babecf46afbd45ae2671c934f2014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 08:47:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3b438b4697b4195b3f574e5bac2be56\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/262eb95f719c4c8f96400ba5ccd91354 : SUCCESS in 17m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/13d7603a79684913969548658964b027 : SUCCESS in 49m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/61f21c72ca34486db936d36ed721c5cd : SUCCESS in 17m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/891648ca79d24e4cb7716cf0f7e0f6fe : SUCCESS in 41m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b51ef4fd3ef8403cb3671155d6b629cc : SUCCESS in 1h 25m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13bd650931bc4325b08cef2e2a3555b7 : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdc78bb1c6b94a6987556a52643ff049 : SUCCESS in 2m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c02514bb569494d8da03a7c5976b5a1 : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43306b54901d4462b8b9acf164a51a30 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f5e2b70ceb54224ae8f7093a6dd4807 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a34fb447c5644fdbd5f831cd8466888 : SUCCESS in 3m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/17d7d4d14bb14d59a710892ab15b6474 : SUCCESS in 16m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e5b224b8c4fe438ab6cb78a1fc039737 : SUCCESS in 57m 37s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d5e3cb48bd624ce8b6f0a11bacbdefc5 : FAILURE in 1h 17m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7a7ca7dc827844ec9b024d7f16c4c7c7 : SUCCESS in 20m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/709cd71d989e45f7b2285f82b04254dd : SUCCESS in 48m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a97da966878e423ea7b06ca1695629bc : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":1},{"id":"32bd75b13b22b34dac5a42df38526b492a899be5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-30 09:52:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aead203d8a6a65ba319178720275ee84243b9d36","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-06 05:32:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3fe4acb9076ad124fde5dff06476569bf7bf2a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 06:22:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/693619fdc1024b7dbda2f8ecf5862730\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6c1dd189c1134cc49c71cc0c00dba677 : SUCCESS in 20m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bdd05aae9a1e4681a4996bbcf91399be : SUCCESS in 20m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5d0308db8d244849d8fdeade67f109f : SUCCESS in 2m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fda7eeb391db497ab5b98fc8f68bc5e0 : SUCCESS in 2m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/401fb4c769734e92865821710ea1ffc8 : SUCCESS in 47m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9b1c9ea92a7c4408b3ccc06ac63e776b : SUCCESS in 17m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60e2e183342fc0d31213874c5ef3a0421b0fc4a2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-06 06:33:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"72c39ada053ba370c84e59377e2707bb2c1a0271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 06:59:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5e995250779426bbb8cec6693fb1e7b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d5ef04ca583b45bfb285f5986157d10c : SUCCESS in 22m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/995e5d69a1304de3872aabb187cd0386 : SUCCESS in 49m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dcfc6a1d1357450497f6afbcf25a08cc : SUCCESS in 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd0d107bc1c6479ba5768f69617544c4 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7422111733c547cab099701d906b2ee7 : SUCCESS in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb13698df99f4b1e811b5cc4e0977c00 : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1156a21141d34c28b8ac0cd296c36928 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9da4596400c14b9580cfb6c0d7f465e6 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b696160955a1489ba407bf6eff127ff5 : SUCCESS in 3m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/052cde74ea2046c99cce10c2eae4c73e : SUCCESS in 21m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c7fec702e0ee4da891af1c9caa2b147a : SUCCESS in 1h 06m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0852e65078fe4bb7ae7e2e6a50b4ce90 : SUCCESS in 1h 20m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ec8eeec5718149b0b39020171b078513 : SUCCESS in 18m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1dc30abc159a4171bfb2faa4d5cf18dc : SUCCESS in 44m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cf193332a40a436ea171e6c3a60e1b28 : SUCCESS in 50m 12s","accounts_in_message":[],"_revision_number":1},{"id":"86c48a5d39937d627e82b6b0ee3962308176fc79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 06:59:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7c052df350f1f8bfcd6aae30299f38ed5733614","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-06 07:00:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48c92105b387efc76f1a43825591fc58db6bebf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 08:19:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d14995fb4304fdeaddb799a3fe3d8d7\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5666c2ed9a76412ca235ffb4f7c22baa : SUCCESS in 22m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a793bcbaab294ddc9d53b10cd88abf67 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acbb37f36ed44445b5eb8d7bb10b45b5 : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14a0a418cdca4578b2a84b90e7fe5a75 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a96edb52791d4b79b1cc96cbb97815f0 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5b9611ecdb94bb4b217cfc34621122d : SUCCESS in 3m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3388aaef9b454255ba1dba602e93e118 : SUCCESS in 21m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7bfd0b52f1c841059635a069e3a2d6ba : SUCCESS in 1h 05m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cb202d865a3d45a98025324bb90ff8d1 : SUCCESS in 1h 14m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c62329946f7449dfba4292ca2b66a84e : SUCCESS in 18m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a1818066452c40208a95c8c48b34d1d1 : SUCCESS in 1h 00m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3f913db1339e4966a7ce05b1e847792f : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1b25e337a16cd19757e0eeae640bada3c7e20398","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 08:19:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"528646aa81fa99b9c442c34724c0f645b7468e9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 08:20:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f296e555611b4f108fd280ac14be3a8a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4788b38f18b94858aa99841677dffd45 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75005db0549c4fdd903f30b053ad9934 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"288a400dbf30797b70e64af64f6deadcca582851","revisions":{"288a400dbf30797b70e64af64f6deadcca582851":{"kind":"REWORK","_number":1,"created":"2023-08-30 07:13:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/76/892876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/892876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/892876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/892876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/892876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/892876/1"}}},"commit":{"parents":[{"commit":"cad045b264ac0c041f38538047d026185c321636","subject":"rabbitmq: Fix repo for ubuntu aarch64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cad045b264ac0c041f38538047d026185c321636"}]}],"author":{"name":"Dawud M","email":"dawud@stackhpc.com","date":"2023-02-15 18:02:05.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-30 07:13:49.000000000","tz":0},"subject":"Add comprehensive checks for container restarts","message":"Add comprehensive checks for container restarts\n\nWhen adding a dashboard to grafana the containers aren\u0027t restarted when\nthey should be. This is due to a bug in Kolla where the logic to\ndetermine whether or the container needs to be restarted fails in the\ncase where the file does not exist in the container. This patch adds\nmore comprehensive checks for container restarts in the set_configs.py\nfile. This patch also adds a test to ensure that the functions work as\nexpected.\n\nCloses-Bug: #1997984\nCo-Authored-By: Will Szumski \u003cwill@stackhpc.com\u003e\nChange-Id: I67f5f12700d7b55f26bff81e9b54559303da6d83\n(cherry picked from commit d9a6c5f3901e078a30d341397ac57d867d8191cc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/288a400dbf30797b70e64af64f6deadcca582851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/288a400dbf30797b70e64af64f6deadcca582851"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
