)]}'
{"id":"openstack%2Fkolla~892878","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I67f5f12700d7b55f26bff81e9b54559303da6d83","project":"openstack/kolla","branch":"stable/yoga","topic":"bugfix/kolla_container_restarts-stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-08 09:05:41.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-09-08 10:35:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I67f5f12700d7b55f26bff81e9b54559303da6d83","subject":"Add comprehensive checks for container restarts","status":"MERGED","created":"2023-08-30 08:21:26.000000000","updated":"2023-09-08 10:36:22.000000000","submitted":"2023-09-08 10:35:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892878-bugfix/kolla_container_restarts-stable/yoga","cherry_pick_of_change":873972,"cherry_pick_of_patch_set":7,"meta_rev_id":"0d8cec1fadd8cfe540dba46f95b557e0457fc664","_number":892878,"virtual_id_number":892878,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-08 10:35:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-09-08 09:05:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-09-06 05:33:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-09-06 07:23:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-06 05:33:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-09-08 09:05:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-30 08:21:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"state":"CC"},{"updated":"2023-08-30 08:39:24.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-08-30 09:31:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-30 09:45:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 09:53:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-09-06 07:23:28.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"58ceeed7f04b0f572e646e85ce7569895df06f76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-30 08:21:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e98e3304e7c97123e5192ea3e8fedaf3807f42b8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-30 08:24:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4de595bf3700e83cdff45b728451daa65f1064a4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-30 08:39:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4caed300c3903b5aa32e47e3e17960ea7fd0df47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 09:31:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/906bf0e23c2b43a88423e36729ce1a80\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d9731d93cd6414e9d795f1582bd51f5 : SUCCESS in 27m 35s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f8e2853cf7b84780922fa0b394f22258 : SUCCESS in 19m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7d7ddf5589c741a7bdd81dd780dfc9f6 : SUCCESS in 20m 42s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f7fb0f9a64a446699bc135099419dea4 : SUCCESS in 1h 07m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26ea87077d704e1ca3f3c72cf77e7e52 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e7bab7e4a5b420bbd5d5ed8ac0b5b89 : SUCCESS in 3m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a9ace74765764b6d975d628a4fda7791 : FAILURE in 59m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b405017a2c4385c4e59aac6b5b453da09c266105","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 09:45:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2241ae13b5534dce86bcba27e58f50ba\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1fade486eaad4a428f14cf0af568975e : SUCCESS in 34m 46s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/145f9653e9414d7298ecba640149672e : SUCCESS in 37m 56s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/fc088931325e4046b110e860c5f52372 : SUCCESS in 24m 09s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/faac9c7544df4c68b390172b0092d2d3 : FAILURE in 58m 54s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/852a369fb8fb433eb487a0469a51ec60 : FAILURE in 57m 58s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/337df77abdb44c4e9cc657f57f150784 : SUCCESS in 52m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/145740ae14ae4956bb21bdcca7ae8a8e : FAILURE in 1h 03m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/eaadeaf4222c4b1db84701cef4ca0367 : SUCCESS in 21m 03s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d0d95ed765634cb5b60b159f5ebdbcfe : SUCCESS in 16m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f7822b2b3abf4cd0b82a2ad93951bf8d : SUCCESS in 51m 32s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d121b2630029428687813cc70d2e2e8e : SUCCESS in 1h 00m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f70ce32aa0947c78344ac5bcb447ad5 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73fe020865c04b2f9374a037c808ab5b : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afda16443be74bb283f4866c4a9e503b : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8703b2c5f70a4cbe9a896388b638fd44 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/befe4b0194564d04bf71f9884d240190 : SUCCESS in 4m 06s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/7354a5d54b3f4cdbbe998255ae344ed3 : SUCCESS in 26m 21s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/1dedbc6d94104212964f94bdc390d433 : SUCCESS in 1h 17m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dada65a2578c4650b8d62ace638877a1 : SUCCESS in 24m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b3a7ed9a19d441c49ff2913995db0e5b : SUCCESS in 26m 39s","accounts_in_message":[],"_revision_number":1},{"id":"818771d65dbae68a64c95bc272bc01d86cd8478b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-30 09:53:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1d17f7e7dc866b7a6d475d6ef13fb90e1933f3a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-06 05:33:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8e40e6f981bd0da1e4e0c59cd4516962ad298512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 07:12:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a88c6e6ee58438195048ade034479d8\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8919748f160a4828a952121da2acaa9b : SUCCESS in 35m 05s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2fb0f10eb5864e7299712644bde79aea : SUCCESS in 38m 15s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/47b66f448f9649838653f23616db961f : SUCCESS in 26m 02s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/82936041072e4cfdac68fb21c7541a44 : SUCCESS in 58m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a82ca32980ba4b6790b08ce14c3a373a : FAILURE in 53m 29s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/0f25fbe5fd9f47829585c2ad6679a09a : SUCCESS in 1h 04m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/296cbe99699749bdb79c28d57ed777c6 : SUCCESS in 1h 32m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/572746db84354082aca592592654180f : SUCCESS in 26m 11s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dd89a44c8e2245bda7217abf9809e46a : SUCCESS in 15m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d89bfe47548e413b9669da877224c3a8 : SUCCESS in 57m 59s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0590a36460b34ea3b0388060911fc9ac : SUCCESS in 1h 01m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4cdaf2b7e9944dd96acfe998e2c2c40 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/344f585b50c94576a3070d3f69bde702 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37cf37124140420484687a9e5186b79c : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0ecb08a5bbf4761a3a427546b3f9268 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0d5833a79924be7aac81cb8feba2309 : SUCCESS in 4m 00s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/def192cebb8c48a88770ad9b12e49901 : SUCCESS in 25m 36s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/68a6515f7c164780860cd3ac4bd4e5c2 : SUCCESS in 51m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/93545738118a4942b93dc1e09f1f7b29 : SUCCESS in 17m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b35455634249414ab7b11551d50de87e : SUCCESS in 24m 04s","accounts_in_message":[],"_revision_number":2},{"id":"0f27283bf36cae5dde37655a3dad602cd5c351b0","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-09-06 07:23:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"016d7d8cddc282cc421a800452fd0333d59b8259","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 07:46:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d5eaa04c3684a17802a33deb7340368\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2dd36829d7764506941cb5e18e23dd79 : SUCCESS in 1h 16m 33s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/24a143ef2de144cca6e6c0f6e30b617d : SUCCESS in 19m 50s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/888ff64a10ac4467b1cea4aed0f90840 : SUCCESS in 26m 44s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/57f6f195e96940298e5581567c1f1cb8 : SUCCESS in 2h 11m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a16c5165b4354a69a9741006f4a4e5d2 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8cd3ce75487b40e1abc49627d16a9ccf : SUCCESS in 5m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b96217c9a6024fa88a01f926e62e023b : FAILURE in 21m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6e6f6f190a0242081054857d339186783ac8e9fc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-08 09:05:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"566ab25cb2ddc1a5d3581b81e9a9e3d3b16c5cd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 09:06:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1273036a58ab1cc396c216e61bb3594a3919705e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 10:35:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a54d04355234c198f6ffe578928c2ff\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/31b81149a77d41158c57bffeda04488b : SUCCESS in 37m 59s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f8eb7989c9d5425fa5220461ff977d76 : SUCCESS in 1h 14m 43s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e2d224bf6fa247c4bf0916268d9e23fa : SUCCESS in 1h 28m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/52983aa8468c4e6eb63f216c9d6aa90d : SUCCESS in 23m 20s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/83f5343368df47c88f62e371103568da : SUCCESS in 1h 11m 48s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/25d351374b4d42f7a623fed1ba2d42f9 : SUCCESS in 59m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7446f0bc27174baa9e438e7f2c025529 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e1fc7df3f6a4dbab6d98fc044ec2f67 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e38963d501af479396a1c2ecc76866bd : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d3a8417f48242c6b526e4f9153789ca : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a18adcf390844e592afb02e07d11f8f : SUCCESS in 3m 29s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/3a4438c6140641a9a677cde1cfcb1df8 : SUCCESS in 1h 04m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ecac3f03c2f484fb3d788d721f3a862 : SUCCESS in 23m 56s","accounts_in_message":[],"_revision_number":2},{"id":"31d5a915391ffd656828eaedad6defb56eca6b4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 10:35:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0d8cec1fadd8cfe540dba46f95b557e0457fc664","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 10:36:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d7b0d9eb749492bb55b9c3186c18cc8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/288f3e7c53a4497191f6d66cf136cff5 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/305fa95ff89f4a6c96996170104fb632 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0353c51c68c30632ecb29fd0ae4372a375086865","revisions":{"1ac73afe360b9941bd5b99ba94b7a86b45309a7a":{"kind":"REWORK","_number":1,"created":"2023-08-30 08:21:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/78/892878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/892878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/892878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/892878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/892878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/892878/1"}}},"commit":{"parents":[{"commit":"f1d1881e9532b4b342ef2a4175e8248181d4eaf1","subject":"Merge \"Pin elasticsearch-curator\u003d\u003d5.8.4\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1d1881e9532b4b342ef2a4175e8248181d4eaf1"}]}],"author":{"name":"Dawud M","email":"dawud@stackhpc.com","date":"2023-02-15 18:02:05.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-30 08:21:26.000000000","tz":0},"subject":"Add comprehensive checks for container restarts","message":"Add comprehensive checks for container restarts\n\nWhen adding a dashboard to grafana the containers aren\u0027t restarted when\nthey should be. This is due to a bug in Kolla where the logic to\ndetermine whether or the container needs to be restarted fails in the\ncase where the file does not exist in the container. This patch adds\nmore comprehensive checks for container restarts in the set_configs.py\nfile. This patch also adds a test to ensure that the functions work as\nexpected.\n\nCloses-Bug: #1997984\nCo-Authored-By: Will Szumski \u003cwill@stackhpc.com\u003e\nChange-Id: I67f5f12700d7b55f26bff81e9b54559303da6d83\n(cherry picked from commit d9a6c5f3901e078a30d341397ac57d867d8191cc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ac73afe360b9941bd5b99ba94b7a86b45309a7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ac73afe360b9941bd5b99ba94b7a86b45309a7a"}]},"branch":"refs/heads/stable/yoga"},"0353c51c68c30632ecb29fd0ae4372a375086865":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-06 05:33:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/78/892878/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/892878/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/892878/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/892878/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/892878/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/892878/2"}}},"commit":{"parents":[{"commit":"a76d93dfb9592830623cce46e2033d3542fd1da4","subject":"Pin iptables to 1.8.4 in Centos Stream 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a76d93dfb9592830623cce46e2033d3542fd1da4"}]}],"author":{"name":"Dawud M","email":"dawud@stackhpc.com","date":"2023-02-15 18:02:05.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-06 05:33:31.000000000","tz":0},"subject":"Add comprehensive checks for container restarts","message":"Add comprehensive checks for container restarts\n\nWhen adding a dashboard to grafana the containers aren\u0027t restarted when\nthey should be. This is due to a bug in Kolla where the logic to\ndetermine whether or the container needs to be restarted fails in the\ncase where the file does not exist in the container. This patch adds\nmore comprehensive checks for container restarts in the set_configs.py\nfile. This patch also adds a test to ensure that the functions work as\nexpected.\n\nCloses-Bug: #1997984\nCo-Authored-By: Will Szumski \u003cwill@stackhpc.com\u003e\nChange-Id: I67f5f12700d7b55f26bff81e9b54559303da6d83\n(cherry picked from commit d9a6c5f3901e078a30d341397ac57d867d8191cc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0353c51c68c30632ecb29fd0ae4372a375086865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0353c51c68c30632ecb29fd0ae4372a375086865"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
