)]}'
{"id":"openstack%2Fkolla~893241","triplet_id":"openstack%2Fkolla~stable%2Fzed~I14ba7e0319b7eb0390361d397d4baad0e9a3f0b1","project":"openstack/kolla","branch":"stable/zed","topic":"add-server-status-redhat-stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-06 05:37:54.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-09-06 10:08:50.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-09-06 10:08:50.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-09-06 08:35:15.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I14ba7e0319b7eb0390361d397d4baad0e9a3f0b1","subject":"Add server-status handler to Rocky/Centos Apache conf","status":"MERGED","created":"2023-08-30 16:09:30.000000000","updated":"2023-09-06 10:10:19.000000000","submitted":"2023-09-06 10:08:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893241-add-server-status-redhat-stable/zed","cherry_pick_of_change":891098,"cherry_pick_of_patch_set":3,"meta_rev_id":"030c270472fe8a1a1544901a35a61f4c4b767b13","_number":893241,"virtual_id_number":893241,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-06 10:08:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-09-06 05:37:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-31 08:57:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2023-09-06 08:35:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-09-06 08:35:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-30 16:10:50.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-08-30 16:10:50.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-08-30 16:10:50.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-08-30 17:49:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-31 08:57:29.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"440321a436608d827d079057a52e51be6b81faa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-30 16:09:30.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a03b40c58568868274eda257475691059874d395","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 17:49:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1e131bcb3d843948026499e896b1d83\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8efb0a814c664a51b4459e916bf33400 : SUCCESS in 22m 30s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/157a00a8ca1047659ae7c94ce5503049 : SUCCESS in 1h 20m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d278a7d2ea4d4938a04f47ae870bfb69 : SUCCESS in 19m 54s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/26afc99d460b4346b9ecd4e36c66e03d : SUCCESS in 1h 10m 56s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/042b9bef56944a5b991c750741e03114 : SUCCESS in 1h 31m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38d4c0efa2124fb8b9bef86bd711544d : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c898df657583473e956341bf106485dd : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2089e637c32d4be9848d6ef2f95e24b4 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90000bebf4b548a1998e14c9237429ad : SUCCESS in 3m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e9cf2cdbc1449b6bff4d21549a6cdf7 : SUCCESS in 5m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cf46ca3c48f84bb28de97732187a5ffd : SUCCESS in 23m 24s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a27a02685a894168ae5c9225b10ebcbc : SUCCESS in 1h 17m 10s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cb46c217d7104302adce77c6dad00c98 : SUCCESS in 13m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d379580853ba4889942abc25902b00b9 : SUCCESS in 41m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/077ebc722640440096c7c304b24a4144 : SUCCESS in 1h 06m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d063211f1503d532b34093bfe8da584a9a631066","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 18:57:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21b628f5c6694ec4aefee2d9902d6f7e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/bfe30959a9504951a0ee85496a7687f4 : SUCCESS in 1h 00m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/450cec8c24bd42caa726a78651853e41 : SUCCESS in 1h 01m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dbb715ddf9ce45c5a161268a5f805f0f : SUCCESS in 2h 08m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c698348b3fd84d28882d5e7dbbb32e29 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d8364c9a7e4048a591d9c53e0f63f95b : SUCCESS in 4m 35s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/dddf35aadbfa42b393a07414106b84e4 : SUCCESS in 1h 09m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/321acf6a90d4452498aa49ff53c30793 : SUCCESS in 1h 43m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6987d7d52acd1fe0a118183544a13e5b71f05396","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-08-31 08:57:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"db46cc2868610e12b7f6a8481c52977cb2becea4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-06 05:37:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"96473f9e1f3e1e14aa3175e00f329ce621f612de","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-09-06 08:35:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d6b688c54e8fe9d281ecfabe615908b3dd387b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 08:35:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf81d01b79d9983e803edbf00c6265ccc00f8c77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 10:08:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e484297b02cf482b9b7e23ff29fd5330\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/108b5f0fdb95488ca900a100c1c86d5d : SUCCESS in 22m 21s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bcda8b07b548417fbcd4374084c01d58 : SUCCESS in 1h 04m 32s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f636f90b7a124230b27ce583a2a34a08 : SUCCESS in 1h 17m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb1e908cbc7a452c8c052b320015b5e3 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88dfbf364f584846930727b39b5ffe03 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e51e5f898754b1eb38ed8fbc5ffcc00 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/485bef5dcfff418ba0659bfe0a7d1572 : SUCCESS in 5m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/891e616dbafc4d50a00910069bc34b51 : SUCCESS in 17m 50s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/2eae8e04fcdc4753b51c4e65aa2c27a2 : SUCCESS in 1h 09m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2caa90889bcc4a1b8e81d582e70a2f9b : SUCCESS in 13m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/184e759358264e4098447726aa5e531a : SUCCESS in 47m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/529b799c832e4f0ca07d10e43e18e412 : SUCCESS in 1h 10m 15s","accounts_in_message":[],"_revision_number":1},{"id":"76f0f20fd010330bbf85d636c9bd4e57d6e57158","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 10:08:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"030c270472fe8a1a1544901a35a61f4c4b767b13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 10:10:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0031098fb66345c2b481c87ae655892b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa8af0f57c55452c9152e88fa1083fab : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81dd5d66d83b639462c978681a5fac972c218a87","revisions":{"81dd5d66d83b639462c978681a5fac972c218a87":{"kind":"REWORK","_number":1,"created":"2023-08-30 16:09:30.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/41/893241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/893241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/893241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/893241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/893241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/893241/1"}}},"commit":{"parents":[{"commit":"3684f67fee9026ff85771517af32f12d48e9c625","subject":"Add comprehensive checks for container restarts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3684f67fee9026ff85771517af32f12d48e9c625"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-08-09 14:34:32.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-08-30 16:09:30.000000000","tz":0},"subject":"Add server-status handler to Rocky/Centos Apache conf","message":"Add server-status handler to Rocky/Centos Apache conf\n\nThis added to make Rocky/Centos similar to Ubuntu/Debian\nto provide /server-status handler for local monitoring of the Apache.\nSince I3ec1af6353c3ecc64589599abe375b0ae9b14d5c is merged it safe\nto add this change.\n\nChange-Id: I14ba7e0319b7eb0390361d397d4baad0e9a3f0b1\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit daf9fdd3a6e96d417f19440d59452e3b271ec9d9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81dd5d66d83b639462c978681a5fac972c218a87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81dd5d66d83b639462c978681a5fac972c218a87"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
