)]}'
{"id":"openstack%2Fkolla~893423","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I87c5c3792c400f9bc90d00ebc5fc858f3a2b02e2","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-09-01 10:22:37.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-09-01 09:03:19.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I87c5c3792c400f9bc90d00ebc5fc858f3a2b02e2","subject":"Pin iptables to 1.8.4 in Centos Stream 8","status":"MERGED","created":"2023-08-31 17:57:48.000000000","updated":"2023-09-01 10:23:34.000000000","submitted":"2023-09-01 10:22:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893423","cherry_pick_of_change":893364,"cherry_pick_of_patch_set":2,"meta_rev_id":"24c612f199aec45d4d13937259d671b1b5cc0e7c","_number":893423,"virtual_id_number":893423,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-01 10:22:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-09-01 08:32:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-09-01 09:03:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-09-01 09:03:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-09-01 09:03:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-31 17:57:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2023-08-31 20:02:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-01 08:32:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-09-01 09:03:19.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"725cb081031cb203c2b2b0b60911d8a9ffcde664","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-31 17:57:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74e1171452a5066a989ecbb15e949e5e1074d75a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-31 18:29:53.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"4e1364bda240c4b298bbea913d61836b391c342b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-31 18:36:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e8921b9204e60d89511c3f09b9942c4ca59ed0d2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-31 20:00:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43525f85c9907c41dfa443b8d78f3051525c4170","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 20:02:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebed0209492342fda150dcea8082f2b7\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/24e1a48197be4979b4854790ec2bb680 : SUCCESS in 33m 47s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c3649ef21b2f437baf9ed1f2a8c67bc3 : SUCCESS in 36m 21s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/355f1a4b22ee4efb8cd59f2626acf781 : SUCCESS in 27m 23s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/483f8b456e5e40df9e05d4fdea9b9331 : SUCCESS in 1h 07m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d5b58286a3ed4f068f46fe7a02746aa4 : FAILURE in 51m 41s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/9ced3ffc6b6b42abb95c92935de4c177 : SUCCESS in 52m 06s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/88ceae0683d84040b5ca739f31a277bc : FAILURE in 1h 04m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/95dd8b285dad4dcc9217ff7c2b8c956a : SUCCESS in 33m 22s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a0137d37f5ee4c4c975d2109a985b1bf : SUCCESS in 15m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5f723379c5c34ac39413f7102199ad9e : SUCCESS in 1h 06m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b1582c8782c44951bfc8991a2b8ddfe4 : SUCCESS in 1h 21m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/843158d0f04944acbfc344b844700764 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a7adef0984c483ab2b5597fc8a0918e : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfba815ea1c04398bbac56a4704053cb : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffa520ca744148038d01ac5ab393709b : SUCCESS in 4m 50s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/05f1661f32e24f17aaa046b91fb523d2 : SUCCESS in 22m 50s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/e831ece78bd942deb8abad40b787f9aa : SUCCESS in 1h 00m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/74ecb7664fce431e85bf8d9e07946e5b : SUCCESS in 25m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d1348826ea6f48b78f95d32be4dee42f : SUCCESS in 22m 57s","accounts_in_message":[],"_revision_number":2},{"id":"321d7bf24b740cac5ff2ebe3cb3931ee2b9b8f5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 20:49:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd883be776df40fdb45359ebc958cf58\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a3c255268346491b875a588f59fced0d : SUCCESS in 2h 10m 01s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/aefa0b0f653043f0bebff9a80ff58848 : SUCCESS in 25m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/96ab0c6e80a64ee6a432809bd0a946e4 : SUCCESS in 1h 40m 32s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f6f6e6134ada49f9b62b2bc34b3bf5f2 : SUCCESS in 1h 14m 27s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5a65868ddf34165bf8cc2e04f34591f : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9068e46af75240a4ae3015dfe6653ae6 : SUCCESS in 4m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4c5602ae6bf240c0b21f3cb4682ca9d6 : FAILURE in 1h 37m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aced82c41aea8b3859245cc687bdbe9a1372a40d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-01 06:27:33.000000000","message":"Abandoned\n\nduplicate of 893359","accounts_in_message":[],"_revision_number":2},{"id":"d8f9aa5f9c418341c9bbae417e3b3e1c62e8c9c4","tag":"autogenerated:gerrit:restore","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-09-01 07:21:26.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"4e4c7ea4e832330a09bf2754e1dea02019489f88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-09-01 07:22:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"82ad5be4e3057784db7bdea64c199bf223df5028","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-09-01 07:24:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5c5089e8d6c7280efbb6ca05ece4fc6c2b0bbdef","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-01 08:32:35.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"40df8869f73630d04f36cacca86b735d99503cba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 08:52:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ae660c7eb9e431ba6b1cd844616dce6\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/511e1fe32e954eb2a43839f38a8aa2cd : SUCCESS in 49m 23s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/cb7ebddff8244d318c721fc1f704b1c8 : SUCCESS in 47m 52s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/7aed781c0e2640b69bae0e6a88271bc3 : SUCCESS in 37m 39s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/07bed6a52e724ae3b426846a0fa0c259 : SUCCESS in 1h 08m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/430b8d147a124e0fbe2607a727128326 : FAILURE in 1h 01m 21s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/c9eecfe68e9d48e3a702d8560f1a6a71 : SUCCESS in 58m 15s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/85e3b441ef4b488f88fe6ed23116f054 : SUCCESS in 1h 18m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6a7f40eb91b944c1a0481ba5e0729a53 : SUCCESS in 19m 14s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a92852bf65db44328244fa129f191b74 : SUCCESS in 21m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f0082b2c3d504669b60ffc3c363fddaf : SUCCESS in 1h 00m 18s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a47a96dbfe9d4a388400f01761df84b9 : SUCCESS in 1h 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00e6ab682038488e8060c96d0dd682e6 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2ba65d5b2bb4befb434e3de03ae6818 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/441c5864869c4c0fb88c86592a22ac9c : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0851b9cd536d44bbbaba8fb515a025a7 : SUCCESS in 5m 06s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/79388c4669c842bdbdc7b70ed88a629e : SUCCESS in 26m 08s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/d38cf752373d4e00af1fa290f7cf6388 : SUCCESS in 1h 08m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ca38daab98334e719c56a85332d97349 : SUCCESS in 20m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b5eb18bdb4a144ad9c95b731da510a87 : SUCCESS in 20m 30s","accounts_in_message":[],"_revision_number":4},{"id":"d6d007cda9eeae5b3d2c2719bbbf4edf9dddfd35","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-09-01 09:03:19.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1 Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"54394a60878b39029bf4273abd8cd7e70d74adff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 09:03:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"977404accd04a5887e1b69223eaaf4b68d17aed3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 09:56:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fcf659147f04b22bf058ef49cfda8c7\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d19eda49d0da4a7a864cdc91ed4ab3a0 : SUCCESS in 2h 29m 09s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/765782bb149a47dab02e50a2c3de1115 : SUCCESS in 22m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a0ca81b4db024801a2cbd09ed7206421 : SUCCESS in 1h 42m 26s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fbdc0df7dd2248f292f650306f533d51 : SUCCESS in 1h 13m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/489e8ccf703d40129c8527c675502a0d : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/58c7dcd3fa9b460b816cefa26e73d89d : SUCCESS in 2m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e08bb7108df14187b0eedfc539f29e83 : FAILURE in 1h 34m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7049ca627160cc9c33ad83d641ce2c6e9bdc1798","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 10:22:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae98bb6dc53141b5b0dd1c7227ac1641\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a8d17f22aeb94c43829aa7939a01bf6e : SUCCESS in 39m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a4a1a6960da145ca9f8e4c53f38d739d : SUCCESS in 59m 29s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b60efe7c85d44897933df3bb3bd9ca7b : SUCCESS in 1h 07m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/abab0a4ea20248eb9d936011c365630b : SUCCESS in 25m 06s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a98da716fca244ebbb1adace6ca5e1f6 : SUCCESS in 1h 12m 59s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d9f59917d2244d98adadecd91c0fe458 : SUCCESS in 1h 08m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/821ba35a644242ecad54afb640db5901 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b2618624adf41a38bc1955141a31628 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d15256a625f40b8b074b7521f3fe9a0 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0d5b55f37894995a4029073918a08d0 : SUCCESS in 5m 43s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/875af6a377a1429284223b8930da6784 : SUCCESS in 1h 06m 40s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2be14de36104006b8a86893baebc34f : SUCCESS in 25m 58s","accounts_in_message":[],"_revision_number":4},{"id":"1099b7bd9aeee8f1afc1d0bb95216de3e9c6a683","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 10:22:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"24c612f199aec45d4d13937259d671b1b5cc0e7c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 10:23:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b0c8cedc3f142fdbe463852792173be\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52911a9332e44981afddd5cd6774973d : SUCCESS in 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a76d93dfb9592830623cce46e2033d3542fd1da4","revisions":{"20f5159456e295aac202f0a1f4dac6aea90559e4":{"kind":"REWORK","_number":1,"created":"2023-08-31 17:57:48.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/23/893423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/893423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/893423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/893423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/893423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/893423/1"}}},"commit":{"parents":[{"commit":"f1d1881e9532b4b342ef2a4175e8248181d4eaf1","subject":"Merge \"Pin elasticsearch-curator\u003d\u003d5.8.4\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1d1881e9532b4b342ef2a4175e8248181d4eaf1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-08-31 12:45:02.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-08-31 17:56:34.000000000","tz":180},"subject":"Pin iptables to 1.8.4 in Centos Stream 8","message":"Pin iptables to 1.8.4 in Centos Stream 8\n\nBug in Centos Stream 8:\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d2236501\n\nChange-Id: I87c5c3792c400f9bc90d00ebc5fc858f3a2b02e2\n(cherry picked from commit 8f076857fde79477432c1377559a452c265297ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20f5159456e295aac202f0a1f4dac6aea90559e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20f5159456e295aac202f0a1f4dac6aea90559e4"}]},"branch":"refs/heads/stable/yoga"},"5ea1e8d41bf9a40a9d889caa5e83b5fe7d6cacdf":{"kind":"REWORK","_number":2,"created":"2023-08-31 18:36:22.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/23/893423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/893423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/893423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/893423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/893423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/893423/2"}}},"commit":{"parents":[{"commit":"f1d1881e9532b4b342ef2a4175e8248181d4eaf1","subject":"Merge \"Pin elasticsearch-curator\u003d\u003d5.8.4\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1d1881e9532b4b342ef2a4175e8248181d4eaf1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-08-31 12:45:02.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-08-31 18:35:59.000000000","tz":180},"subject":"Pin iptables to 1.8.4 in Centos Stream 8","message":"Pin iptables to 1.8.4 in Centos Stream 8\n\nBug in Centos Stream 8:\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d2236501\n\nChange-Id: I87c5c3792c400f9bc90d00ebc5fc858f3a2b02e2\n(cherry picked from commit 8f076857fde79477432c1377559a452c265297ce)\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ea1e8d41bf9a40a9d889caa5e83b5fe7d6cacdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ea1e8d41bf9a40a9d889caa5e83b5fe7d6cacdf"}]},"branch":"refs/heads/stable/yoga"},"65d140e6af18535e3e005b1e416db576ffb73284":{"kind":"REWORK","_number":3,"created":"2023-09-01 07:22:01.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/23/893423/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/893423/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/893423/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/893423/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/893423/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/893423/3"}}},"commit":{"parents":[{"commit":"f1d1881e9532b4b342ef2a4175e8248181d4eaf1","subject":"Merge \"Pin elasticsearch-curator\u003d\u003d5.8.4\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1d1881e9532b4b342ef2a4175e8248181d4eaf1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-08-31 12:45:02.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-09-01 07:22:01.000000000","tz":0},"subject":"Pin iptables to 1.8.4 in Centos Stream 8","message":"Pin iptables to 1.8.4 in Centos Stream 8\n\nBug in Centos Stream 8:\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d2236501\n\nChange-Id: I87c5c3792c400f9bc90d00ebc5fc858f3a2b02e2\n(cherry picked from commit 8f076857fde79477432c1377559a452c265297ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65d140e6af18535e3e005b1e416db576ffb73284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65d140e6af18535e3e005b1e416db576ffb73284"}]},"branch":"refs/heads/stable/yoga"},"a76d93dfb9592830623cce46e2033d3542fd1da4":{"kind":"REWORK","_number":4,"created":"2023-09-01 07:24:48.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/23/893423/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/893423/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/893423/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/893423/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/893423/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/893423/4"}}},"commit":{"parents":[{"commit":"f1d1881e9532b4b342ef2a4175e8248181d4eaf1","subject":"Merge \"Pin elasticsearch-curator\u003d\u003d5.8.4\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1d1881e9532b4b342ef2a4175e8248181d4eaf1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-08-31 12:45:02.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-09-01 07:24:39.000000000","tz":120},"subject":"Pin iptables to 1.8.4 in Centos Stream 8","message":"Pin iptables to 1.8.4 in Centos Stream 8\n\nBug in Centos Stream 8:\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d2236501\n\nChange-Id: I87c5c3792c400f9bc90d00ebc5fc858f3a2b02e2\n(cherry picked from commit 8f076857fde79477432c1377559a452c265297ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a76d93dfb9592830623cce46e2033d3542fd1da4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a76d93dfb9592830623cce46e2033d3542fd1da4"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
