)]}'
{"id":"openstack%2Fkolla~893605","triplet_id":"openstack%2Fkolla~master~I17683bdbabb72a1a8c64f3dfec14870b3435c829","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-09-13 09:52:26.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I17683bdbabb72a1a8c64f3dfec14870b3435c829","subject":"[DNM] Test kolla-toolbox permissions","status":"ABANDONED","created":"2023-09-04 09:04:35.000000000","updated":"2023-09-13 09:52:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ee7ab87d78596d080c3da6062380ed222f6a8986","_number":893605,"virtual_id_number":893605,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-05 15:03:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-04 10:22:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"222ca67c804c517c92de422f1acead8b9dea707e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-04 09:04:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8e901fec4fdc40700ac24292fcb04140c79b3bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 10:22:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69f2859503864fd3b1fa82cc878d608a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/447043ebec2445308a4ef9fbbc9b0fda : SUCCESS in 14m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1a2ace25f4bc43a498f57aec668ae2e8 : SUCCESS in 1h 03m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9cc7f885598949cc91362fac3e3d593f : SUCCESS in 17m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0fae4cf3a7114393907153bc9ff460f3 : SUCCESS in 1h 00m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dbebca5c434a416da3008c857a21b80d : SUCCESS in 1h 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a423b1bce31d4fbcb1e81365463c4767 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11500146d6cc4041a73bbcfdf9a9f19d : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d5bb765c481427d88899dcad9e815f7 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dba0bfc039f246f3bdc2f53710e301f7 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cdfd8f02286543e3b4ccb5d797fd9702 : SUCCESS in 2m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e883cb4b1654a04a48ea78f8887f119 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4a272342e19458b90f274055f773eaf : SUCCESS in 4m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9cb30c2ad48b4328a43938e25b2d0d99 : SUCCESS in 16m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b12e8ff98dd549b7946defd989a81abd : SUCCESS in 1h 09m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e7d0b63765a04e99a31bdcf56f2ebc60 : SUCCESS in 1h 12m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bfb10163329c490d928ea46860ae721a : SUCCESS in 19m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e2c6176dbb0b46faa26d7ca96de70c4d : SUCCESS in 58m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2857b9e51d674555b88a01daf6ebf7a0 : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1405f8d01e50cfec1fc21fb09296df1c4420394a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 11:08:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5028c3f1ce934b09aa89edc92a49d835\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0b0eaf4769ad4eaf8339dffbf129dc83 : SUCCESS in 1h 09m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/923eacb3ae2a4d8287ec3bc778c6ed44 : SUCCESS in 22m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d0086cec5c05444eb15f6c5dbc9499a2 : SUCCESS in 2h 02m 21s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/539470402f3d40ecb5771e88d27cc514 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2173f11cf3154de89e7763c1e4518d8e : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e557e9d256ed4a378fa920eacabff7f8 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a16f43016aa4d0e8e3704326f0ea323 : SUCCESS in 5m 18s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7160c21e4e724720a0fe666932e6f96b : SUCCESS in 22m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/65218821c94b4f5d8a15496bb5f68b5d : SUCCESS in 1h 10m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f67704a3e95fde049ed0d419e5a9a8f6172f75b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-05 12:18:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"531d5fea64d2295db0fad70afb7bf8ba853852b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-05 13:15:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ca822a4ce9cb393b6db66f88a1782ca285a90de6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 14:34:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddf84c8976e84e84baabe89ae0a747f1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6530f470cb2b4284a8650b519fe07809 : SUCCESS in 20m 13s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c28bf737e1ea4b609e62a1ac24acb337 : SUCCESS in 19m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1ef703bc04064ae9ac0c76239ff16a7c : SUCCESS in 1h 17m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6775341c7914063a8a390a2a88873db : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d837b5532e7452ba9769af307b80fc8 : SUCCESS in 2m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0aaac9e06435489c9e6c76871662f66d : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/154bb9bd1d8e470ba1b6cd7c4092a852 : SUCCESS in 3m 10s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1f4e2c33ef9e42138f5deaddb90b8a86 : FAILURE in 22m 51s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/566f9425f9304360bb6b726002391de2 : SUCCESS in 16m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c6da94dd9ea8c986be863b5ad749e289934d25ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 15:03:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d68a90a686a04d1191a297080884817b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0919336a61ad426cacbcb5d44ec117f1 : SUCCESS in 20m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5a964015fab64d819b30e4dbf1bbccce : SUCCESS in 1h 01m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8a61eb33ddb34d6ebfceb98cb000b22b : SUCCESS in 18m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/63177b127bc94ea0ab67b79cb474408c : SUCCESS in 56m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/846e8e11e97e4940a33ffc55c019bd38 : SUCCESS in 1h 08m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a08de819e6a242dba30b53535c82fc99 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1954ed71903f40acaf18a41e7b1f97f4 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e463ab0bdc6e4b16b5ab14c9bdbf5c6c : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e67a16616674b8c93c613dd1956d264 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b68d6cac6914dbdb4a61157382387b1 : SUCCESS in 3m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b94d2662531e4b5ea17a80fa171723c3 : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/254d17dfee3f4fc0a274c7726f14cf66 : SUCCESS in 4m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f1e1d5cd43ce40418b2dbff7cb591f98 : FAILURE in 10m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d9754ce86d4147c885b841166324e0d2 : FAILURE in 9m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/913db01276c740a3a0727ee39d0337af : SUCCESS in 1h 28m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/23f61d4fdc8f4132a54dca46838dafac : SUCCESS in 20m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cea8b0096c394db89cb27b301e56822c : SUCCESS in 44m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6806c80d3b4740ebbb518ac62d08c0f9 : SUCCESS in 1h 11m 32s","accounts_in_message":[],"_revision_number":3},{"id":"ee7ab87d78596d080c3da6062380ed222f6a8986","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-13 09:52:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e4bd77b7e14923bc126374261db70555535c4883","revisions":{"067fc1b4d2d5e6df9d7dd4135f9e37315a36bb5d":{"kind":"REWORK","_number":1,"created":"2023-09-04 09:04:35.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/05/893605/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/893605/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/893605/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/893605/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/893605/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/893605/1"}}},"commit":{"parents":[{"commit":"c8625600a5245e464bc5ef012cb6e60fed14690a","subject":"Merge \"Add server-status handler to Rocky/Centos Apache conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8625600a5245e464bc5ef012cb6e60fed14690a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-04 09:04:25.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-04 09:04:25.000000000","tz":120},"subject":"[DNM] Test kolla-toolbox permissions","message":"[DNM] Test kolla-toolbox permissions\n\nChange-Id: I17683bdbabb72a1a8c64f3dfec14870b3435c829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/067fc1b4d2d5e6df9d7dd4135f9e37315a36bb5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/067fc1b4d2d5e6df9d7dd4135f9e37315a36bb5d"}]},"branch":"refs/heads/master"},"77ce7f57b77428a6c398c6be86ef6c33f8f680fb":{"kind":"REWORK","_number":2,"created":"2023-09-05 12:18:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/05/893605/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/893605/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/893605/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/893605/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/893605/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/893605/2"}}},"commit":{"parents":[{"commit":"c8625600a5245e464bc5ef012cb6e60fed14690a","subject":"Merge \"Add server-status handler to Rocky/Centos Apache conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8625600a5245e464bc5ef012cb6e60fed14690a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-04 09:04:25.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-05 12:17:51.000000000","tz":120},"subject":"[DNM] Test kolla-toolbox permissions","message":"[DNM] Test kolla-toolbox permissions\n\nChange-Id: I17683bdbabb72a1a8c64f3dfec14870b3435c829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77ce7f57b77428a6c398c6be86ef6c33f8f680fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77ce7f57b77428a6c398c6be86ef6c33f8f680fb"}]},"branch":"refs/heads/master"},"e4bd77b7e14923bc126374261db70555535c4883":{"kind":"REWORK","_number":3,"created":"2023-09-05 13:15:14.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/05/893605/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/893605/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/893605/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/893605/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/893605/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/893605/3"}}},"commit":{"parents":[{"commit":"c8625600a5245e464bc5ef012cb6e60fed14690a","subject":"Merge \"Add server-status handler to Rocky/Centos Apache conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8625600a5245e464bc5ef012cb6e60fed14690a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-04 09:04:25.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-05 13:15:07.000000000","tz":120},"subject":"[DNM] Test kolla-toolbox permissions","message":"[DNM] Test kolla-toolbox permissions\n\nChange-Id: I17683bdbabb72a1a8c64f3dfec14870b3435c829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4bd77b7e14923bc126374261db70555535c4883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4bd77b7e14923bc126374261db70555535c4883"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
