)]}'
{"id":"openstack%2Fkolla~893854","triplet_id":"openstack%2Fkolla~stable%2Fxena~I169e51c0f5c691518ada1837990b5bdd2a3d1481","project":"openstack/kolla","branch":"stable/xena","topic":"bug/2020761","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-09-08 15:00:53.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-08 09:18:30.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-09-08 15:00:53.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-09-08 15:00:53.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-09-08 13:27:08.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I169e51c0f5c691518ada1837990b5bdd2a3d1481","subject":"Improve kolla-toolbox offline support.","status":"MERGED","created":"2023-09-06 09:35:11.000000000","updated":"2023-09-08 15:02:14.000000000","submitted":"2023-09-08 15:00:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893854-bug/2020761","meta_rev_id":"a0d1215660ef1f3eaed06f69125f923176c4beba","_number":893854,"virtual_id_number":893854,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-08 15:00:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-09-08 10:03:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-09-08 09:58:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-09-08 13:27:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-09-08 13:27:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"reviewers":{"CC":[{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 09:35:11.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"state":"CC"},{"updated":"2023-09-06 10:49:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-06 11:30:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-09-06 11:30:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-09-06 11:30:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-09-08 11:12:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"bc95dc3b38a3bbb9c7213b01010b49e7b586f9c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-06 09:35:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5153348be0ed7196c5f3ffd9464363da35d65314","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 10:49:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a96a88c23494190b99267155cfa5696\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/71755e12962b48338e780d48201c0d50 : SUCCESS in 26m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a1b011221a064792bb89b49a44884755 : SUCCESS in 33m 24s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/11a3453350374e408e514efec0d3d318 : SUCCESS in 1h 12m 22s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1c15e5322f9444529e4da16e61e2e573 : FAILURE in 58m 14s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e404403eac73428fb74006ea61bbe904 : SUCCESS in 16m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7348c6605f5645e0b82cc4626f17b362 : SUCCESS in 18m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c2131fd0bc8049ce9dd83d0f7b26e2f7 : SUCCESS in 1h 07m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1132965d06641bc8b4564120768a8f8 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fb0d96b203b4b1287815bc99c9cd353 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1589a78af7fa4990ba41af4e412bba42 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10a30631b64b41e9ad286b32b4558455 : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d75c6317f7774226ba48dfde000afe86 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac65035855c84d3186185245ce8b5d75 : SUCCESS in 3m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/90f05888eebc4d51814f0081652ee7a7 : SUCCESS in 20m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6830d4f102374466bb92051bd7375375 : SUCCESS in 20m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a72e62fe54143d892ca9bc56dcf75f0 : SUCCESS in 57m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f185efeb991147138c474bd5e96a80f2 : SUCCESS in 48m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e0c1fa8a554341919bbf4eb95a40c981 : FAILURE in 7m 24s","accounts_in_message":[],"_revision_number":1},{"id":"f3a9dd0b22bb290bd68acae54f83489bf3d4ce1b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-06 10:55:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f385ca303a1fe387fc3db3381386bab587bcbe8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 13:37:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7da52d2297d54aa1aa952102dec66678\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4215385b58ef4d9a854ea4b15e068a54 : FAILURE in 30m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e01036436d334e8eb4f24f9b195db256 : SUCCESS in 1h 45m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c0c4e1cc1ba24f36be4e1169cffc8f96 : SUCCESS in 2h 31m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ec2ce97be7846de8ff58891603a6643 : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b7f106b9758471db61ab805ceaee731 : SUCCESS in 2m 56s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ffda55cb1cf44ac8af6ce79be96d70d8 : FAILURE in 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"010f4bb2a7a8a855d2627a5cf8804471354753e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 14:05:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76808979d4df4d629f5a42bc0363969a\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6b5ca991efc344de98db1e536b171e63 : TIMED_OUT in 3h 02m 34s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3b5393407ce0407099086a9e1bc286a3 : SUCCESS in 35m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9d4f4d5e9e45411ead72c39f09e831e2 : SUCCESS in 1h 15m 51s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d135fd8edf5142ada2c39414d5a8646a : FAILURE in 54m 01s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/82fc2c6b84314cccae9575c1bb85d5c3 : SUCCESS in 14m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f13d7160e97540339de7b69328792888 : SUCCESS in 17m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a44b030f2aa24eb3b7b54e0ddc388efd : SUCCESS in 1h 09m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f067c434e5e448b98ff0c31e76b67bb : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cc3fae9ccf84bcfa65082fd6b06c511 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9f661e05f004af49a74d0a2e5d8d428 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf0299f5b479401c83718ba5d527394a : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9671387fcbfe4a34a5c1c339f84ac4a7 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c87009bfefbc4e84afe3e465665905c8 : SUCCESS in 3m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/def578b4fd1644e595307cbc59ade5bb : SUCCESS in 21m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f7e45a307e6544fc97c3c76f2b0517d1 : SUCCESS in 20m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a679b00f94174be6b66c42ee9087cca9 : SUCCESS in 53m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ca6f4cf612314c91860e9e97b92d070c : SUCCESS in 51m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7c83416b69cce36d7405b9c3bd62550012db8dc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-08 09:18:30.000000000","message":"Abandoned\n\nxena going EOL","accounts_in_message":[],"_revision_number":1},{"id":"cd74bc173adb6bdeebaceacf8203a50fd114ad13","tag":"autogenerated:gerrit:restore","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-08 09:57:42.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"2f3f965a2d0362dcedfee8fd96a3f701a0074d33","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-08 09:58:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a52297262fff1998079b327fc875476fccd3017c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-08 09:58:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7e4348eb6907506745a5e6b5aae866c35d972ddf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-08 10:03:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b395b095825880a3a577a16e361745ba5717ed51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:04:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d10d65e69624d99acaba46ecf1b3c67\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b06c395d7c96400b9081579c1f49de58 : FAILURE in 25m 10s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5ba26d987bc842cb90e1a969e6d36b00 : SUCCESS in 1h 02m 26s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/840c9838cd89430ab4daccbe88c996b4 : SUCCESS in 1h 05m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e796799784d04bc2a4fca7cfd0298286 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/211f545ff64742cd9a3d15463ca392a6 : SUCCESS in 3m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e14e45c16edb42d19f5fe2fa0548202d : FAILURE in 1h 04m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"76fc8ffbc575810a873ea055c39b3dad9aa71249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:17:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f6287215ef94f7186d862d1efe1e013\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/55a6d851eda9466fa898a54a7992ff2b : SUCCESS in 32m 16s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f8d5367333424b859cb13ff0a4a4f02c : SUCCESS in 33m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5ed5c419cbef4d5692dd4163d643dc26 : SUCCESS in 1h 14m 42s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/72d82502ef9744fc85e05b0cf66fca02 : FAILURE in 52m 52s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/306b019483474a03a657333c68417fcf : SUCCESS in 21m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/306fd6a9dfd549738c83f8f320646da9 : SUCCESS in 20m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/30d27c2b97f14fc9b38aac9474e4fed8 : SUCCESS in 1h 12m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32ee2b7a819049988e8ba030f54d5f70 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b72dc92d2a7144828109db823ffb7730 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c762399dd6494fb697071a6bf2c6d3c3 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a516ab925844cf8b622780fff130b80 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c5d314b6c94835821dc81f4c183d88 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89f11c588e154ba4b4e1d89aba27d6ea : SUCCESS in 3m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5e27c1643ed6429494ce93d60a40ab98 : SUCCESS in 20m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca8d404d945c40e19c60f252d1a42df7 : SUCCESS in 19m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b2cdb11d07054ac8b0f3c0e1127f846f : SUCCESS in 55m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5cb323a7a7594b5dbe22f9846f6e6fe5 : SUCCESS in 50m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e69d78d106762d6875b34e6120bfefc7e1a14406","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-09-08 13:27:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a8ddd6fb411a21006232d6e739f5fc8efe717a77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 13:27:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da36fc5e982047570bad9b6e3c4b6360678a2310","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 15:00:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cc097403c8a4b0c92f20d405d009e30\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e9b7bd3686354a5bb3543af26acbb577 : SUCCESS in 34m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9e54d753a2ed48c7a0304b3ae838616a : SUCCESS in 1h 21m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1264242e6d9248e2913f56f433e8460d : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02ab1f42b8574264b0a5d84369e092f9 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1875caf36e9c4812a9ecb13afad1ef51 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255e8b8ca8af4565a1ec4bcfabc25037 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a0ec4a392e74eaba347078437f15753 : SUCCESS in 4m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/60897198eff1467b97af4cde836235db : SUCCESS in 22m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3642c42cb2164c9ab986ce908c4ac76e : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":1},{"id":"6f0659cfef128f870e601ca49909fa6d9e845a55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 15:00:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a0d1215660ef1f3eaed06f69125f923176c4beba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 15:02:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/886470a820c947f88e921d47fde8c6a0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f62b765a704ad6a4c8f84e57bda1b1 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/733fe7c8c8954eb09f435e2209a9cc9f : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4835a3d607b67259f85599114ca98d4afd35b5a8","revisions":{"4835a3d607b67259f85599114ca98d4afd35b5a8":{"kind":"REWORK","_number":1,"created":"2023-09-06 09:35:11.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/54/893854/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/893854/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/893854/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/893854/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/893854/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/893854/1"}}},"commit":{"parents":[{"commit":"8f076857fde79477432c1377559a452c265297ce","subject":"Pin iptables to 1.8.4 in Centos Stream 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f076857fde79477432c1377559a452c265297ce"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2023-05-25 14:39:25.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-09-06 09:34:59.000000000","tz":180},"subject":"Improve kolla-toolbox offline support.","message":"Improve kolla-toolbox offline support.\n\n  * Use structure that allow for custom package version and base URL.\n\nChange-Id: I169e51c0f5c691518ada1837990b5bdd2a3d1481\nCloses-bug: #2020761\n(cherry picked from commit 7f38e1df71eafb250b14c4e125c7313bd9218f7d)\n(cherry picked from commit 8558004ab70919cc2a822107c953a24710227f9e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4835a3d607b67259f85599114ca98d4afd35b5a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4835a3d607b67259f85599114ca98d4afd35b5a8"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
