)]}'
{"id":"openstack%2Fkolla~895822","triplet_id":"openstack%2Fkolla~master~Ic34a150282b0e77282c31ad03e6a9839b6636cc1","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-11-24 10:46:34.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-11-24 14:09:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic34a150282b0e77282c31ad03e6a9839b6636cc1","subject":"CI: drop temporary vars for Bookworm switch","status":"MERGED","created":"2023-09-19 15:16:44.000000000","updated":"2023-11-25 14:51:02.000000000","submitted":"2023-11-24 14:09:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895822","meta_rev_id":"e4877fc104c0304c2ff6f62972241c6246f2ac46","_number":895822,"virtual_id_number":895822,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-24 14:09:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-11-24 10:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-11-25 14:51:02.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-11-22 06:49:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-11-24 10:46:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-19 16:30:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-20 13:33:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2023-09-26 10:46:10.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-09-26 11:13:07.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-11-25 14:51:02.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"d0ce5d96ff59cc50335a0b7ab34ce98256b1482d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-19 15:16:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d885bd0d33126d8e46886eb19452a704b829c94d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 16:30:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e42677337b146fd9a73d575634f6b10\n\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a27a4bf19d6e49b9aa5b715b37aba591 : SUCCESS in 1h 06m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2c042a6f2cbe402393d27a8ba3a11e28 : SUCCESS in 1h 03m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39ac3e5fcf0d47aa92a91e118fab1e61 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08a766c1c50b49fa8ebdb7789d1475fb : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/499b889432464686860cdfda7269b135 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26526146d08f4935b2851f51d7817e0d : SUCCESS in 2m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c84a355e930442ebb0dfa6001b4c0faa : SUCCESS in 2m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90071f61237943a193432445131c4c5b : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f84ac3e60044a378a130a2384731efc : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"368452400f6d2ab683713e5c84aa1c58e017dc72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 19:10:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e19b8c2b13e54a2e9c2d0ed9588dec4e\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ab3efbe10d0a40858d3d787689f6d5ed : TIMED_OUT in 3h 02m 16s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5f01d74a49148a48ca4d42c1b01da26 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb5998413f2046718a23669847a054de : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dcbd18483bfe4938921b829ce4a30bb7 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f904b2958ed24b4f9bb5994558ef73f3 : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5bdeeaeb158ab1cc62aa2fc0385ee7b6a2df94a1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-20 13:33:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bfce18fc0ca470d22546133ea8bec2cda30e1aa5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-26 10:30:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nThe following files contain Git conflicts:\n* .zuul.d/debian.yaml\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1ca7a5ea15bd1f8d381b2c116f07e38397b38200","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:30:44.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  \n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/kolla on branch master.  The error was:\n  \n    while parsing a block collection\n    in \"openstack/kolla/.zuul.d/debian.yaml@master\", line 2, column 1\n  did not find expected \u0027-\u0027 indicator\n    in \"openstack/kolla/.zuul.d/debian.yaml@master\", line 6, column 1","accounts_in_message":[],"_revision_number":2},{"id":"2b3b441fa2b04931ef3b306e91a1bf1650377d10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:30:51.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  \n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/kolla on branch master.  The error was:\n  \n    while parsing a block collection\n    in \"openstack/kolla/.zuul.d/debian.yaml@master\", line 2, column 1\n  did not find expected \u0027-\u0027 indicator\n    in \"openstack/kolla/.zuul.d/debian.yaml@master\", line 6, column 1","accounts_in_message":[],"_revision_number":2},{"id":"19b7c37a9b407cc949691d203edbe4927f7956ff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-26 10:31:09.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"436354973f727107cd28397086fb7596c6d405fd","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-26 10:31:25.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"bd4285a7644db9cb3e65f8a22bed67cd5f5e0322","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-26 10:46:10.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b42e3511f53da2f20e6ea1503338ab405e43885f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:56:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c80fba8621e4773b459f55d6fc2f329\n\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/275a0521c26045339d0aabae9949ca82 : FAILURE in 18m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a843b1a6c82b48d9812424fe06efd8e8 : FAILURE in 22m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/656906e502ea44dfa9ae9ce7f911589a : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16f81bd848584e8d9ebb0d14523c191d : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd03f26153834d39b136b723e667a49d : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20af7a77aff8472b8a0d8b985a0ac331 : SUCCESS in 2m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64219eeea0d540b989ddb77fe1c27d48 : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1f30cd62bd147ad8da0d4da838cc671 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7573e07a6a10412f8e63c1b2442ae80b : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":3},{"id":"b94949a9c4a2797fbb41f2af34eaafe62f272a42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 11:16:34.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4765325dddc94ef0af8f7a1e8d7f92fb\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6730ccdb88424fdaa038f8665f1fc2bb : FAILURE in 43m 20s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c18ca35cef147c593f85e1ee9ae70ec : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ff72cc3fa9a4fa288006376b3f44e98 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/755fd18a826d4e1aaf6b562f940604b3 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7b29dbd0d7764db0962ed76e0c05e33d : SUCCESS in 5m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5428bfded366e6f9af3366beedb3f5bc9db4f65b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 06:49:04.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"aea7ad345370de101842032616b8dd665c518a87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 08:40:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eee6f0b5894b4c07ae03b31a0cfcf722\n\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/69c6783e752443b895496982a4e5f400 : SUCCESS in 1h 07m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6ce70f80afc44fa19ee5e000256483ab : SUCCESS in 1h 33m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/794956a554e54e8a8b6633e757700b33 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44f75ad9f47d4df78818e6866d77a289 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23835bc4aef448f2a4727676b2051a7f : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/144616447ac54a9cbfb3f506169ffbce : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10974ba7173f4c41939599ad952c1852 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":4},{"id":"cd95ea1daf2da5579dd8371004e1b5f657cf963c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 09:35:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98586aae124c498d97b70507695ea1a4\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6760f0c16b66481d961693e13a552e02 : SUCCESS in 2h 37m 25s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e9d5eb6e1554a91a5e5160cfedd49cc : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/922711cb90b441e6a080fbde7c5eff18 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d66c685b57d1458f9a913b9e2f28092f : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d3dbb740f273bc7e0611c1e42ad0342dc8481544","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-24 10:46:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ebbd8f3e28ee6e759eb1ad7e827838b27f2e28a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 10:47:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c3278eec9e5491b4b3f08a5bb842fbf75d30ef38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 14:09:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac68746fdc804f5087b2ed17f9449f34\n\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/028c4ac7f4a44e238f13f37ffd27949d : SUCCESS in 55m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ccd571382af6491d96eb3d1bf9fb4c50 : SUCCESS in 1h 30m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81c73971e25a43da99823d4acf1817b0 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4dc1dda933440e294a092cc03fa600b : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/370e3b07a65840f3a70ea1e014b0afd9 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa60bb0a902a4020b65f996b2231a824 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":4},{"id":"6e2a532c92febcf2c3891b48956604f2379d2f1b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 14:09:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7f3f4d3fd180fa3aa1bb051d6e866e068c376c6f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 14:14:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd3e265644794ee5b881048630139b91\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0caab4d64c014c0f9970913ce95c9fc9 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":4},{"id":"e4877fc104c0304c2ff6f62972241c6246f2ac46","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-25 14:51:02.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"88bf4ba384e1b7f53408d6bff556b0279cc16d3a","revisions":{"d89e1b09dd31ddb257c13495a042b5c89b21b987":{"kind":"REWORK","_number":1,"created":"2023-09-19 15:16:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/22/895822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/895822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/895822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/895822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/895822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/895822/1"}}},"commit":{"parents":[{"commit":"3d703b60e6fb1062313b57293696f1c34151c27f","subject":"Merge \"toolbox: Add sleep script handling signals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d703b60e6fb1062313b57293696f1c34151c27f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-19 15:16:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-19 15:16:42.000000000","tz":120},"subject":"CI: drop temporary vars for Bookworm switch","message":"CI: drop temporary vars for Bookworm switch\n\nChange-Id: Ic34a150282b0e77282c31ad03e6a9839b6636cc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d89e1b09dd31ddb257c13495a042b5c89b21b987"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d89e1b09dd31ddb257c13495a042b5c89b21b987"}]},"branch":"refs/heads/master"},"d7b2385659be336122b4022a14dc34a6179d99ec":{"kind":"REWORK","_number":2,"created":"2023-09-26 10:30:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/22/895822/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/895822/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/895822/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/895822/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/895822/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/895822/2"}}},"commit":{"parents":[{"commit":"91dcdeffb863f77fac6b640730a40cca28f0c47c","subject":"Merge \"Drop openvswitch-netcontrold\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91dcdeffb863f77fac6b640730a40cca28f0c47c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-19 15:16:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-26 10:30:22.000000000","tz":0},"subject":"CI: drop temporary vars for Bookworm switch","message":"CI: drop temporary vars for Bookworm switch\n\nChange-Id: Ic34a150282b0e77282c31ad03e6a9839b6636cc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7b2385659be336122b4022a14dc34a6179d99ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7b2385659be336122b4022a14dc34a6179d99ec"}]},"branch":"refs/heads/master","description":"Rebase"},"960c1a3102af2132ac27bb0a5108eb3167257575":{"kind":"REWORK","_number":3,"created":"2023-09-26 10:31:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/22/895822/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/895822/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/895822/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/895822/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/895822/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/895822/3"}}},"commit":{"parents":[{"commit":"91dcdeffb863f77fac6b640730a40cca28f0c47c","subject":"Merge \"Drop openvswitch-netcontrold\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91dcdeffb863f77fac6b640730a40cca28f0c47c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-19 15:16:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-26 10:31:09.000000000","tz":0},"subject":"CI: drop temporary vars for Bookworm switch","message":"CI: drop temporary vars for Bookworm switch\n\nChange-Id: Ic34a150282b0e77282c31ad03e6a9839b6636cc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/960c1a3102af2132ac27bb0a5108eb3167257575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/960c1a3102af2132ac27bb0a5108eb3167257575"}]},"branch":"refs/heads/master"},"88bf4ba384e1b7f53408d6bff556b0279cc16d3a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-11-22 06:49:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/22/895822/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/895822/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/895822/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/895822/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/895822/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/895822/4"}}},"commit":{"parents":[{"commit":"d86cba116ef686a7bad6153a5ce865ba77850cc7","subject":"Merge \"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Bobcat)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d86cba116ef686a7bad6153a5ce865ba77850cc7"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-19 15:16:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-22 06:49:04.000000000","tz":0},"subject":"CI: drop temporary vars for Bookworm switch","message":"CI: drop temporary vars for Bookworm switch\n\nChange-Id: Ic34a150282b0e77282c31ad03e6a9839b6636cc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88bf4ba384e1b7f53408d6bff556b0279cc16d3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88bf4ba384e1b7f53408d6bff556b0279cc16d3a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
