)]}'
{"id":"openstack%2Fkolla~896616","triplet_id":"openstack%2Fkolla~master~I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-29 10:11:42.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-09-29 07:30:06.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-09-29 10:11:42.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-09-29 10:11:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36","subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","status":"MERGED","created":"2023-09-27 08:24:30.000000000","updated":"2023-09-29 10:12:53.000000000","submitted":"2023-09-29 10:11:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"896616","meta_rev_id":"8edea62825cff42806a2e36d5a2716bda4d3341c","_number":896616,"virtual_id_number":896616,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-29 10:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-09-29 08:28:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-09-29 07:30:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-09-29 08:28:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-09-29 04:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-27 08:37:55.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-09-27 08:49:06.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2023-09-27 10:53:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-27 14:07:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-28 08:52:06.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-09-28 10:48:21.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-09-29 10:02:29.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"}],"messages":[{"id":"93e0f67320c2123012cac0fe1841cfe4f0b541ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-27 08:24:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18c59a2136373f2a5a097825e67e805076332ba2","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-27 08:49:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c41f945fda06ca19a48207b27f05f9145b117146","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-27 09:06:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e87bc7eeca9f8987f07cb57966bf9717c1c359d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-27 09:13:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"63051a571d85e23a08d4d82391b33eb97ffbf42a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 10:53:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80ee34574f2443cbb2e26bfdeb28e6c4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/05f56a0e8fde40e69d6278b6e624ff47 : SUCCESS in 1h 18m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/12002baf476943089504485239c443fd : FAILURE in 1h 38m 23s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/762359f1a5604f4fb909322b067edea4 : FAILURE in 55m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ce311963b0e4875bef15b728f94c2b7 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2c51e51016c4440a2a13ee9899339d0 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/493ec1485ed74f2b806504e7391493b2 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8c919c4a58f34810b6e5685588bb8fbd : SUCCESS in 4m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/da3c7a70d97849d48517572173da4be0 : FAILURE in 30m 40s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4a681145f2cd47f6844b13f1d8ae9a8e : FAILURE in 1h 37m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca0e42d0b2661be6cfcd09dd50157d7ae97eff66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-27 11:15:15.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"1851e5749cbd0ee06bcf14527d367c1709b4e576","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 13:02:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d90df8b85c744092ac9242bf905799fe\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8c0d0b98396e4492b2c40896bf10a940 : SUCCESS in 1h 34m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c36108655b3c4c91a122291a5813038c : FAILURE in 1h 33m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5c7c5f2c17ad42dcb51e3fe2bcb8a4d4 : FAILURE in 1h 08m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3bb253c7afe94ce6bcd10c132a8b3a23 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/06441f87b7e24372a3ed7c5b40e00ca5 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0774bdf6bf9b4650a59a8417fdca8660 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/870d34d78aef4175b3bd6793bf2ec8f0 : SUCCESS in 6m 13s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6019d4c01a9c4663b35b801ccf64e1e3 : SUCCESS in 1h 29m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2f1f7a5c5c064c07b907b0389ee092d4 : FAILURE in 1h 25m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c569542fb58eccf18ed1226e9a23ec4fb0cfad43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 14:07:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98ff3622d90c46a8a3cfb5a44c384004\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/926ad168692c4482a5917807fe152bed : SUCCESS in 17m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8ab56ef648654b5a98c3e20a759b7227 : SUCCESS in 1h 21m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/699ef66b5e8240a586e8db7299a82ef6 : SUCCESS in 20m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/37dfd84e623641eda4c558f29605412a : SUCCESS in 30m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/710ed15fb4ef4d83bd95f4b3fc77192c : SUCCESS in 1h 06m 45s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5b1894796aa14bc992d011ba06510fd7 : SUCCESS in 1h 35m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e01cda3e66f4a8b9147080ed6d39f64 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/872f0d44bfed4a9c84b44a61f14807d3 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e117ba76c1a4e8ca3913fe6a91531d9 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/349b3588140d4e83b02d7ccffcf41f92 : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2dfedf3bda1649a4bea299d1e58f8e07 : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca196f9ad3954648b536e395b148dd30 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa62ed73cf8457285fc92f41fc9da98 : SUCCESS in 5m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4f14c94be813470fa456a2846e88ba83 : SUCCESS in 24m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/55f871e4f9a2472d9f39732a0704cb3d : SUCCESS in 1h 21m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/db4f40892493484ca4a269f5a0d0f7d8 : SUCCESS in 1h 12m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c0b14efba99344f3a9770523eaf701e6 : POST_FAILURE in 1h 42m 26s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/7e258b701dd244289d28b443d877398b : SUCCESS in 1h 29m 47s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5c9e61d086f649b5824f92130f5a3b38 : TIMED_OUT in 2h 34m 46s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/39294ba5171743dd8453ae79976a9d3e : FAILURE in 56m 26s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/e6837090bd13434193eb9cfc999d2b13 : FAILURE in 1h 07m 34s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/5e5e0eaed7b54aa096bb36284eb81927 : SUCCESS in 1h 40m 15s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/082693c5fca142f19cd97e72ef9ccef4 : SUCCESS in 30m 24s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/530746a462934697828908e7443527c0 : SUCCESS in 58m 44s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0acd06669a314ff98388e7119b9adf49 : POST_FAILURE in 14m 05s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9612ee32d66e43b2bff4e207d079b433 : FAILURE in 1h 40m 30s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/bf4b16732be84475853dc74ae19e51c6 : SUCCESS in 37m 07s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/750545dfd7d8410491fb0859258e5d4f : SUCCESS in 30m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/696bd653def6487ca259895dda9981b4 : SUCCESS in 31m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b25b01ff53474a92b573df1a706145fb : SUCCESS in 57m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4b4a8b7c2c0940afa0410e80b3f45249 : SUCCESS in 1h 26m 06s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/84faa5c7e00a4cb99be2397ae5da56d4 : SUCCESS in 1h 49m 41s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/525e2c9df4904cdca0c87ba681f65281 : SUCCESS in 2h 23m 23s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/3c789c053c5c439e98e59f135c08daa7 : FAILURE in 1h 07m 11s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/2d5fbb7c395b484783c4d818fade397c : SUCCESS in 1h 18m 53s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/8cfb08e48c444490bbbf0a1aeef8d9ec : SUCCESS in 1h 31m 44s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/0e64cf1d8542494891ce83052e535f4f : SUCCESS in 25m 18s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/e4e0537b68034e6188f8a1d36dc73dd6 : SUCCESS in 51m 25s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/de7a8ece939f48a3b61f6e8841843d8f : SUCCESS in 56m 18s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/762442e864734af88c6a5d7c4ac64a3b : SUCCESS in 1h 17m 38s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3a222cf70e0f42deb633d4658996b263 : SUCCESS in 32m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6d7c9da2e52409e6918aac13e5c76ae3f3fc3e42","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-27 18:39:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"010c5308e09b8689bb36e8e83d86a384b34b5eca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-28 08:38:16.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"07e329a4169679cbdca9f924095640c66310f64a","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-28 08:52:06.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"360a423e56881d843180bd9eda08e1128b40199c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-28 08:55:05.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"476567659e5e82c8362ebfb3739555eea7def240","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-28 08:55:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ffc2f9220495b50206f3e9c9dbd68829277ebb2c","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-28 08:56:35.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"99c33addd8746b98b66a23a0c0083f0529867d85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-28 10:26:54.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"076d2d0c1a53d4c51b140b0e4a4dc64553bf8134","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-28 10:31:48.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"a80a470cea4b210f417c36caff50b5031d8e20ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-28 10:32:18.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"a27be0dccb04ef9de03786971974dd8b98ad6957","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-28 10:38:56.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"558352ecd851828d899f7380eb80fad4b4162b1c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-28 10:39:46.000000000","message":"Patch Set 9: Review-Priority+2","accounts_in_message":[],"_revision_number":9},{"id":"95369a662636c9f0fae2632c9dc2f162fdda6328","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 13:24:47.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b09c67bdebf49ad83f1ab4e17db77b1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/07a6fe00df674ad08c7c556a6ac01560 : SUCCESS in 1h 23m 22s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6212425d2d644ed2a9f9f043796e19a4 : SUCCESS in 1h 16m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/51956f71aa704510bee71bfc9467c3b7 : SUCCESS in 2h 26m 12s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/960a3801a6204a46a79d6979e6277afc : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d2d0d30c3cc4984852808de94a2523d : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d0fa0b01ec574eb49a00796db134c5bb : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/da9699aebafe45b2b459f7135a4b1503 : SUCCESS in 5m 22s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b08a393d072e43ae8da9f39bf27bfbcd : SUCCESS in 1h 28m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b055073c8627439b92c3200b9920c74e : FAILURE in 1h 39m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e38acc8ac571af372add0c38b9572ebb273a400d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-28 13:24:52.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":10},{"id":"c21cd8576eb3e1aedf29ca42e39a3ce3a27f63c3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-09-28 14:00:36.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"49b54e94573c96bd84f48c4844d890a8965ebe88","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-28 14:40:57.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"096e8c31b2369e0ca93f8249e88ae657a3a5775b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 15:47:41.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c4b1ae010a34f4ab622920b269fda5c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/11de7f381dfe47a5a7891060effb1d71 : SUCCESS in 21m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4fff9461687c4a9ba39a8f53f3c8155e : SUCCESS in 59m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/92ab8f7c79b64247995a2813784abe59 : SUCCESS in 17m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/48ac8a13c55f465eb99348c67f9dde4f : SUCCESS in 26m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0727772051eb48379d2d5ebf8c5d16df : SUCCESS in 1h 07m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b1f1c676b0ed435d909004371cd5f546 : SUCCESS in 2h 15m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50dcebf974ec4e11af80e9ae58aecaaf : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4d8efb4f6e74fa7876e6702d437fbb1 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e4b53474ce141b3b70087f60c5292b9 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c0fe6257f5e4ec3bdc5e3dd589a03c7 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b89d5489044a4fd8a49d99e25011d11e : SUCCESS in 2m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01c4da8854fe46a89d80a048c1e3d6d2 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeab1fd199344a77a05d1d269ea2a6cb : SUCCESS in 4m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/072ba1ea9c284d909b24debcc96dd1a0 : SUCCESS in 22m 48s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/aab61aef6d81427da03103ef2170ea98 : SUCCESS in 1h 24m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/273586d7b7074c9387db92316e662160 : SUCCESS in 59m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c2c2562bab1d4cf5bcc39540926a03a6 : SUCCESS in 1h 38m 20s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/136815a0b6944c9d9c3ccc5b8a800ce4 : SUCCESS in 2h 10m 58s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/de0201ba668341aca3900a51391bc39f : FAILURE in 39m 15s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/f365ff17aab144dc9110ae1dd8186912 : FAILURE in 28m 12s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6c9c816b3df24df695897318d9770dc1 : FAILURE in 46m 23s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/af8ef3eba0474d3d9b77291e47b5600f : SUCCESS in 1h 18m 32s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/8b46f999e71e46fca694c21d48aac45b : SUCCESS in 23m 56s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/a574d7d95d0246f3a17bb1eb30970ec0 : SUCCESS in 1h 12m 44s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/f83df8b781b54143a375b5c9911399b8 : SUCCESS in 1h 18m 08s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/0f328ac3bb6f4a9d93300c716dfc4f45 : SUCCESS in 1h 46m 24s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/db916a75c7e442ce9f2fa4cd599f43e5 : SUCCESS in 28m 50s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f786c2b4134d4a6db34163921b378886 : SUCCESS in 23m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5ea07e1a2beb4053a7918beca23a8bd7 : SUCCESS in 26m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bba26feea90a4db28e38c44ebf1ade24 : SUCCESS in 1h 02m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2d21e64ecaad4b77a5b3c559e8b3b1bc : SUCCESS in 1h 21m 09s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/795f218085dd4a25945d5b9496e62c7e : SUCCESS in 1h 49m 31s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1d2dc32dd51d4da3a58b04aee287f976 : SUCCESS in 2h 09m 10s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/fe85bb1d083f4fb38b1a8f90158d1293 : FAILURE in 22m 40s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ae3f19c2c1ae4bbba2e3f2e9f44a10d8 : SUCCESS in 1h 27m 31s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/d708778e311745bfa27f43f5c32ad4ed : SUCCESS in 1h 23m 17s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/7cb3349d3d5349adac8a509e832fff53 : SUCCESS in 23m 18s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/abc7519c480146a7a877d13853e12077 : SUCCESS in 55m 26s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/77617c6af12e4b4fa3d5df102cc6d67c : SUCCESS in 1h 05m 33s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/58e3e1a5f21b47939f00b68f69f4fe3b : SUCCESS in 1h 35m 40s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/99a8f7a3b19f447f8eb8d571214cf5be : SUCCESS in 1h 06m 12s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b4b81074f5331eefba8ac080b154dd1a301fda9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-28 16:06:37.000000000","message":"Patch Set 11: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"d27ec8f208275553306d80c2300b9dbd70c76b8e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-28 16:07:50.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e42bb405936fbf85af49e3c4f54de51fa54a8547","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-28 16:08:01.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1 Review-Priority+2","accounts_in_message":[],"_revision_number":11},{"id":"89c81138234ae3db34db155d8add4697f872ba38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 19:08:07.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c76488ac58904dabacd7ff73c9617f0a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/61fa6e6280064d2c959794487d3f0440 : SUCCESS in 1h 18m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/83b6a6dfe6764590971db660cf41c645 : SUCCESS in 1h 45m 21s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/38033817473147efaf7a16ca709aeb46 : SUCCESS in 2h 57m 57s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b87decaf3a3542018cd3374bbf380e3d : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f380f4add28e4f398da9a1f64e8ab0ea : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7935b62f65ab4aa9965e7a8f152e36c0 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8ff813e8bf2548a98399f7950572ee5a : SUCCESS in 5m 05s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/84d1a38d47334d92b7c64d9210ca5d83 : SUCCESS in 1h 31m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/348a6a7244004b1cb2a68ce29efa031f : FAILURE in 1h 54m 28s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ea694a035d44af15227e862dba7b775d15aa1319","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-28 19:13:21.000000000","message":"Patch Set 11: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"ab11060e183d4d56f02594314cb8872dd0d53e8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 19:15:09.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5498f14f892a4de98f78284b74089c74\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b75a309fcd3942d89169ec9bb40b8cb2 : SUCCESS in 21m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3ab8059b2b994ae791f5a7925f73d9d3 : SUCCESS in 1h 07m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4adae960f21249f6b9368530ea76ffc3 : SUCCESS in 16m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/33a5637d4f4a41a8b4d87ef665484cf4 : SUCCESS in 31m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5a526613fc0f489b81a3113d407748a4 : SUCCESS in 1h 06m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3110b9a8aef845e28da7f2173cf7d455 : SUCCESS in 1h 22m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f30473eb39e46d9b24ed30af2603160 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec9c81c5fcf841daab331be9874a78af : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8bf921739634ae7a1a3615fe358b001 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b31f1ed86f949e8b1d7156d6fff59d1 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c23cb6e7aeaf46ccae8c9e7af09db10c : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17c6681ca1384f1394be43b19fb21382 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52c98593297e4b3d96ddb0cd349fd543 : SUCCESS in 6m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fa624ca9ef6f4b9eadda83fef383fb28 : SUCCESS in 18m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0bc1871965374adda5b5009359307d6f : SUCCESS in 1h 32m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ad8728eb40204e4ab988cc92313aba57 : SUCCESS in 1h 06m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e95b544ac8714786a4257620aa3cccbb : SUCCESS in 1h 31m 15s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/bf14a766a60f4605b5bff5b05455dcf3 : SUCCESS in 1h 52m 50s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ed947b0e7ef34423a020e8a385080f33 : SUCCESS in 2h 15m 30s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/415e2da1c6fc41f8be2f34687e414289 : FAILURE in 23m 06s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/0269237f6c1e41f895c454a2c452e489 : FAILURE in 43m 04s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/897caaba608b4124951dfd004f48e954 : SUCCESS in 1h 28m 16s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/fbf877e7b6e04d03abf4c4ad7335d7f8 : SUCCESS in 32m 49s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/2df38bb4587840b4a670df978b012c91 : SUCCESS in 53m 08s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a9602540725746efbf3da7884455bf6d : SUCCESS in 1h 11m 00s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/f15d1e59231441319f3d3cec63c22b21 : SUCCESS in 1h 28m 14s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3a924630fbdb415fb985a6c01f69e60f : SUCCESS in 36m 36s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f2c0d39dc4b04fbbbc6df87adbc325ad : TIMED_OUT in 3h 02m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9ebd19da765e44aa85a2e043515092c1 : SUCCESS in 29m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/de59762fb983433bbecf5f40c8be02b1 : SUCCESS in 59m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4d9be88e822c4ca88c5b185b12f5371b : SUCCESS in 1h 08m 50s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/58ec7ca97fe341edbd87543a92187c06 : SUCCESS in 1h 51m 07s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5cf7214e9a5642c3b7efc0bb597690e3 : FAILURE in 1h 12m 36s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/2bb2b0f3605c4745bb75b8609612cf77 : FAILURE in 25m 41s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/f4aeef9ec5374efab7608bd224b38df9 : SUCCESS in 1h 11m 48s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/74edc11dda7f40bebba896c56cee5ac6 : SUCCESS in 1h 05m 07s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/8cb404eca1944b228d46f3c05fcac153 : SUCCESS in 26m 44s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/ff0d7eaaab9941c0a7d04628af698d6f : SUCCESS in 1h 04m 22s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/4ac04217017c46f4a3febe7a4ccbbfdb : SUCCESS in 1h 02m 01s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/0c0361e1125a4fc493d17ac9478efbcf : SUCCESS in 1h 29m 54s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e8bb4c36d1424f0f91867b025418d17c : SUCCESS in 28m 29s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ff1abf41d5beb2535d61f0d4d24ab1e44621def0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-28 19:21:46.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"18eca2add6e741a2889b7659eb1c28855caa4a09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-29 04:58:34.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":12},{"id":"99c87424d7be52c1ad3738da883800ab414ed91f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-29 04:59:08.000000000","message":"Patch Set 12: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"795f8a4772fe668b2f48a97185009288b13d8c5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 06:48:25.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce085cadbb5647afa4755fa1687372e8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/99edea1a41764234936e2b8849d66570 : SUCCESS in 14m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/fea5947663b24d939a9a304e5d1c849f : SUCCESS in 1h 00m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a97ecb0cea5242ab9f543a3437bf19aa : SUCCESS in 15m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a59ad2cb10fb4a669eee4c1b36b15bb8 : SUCCESS in 25m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9f77a470562e41fa956a64827f77fb09 : SUCCESS in 59m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1589f647bc8542e58de42d49ad6b2f0b : SUCCESS in 1h 12m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb76017a309e4de1900ab529651824c1 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/668124badba24fc5ae713ef063560404 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4d2461e4d714be2af8cd6fd2c10ea5d : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c6c82cb61a14cf3895b7c91968dfaba : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d12145cc235e4b79a4ddc0c663e83675 : SUCCESS in 2m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6730700cf3845719f3760b4da24b457 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6718131917f14b50b1bfa6acedd2c9de : SUCCESS in 4m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6d721a135dd1418aa079db471c673407 : SUCCESS in 24m 04s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2f548235526d4ba3b489f82836768987 : SUCCESS in 1h 15m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7a2b1ab23efa417fb2b347cfd76528d8 : SUCCESS in 1h 11m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0625008c9f5447f0893a43b6fb997430 : SUCCESS in 1h 24m 51s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/b8095c78695f44c5a29eaa890830706f : SUCCESS in 1h 45m 20s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8153be4a74744bb9a4eea0d1a78c8a46 : FAILURE in 1h 15m 15s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/f37329650e754ae8b7f29efc5e882800 : FAILURE in 21m 21s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/662f295b5b43459b9d3a62d9a34bcb11 : FAILURE in 1h 02m 38s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/05abc2d21285457984a0c5a062f7f98e : SUCCESS in 1h 23m 42s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/752b44305fe343b3bd8c70123a497378 : SUCCESS in 28m 18s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/58e6dc2d1a9a4da38d7cceb6739252cc : SUCCESS in 1h 07m 31s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/03acaaa5dae64285b8fb2bf89288374c : SUCCESS in 1h 10m 39s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/86c71c37fb344b20bfd9ae079b4c8628 : SUCCESS in 1h 20m 03s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d9054efdbde54bb5a839bfaf1e87b564 : SUCCESS in 31m 42s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/056b051d8f7e491380aeb33e2888aa07 : SUCCESS in 18m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6a5a85a449ca4421b1f3f202a2530e14 : SUCCESS in 27m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/65349f04978e467dacbf5ec5b46ad596 : SUCCESS in 54m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c0f00fa4f73b4f42ac7d75cfe765823a : SUCCESS in 1h 08m 43s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/aa9724249b2848a7bf3cca0ca0a8d9bb : SUCCESS in 1h 15m 49s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c7461be5bb63446cbe5250408620e85d : SUCCESS in 1h 46m 37s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/f452ddb6c8a549779a9f768758e3062c : FAILURE in 16m 17s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/f7a4f54773ad4d0788059f774f6e05f1 : SUCCESS in 57m 34s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/30ed74ab946e4f75bad63f68ef231506 : SUCCESS in 1h 06m 12s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/60b3c60c1cc54e1e88a83b1aaec8a02f : SUCCESS in 21m 39s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/24b1529cb6da47b5bb8ee0a0fc95780c : SUCCESS in 47m 44s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7dfd28789a7c4dceaed9c3cc02aceacd : SUCCESS in 53m 41s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e9d400c791c046e0b67b73fb5323a97c : SUCCESS in 1h 19m 49s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/69ed114935b34f09bbbc77464ad2a60a : SUCCESS in 23m 59s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f6a03ee26acee428f367637a9feb8c7bd212f997","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-29 07:30:06.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"3776d525ea372c0fd01f8d3cca9315b91fc077f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 08:16:46.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e456fa39aee4031ab8fc14c72828e64\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/26ad3089177247568ceeee427ae785b1 : SUCCESS in 2h 16m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/876380c3c41a402795079c0f7970a3ea : SUCCESS in 2h 27m 50s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a9317202fd1d475284ac93a5c29ef94e : TIMED_OUT in 3h 02m 33s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd0a1a9f0ade49ec80aa3fa830f9db34 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fab291d93da94d97b00c55a48ad893c1 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/136af9f5af2e4e608d8967ee4b72396a : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5a7972a52c094673b204206681cc8460 : SUCCESS in 5m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7f761276571349f0ae97cf6fdd520872 : SUCCESS in 2h 04m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ff89182424994b1f999ca36d73031f5a : SUCCESS in 2h 05m 11s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"86be652cdbf4dd13d4a1f81d153a7c7e4ead9d19","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-29 08:28:56.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e3c7c781eff4b9bcda3492cc63421026b1d490a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 08:29:18.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"fa7fee0c9a67f4757db3580efab95cce7eeea17d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-29 10:02:29.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"8c5ed5ca5961ab933e56e17a24960410780de438","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 10:11:42.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd645566ded6457e81763f91c77445fc\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7606183ae5df4c08a279cf1458f708e5 : SUCCESS in 19m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0a3b872fe570464d8c36c531c345065c : SUCCESS in 38m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e867d62011654f3782506afa4427876c : SUCCESS in 1h 08m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/70073a8c16b740ba841b3a1ca6cc67dc : SUCCESS in 1h 38m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1973c5e8d8c4d818c2567598095f6c8 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/907d650efb824eb29baf8b38257b1c5a : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/478d6335038a4f61b8cdfee6124c5c0f : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/344f619ee87a44fbb15382963e35f552 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87ec1a7e6fe447e8ae52c6e301b0afdb : SUCCESS in 4m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/318bd3de685940f4a44e5e3840959db0 : SUCCESS in 26m 43s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7f977427516b44649c2c7ddbd1764452 : SUCCESS in 1h 32m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/28419cdb50ef4aa6a30fddeb85c8dc61 : SUCCESS in 1h 01m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/adabe24e2b844a27b28b57a22d2ecd4d : SUCCESS in 1h 32m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d1020c472e67448186007caf46725c72 : SUCCESS in 19m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b085a78e2f414e118bb2d48daa318941 : SUCCESS in 28m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dc16183e0bee4f92b8c354c909fe327b : SUCCESS in 1h 07m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2f08f3c3dfd44ab392bac69265c580e7 : SUCCESS in 1h 17m 31s","accounts_in_message":[],"_revision_number":12},{"id":"1098af98333bb365cbe6dc90f7b94718bf520ff4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 10:11:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"8edea62825cff42806a2e36d5a2716bda4d3341c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 10:12:53.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0f268cc571f49dd92cc9f5089b839d0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc6db03bc2514832af5a7538db4a2343 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"47b593d1e5531c693dde3e111ed3be91c60c41be","revisions":{"4d042670fe2835e6de22a6340dcadaa7f12dd700":{"kind":"REWORK","_number":1,"created":"2023-09-27 08:24:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/896616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/896616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/896616/1"}}},"commit":{"parents":[{"commit":"95a191e207b1e34fb9a26d8e1909153c0956e18a","subject":"Merge \"[docs] fix docker best practice link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95a191e207b1e34fb9a26d8e1909153c0956e18a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:24:22.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos","message":"rabbitmq: Use novemberain repos\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d042670fe2835e6de22a6340dcadaa7f12dd700"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d042670fe2835e6de22a6340dcadaa7f12dd700"}]},"branch":"refs/heads/master"},"ac02972dc7fa91d8e88ca680695bdd7835b6d4ee":{"kind":"REWORK","_number":2,"created":"2023-09-27 09:13:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/896616/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/896616/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/896616/2"}}},"commit":{"parents":[{"commit":"95a191e207b1e34fb9a26d8e1909153c0956e18a","subject":"Merge \"[docs] fix docker best practice link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95a191e207b1e34fb9a26d8e1909153c0956e18a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 09:12:21.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu\n\nCloudSmith is over budget, see [1].\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac02972dc7fa91d8e88ca680695bdd7835b6d4ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac02972dc7fa91d8e88ca680695bdd7835b6d4ee"}]},"branch":"refs/heads/master"},"576e99bf3b1fe48a80b99488ccd52dc532ae3a68":{"kind":"REWORK","_number":3,"created":"2023-09-27 11:15:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/896616/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/896616/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/896616/3"}}},"commit":{"parents":[{"commit":"95a191e207b1e34fb9a26d8e1909153c0956e18a","subject":"Merge \"[docs] fix docker best practice link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95a191e207b1e34fb9a26d8e1909153c0956e18a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 11:15:15.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu x86","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu x86\n\nCloudSmith is over budget, see [1].\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/576e99bf3b1fe48a80b99488ccd52dc532ae3a68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/576e99bf3b1fe48a80b99488ccd52dc532ae3a68"}]},"branch":"refs/heads/master"},"b8e5208562fa6fa3882f688c8c62e1193cd56d03":{"kind":"REWORK","_number":4,"created":"2023-09-28 08:38:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/896616/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/896616/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/896616/4"}}},"commit":{"parents":[{"commit":"95a191e207b1e34fb9a26d8e1909153c0956e18a","subject":"Merge \"[docs] fix docker best practice link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95a191e207b1e34fb9a26d8e1909153c0956e18a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-28 08:38:16.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu x86","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu x86\n\nCloudSmith is over budget, see [1].\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8e5208562fa6fa3882f688c8c62e1193cd56d03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8e5208562fa6fa3882f688c8c62e1193cd56d03"}]},"branch":"refs/heads/master"},"97fbd8438505367b9ec166bacda3997361444f76":{"kind":"REWORK","_number":5,"created":"2023-09-28 08:55:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/896616/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/896616/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/896616/5"}}},"commit":{"parents":[{"commit":"95a191e207b1e34fb9a26d8e1909153c0956e18a","subject":"Merge \"[docs] fix docker best practice link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95a191e207b1e34fb9a26d8e1909153c0956e18a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-28 08:55:04.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu x86","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu x86\n\nCloudSmith is over budget, see [1].\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97fbd8438505367b9ec166bacda3997361444f76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97fbd8438505367b9ec166bacda3997361444f76"}]},"branch":"refs/heads/master"},"db49544efcd65ccc2e97afdbfa1402458417ecbb":{"kind":"REWORK","_number":6,"created":"2023-09-28 10:26:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/896616/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/896616/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/896616/6"}}},"commit":{"parents":[{"commit":"95a191e207b1e34fb9a26d8e1909153c0956e18a","subject":"Merge \"[docs] fix docker best practice link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95a191e207b1e34fb9a26d8e1909153c0956e18a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-28 10:26:47.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu x86","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu x86\n\nCloudSmith is over budget, see [1].\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db49544efcd65ccc2e97afdbfa1402458417ecbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db49544efcd65ccc2e97afdbfa1402458417ecbb"}]},"branch":"refs/heads/master"},"b6aed14268703c6216adb546a89e8a184e1950ed":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-09-28 10:31:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/896616/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/896616/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/896616/7"}}},"commit":{"parents":[{"commit":"95a191e207b1e34fb9a26d8e1909153c0956e18a","subject":"Merge \"[docs] fix docker best practice link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95a191e207b1e34fb9a26d8e1909153c0956e18a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-28 10:30:51.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu\n\nCloudSmith is over budget, see [1].\n\nAs mentionedin in [2] - we\u0027re using amd64 packages on aarch64,\nsince the community mirror does not mirror aarch64 - and the packages\nshould be really marked as noarch.\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6aed14268703c6216adb546a89e8a184e1950ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6aed14268703c6216adb546a89e8a184e1950ed"}]},"branch":"refs/heads/master"},"130a59cfbb62e5d8a61b975c691ed93f2a7ba481":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2023-09-28 10:32:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/896616/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/896616/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/896616/8"}}},"commit":{"parents":[{"commit":"95a191e207b1e34fb9a26d8e1909153c0956e18a","subject":"Merge \"[docs] fix docker best practice link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95a191e207b1e34fb9a26d8e1909153c0956e18a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-28 10:31:58.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu\n\nCloudSmith is over budget, see [1].\n\nAs mentionedin in [2] - we\u0027re using amd64 packages on aarch64,\nsince the community mirror does not mirror aarch64 - and the packages\nshould be really marked as noarch.\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n[2]: https://github.com/rabbitmq/rabbitmq-server/discussions/9553\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/130a59cfbb62e5d8a61b975c691ed93f2a7ba481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/130a59cfbb62e5d8a61b975c691ed93f2a7ba481"}]},"branch":"refs/heads/master"},"3a420134e7ff2c0d6a829dfdf627d31e8da655be":{"kind":"REWORK","_number":9,"created":"2023-09-28 10:38:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/896616/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/896616/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/896616/9"}}},"commit":{"parents":[{"commit":"95a191e207b1e34fb9a26d8e1909153c0956e18a","subject":"Merge \"[docs] fix docker best practice link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95a191e207b1e34fb9a26d8e1909153c0956e18a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-28 10:38:49.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu\n\nCloudSmith is over budget, see [1].\n\nAs mentionedin in [2] - we\u0027re using amd64 packages on aarch64,\nsince the community mirror does not mirror aarch64 - and the packages\nshould be really marked as noarch.\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n[2]: https://github.com/rabbitmq/rabbitmq-server/discussions/9553\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a420134e7ff2c0d6a829dfdf627d31e8da655be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a420134e7ff2c0d6a829dfdf627d31e8da655be"}]},"branch":"refs/heads/master"},"fb1273bbca365d06494579a36d44de5eafa419b2":{"kind":"REWORK","_number":10,"created":"2023-09-28 13:24:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/896616/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/896616/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/896616/10"}}},"commit":{"parents":[{"commit":"95a191e207b1e34fb9a26d8e1909153c0956e18a","subject":"Merge \"[docs] fix docker best practice link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95a191e207b1e34fb9a26d8e1909153c0956e18a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-28 13:24:52.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu\n\nCloudSmith is over budget, see [1].\n\nAs mentionedin in [2] - we\u0027re using amd64 packages on aarch64,\nsince the community mirror does not mirror aarch64 - and the packages\nshould be really marked as noarch.\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n[2]: https://github.com/rabbitmq/rabbitmq-server/discussions/9553\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb1273bbca365d06494579a36d44de5eafa419b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb1273bbca365d06494579a36d44de5eafa419b2"}]},"branch":"refs/heads/master"},"8753401494f8e50b1593a4c54a98a28463ec602d":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2023-09-28 16:06:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/896616/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/896616/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/896616/11"}}},"commit":{"parents":[{"commit":"95a191e207b1e34fb9a26d8e1909153c0956e18a","subject":"Merge \"[docs] fix docker best practice link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95a191e207b1e34fb9a26d8e1909153c0956e18a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-28 16:06:37.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu\n\nCloudSmith is over budget, see [1].\n\nAs mentionedin in [2] - we\u0027re using amd64 packages on aarch64,\nsince the community mirror does not mirror aarch64 - and the packages\nshould be really marked as noarch.\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n[2]: https://github.com/rabbitmq/rabbitmq-server/discussions/9553\n\nCloses-Bug: #2037469\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8753401494f8e50b1593a4c54a98a28463ec602d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8753401494f8e50b1593a4c54a98a28463ec602d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"47b593d1e5531c693dde3e111ed3be91c60c41be":{"kind":"REWORK","_number":12,"created":"2023-09-29 04:58:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/896616/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/896616/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/896616/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/896616/12"}}},"commit":{"parents":[{"commit":"95a191e207b1e34fb9a26d8e1909153c0956e18a","subject":"Merge \"[docs] fix docker best practice link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95a191e207b1e34fb9a26d8e1909153c0956e18a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-29 04:58:34.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu\n\nCloudSmith is over budget, see [1].\n\nAs mentionedin in [2] - we\u0027re using amd64 packages on aarch64,\nsince the community mirror does not mirror aarch64 - and the packages\nshould be really marked as noarch.\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n[2]: https://github.com/rabbitmq/rabbitmq-server/discussions/9553\n\nCloses-Bug: #2037469\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47b593d1e5531c693dde3e111ed3be91c60c41be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47b593d1e5531c693dde3e111ed3be91c60c41be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
