)]}'
{"id":"openstack%2Fkolla~896851","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-10-03 10:38:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36","subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","status":"MERGED","created":"2023-09-29 13:47:32.000000000","updated":"2023-10-03 10:39:53.000000000","submitted":"2023-10-03 10:38:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896851","cherry_pick_of_change":896850,"cherry_pick_of_patch_set":1,"meta_rev_id":"22c514bc1ed897c43d43513c2e6ba69ba0f7826c","_number":896851,"virtual_id_number":896851,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-03 10:38:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-03 06:31:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-03 08:01:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-03 08:01:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-29 16:35:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-03 08:01:38.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"f112652f58addd378fd398cca6add18cb4e5b7ad","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-29 13:47:32.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.\n\nThe following files contain Git conflicts:\n* kolla/template/repos.yaml\n* kolla/tests/test_methods.py","accounts_in_message":[],"_revision_number":1},{"id":"224827c69dcca8e0dbd9335f6a692d9daf711339","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-29 13:48:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"98eb46e8648b0f490ed421a9fa0f5fdadb63ea4e","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-29 14:15:37.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"7164ed4655aa7de61130ccf36ad5bf1fa2afcac3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-29 14:30:31.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e40584f95091f3cc7ad15597797cc440e35eb1b2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-29 15:06:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"32c6898b68ffd32dfdcec2772e8d0bbcabe596ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 16:35:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa8eaf0c47e34ee39006884f5b3e107b\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/76296428c4e54b39a4aca98f9c9836d9 : SUCCESS in 39m 57s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/94d4fbf3d33543d5a9210f0ccd14a316 : SUCCESS in 34m 20s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/720b149b22c849a482d4c95ee8ff5c42 : SUCCESS in 29m 22s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9ac2560b3e9a44e99fcc2ba7887e35eb : SUCCESS in 1h 17m 41s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d8710fedf30c498d9d2ef246f7c89a2a : SUCCESS in 1h 06m 22s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/04dc13dc2b394fc598b4f95dc6807119 : SUCCESS in 1h 05m 37s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/694cf8e46cb54143961468614dae6995 : SUCCESS in 1h 57m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/25d94c842d784e5cbca67923fe9196f7 : SUCCESS in 25m 08s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c1a9ad65a7694fe58ea8a9cf2115722d : SUCCESS in 23m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f047bd73402042dd838506a0f772de20 : SUCCESS in 53m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/19a8bbe74c9b45d8acb6a8047b5dbe83 : SUCCESS in 1h 02m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/addec052eee8415282dcbb33c1953e20 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41b6d97a030949acb734e04d84934b05 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be09802c5cd241129def432acaa23bd3 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1582670562a74d7fa9873f7c08f48bdc : SUCCESS in 5m 24s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/96217a4f601f454596e12162fd2331da : SUCCESS in 26m 15s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/172d4cd336d340f6b627faf309e70737 : SUCCESS in 1h 09m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/22207d20e72a4d819355306a8f6742d4 : SUCCESS in 21m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6e89112cd9d64b4a834ae1814f1dda56 : SUCCESS in 19m 38s","accounts_in_message":[],"_revision_number":3},{"id":"cdd35b2ed74a27b9e569bf9d2c779a1b7d4a02be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 16:35:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ed901b34c704f8408034dbf806b6a785509f4a8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 19:51:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f374da6d2fdd4993a06716145b3ab3f9\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/14f5d68dd4c14ab98eebcc337e1fb1e8 : SUCCESS in 2h 16m 59s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/035acba45b3941d4a2bc077671da22b7 : SUCCESS in 1h 30m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/69801592019e43fb861314a584e710ac : SUCCESS in 1h 23m 29s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0fe53b9d32394896b41feabe2919b25a : FAILURE in 2h 16m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d30a04cba804776b401fdb77ab6a24e : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/49fb8866757f42faa7b29295303d20fe : SUCCESS in 5m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/88a7f9c1eaa04317a944daa92f5c7261 : FAILURE in 1h 52m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"796e6af2955b7084b99564c7e686a5c8a02106ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 20:59:34.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f0dd0cf566740b5b5589bb6d0ef0cf8\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/55fe87250d81482a8777a865e30b666c : SUCCESS in 36m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f0ed63dd273149e5b4da649ffbe3f9a0 : SUCCESS in 1h 13m 53s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3fee50942ce1431b9aa6362f1788e250 : FAILURE in 10m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/948d9e9888a14511924f287ca3a30bde : SUCCESS in 19m 20s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f6bb8e4a08f542cf9027e2d98aa23fed : SUCCESS in 1h 00m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2a6d0cc7726240b1a765a70d9defbb0d : FAILURE in 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42e9639a39ed4cca981272191a43ac8b : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c6f6b3855e449e8a40907883cbb3f8e : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d61f10f5b4a461cb00880f453698503 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2af4ac49851405aafbd4935338fbba5 : SUCCESS in 6m 47s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/eaaa8745c73f49759102003fc57739de : SUCCESS in 1h 08m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/02117910fe9a48bb9d8406984e84d4d1 : SUCCESS in 25m 39s","accounts_in_message":[],"_revision_number":3},{"id":"804d22e2dffd92ab1e94168239a59f32771ca66d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-02 05:43:07.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"5f8b80c1922ae23bf6e3f75922288aab13ad5f7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 05:43:28.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":4},{"id":"34bfd890219d9915f2a123f9ce6ad879746420e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 05:43:41.000000000","message":"Patch Set 4:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":4},{"id":"08bace0854e0ef3be4be7508b99c5bc2cca76aeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-02 05:51:01.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"52ef3ef2a32e3ada58425a99c8a7229e4ffd27d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 06:25:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a67e10e8b9cf401a9fac451b71d68b69\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/bf578ab123d14af79b239b62abcf2457 : FAILURE in 30m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a2e781cf71ef4f09905c9ca3b4212b5e : FAILURE in 31m 24s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/58c91e0b674446128a286525e2ca9689 : FAILURE in 19m 59s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d76225da34934689aee8369c5d53393b : FAILURE in 27m 41s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/18b4ad6514c5427f813bc22c8cbf9e8b : FAILURE in 24m 02s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/98759bcf81fa41e583af89a01905834e : FAILURE in 16m 28s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e50a5d14d08541e2b193f8ede4b0db13 : FAILURE in 23m 30s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e35a0e1fafac4f25ba967a0f509ad51e : FAILURE in 13m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ee2ec8c7594641358932a8cd0eaf08a1 : FAILURE in 20m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f8243acdc674e1a90d12eacf98a68cf : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edd9c58781e54917acea3df8f4747ac1 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87eaf7c7f8da4cd787c9d22032550025 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0e41b00719a4587b400074230f96b9f : SUCCESS in 4m 59s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/636dc43eb5654accae9134a3b22c0778 : FAILURE in 16m 49s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/dfe62fdd94304f4daccecd4b741700f8 : FAILURE in 22m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6dfec71616f8469bb145abd88e855a75 : FAILURE in 16m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ccd7667b1de44252a18f64c8d71076a6 : FAILURE in 24m 23s\n- kolla-ansible-ubuntu-binary-focal https://zuul.opendev.org/t/openstack/build/5027ec703f494b048e205d2f13d309e4 : FAILURE in 21m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"503b45183a30f9482625538369d10832b7e0fbec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 06:25:56.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/896993 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"1ba805473a7d036ebdd0ab281d10a15a7eece751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-02 07:18:35.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"86d65ae0d92bd4e31638014662d49eb45d48b335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 07:56:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c65a67bf51a4f72abfa08e6322b1df5\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4e5fa7c879a74fcd9621b74e576f503b : FAILURE in 27m 37s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2434562f8df54d7ea9c61d5eefeba87a : FAILURE in 30m 14s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/f6a90d28458e4e5f80b21a25f6ebd8d5 : FAILURE in 27m 32s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/63170951332e40cf8c196e6eedce925a : FAILURE in 30m 29s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8f7e1bba7898460fab421beabed66952 : FAILURE in 28m 50s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/4fde7b6006a54079a93ead368a449030 : FAILURE in 21m 30s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b72db8f2aaf04bbb9d8b1acb122ae108 : FAILURE in 24m 43s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e280217517764cbfa26d21412c1fc4e6 : FAILURE in 14m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/02bae344ada1437586c6b6519302ba6e : FAILURE in 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c869a4fd2f74833a9ab3a7d2d02266c : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ed7af85bceb4daa8110aa5e84d1cd3b : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86980c8513b9458aa7bb673bcdc3d9a8 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6282229f70334e0d917f4f726c570893 : SUCCESS in 4m 35s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/e15022d54e6d4098ac7491f509e428fc : FAILURE in 24m 23s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/85c1511e0328495c9a46b17b9150a424 : FAILURE in 21m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e277b11cc914a1491baab68701e413e : FAILURE in 18m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/28ac78db45d644c9a13275b12b520aea : FAILURE in 25m 29s\n- kolla-ansible-ubuntu-binary-focal https://zuul.opendev.org/t/openstack/build/aa6de37ddca14af881cdfd89be310910 : FAILURE in 24m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5228cc6874cd535020400b24b5f923a02a68634e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 08:54:43.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b7896e0317c466dbb9a543e962aad51\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5723028ae71e470ba22758347a782829 : FAILURE in 1h 33m 43s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a509b5aa32ae413aa23d2d8b956b0d84 : FAILURE in 1h 01m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/aee166c89d8a41fc8bb8eed0fe659a07 : FAILURE in 1h 14m 23s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6313584eced04fda8f4d3b69c0386354 : FAILURE in 51m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f21de8401e54fb9bc51e83c1af28a2b : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/421193a986a840b4a18bd44d85cd5985 : SUCCESS in 5m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/69b026a179a34730ad490ba08138725b : FAILURE in 59m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"eb283c78fe6a99681fccd11989d60421d37aabb0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-02 10:49:28.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e0620f2a26ef0d5e19de37533864b84cdc38176c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 11:48:03.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23587238aa39453ea31e04554c03c07a\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/eb69a2c784ea46c79b3c711546cae65e : FAILURE in 45m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6c9f00e0d5dd4cc9aa733e7a353b1e9b : FAILURE in 31m 37s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/f3d6815f4721478d96d3e56c4c02fc83 : FAILURE in 28m 41s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8b41a3a02f3542cf8f7d1419b1d1378d : FAILURE in 28m 25s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/88130305ef0a47659670825df6ed83a1 : FAILURE in 28m 18s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/10e65c1094f04f0695cb8e5e39b619c1 : FAILURE in 20m 32s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/23db93b54b414380bfd9c2bce7c7a699 : FAILURE in 18m 35s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d531383329414d0da9c050d1d6c91962 : FAILURE in 17m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f55e1b9e3bc74e6692ab898a99e6cbcc : FAILURE in 19m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9402527b4fc1469288ec6775908cda80 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/998c91b766de4c89af22d9f65e86254e : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd2e9361a0414aafbf2b30466cf37153 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b66fd7756834ed994e57ea2a8af893c : SUCCESS in 9m 30s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/6d4fe527070a42b98a1284aa9e34ac61 : FAILURE in 21m 11s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/c66f650cf98a44d0b890b191c1067974 : FAILURE in 23m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/79e424f24d9c40cf96e7e1f88f66067d : FAILURE in 21m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7632e51e718f45938727f5b6263977aa : FAILURE in 15m 45s\n- kolla-ansible-ubuntu-binary-focal https://zuul.opendev.org/t/openstack/build/c0daa1254bcc404e878b0e4c776ec12c : FAILURE in 27m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"46bfda63f800004e3005bedfd97751b9341c3c73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-02 11:59:11.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"7e364cc282534286360fae8d9a2c5f22029e6a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 13:34:16.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/68dd664ed6a347e7b309ddafa53a63f0\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3746b7bfd31943749e050f282d4504c0 : SUCCESS in 40m 09s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ce7725dc3b004881ad3dd5552ab810c3 : SUCCESS in 41m 49s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/f84d2937b1e0436e984b71c57e726887 : SUCCESS in 26m 28s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/497608043bdc4aeca914a8195db7b95c : SUCCESS in 1h 28m 12s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/96f996fb12df4d86927075986b56c64a : SUCCESS in 57m 50s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/10b8c89d539144fc9f5626149c822422 : SUCCESS in 1h 11m 31s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6ea3fe05b90c4b658c3dd3c535a93d2b : FAILURE in 4m 11s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2562d00ad8404e7e877bb2d4e755515d : FAILURE in 7m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7571731c066945c1ba85625de4779a4d : FAILURE in 12m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ab180409f124ece983e264192f140fe : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f88a1fd591143fdaf92b278f0a43f06 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/105a6b58af3b4494927c56040f78da19 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33b41b0c9f594415999128281153f648 : SUCCESS in 5m 24s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/997b8c1549684269bf11cd021ac657b1 : SUCCESS in 26m 36s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/9c19d8544a1f4fc2ab4ca82287b8545f : SUCCESS in 1h 08m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/730abce9e7574bd7971c1ae2f6c336da : SUCCESS in 22m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/788c98cf1fb24cfe9c47f9c620791144 : SUCCESS in 24m 28s\n- kolla-ansible-ubuntu-binary-focal https://zuul.opendev.org/t/openstack/build/7183cfa479e947ebad87254876420510 : SUCCESS in 56m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c8f51e6646c5ca123c215e1cc45ccab0d6be7f0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 16:49:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bbc4c3b5c9045e9874df49881b2a756\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0b7fa8e8795e4363aec5ad8e18f20dfc : SUCCESS in 2h 04m 12s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6b719d1f5c3343e1b197234c7431cfd5 : SUCCESS in 1h 38m 58s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/05682d1acb494504b2af9ba8e62ad361 : SUCCESS in 1h 41m 29s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b390f83753d46fc98fe13aa489ef56c : SUCCESS in 2h 34m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2ef13b953754b1c863d228be40b47b0 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c5cee5dd5dbf4cc9b75c48b521b83f5c : SUCCESS in 6m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5362da3933b6488e8ab3ce45eb963e6e : FAILURE in 1h 31m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bdee009cc57b963686863934d5e3fcc9332e7479","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-03 06:31:28.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"70f230d5138b1299c2b68fe6e4122cae01b2645f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 07:59:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22fda878b7844a1bba7eea175f8fe135\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f9ae0c8b978247aea974d769bbca6b77 : SUCCESS in 30m 58s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b514a5dd0f794472a465ad06d1ce39be : SUCCESS in 33m 03s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/4740018a12014dc1b6790b8d43a322ed : SUCCESS in 27m 16s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ad336fab64394741ad28cda270688c60 : SUCCESS in 1h 13m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8ea53654f40a49f2940bf3cb871ce7e1 : SUCCESS in 1h 13m 01s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/34ef2ae3e22846fd95fa0496cea54488 : SUCCESS in 1h 09m 28s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8b5323fa97064b9f96e01cb208a9cbf2 : SUCCESS in 23m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0f8c8e067838468785b3472c38205828 : SUCCESS in 18m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3116c62df4cd43e39a7e4918ff3e354f : SUCCESS in 49m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45f3d7c9b3bc409782cdcb2bd4d73ee4 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37bbc5b76b35433e8cf52beb1b21a04d : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d640abacd284b158a6b0d2a5853dddc : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcfa305d2bfe4e7a95f8e105845a2c9e : SUCCESS in 4m 23s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/eee3f69e2e8a4b32bf1d3c86a6dceb4d : SUCCESS in 19m 34s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/4c50cff7fb1b4234b41a92fd10042491 : SUCCESS in 1h 08m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/60e7af97d8a24402a642b29b4f5b1c7b : SUCCESS in 20m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a68b4f7fe33f423889258d2870636b3b : SUCCESS in 24m 12s\n- kolla-ansible-ubuntu-binary-focal https://zuul.opendev.org/t/openstack/build/d4e866697ad944bbac16ab5d70b92450 : SUCCESS in 41m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b6c15d3da2019c400ba3b1c74ee2e4c8403acb40","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-10-03 08:01:38.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"d27aa67a049cb632b1c5591409189223d59ec415","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 08:53:32.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e2974ee1898c4a397b9f8c26d0775329c81504d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 09:21:25.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8b412b0389e44f7839a483dd5d99eac\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d067591f1ec741bea82409ae9c78a89f : SUCCESS in 2h 16m 11s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/32e32c2f7273466aa9a39b70263c2544 : SUCCESS in 1h 30m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d2317be75fc644538ab4b22cd0f071ec : SUCCESS in 1h 40m 06s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/83bd5f4fe3c94bb3a570860cace1a123 : SUCCESS in 2h 47m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ab3f0caa92c4c2e989a1ec08c56dc4c : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29c7d457814b43f2a6df4f0c526a6b7f : SUCCESS in 5m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8290878e09fd4a78804d73c3fd264002 : FAILURE in 1h 23m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"405ce5e8dbde5eeb6e6f230a84b4dca8b570ac45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 10:38:09.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c1b4b7632a642ceb5a94f3f63db4565\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/cdeb01ba60644e7e90e46c491082dc91 : SUCCESS in 32m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/cb1590aaa2a749bcadcc19cc2a04aa0f : SUCCESS in 1h 17m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5fc45531fb274336af3691271fb18ca3 : SUCCESS in 17m 55s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7413e9bdd3a64080a7e0e3aeded593e4 : SUCCESS in 1h 16m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d8c456e5aad43f886b09e76d985c361 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be974d49df1241a7a2aa357a56518136 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31bdee80c1c64688b3488120bb120f64 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c66d311d3b4208926a196ef3ded4c0 : SUCCESS in 5m 54s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/397feca099d64646ab3359d8ccfe821b : SUCCESS in 1h 11m 21s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5e1d2a5cc43a4647a97b310d13e4eb2d : SUCCESS in 25m 42s","accounts_in_message":[],"_revision_number":7},{"id":"e1ada7873ed2a97e5543a6a2532c9f9b4bf6bedc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 10:38:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"22c514bc1ed897c43d43513c2e6ba69ba0f7826c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 10:39:53.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a18d7a0803ba4a92a69e3e8a62930d8e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b8b78688dfc490888b157ae75bf7e59 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"06f9ddef875ee8eee2c7165b90b1ba979367a3a0","revisions":{"d8a2c9dec3486215c73a77bae0adcb63cb11c995":{"kind":"REWORK","_number":1,"created":"2023-09-29 13:47:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/51/896851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/896851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/896851/1"}}},"commit":{"parents":[{"commit":"87f001fd3f6469ee08573f1e102e6897ae9dcb95","subject":"Merge \"Improve kolla-toolbox offline support.\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87f001fd3f6469ee08573f1e102e6897ae9dcb95"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-29 13:47:32.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu\n\nCloudSmith is over budget, see [1].\n\nAs mentionedin in [2] - we\u0027re using amd64 packages on aarch64,\nsince the community mirror does not mirror aarch64 - and the packages\nshould be really marked as noarch.\n\n[3] has also been incorporated in this backport\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n[2]: https://github.com/rabbitmq/rabbitmq-server/discussions/9553\n[3]: https://review.opendev.org/c/openstack/kolla/+/896927\n\nCloses-Bug: #2037469\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n(cherry picked from commit 47b593d1e5531c693dde3e111ed3be91c60c41be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8a2c9dec3486215c73a77bae0adcb63cb11c995"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8a2c9dec3486215c73a77bae0adcb63cb11c995"}]},"branch":"refs/heads/stable/yoga"},"46208eceacfced3c44c61badefee6c27cfb1ae3b":{"kind":"REWORK","_number":2,"created":"2023-09-29 13:48:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/51/896851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/896851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/896851/2"}}},"commit":{"parents":[{"commit":"87f001fd3f6469ee08573f1e102e6897ae9dcb95","subject":"Merge \"Improve kolla-toolbox offline support.\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87f001fd3f6469ee08573f1e102e6897ae9dcb95"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-29 13:48:23.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu\n\nCloudSmith is over budget, see [1].\n\nAs mentionedin in [2] - we\u0027re using amd64 packages on aarch64,\nsince the community mirror does not mirror aarch64 - and the packages\nshould be really marked as noarch.\n\n[3] has also been incorporated in this backport\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n[2]: https://github.com/rabbitmq/rabbitmq-server/discussions/9553\n[3]: https://review.opendev.org/c/openstack/kolla/+/896927\n\nCloses-Bug: #2037469\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n(cherry picked from commit 47b593d1e5531c693dde3e111ed3be91c60c41be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46208eceacfced3c44c61badefee6c27cfb1ae3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46208eceacfced3c44c61badefee6c27cfb1ae3b"}]},"branch":"refs/heads/stable/yoga"},"bc833c3e3dfd877ed0827d9b9e86fe1e8fc55faa":{"kind":"REWORK","_number":3,"created":"2023-09-29 14:30:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/51/896851/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/896851/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/896851/3"}}},"commit":{"parents":[{"commit":"87f001fd3f6469ee08573f1e102e6897ae9dcb95","subject":"Merge \"Improve kolla-toolbox offline support.\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87f001fd3f6469ee08573f1e102e6897ae9dcb95"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-29 14:30:31.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu\n\nCloudSmith is over budget, see [1].\n\nAs mentionedin in [2] - we\u0027re using amd64 packages on aarch64,\nsince the community mirror does not mirror aarch64 - and the packages\nshould be really marked as noarch.\n\n[3] has also been incorporated in this backport\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n[2]: https://github.com/rabbitmq/rabbitmq-server/discussions/9553\n[3]: https://review.opendev.org/c/openstack/kolla/+/896927\n\nCloses-Bug: #2037469\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n(cherry picked from commit 47b593d1e5531c693dde3e111ed3be91c60c41be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc833c3e3dfd877ed0827d9b9e86fe1e8fc55faa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc833c3e3dfd877ed0827d9b9e86fe1e8fc55faa"}]},"branch":"refs/heads/stable/yoga"},"dfc26a0aec1b0b7cdb2c8ad4f423643287f62b09":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-10-02 05:43:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/51/896851/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/896851/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/896851/4"}}},"commit":{"parents":[{"commit":"ec70456a95a5cdcd89e4cc834bf67701268112de","subject":"CI: Drop upgrade jobs since Xena is EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec70456a95a5cdcd89e4cc834bf67701268112de"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-02 05:43:07.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu\n\nCloudSmith is over budget, see [1].\n\nAs mentionedin in [2] - we\u0027re using amd64 packages on aarch64,\nsince the community mirror does not mirror aarch64 - and the packages\nshould be really marked as noarch.\n\n[3] has also been incorporated in this backport\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n[2]: https://github.com/rabbitmq/rabbitmq-server/discussions/9553\n[3]: https://review.opendev.org/c/openstack/kolla/+/896927\n\nCloses-Bug: #2037469\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n(cherry picked from commit 47b593d1e5531c693dde3e111ed3be91c60c41be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfc26a0aec1b0b7cdb2c8ad4f423643287f62b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfc26a0aec1b0b7cdb2c8ad4f423643287f62b09"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"7eb17137fecfdcee3d73d82311af08866ba24ae9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-10-02 05:51:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/51/896851/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/896851/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/896851/5"}}},"commit":{"parents":[{"commit":"61fb179c4cbefd44dff79e5435445e36a172e5ba","subject":"CI: Drop upgrade jobs since Xena is EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61fb179c4cbefd44dff79e5435445e36a172e5ba"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-02 05:51:01.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu\n\nCloudSmith is over budget, see [1].\n\nAs mentionedin in [2] - we\u0027re using amd64 packages on aarch64,\nsince the community mirror does not mirror aarch64 - and the packages\nshould be really marked as noarch.\n\n[3] has also been incorporated in this backport\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n[2]: https://github.com/rabbitmq/rabbitmq-server/discussions/9553\n[3]: https://review.opendev.org/c/openstack/kolla/+/896927\n\nCloses-Bug: #2037469\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n(cherry picked from commit 47b593d1e5531c693dde3e111ed3be91c60c41be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7eb17137fecfdcee3d73d82311af08866ba24ae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7eb17137fecfdcee3d73d82311af08866ba24ae9"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"3d5ee682f553c713461519def891b238dd8f9f22":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-10-02 07:18:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/51/896851/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/896851/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/896851/6"}}},"commit":{"parents":[{"commit":"c2be45ca14f1ddeed47be35179c7e1f5c7fce358","subject":"CI: Drop upgrade jobs since Xena is EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2be45ca14f1ddeed47be35179c7e1f5c7fce358"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-02 07:18:35.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu\n\nCloudSmith is over budget, see [1].\n\nAs mentionedin in [2] - we\u0027re using amd64 packages on aarch64,\nsince the community mirror does not mirror aarch64 - and the packages\nshould be really marked as noarch.\n\n[3] has also been incorporated in this backport\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n[2]: https://github.com/rabbitmq/rabbitmq-server/discussions/9553\n[3]: https://review.opendev.org/c/openstack/kolla/+/896927\n\nCloses-Bug: #2037469\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n(cherry picked from commit 47b593d1e5531c693dde3e111ed3be91c60c41be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d5ee682f553c713461519def891b238dd8f9f22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d5ee682f553c713461519def891b238dd8f9f22"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"06f9ddef875ee8eee2c7165b90b1ba979367a3a0":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-10-02 11:59:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/51/896851/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/896851/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/896851/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/896851/7"}}},"commit":{"parents":[{"commit":"1f11f5c23c92ed175e2e1b8c7e7e51810dfdec47","subject":"[yoga-only]: Fix kolla-toolbox after Galaxy migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f11f5c23c92ed175e2e1b8c7e7e51810dfdec47"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-09-27 08:12:10.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-02 11:59:11.000000000","tz":0},"subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","message":"rabbitmq: Use novemberain repos for Debian/Ubuntu\n\nCloudSmith is over budget, see [1].\n\nAs mentionedin in [2] - we\u0027re using amd64 packages on aarch64,\nsince the community mirror does not mirror aarch64 - and the packages\nshould be really marked as noarch.\n\n[3] has also been incorporated in this backport\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/9526\n[2]: https://github.com/rabbitmq/rabbitmq-server/discussions/9553\n[3]: https://review.opendev.org/c/openstack/kolla/+/896927\n\nCloses-Bug: #2037469\n\nChange-Id: I83179ce5d53fed6cfe7f5fcbf6fdecd7199baa36\n(cherry picked from commit 47b593d1e5531c693dde3e111ed3be91c60c41be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06f9ddef875ee8eee2c7165b90b1ba979367a3a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06f9ddef875ee8eee2c7165b90b1ba979367a3a0"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
