)]}'
{"id":"openstack%2Fkolla~897843","triplet_id":"openstack%2Fkolla~master~Ie117d68310f1093d80cf1a99cbae85d9cd265112","project":"openstack/kolla","branch":"master","topic":"workaround_ansible_bug_81945","attention_set":{},"removed_from_attention_set":{"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-10-11 08:37:25.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}},"hashtags":[],"change_id":"Ie117d68310f1093d80cf1a99cbae85d9cd265112","subject":"fix broken ansible-core version","status":"ABANDONED","created":"2023-10-10 14:50:56.000000000","updated":"2023-10-11 08:38:51.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"0740c81dcad368d29333ae90dbf9e5d05af6d8cb","_number":897843,"virtual_id_number":897843,"owner":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-10 16:36:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 16:36:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f068241ecb6f6b27bc3a8ffaa6583ae6f7a4d26a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-10-10 14:50:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b69cc1afe0aa93d9d2046381eba235ecb60e0ad3","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-10-10 15:05:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9890e159334fd411a7730e7872fbd817b7fde2ab","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-10-10 15:23:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aa037a90b8df4f17cd26e92a25c178da62cc4d81","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-10-10 15:51:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"529f1067848b028fa8d59799c202c4e24fdef2b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 16:36:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1c7e8491c9043fd90a2bb25149ae4a7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e42e27f311fb4d4c984cdf27dc458f6f : FAILURE in 23m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/de27ceb71c394e1c96eef2f8b6c5dcbe : FAILURE in 19m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d97e12c90954419e80f44240af0595da : FAILURE in 26m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9e5fef35401846479d51f824368b6b0c : FAILURE in 9m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/690c0b19e4054bc0a1fb06b412fea548 : FAILURE in 26m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3829954a25794eebb9d0df5e6f9c017d : FAILURE in 30m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a0e2fa0d0af4dd7871562b658454f4e : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/811075483717488c88e3e4951b8acb2d : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b645ffc08614124bd89191b4c9ae663 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7494294a1e7d471d91e27974ffecbec3 : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b3463f6851448eeb5c3e60b9fdf4261 : SUCCESS in 3m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/163d9e0cefcf4589bd2a252adc9648a8 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b7a5a9b6fb541a280951ff516c9e6f8 : SUCCESS in 5m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a0dd680c70ec4ac98264772ee030849d : FAILURE in 22m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d33b27d485aa45a8b736a9968e403087 : FAILURE in 1h 39m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7b4bc3cd15d941b78bef994d89373962 : FAILURE in 21m 47s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/84024a3c32654b9ea0864c2b9a1b3d69 : FAILURE in 26m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0fe18a05d2b34ec98e6102094746ac46 : FAILURE in 23m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/58dcc4371a2249f78960908eab81d5c9 : FAILURE in 30m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/598a23bcec0348f19484b74fc4934fcf : FAILURE in 18m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ec5a5f5e573946709c9e5b628b882114 : FAILURE in 22m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2514d1537d685b7e1d3da32951527c96e2241355","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 17:32:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fa0fb84c9614d1c8c3aa18fd42407fa\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/dad150aab11e4df68d3a11bf22a2d35a : FAILURE in 1h 43m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f79aca0166934543a1b13e76445997f9 : FAILURE in 1h 21m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cd225e62345c42e7b6897ca36eb73a2a : FAILURE in 1h 24m 27s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4f8ef6344ce4ef4b1afd419b858ca46 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4768098cf74c4cdb80685767f2d39f72 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f138d0cafc1045faaeaa12c87f050885 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/984691ade9a74a9eab2adfde2d468a97 : SUCCESS in 4m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/edffc0aef34b4163a99b2ef66b4ca952 : FAILURE in 1h 38m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e19c47db1ab94c489653abe431dd3f5a : FAILURE in 1h 18m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02c143f27cd74d4a065847abd458771f9bbe9a20","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-10-11 08:37:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0740c81dcad368d29333ae90dbf9e5d05af6d8cb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-10-11 08:38:51.000000000","message":"Abandoned\n\nsuperseeded by: https://review.opendev.org/c/openstack/kolla-ansible/+/897926","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"464c5db4b2341f40e8b81d1369bc0d4a0acfb2f2","revisions":{"464c5db4b2341f40e8b81d1369bc0d4a0acfb2f2":{"kind":"REWORK","_number":1,"created":"2023-10-10 14:50:56.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/43/897843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/897843/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/897843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/897843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/897843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/897843/1"}}},"commit":{"parents":[{"commit":"a1261160da74e39b066401e315f19134d5b5acf6","subject":"Debian: Switch from antelope to bobcat APT repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1261160da74e39b066401e315f19134d5b5acf6"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-10-10 14:49:20.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-10-10 14:49:20.000000000","tz":120},"subject":"fix broken ansible-core version","message":"fix broken ansible-core version\n\naccording to: https://github.com/ansible/ansible/issues/81945\n\nkolla breaks with ansible-core\u003d\u003d2.14.11 and works with 2.14.10\n\nwe see the same error message in various build pipelines\n\nChange-Id: Ie117d68310f1093d80cf1a99cbae85d9cd265112\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/464c5db4b2341f40e8b81d1369bc0d4a0acfb2f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/464c5db4b2341f40e8b81d1369bc0d4a0acfb2f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
