)]}'
{"id":"openstack%2Fkolla~897924","triplet_id":"openstack%2Fkolla~master~If374f086e54be5b9799db7661c88b0204c4f507c","project":"openstack/kolla","branch":"master","topic":"workaround_gnocchi_unbuildable_bug_1345","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-10-11 16:40:47.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-10-11 18:31:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If374f086e54be5b9799db7661c88b0204c4f507c","subject":"mark gnocchi-base as unbuildable","status":"MERGED","created":"2023-10-11 07:59:05.000000000","updated":"2023-10-11 18:32:09.000000000","submitted":"2023-10-11 18:31:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897924-workaround_gnocchi_unbuildable_bug_1345","meta_rev_id":"387bb92ff392806af8ccc0022562296f16a7f156","_number":897924,"virtual_id_number":897924,"owner":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 18:31:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-11 16:40:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-10-11 16:40:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-11 16:40:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-11 09:22:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-11 13:07:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"2855cd8a0fddf5618422150e10dce7e682462d66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-10-11 07:59:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd266e48407e68f9eb3b35860fadc343e9c01eae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 09:22:24.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b0736730d8c4430bd0efb0dac32089c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ad5999b98c5447ed99b6e5c5e4809c26 : SUCCESS in 18m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/08f8e9f5b7da4f32a22d790fec1d6f6e : FAILURE in 27m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/33b1e574fad44eb88c8ffe59e1c73d23 : SUCCESS in 21m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/eea5df7381c14b938b90a2c76818dbcc : SUCCESS in 34m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1abb17168f1944aba2c45f9a0b13011c : FAILURE in 31m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/df7157ea945b482f88e71a0350aa50d6 : FAILURE in 35m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3169466027054d9fb2b4ae48c609c813 : FAILURE in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd02c68e95854fdaa01b31a0e275fff9 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa7abae013ba416bb06447891849452b : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4d1b68b169c416cbbdbdc5349d1e803 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ccf7eab058641f8a68b1288aa254373 : SUCCESS in 2m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2e6a7f0d1d64e08b5bb62ab4e854cfb : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f213be24fd949c58e602e583e4ba0de : SUCCESS in 5m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/693303f357464abab4281ddcdea3cb92 : SUCCESS in 21m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ddf9dd6190824c9ea8cc6a85fad73eae : SUCCESS in 1h 18m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f06526044728453d85f7ecca9fe947d8 : FAILURE in 23m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0d265c42f4d941849eca3d852925b1fe : FAILURE in 33m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3472f8c7ed0049e28fc79ec7884f5baf : SUCCESS in 19m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/07f62c0835364620823bf8dfbdb5d874 : SUCCESS in 28m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8bf2df574d214c7a97a21c28e455518d : FAILURE in 20m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f55256e283c541a1b3661ddfc21073a2 : FAILURE in 26m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b2dd33596fb85739c1f92595d2e6be8f175d4a9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 09:44:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/477e250a73bd41278947484a734699e1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c62f1e4e40c5468c8affbaba8b69249f : SUCCESS in 1h 14m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0a932af532824f43af73de9f42eebe6d : SUCCESS in 1h 19m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/32e9243b2b834a07addb976bd2da3fce : FAILURE in 1h 11m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e798d5c3464d4b4093a4dbddc2809a23 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98607d6bf2964c7fa511b4a6b4359910 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f80f7a9a973143d9b7657b8864d37c4a : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cc50c3d22532462884d80340058c8ac5 : SUCCESS in 4m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d666f302bb204bbe8544e8489df4595f : SUCCESS in 1h 13m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f82c3c40efc349efa157f01f37ece3df : SUCCESS in 1h 04m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0a14e2ee0c482f25631f5b171dff2d92a9f3bac","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-10-11 10:01:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77530a4edc1914d5e93c44cc04e32ac018489eea","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-10-11 11:45:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c33a5af1eabbd42a04b15c6dad1e3a808b00e73e","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-10-11 12:46:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68ec6c436700a9847eb23a5612504efa8e927bd4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-11 13:07:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae8b0516d3e3597ee34c0f6d9a331c456182a798","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-10-11 13:20:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8399be74bd94afdbfb9bcba90ff739634ab82b44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-10-11 13:27:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"42bf66e8a5b7f23412afb1a67d12c73ed464220b","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-10-11 13:32:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"20e2d8dbdf7e6cfae5167c1216aca22790f41355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 15:09:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e81229f173649f0bf9fb7d6dbc860c2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ac33172c9dd74d94a117fad53a1416f4 : SUCCESS in 27m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b2eb40ab4cd142e99b4bcc9ac2d5c5d2 : SUCCESS in 1h 05m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f0362a5037df4dd1a7242d067b6743c3 : SUCCESS in 21m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/86113e01c4f0409385d07778aa4c3226 : SUCCESS in 30m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fe9a2a0f58324d5b994097585e178cb6 : SUCCESS in 1h 07m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4394c81a0d6b4480af52c1f901258932 : SUCCESS in 1h 32m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15bce21ce245434788150e702e7aa089 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/005a3e5c2bf543d8812f2e0082df567b : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b19e975785084096931b4708ac12db2a : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f208d1f78f864a718ac5cf11b7e6b319 : SUCCESS in 2m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5cf4e0105cfd46db8f3a85f63d83e1ee : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fc24127627348d9a3c9e8b76fc1c40e : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dadfd211c00045a9b104e61efea2753e : SUCCESS in 4m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/db71426000564520af1c10f16dcc5c9e : SUCCESS in 24m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d0ebc53a31d3449f89c490062ed0b7fe : SUCCESS in 1h 09m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/addd38a3e81e48a68d0be2fa764103b5 : SUCCESS in 1h 11m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/59cad3c490094ce59991ae8e68f87b2b : SUCCESS in 1h 21m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8ddc74b35e8c41f287548fe1ac981f59 : SUCCESS in 22m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8e9ab73b1b6c4a2e8b0960f2fdf8d160 : SUCCESS in 22m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2361338bec0b4c2d9c4e6fdd422c2aec : SUCCESS in 1h 02m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ecdb189dfe354662a6626f00d6237478 : SUCCESS in 1h 22m 00s","accounts_in_message":[],"_revision_number":3},{"id":"ad34a25711cc6257d51d8c4685176079dd0759af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 16:14:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/206bc3903e854933a4fa6a6025255d31\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b80eed3e62a945ea85ea9c2f9e1efcf0 : SUCCESS in 1h 25m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e8029c93b4704b37ba22882796a42368 : SUCCESS in 1h 16m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3835ec423c634fb7ac69c17d5488fdc5 : SUCCESS in 2h 43m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e08614e5fcde4d03bf9352cd1874faae : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/03a15426e2d84c2788963ea9dfd9a382 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b1764bfdf94a4311807c28c6e722b01e : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a3d5b78d50de4430bfb9087daf10d73c : SUCCESS in 5m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ae4999db1a93450499e1f059b4c46bb2 : SUCCESS in 1h 20m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/80b3ce4bc5544438900972494e049e4f : SUCCESS in 1h 22m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"90d0301d4460a9dc0ff2ea5ac58992c665e9dbc9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-11 16:40:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1 Review-Priority+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"58342cf2f12760aba493e3a64894dd47f500b7e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 16:41:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7b23fbf75cfdbf3e67c299f1daffacbbec5fe44c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 18:31:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e2514a1a99b46ea90d7dd23d5f0a01b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9ffdcff15937419f8293e134a50c4f91 : SUCCESS in 22m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/92490750da234eed84473115b6446a8d : SUCCESS in 37m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d31315a2c48f448d8c90c8d81f883485 : SUCCESS in 1h 07m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7969f27f04f5468b879671d78b0a40bf : SUCCESS in 1h 31m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/256e2507b68e4dd189cbaed07815e6f8 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1abe49ecd4a34db2871be3ead06dd3d1 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32f4a0409b904db8887e69180e0b8e76 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4f4f095c03b41e58f43bcd20d4ac35b : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca4e6fb55e334dfcbec46a693e698e47 : SUCCESS in 5m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/db00ea98909b41ee9329a3130ddd14eb : SUCCESS in 19m 51s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c731476dcfcc479b85ecb3f16f70fccd : SUCCESS in 1h 08m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/39b08838209248eeb98d523ef61c87e2 : SUCCESS in 1h 18m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/37b3ddf5487c423fb1f750993d5b4c9e : SUCCESS in 1h 43m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5fcf762959f242af88ebc06143148788 : SUCCESS in 28m 35s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/25d3e3b5957d43bbbdf3a9208417f9dc : SUCCESS in 27m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8674015d265541da97ae151b8c8953c5 : SUCCESS in 58m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1bdef16a75584a3881dcee72fc8d222d : SUCCESS in 1h 23m 38s","accounts_in_message":[],"_revision_number":3},{"id":"4837dee8573b3a95f5e69927a42f7ff47876b528","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 18:31:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"387bb92ff392806af8ccc0022562296f16a7f156","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 18:32:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8783e41a9ff8487f8f40b75636f5a036\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d7824205a046b0aa8c8b17243dd2a2 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e4df04e9615b19e8fb0bcd1b192928815b027b82","revisions":{"86c74656b0a3d699673d1626223fdb8ea2630b17":{"kind":"REWORK","_number":1,"created":"2023-10-11 07:59:05.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/24/897924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/897924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/897924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/897924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/897924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/897924/1"}}},"commit":{"parents":[{"commit":"a1261160da74e39b066401e315f19134d5b5acf6","subject":"Debian: Switch from antelope to bobcat APT repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1261160da74e39b066401e315f19134d5b5acf6"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-10-11 07:58:16.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-10-11 07:58:16.000000000","tz":120},"subject":"mark gnocchi-base as unbuildable","message":"mark gnocchi-base as unbuildable\n\nsee upstream issue for the details: https://github.com/gnocchixyz/gnocchi/issues/1345\n\nChange-Id: If374f086e54be5b9799db7661c88b0204c4f507c\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86c74656b0a3d699673d1626223fdb8ea2630b17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86c74656b0a3d699673d1626223fdb8ea2630b17"}]},"branch":"refs/heads/master"},"35d80f0ea425f99310da7455e0d8d9e32cf1d47f":{"kind":"REWORK","_number":2,"created":"2023-10-11 13:20:44.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/24/897924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/897924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/897924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/897924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/897924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/897924/2"}}},"commit":{"parents":[{"commit":"a1261160da74e39b066401e315f19134d5b5acf6","subject":"Debian: Switch from antelope to bobcat APT repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1261160da74e39b066401e315f19134d5b5acf6"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-10-11 07:58:16.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-10-11 13:19:32.000000000","tz":120},"subject":"mark gnocchi-base as unbuildable","message":"mark gnocchi-base as unbuildable\n\nRelated-Bug: 2039051\n\nChange-Id: If374f086e54be5b9799db7661c88b0204c4f507c\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35d80f0ea425f99310da7455e0d8d9e32cf1d47f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35d80f0ea425f99310da7455e0d8d9e32cf1d47f"}]},"branch":"refs/heads/master"},"e4df04e9615b19e8fb0bcd1b192928815b027b82":{"kind":"REWORK","_number":3,"created":"2023-10-11 13:27:34.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/24/897924/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/897924/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/897924/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/897924/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/897924/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/897924/3"}}},"commit":{"parents":[{"commit":"a1261160da74e39b066401e315f19134d5b5acf6","subject":"Debian: Switch from antelope to bobcat APT repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1261160da74e39b066401e315f19134d5b5acf6"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-10-11 07:58:16.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2023-10-11 13:27:16.000000000","tz":120},"subject":"mark gnocchi-base as unbuildable","message":"mark gnocchi-base as unbuildable\n\nRelated-Bug: 2039051\n\nChange-Id: If374f086e54be5b9799db7661c88b0204c4f507c\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4df04e9615b19e8fb0bcd1b192928815b027b82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4df04e9615b19e8fb0bcd1b192928815b027b82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
