)]}'
{"id":"openstack%2Fkolla~898107","triplet_id":"openstack%2Fkolla~master~Iced4520e4cd8811c0e16227e130e094f5eba23e8","project":"openstack/kolla","branch":"master","topic":"bobcat-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-10-17 11:36:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iced4520e4cd8811c0e16227e130e094f5eba23e8","subject":"[release] Use Bobcat sources by default","status":"MERGED","created":"2023-10-12 09:53:26.000000000","updated":"2023-12-20 17:04:34.000000000","submitted":"2023-10-17 11:36:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898107-bobcat-release","meta_rev_id":"2e9ef7fdd63f04ed5181a4b5c89455c4c3b56d06","_number":898107,"virtual_id_number":898107,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-17 11:36:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-17 10:03:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-10-12 10:21:26.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2023-10-12 13:58:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-10-17 10:03:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-12 10:21:26.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-10-12 13:01:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 13:58:33.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-10-17 10:03:09.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"6d3042a21f09587a781658ba2bf659d9d4d5a876","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-12 09:53:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c154a3b9d24b7b18980690c1e5e3c311fc9b2836","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-10-12 10:21:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0144032365e035aee764b7416091af8cb992fdd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 13:01:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3e5e2eecfd6419a8ffd917e8ec3df5c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3008251b4a4c49a0b2cb28a841f00e76 : FAILURE in 21m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/96dae7f9b1fa4d9ab463bcd78385a81f : SUCCESS in 1h 02m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/abada215e6cf4782ba8e5ca6b472aff7 : SUCCESS in 17m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8169d89f9c0e4681aa02e6e43d7576ec : SUCCESS in 23m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7526d66707f842929ecb35cb4dc77052 : SUCCESS in 1h 01m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0868b3ca5ed546b5b94e7adb384a8fa6 : SUCCESS in 1h 32m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22647b32f36e4ba190044117df33dfdc : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79f80ac3f96144eea0f7ff4368c8c1b2 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a15b02e6b484a3b9b4591d8a688c5e8 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1207d472c53845f783d5f42fc69514f5 : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0ef76c9a421464997324356c2f0a8d4 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca65e3be5f574db29d8bc7abef282a2c : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c635746691c746e796931c35c527c9a3 : SUCCESS in 4m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/05ac72f87f464d119169b0e70e2f4e8c : SUCCESS in 18m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3d27f2f046e64c7a9f45e077d1222e89 : SUCCESS in 1h 21m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f22d60e1012d47d48504f80c79f0e54f : SUCCESS in 1h 13m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3b021cf372d3450a9afbc847ebcd9422 : SUCCESS in 1h 33m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/781f3108a3c94ef888c1864bd71340ee : FAILURE in 20m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/de41f668a8f54c2ca353c2632fe07074 : TIMED_OUT in 3h 01m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7fda24788fd44f519508b07b1001a147 : SUCCESS in 1h 00m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b6774e71a40d43b4af675880d6dccd30 : SUCCESS in 1h 16m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b57182003cc11aa1d0ee21bfcb25147e6a5a73ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 13:38:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bd54add8b5544a1992766f4b600d564\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/67aa8a166da14f2fa1640971f4db6261 : SUCCESS in 1h 46m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/15c5c777360d4bcd942253b2e2e8c44e : SUCCESS in 1h 32m 13s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/08d2d04faa72491081d26633d423bdb7 : SUCCESS in 2h 36m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d65278c9e6f4c27b63941f554aff97d : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cdd03c05170747dc8f73c3b1f7ff2984 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f62204b28c0f4590a322e55556870d4d : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a84ce4a5190340d8ab2acad1ad8a2469 : SUCCESS in 4m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1da17f94eeda43b482b0ece12cc0e1fe : SUCCESS in 1h 29m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3a42433e871a4b6994aa4282f886233d : SUCCESS in 1h 30m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"18b4be19e11bc922b227c2c3e1ccd5a775e4f7fc","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-10-12 13:58:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ab8870b2af57a0675477ab248849d48bde847fbd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-14 08:02:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ada29975718393d60e93755f0ae8cc2b9f48698","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-14 09:46:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e175bad9c1f04bb7bfeda9f9924e61d8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/99fec2fbe81144c88e5b7983cb79ebf8 : SUCCESS in 23m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4ce65ca71e4f4d77851187c11b907ea9 : SUCCESS in 1h 00m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8025f8a385a7499fa21f6d83812246d4 : SUCCESS in 18m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0540e5d9b54f4726bc37e800b0bcbc95 : SUCCESS in 31m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c5229834db624090a85c78684a722c57 : SUCCESS in 55m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fe9b68b19407431fbd23653d5418d94d : SUCCESS in 1h 34m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af3b1df72ed544828ba185b027a3ef33 : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37cc25f0e56b4c34a778b815f5db0dda : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f29151fcfc7f48e2b20315ea96fc28b0 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a030ab8b5be348d99067f0ca606cb80a : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4eae727d0be4753bf33a8707b480225 : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d33f512a6e2748ca9fe21dac3cdbc931 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7914841f311437dbc6cddc0b5de3bc8 : SUCCESS in 4m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0561af6bff884f459a62962b89a4bb5f : SUCCESS in 18m 34s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/dd4dc42c4bb540e28234fdc8ba44595f : SUCCESS in 1h 27m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9d3686e81ac047daa16a965a8f3590b2 : SUCCESS in 1h 07m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d55456acbeb946b5a984a6b25c6f70cd : SUCCESS in 1h 29m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b245b5d947544a1a87da8054fb997d97 : SUCCESS in 18m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/defb9e775a40496684f6bc59fb562e49 : SUCCESS in 26m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cf52ff11513a46e096ba83d01ff13e73 : SUCCESS in 1h 00m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fe2983e39ee841bc875d7e047dc36a7c : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":1},{"id":"da13f238fa360ed9cdb7008511bb3c15414c2e2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-14 10:24:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78cfa7621154431cb7abc06735495d6e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d1af71ff45d849e294cc2774b579971d : SUCCESS in 1h 18m 19s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/deb914bd54b745f98bed37ec3996fbf7 : SUCCESS in 1h 17m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/61c0d7adf5c94c84a204d673e7f9b467 : SUCCESS in 2h 20m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b878f100f19435ba098b349406964a8 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0872fbbab7f407face9e1f574ea3bd2 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e1d616c154704e98bd59b8f216015607 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/33d9f419e8db4949a40850952d86e1d3 : SUCCESS in 4m 40s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e9971a9df3434317962cda92aa22ba2a : SUCCESS in 1h 12m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0d27c929381243679b43bd5b514a5e2d : SUCCESS in 1h 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"14651909abc5c942550164e8bfa7998c1eec02a6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-17 10:03:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c70b5c45c002ad8ffcd1b9a5db15ff906028752","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 10:03:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b74bcda54df99fd2da95a94ac60e3997b2967ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 11:36:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50141fb493174285941965d320bcc846\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b885ba9d33e24ee2a818d85c219e15a9 : SUCCESS in 19m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bc517d6f3df44868b7a88bd351cd9ebe : SUCCESS in 20m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c1c3c412b3bc45cca433491337d342cb : SUCCESS in 55m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0175e8eaee494d309ebe116c3d678b66 : SUCCESS in 1h 17m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da5f2f838028455cac35309a73d8c344 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55890c5c05b549d38cc9791361848258 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7042a95fc0d4b29af157db99dfabdab : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b23281c12b5b4607801b778360da879b : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e92c23d426414c769985d362df8fe86a : SUCCESS in 4m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/512784748d9d4cb7b5ab072abc3f261e : SUCCESS in 17m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/eaee0b6e73d849a2ae7248eb1c14d7f0 : SUCCESS in 1h 18m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/af7a6abc33be47858e1ceb9f64e87d6c : SUCCESS in 1h 15m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ab714a1ac37e4d008e0683fdd056605e : SUCCESS in 1h 28m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/98e7805a23bd4d8e83b0138ee7b46d7b : SUCCESS in 20m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/75274836ffeb4ae8a457524ea496398b : SUCCESS in 28m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2c8523e9c2694ffa85d4e4b12c7bdc3d : SUCCESS in 53m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/89b0957e7b314790a64d5a15b28a53bb : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ebf03c71b8a00d37500736d99cce244d54578367","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 11:36:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c618171108d7cc5086a15692e4700fd78bb97638","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 11:37:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/779a1f77fc164a6984503b629ded9ea1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb5d43f8eb4e4ce3a5f7abfc3a4ac5dc : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1},{"id":"2e9ef7fdd63f04ed5181a4b5c89455c4c3b56d06","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-20 17:04:34.000000000","message":"Created a revert of this change as I635d2bc20d897d037739702bf15cbb2e2bdcb0b1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"033286f4c15bb80a91b4e8dcb9a0be283daf9535","revisions":{"033286f4c15bb80a91b4e8dcb9a0be283daf9535":{"kind":"REWORK","_number":1,"created":"2023-10-12 09:53:26.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/07/898107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/898107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/898107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/898107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/898107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/898107/1"}}},"commit":{"parents":[{"commit":"e4df04e9615b19e8fb0bcd1b192928815b027b82","subject":"mark gnocchi-base as unbuildable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4df04e9615b19e8fb0bcd1b192928815b027b82"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-10-12 09:53:04.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-10-12 09:53:04.000000000","tz":120},"subject":"[release] Use Bobcat sources by default","message":"[release] Use Bobcat sources by default\n\nChange-Id: Iced4520e4cd8811c0e16227e130e094f5eba23e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/033286f4c15bb80a91b4e8dcb9a0be283daf9535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/033286f4c15bb80a91b4e8dcb9a0be283daf9535"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
