)]}'
{"id":"openstack%2Fkolla~898289","triplet_id":"openstack%2Fkolla~stable%2F2023.1~Ic86e6c22840f658bb68387aac688918d4db1f766","project":"openstack/kolla","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"35025":{"account":{"_account_id":35025,"deleted":true},"last_update":"2023-11-07 10:00:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic86e6c22840f658bb68387aac688918d4db1f766","subject":"Fix an issue with zun-cni-daemon building images in aarch64","status":"MERGED","created":"2023-10-16 10:43:48.000000000","updated":"2023-11-07 10:02:00.000000000","submitted":"2023-11-07 10:00:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898289","cherry_pick_of_change":885729,"cherry_pick_of_patch_set":3,"meta_rev_id":"a73a7711ca77eef77c5ed83a0f8a122cc6bd8a70","_number":898289,"virtual_id_number":898289,"owner":{"_account_id":35025,"deleted":true},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-07 10:00:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-30 09:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-16 11:12:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-11-07 10:00:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-11-07 08:33:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2023-11-07 08:33:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-16 11:12:38.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-16 12:05:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-20 10:14:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-10-30 09:44:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-11-07 08:33:01.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-11-07 10:00:29.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"31dabe30b87e6ac525215eeb189c17d2613075a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2023-10-16 10:43:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fa16b4183feada7e1b355f37e07025c0d40fc999","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-16 11:12:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aea282f80e6f826930ad906cf9d5e046ffadb0ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 12:05:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/18fd1837ac62473f9d7e131c8a8b377f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/daf0902053a04a8ba23f074fe171f160 : SUCCESS in 27m 48s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/88e000b0ed844d3b914b5ec20ec7deb1 : SUCCESS in 1h 11m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/54a72d13953d4949817c84467021857a : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94c74456b4e74cbfb3eff3b0eaf3c2ca : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d140706b34d4bac8927ae2d53fb2d96 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a56c568b9bd4e6d8e75ad4a2ad4e6b0 : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42cca13a67d149119ae70912c8ee74d6 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e82f71676f44db2b348548743aca722 : SUCCESS in 4m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5690dcd0789d4abd9ab0f01f8e49771f : SUCCESS in 22m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c4f44314260f457ba3be2fcc90b8ac96 : SUCCESS in 1h 07m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/87d5ebf0403047b2857e7532de6013c9 : SUCCESS in 1h 21m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/72ab1be6e72447b6b7f8e39f41110c3a : SUCCESS in 24m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/950f22bced2148fc933d47e82e6d72ac : FAILURE in 20m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0772e1f4ef7b4808a1cb0849575a573e : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6b35d01f1507ab9d5acfdbd59ca59156353ba1f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 12:14:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c76115687aed41f088b5b164d0bf95bd\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/27afa8b0fee14bf59f0187d0beeca4c9 : SUCCESS in 1h 27m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dfa6192b4b4d4f57b25d7ad1bef75dd8 : SUCCESS in 1h 28m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6460ad109d2c4bf1af0f2b68c8ce7c8c : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a7e116121642424eb258a5e9a995f546 : SUCCESS in 5m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/65bb5ded0cd34098812546734e6d03b2 : SUCCESS in 1h 27m 58s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f024d09696e24ce59fa3f4d77a8697d7 : SUCCESS in 1h 18m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f4417e00394bacace5f92be62c80554fd4b18c2","author":{"_account_id":35025,"deleted":true},"date":"2023-10-16 14:13:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90ffef9dfb73b9a7dd0cbf7c94efcc95afbfab20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 15:32:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/27996575e29a4980b8cd0387d83673ee\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8b12db35dd7a48fc8b106945254ada00 : SUCCESS in 23m 48s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4927c3fb1b8d4f7d9986a81c057f769e : SUCCESS in 1h 08m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f1557940f0d248c7936a50484e61122f : SUCCESS in 20m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c228f9e17b8474f94e186797ec26f9b : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75ef598a460e4e2fb10e5b38ff4dac52 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3bdaf41082d04d909d10a85b2463a4a3 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b730d978ee4f4021ba09874381808889 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f1c954236e5457c94e41ce878d7e48b : SUCCESS in 6m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2600fe6416c14dd782124537fe23d8b9 : SUCCESS in 24m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/004cd25515514aa1b6249cf0e08dc8cb : SUCCESS in 53m 39s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d6e9acf7b207476eb0ab32fc8bf9cd23 : SUCCESS in 1h 14m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/457ff0be19b34f91b515e6d7bfce2573 : SUCCESS in 16m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a8fbac43cfff4f0d9433466389e0cc5e : FAILURE in 15m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5ba5f185b533474188ce86b3b4338484 : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":1},{"id":"70a42b3f44a253cd1abf18686752b420eb77c091","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 16:06:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cba6d9adaa342a7862ff76e2015c1ab\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6ecaa3da26fc4280a9c5faddda933e34 : SUCCESS in 1h 38m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a2cb8996b974488884fb0827269e881a : SUCCESS in 1h 40m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa232bc982504d8fb2b5bdb6d4740ab1 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8e84cc6e2fee4451a94061cea0445cb8 : SUCCESS in 4m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/521b25a1e02c46919a34094970623945 : SUCCESS in 1h 37m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/cf5f1c16085e4dbd8a7456ede2cb4983 : SUCCESS in 1h 44m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c93b224456ef4336bb8a6972032e1167ac8817b4","author":{"_account_id":35025,"deleted":true},"date":"2023-10-18 02:39:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6985e4b33f87c2c38e7f17442626a8961e785e11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 04:22:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f0feb70e2ba41c596e2650898663000\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/063802aab19c472ab84c62e7e261c812 : SUCCESS in 22m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e3c405c138af4888b0c962097f6e48aa : SUCCESS in 1h 19m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e65d9cecc00a4cc9af000e3c262023f2 : SUCCESS in 31m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11fe81e6b44646f49a7d99dbe71e7076 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bd177970c76434bb46c7bbaf6d0205e : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/346898394d19407382af77cf36fa0e42 : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9795c373a39044cf8a7a3e8c2482260d : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4902aec427f464a853d46238a210591 : SUCCESS in 4m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/94a8d84fe9d1407ba6af1a94a54502dd : SUCCESS in 22m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3d5b7b1995d3473f9378f71b024c1e29 : SUCCESS in 1h 05m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a8c5199aa61a4939877ff3eafa32418e : SUCCESS in 1h 25m 10s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/312eddac48f046beb2f0ba5101c6bdbb : SUCCESS in 20m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/af2699fb353e4ac3bfbc1f84bbf1b96f : FAILURE in 23m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6875caab665e4ac89dee0112365f4b28 : SUCCESS in 1h 24m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f9b0f280e1b73342cbfc88c814ae4317afcf27ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 04:50:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8ed306b28284b1fa1ee07f47105fb53\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/241db82e2bd24ef595cdb9c01ee593b7 : SUCCESS in 1h 48m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cfc6ae888b1c41158ccc1246a0273443 : SUCCESS in 1h 46m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b8e88706d6641d08db20378db97c5b9 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2b476e76bf2a4874bd04309db3019239 : SUCCESS in 6m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/33ae692a31124e7280762c82b43d6d5e : SUCCESS in 1h 39m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3702e6b97ab14a2980c70a461bf33734 : SUCCESS in 1h 55m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"519309218ee1673a737d6dee169b9ac0f178b0a0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-20 10:14:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27fa9480c4d0b5affb031306bb052f82e0f4c137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 11:39:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4aa021f315546e6b07e74e4af58c16a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/56911fb17aa14694b3aee098f709f03f : SUCCESS in 21m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/56da45e480614b5f8fa0a5421d55231e : SUCCESS in 1h 01m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/35b410bb2d27446a997f9fb8a68daac8 : SUCCESS in 23m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647961b1572e408d8005037f5fd56a5e : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e454639ef7894e1f9f2700944723ab2e : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/060ad548be78429ea08b580f49aee5c7 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79804d40a6ed450babd99a79de95871b : SUCCESS in 5m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/980d42e575384ceebc99907b0bb862f4 : SUCCESS in 31m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/985377003b0649469b92b74f483af64b : SUCCESS in 1h 19m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9e7322b2afcb4690b279e3121dcd081a : SUCCESS in 1h 09m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ba5547eeb23c4822b4f874f4c5540ffa : SUCCESS in 21m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b7fc399ed04e47899af747fb8edea076 : SUCCESS in 1h 11m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/126096731ab3408c8f5484b7b78d08e0 : SUCCESS in 1h 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d4f98de020b6889068d04cb5cf8f325d3a858050","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 12:02:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cac67d84623c4146a971d8b9b3e6194d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/588d35abc0154c0090d6725176b29854 : SUCCESS in 1h 00m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/39464a3bf3eb4541ab19e53d81ac034a : SUCCESS in 59m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4e07d2d403c4dbfafb9b69d0c90cd82 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/69dbfb0a072844579e44d6885fff3c91 : SUCCESS in 5m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4f7897ec311147cdb6def3daef97df86 : SUCCESS in 1h 17m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/980300a8208b4053a18c88359f8efa3b : SUCCESS in 1h 20m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03c2f99ad67ff926a27ee5b0f4a1445c0ffc0a0b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-30 09:44:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"da2c63480d55186d7162324c73de9f260225da77","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-07 08:33:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8f2b8c7c23b71720a7066b92f255f92db8877053","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 08:33:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4eb926dc37b20f664bdaf246a6a1f4a26a693029","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-11-07 10:00:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8e8f498c26d1cb07bf7175f51ade794562e0e173","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 10:00:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf6f0360b4d94116ad2c867650f76494\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/655751a955bb4f7ba2da200030d28bd7 : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2447ae72b8648768dba928e4b9c7287 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/211eba6e150641788ddea176fff686ea : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a5cd1b6f514fb6935c8523f58cb282 : SUCCESS in 3m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9fcd6590796f489db320d8a0ea727542 : SUCCESS in 23m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2f7c510164104f65b56e9f4680dfa91e : SUCCESS in 1h 13m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/128fd71c3ab14c578c469a582ddb20c0 : SUCCESS in 1h 05m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/42087fc4d97046f8bc817083e964c7cd : SUCCESS in 23m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5b7989ed18a54aa3933050d76d9dc079 : SUCCESS in 1h 03m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c7bf3892677447d6a813b3ee11c1009b : SUCCESS in 1h 15m 43s","accounts_in_message":[],"_revision_number":1},{"id":"daf9a9ebe9c312f1de8e4e5d261e57957f6796f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 10:00:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a73a7711ca77eef77c5ed83a0f8a122cc6bd8a70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 10:02:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7814340604f4247a99082831a498981\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a9f23317a004617aea00be964413cd6 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6731028be61e37a281b3207f57c7033d4a96990","revisions":{"d6731028be61e37a281b3207f57c7033d4a96990":{"kind":"REWORK","_number":1,"created":"2023-10-16 10:43:48.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/89/898289/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/898289/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/898289/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/898289/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/898289/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/898289/1"}}},"commit":{"parents":[{"commit":"0e6676cb0b9908e53eb1f5aee4c22aec2a1c1e4d","subject":"rabbitmq: Use novemberain repos for Debian/Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e6676cb0b9908e53eb1f5aee4c22aec2a1c1e4d"}]}],"author":{"name":"renliang","email":"renliang@uniontech.com","date":"2023-06-09 08:25:14.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2023-10-16 10:43:48.000000000","tz":0},"subject":"Fix an issue with zun-cni-daemon building images in aarch64","message":"Fix an issue with zun-cni-daemon building images in aarch64\n\nIn the aarch64 image build,\nuse debian_arch to determine packages of different architectures.\n\nCloses-Bug: #2036874\nChange-Id: Ic86e6c22840f658bb68387aac688918d4db1f766\n(cherry picked from commit 0eed86bb02e87c12098e52359897708e071aa07e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6731028be61e37a281b3207f57c7033d4a96990"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6731028be61e37a281b3207f57c7033d4a96990"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
