)]}'
{"id":"openstack%2Fkolla~898290","triplet_id":"openstack%2Fkolla~stable%2Fzed~Ic86e6c22840f658bb68387aac688918d4db1f766","project":"openstack/kolla","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"35025":{"account":{"_account_id":35025,"deleted":true},"last_update":"2023-10-20 11:47:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic86e6c22840f658bb68387aac688918d4db1f766","subject":"Fix an issue with zun-cni-daemon building images in aarch64","status":"MERGED","created":"2023-10-16 10:43:59.000000000","updated":"2023-10-20 11:51:51.000000000","submitted":"2023-10-20 11:47:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898290","cherry_pick_of_change":898289,"cherry_pick_of_patch_set":1,"meta_rev_id":"0ed6a1b91209eccb96f90da208b0af5da9506db1","_number":898290,"virtual_id_number":898290,"owner":{"_account_id":35025,"deleted":true},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-20 11:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-20 10:14:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-16 11:12:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-10-20 10:14:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-16 11:12:30.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-16 12:11:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-20 10:14:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"6a6be28b91685ce973c88646fea7c1b7eb38295f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2023-10-16 10:43:59.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"0ad771fa655eb80818e5422c8aa0c5728c17ab9f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-16 11:12:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"548f2d3f385739dcd8f8cb608f5a52d6aa18a670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 12:11:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48b8d6b502bb400b8f30fc03b29eb83d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b035c6b9fbfd4eabb67391f61c45ad8b : SUCCESS in 22m 26s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/88b4bf3d452d4ec5a769dcf52a02d1f4 : SUCCESS in 1h 10m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8f04bdad1c644aaba24c01d74d19aecb : SUCCESS in 19m 57s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3cb100cc40fa4c59a52ea954d8d6c333 : SUCCESS in 1h 04m 58s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/fdca7f86f698456d85f3ea07b0750d92 : SUCCESS in 1h 21m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8abd4ebddbd44dddb8f992be011b7c79 : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/891b46f3b60a4decac7fc4bdcfd61648 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acdadc3cbe1b4166af9641441b8208dc : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03295b4bafcc4ceba7ce3ee762704d8a : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/702507034de6488db1cd95f41670d59c : SUCCESS in 6m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2f2eb5f633854abf98a86f245d2854b3 : SUCCESS in 20m 46s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a614d558bf06458489ea3dfdff8f80af : SUCCESS in 1h 09m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5b3226786fe3418a9dcdd3283e2903b6 : SUCCESS in 21m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/82ccbd6cf6fa42099adbc23aa13789f5 : SUCCESS in 55m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b974f2fd6810419db42a89b44d20f216 : SUCCESS in 1h 23m 23s","accounts_in_message":[],"_revision_number":1},{"id":"537b4e73b69beb56f9835e8212dff6e51d9a37ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 13:00:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83558cb2208448be88f17632f2a92842\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/59e8d2edde0d4a0390afb404a2f21f9a : SUCCESS in 1h 22m 20s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/745640b290da4b09b7976711162faa12 : SUCCESS in 1h 26m 18s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b16b9d74444460eb33e4af83af75d94 : SUCCESS in 2h 10m 01s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b070a2cde642443a82594f7727f8f2c6 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a1f292e5dac4e6c8a53be3a7ce17f28 : SUCCESS in 5m 16s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d1dd9058fdf149989994cb1b5962ada1 : SUCCESS in 1h 22m 52s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c8a8ca1b6c6243e58ed352d39fc28df8 : SUCCESS in 1h 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee72a48614acbcedb01eba6fe1278a8b76b9e708","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-20 10:14:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd2fec1c1d52e8f20776591d4994378ecd52fbfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 10:15:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9c301463fd568e221db01b57d3f72a8bc544d77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 11:47:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1903b787309c4c2e8cced941415d08a9\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5c01f16d800842e8aa3c278012ab3067 : SUCCESS in 21m 35s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/47190ce44e0a4723bf6eee950088268e : SUCCESS in 52m 29s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/129507a686734f3e8649aa3c2d1890d8 : SUCCESS in 1h 14m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6629a2af4e4d4383b94e8b79e065af6a : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3c8a4d950214ca28dceb3c6e56264ed : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d63aa10ccd4e495589462249f7130652 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a28771a93f1a4ebf9947e5508dd06640 : SUCCESS in 6m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5b5e6ac0e64d46358a19105428b8fd6a : SUCCESS in 23m 35s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/fb5ff04e13004e87bd9eabf20582ada9 : SUCCESS in 1h 05m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5e521a830dbf48dda674554e7c2825c5 : SUCCESS in 21m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/926765d3fcb74ebaaee8eafd943bb3a9 : SUCCESS in 1h 05m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/da0b7546413e45ab84f87a6175005b99 : SUCCESS in 1h 09m 53s","accounts_in_message":[],"_revision_number":1},{"id":"1bceb1e261a61929ed136f164a7df6fbfbbbe4d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 11:47:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0ed6a1b91209eccb96f90da208b0af5da9506db1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 11:51:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3534626cc2345b1af44a78ce2c7ccf8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1043f354e22491baf93ce0c424f3eb3 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b771157057538498fa3ae79ef2c9a773f4f18d7","revisions":{"2b771157057538498fa3ae79ef2c9a773f4f18d7":{"kind":"REWORK","_number":1,"created":"2023-10-16 10:43:59.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/90/898290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/898290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/898290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/898290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/898290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/898290/1"}}},"commit":{"parents":[{"commit":"8498b43c1ec71ad7cea09123f91835d09f20b634","subject":"ubuntu: mark collectd and telegraf as buildable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8498b43c1ec71ad7cea09123f91835d09f20b634"}]}],"author":{"name":"renliang","email":"renliang@uniontech.com","date":"2023-06-09 08:25:14.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2023-10-16 10:43:59.000000000","tz":0},"subject":"Fix an issue with zun-cni-daemon building images in aarch64","message":"Fix an issue with zun-cni-daemon building images in aarch64\n\nIn the aarch64 image build,\nuse debian_arch to determine packages of different architectures.\n\nCloses-Bug: #2036874\nChange-Id: Ic86e6c22840f658bb68387aac688918d4db1f766\n(cherry picked from commit 0eed86bb02e87c12098e52359897708e071aa07e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b771157057538498fa3ae79ef2c9a773f4f18d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b771157057538498fa3ae79ef2c9a773f4f18d7"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
