)]}'
{"id":"openstack%2Fkolla~901634","triplet_id":"openstack%2Fkolla~master~I8ebcf815d80f3bead25e0078d69b34e17ad013bd","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-11-22 13:45:20.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-11-22 15:51:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8ebcf815d80f3bead25e0078d69b34e17ad013bd","subject":"Update Ceph repo name for dnf config-manager command","status":"MERGED","created":"2023-11-21 22:45:30.000000000","updated":"2023-11-22 15:52:57.000000000","submitted":"2023-11-22 15:51:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901634","meta_rev_id":"77a104871a557946d3ad7138e9b625b9d29b91e5","_number":901634,"virtual_id_number":901634,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-22 15:51:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-11-22 13:45:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-22 13:50:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-22 13:50:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-11-22 13:45:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-21 22:55:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-11-22 01:09:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-22 01:46:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-22 09:11:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-11-22 13:50:39.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"74b3d6049a264dab43bb323135e6fe5008632068","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-21 22:45:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7ac36b93e2cacf9539f1aacc0a15189f653dabf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-21 22:55:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"def62035a4320fd200ff90d35d22a980705b2e4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 01:09:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3900e87285844381b6f45127d482a038\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ca51a28643244d3881a70bbe15a19d6d : SUCCESS in 1h 30m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d8f036eeb0d04840804c1871a81561fe : SUCCESS in 35m 23s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/57ae241c54fd47e0825a01858fae8031 : SUCCESS in 2h 21m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff6113f6fcad4e4da334097e9b762f78 : FAILURE in 2m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eeb09f0759e2492fb00f307c53e82544 : FAILURE in 6m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6cc35588f30e4ad291340a837829d056 : FAILURE in 2m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6c89a12b39464fb5b95557b76670bb6a : SUCCESS in 1h 29m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9e9ba4732b11453e94bb2b9e34fcbf21 : SUCCESS in 33m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c832c8f6c52d4812268b7a90d012f7bb76b9d3c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 01:46:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/49db8211588249e5bab04db31341d623\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/48edf604ce214e7bb93aa94b2f5a065f : SUCCESS in 23m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a4f0ec1dc6884e7399c635513bbd4c35 : SUCCESS in 48m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ad98a9260a3d490e880362ec7695b202 : SUCCESS in 19m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fd0384654a4b4b17bbfb701e9af7b08f : SUCCESS in 25m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c0ee9b22e81749dd95432c366f7c7521 : SUCCESS in 51m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ae0653ed382a47c5b5164ea0ad907206 : SUCCESS in 1h 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50083352faad4b3b8fb43e7087e2ed56 : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5b1d470513049b6988e14f6a588b656 : FAILURE in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/165ed864a2da41d7b9d5f4fb4b47ddc7 : FAILURE in 2m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/497dfc03edbd4878acfcd787dbd91630 : FAILURE in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50b830c6ac6a48efa68247c92a743db9 : SUCCESS in 4m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/61eb37eb57c646fdabc8664f78ef0f10 : SUCCESS in 20m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0702b91587cb47ab8eb8c2c363e3659d : SUCCESS in 29m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6ad6c6036205447a959f9f476eb56435 : SUCCESS in 58m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a1f93a4cc1224d18af096c90ce47a1d8 : SUCCESS in 1h 41m 56s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/5cd58af400ec42dbaa28db63cb486800 : SUCCESS in 1h 38m 59s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a1bd8be1f6ed44e4ad5034e7be93bfe0 : SUCCESS in 2h 54m 37s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/9f8aafe3ff7c44409caf0305f4e51de0 : SUCCESS in 1h 16m 51s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/c7360c38c4044f59ad3983fe262a22d3 : SUCCESS in 1h 34m 25s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/c11a35aa55dc405499e1bc2de78ef616 : SUCCESS in 30m 18s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/aebce944742c4983b6010ac525c6d0c9 : FAILURE in 49m 17s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/b201e88dbaf443798a4da88332a10887 : SUCCESS in 1h 17m 18s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/a25ca760dc9745d2a4dbcf1373ba270f : SUCCESS in 1h 27m 32s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6a286ce948ee4051ac0184da12d7b3bb : SUCCESS in 32m 03s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/49ae83b17564415c8af99fd8ab747c23 : SUCCESS in 25m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4545d548e6e649a2aab2d78423a2b695 : SUCCESS in 29m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/04ed7a55efa04033b73a4aed14756d13 : SUCCESS in 50m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/aa91c00ff2874c1c9f2304e6d50d2215 : SUCCESS in 1h 29m 13s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/64d63d6c69174ab399bb5f49987ca1a5 : SUCCESS in 1h 24m 14s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/72a51a21510f4c53814142acabd31ba7 : FAILURE in 52m 07s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/4a931ea36da747008ab0e568f9726d23 : SUCCESS in 1h 12m 35s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/367307cfcf724b4dbe46d3bd3e30bcd2 : SUCCESS in 1h 11m 27s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/bd6295d9e32b49b59bd4e734898cf7a6 : SUCCESS in 29m 03s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2c2fd9ddbfee43cf8b5b130b1ef67809 : FAILURE in 41m 49s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/f3e0d045aefe49a49363ba857bdc2665 : SUCCESS in 1h 00m 56s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9f570d10b77045b09d7ac267a41c38e7 : SUCCESS in 1h 00m 20s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/be22e11ea3f94c4b8671bdb4d4871b74 : SUCCESS in 30m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b90c74cc17bd5110bc7865860298fa91f3025bd9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-22 05:52:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d67c6e8d50b234f99191d9468549ea9ac9b33fcd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 06:45:33.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4aa85b41cc08f7618f1c4f5242d0b83d8b5c589d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 06:45:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"86d14bc9b98078251f19eb2186718b74b23131e3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 07:46:51.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"369c8069af6adcdf0b31d53c0940ec2be8da123b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 07:48:04.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"5468b267fd4a652381254d6e0aade28c3834debf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 09:11:33.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"6e7eeee69674db399c4937b6e26871e6cc4552f8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 09:11:37.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"8c8938e1991bebefab8b1d21ad6cdc06e3b5b147","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 09:11:50.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ee6e504af64d3485196c9c638777323f2ae01041","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-22 10:06:39.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"f804d8906fa18408d8e2a0b07c53fe1227e0cb41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 11:29:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8e7d18546a14dffaa14c5a9ae0efc37\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/98d1c3ba3aca4caf852a029656ee7150 : SUCCESS in 16m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/66435a6ca60b491bad6e287d32dbbbb8 : FAILURE in 1h 20m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/df5c7e682ab940c8b5635989dfd07717 : SUCCESS in 1h 00m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c6d4336ab2746cf98b0bbfce98b800e : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fad02b38ca2a4c23839e682e6258d650 : SUCCESS in 2m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b8de620545ee4211a64512de443f07ab : SUCCESS in 2m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8f92bce1ecc648a8b0f8d35b4142d697 : SUCCESS in 18m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7c7bd87256464fc1943dc2480b0936fd : SUCCESS in 14m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"232c08ff180667d4c7b3b0bd158cb927718553ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 12:45:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/513a05ed66564f87a3af52c6f7d10d3c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8d4ad0a174464388bbcb7357c9788847 : SUCCESS in 24m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/dc96d6c08881448dbf21a322770b7971 : SUCCESS in 1h 11m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fb6acf4d4e194d199645a7ee9391fb75 : SUCCESS in 20m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7261053cc64a46f280b682c59e8d5de2 : SUCCESS in 31m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4dd88e7328bc4a58baee6e348ed2390b : SUCCESS in 1h 02m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dd1462185c3f4d9b9d894c7f175df36b : SUCCESS in 1h 21m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29c70c04c1f24b6e8d3eb1986ac3e785 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbee3565a9474232a267a2d9e3105a67 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/144d9a8b6ab74820a95b1f1ab45be11a : SUCCESS in 4m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0883a9d579cd4d41ab28b27c03352222 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b8e506eb1a43ddaea392d412e6f23b : SUCCESS in 5m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b6686e01a8c344fa8858ca17b17d09e1 : SUCCESS in 19m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6697bbbede53412ba71c22492a6e8dcc : SUCCESS in 27m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bd00c7e10f4c4ff290f52ecc3a430000 : SUCCESS in 1h 04m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/85157463b335405a9cee0449cc074d69 : SUCCESS in 1h 33m 15s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/f910e268d1384a379aed7768158821b3 : SUCCESS in 1h 42m 00s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/10df18b0d3164ee7865eb7e342c46aab : SUCCESS in 2h 33m 06s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/36fd7aa6385d4930b9dc7a1545738d3b : SUCCESS in 1h 20m 15s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/65234405777b460a9657c986ee3716b8 : SUCCESS in 1h 30m 35s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/ff8df1f389c7466081281cb6b7a1f9ab : SUCCESS in 29m 42s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/4dc2e612c13b467987eb70cdcbb0b7e0 : FAILURE in 47m 41s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/c3b4d2f76ec04b3f9730c4648bcbdc60 : SUCCESS in 1h 16m 06s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/6d76b879175541759f965bdd7a47a85d : SUCCESS in 1h 33m 29s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e9d915f1a3d1480f9bfa9368bc0b17a5 : SUCCESS in 32m 56s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/95f6427d53c4434f9b0828d118669e68 : SUCCESS in 49m 52s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/279f6f11c6774405a2208a83b572285c : SUCCESS in 26m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1d1dd783df8e42ba9b75939ba9f0ceab : SUCCESS in 31m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/88c4952f3fa442b9a346b28851e059cb : SUCCESS in 56m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ed5c8165e15f40e780b682efd49ec46e : SUCCESS in 1h 23m 53s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/970a43fb53bf493eb44a367feb6e80e4 : SUCCESS in 1h 25m 09s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d4f6784382e64e039d399a9c5ddd310d : SUCCESS in 2h 20m 48s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/4706b75c64e149a990f3761f7115cf99 : SUCCESS in 1h 01m 05s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/0d3b73987c2f4acfa993c4429b4a14b4 : SUCCESS in 1h 11m 37s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/f5cef8cf2a1146bfbee5511b1a1c87f4 : SUCCESS in 26m 26s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2df96d61d0ae445ba7778a752a9a2786 : FAILURE in 37m 25s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/dfacd6ff7e69468cb7ceee7480ac34e0 : SUCCESS in 59m 32s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ccebe0e1b39648c89f927966e4e92931 : SUCCESS in 1h 10m 41s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/88350546c2924851bc3f91b7ab9e1862 : SUCCESS in 31m 58s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/a8c445f5d49b416abb2e5092bea3daae : SUCCESS in 36m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"946c7c9dd174bc2f80386c16fff280524df92a10","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 13:45:20.000000000","message":"Patch Set 7: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":7},{"id":"a7e5a24cbd1c7ddcc8822eddbeddd78cabdb6557","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-22 13:50:39.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"2f21e7fa59b1a5fd3501cc16c245f523d453df08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 13:52:54.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"3ec600dd8d3ba98b8760c5ea9a2f8a9b94286cb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 15:51:51.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/210a38509f994dc18e89fbd0dbaeb6c7\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5c365ff79eb74426afdfc7b40519fe3b : SUCCESS in 19m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/210ed8bdc7ef454a9d19ffb853eb66cf : SUCCESS in 27m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b9e454e7da244256beb74649c3ef69a8 : SUCCESS in 1h 02m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/14bd28e68ff94781bfa33aa686201ea1 : SUCCESS in 1h 39m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/169b86eb461d408395c602c4ee76b0a8 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/547a48d0c93649e59f29e86eb134e13c : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c20478fb87c4774aca268ded23a7acd : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e2e613f877b403b8c6b6444bb475186 : SUCCESS in 5m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/73898808871142ce8afd6310bde76f5e : SUCCESS in 18m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c2d3c78245554536a34c1c08f1bb368f : SUCCESS in 22m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/71ab5f2e867e4bd2b38ee2274d868b8e : SUCCESS in 1h 01m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fa2373fcce38497698935df051ebca61 : SUCCESS in 1h 52m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3ff1063ecb24489ba69110f28c45c17d : SUCCESS in 23m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6723378020a4441987b4527a660b8528 : SUCCESS in 33m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f79a3813e3af4f7db2ee6406f5434d46 : SUCCESS in 55m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6b4a824d80e24a9ba1de08c965ecc0a4 : SUCCESS in 1h 33m 45s","accounts_in_message":[],"_revision_number":7},{"id":"b6132f7fabf0a6fdb60a344eaa05703d59345d67","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 15:51:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"77a104871a557946d3ad7138e9b625b9d29b91e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 15:52:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8daaa539b4c4aca91dbc6dc9e888080\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8e50b2608b14d58bdc595a6e3614c76 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"33c988a6ce9fc3fe86d32d0bb97f6cb5130bf2f2","revisions":{"ca060550651ee63f46313871077ae320c0665063":{"kind":"REWORK","_number":1,"created":"2023-11-21 22:45:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/34/901634/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/901634/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/901634/1"}}},"commit":{"parents":[{"commit":"d86cba116ef686a7bad6153a5ce865ba77850cc7","subject":"Merge \"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Bobcat)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d86cba116ef686a7bad6153a5ce865ba77850cc7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-21 22:41:48.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-21 22:41:48.000000000","tz":0},"subject":"Update Ceph repo name for dnf config-manager command","message":"Update Ceph repo name for dnf config-manager command\n\nSince change I1bc19f8198da3f9ab2ae2a8864c3349b21b0249e we install the\ncentos-release-ceph-reef package as a dependency, but some code was\nstill expecting the quincy package.\n\nThis is causing container image build failures in the Kayobe seed-vm\ntest, but it is unknown why kolla CI is not failing in the same way.\n\nChange-Id: I8ebcf815d80f3bead25e0078d69b34e17ad013bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca060550651ee63f46313871077ae320c0665063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca060550651ee63f46313871077ae320c0665063"}]},"branch":"refs/heads/master"},"18e39e26d26ab781ae7d711acda3fb30638bd7d2":{"kind":"REWORK","_number":2,"created":"2023-11-22 05:52:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/34/901634/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/901634/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/901634/2"}}},"commit":{"parents":[{"commit":"d86cba116ef686a7bad6153a5ce865ba77850cc7","subject":"Merge \"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Bobcat)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d86cba116ef686a7bad6153a5ce865ba77850cc7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-22 05:52:05.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-22 05:52:05.000000000","tz":0},"subject":"Update Ceph repo name for dnf config-manager command","message":"Update Ceph repo name for dnf config-manager command\n\nSince change I1bc19f8198da3f9ab2ae2a8864c3349b21b0249e we install the\ncentos-release-ceph-reef package as a dependency, but some code was\nstill expecting the quincy package.\n\nThis is causing container image build failures in the Kayobe seed-vm\ntest, but it is unknown why kolla CI is not failing in the same way.\n\nChange-Id: I8ebcf815d80f3bead25e0078d69b34e17ad013bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18e39e26d26ab781ae7d711acda3fb30638bd7d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18e39e26d26ab781ae7d711acda3fb30638bd7d2"}]},"branch":"refs/heads/master"},"bb4a05008b5ca4ae1c322bf8232ad1ab4fe6da60":{"kind":"REWORK","_number":3,"created":"2023-11-22 06:45:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/901634/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/901634/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/901634/3"}}},"commit":{"parents":[{"commit":"d86cba116ef686a7bad6153a5ce865ba77850cc7","subject":"Merge \"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Bobcat)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d86cba116ef686a7bad6153a5ce865ba77850cc7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-22 05:52:05.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-22 06:45:33.000000000","tz":0},"subject":"Update Ceph repo name for dnf config-manager command","message":"Update Ceph repo name for dnf config-manager command\n\nSince change I1bc19f8198da3f9ab2ae2a8864c3349b21b0249e we install the\ncentos-release-ceph-reef package as a dependency, but some code was\nstill expecting the quincy package.\n\nThis is causing container image build failures in the Kayobe seed-vm\ntest, but it is unknown why kolla CI is not failing in the same way.\n\nChange-Id: I8ebcf815d80f3bead25e0078d69b34e17ad013bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb4a05008b5ca4ae1c322bf8232ad1ab4fe6da60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb4a05008b5ca4ae1c322bf8232ad1ab4fe6da60"}]},"branch":"refs/heads/master"},"2d186cb584d96ae2a2f9e0598f58b5c350385620":{"kind":"REWORK","_number":4,"created":"2023-11-22 07:46:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/901634/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/901634/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/901634/4"}}},"commit":{"parents":[{"commit":"d86cba116ef686a7bad6153a5ce865ba77850cc7","subject":"Merge \"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Bobcat)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d86cba116ef686a7bad6153a5ce865ba77850cc7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-22 05:52:05.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-22 07:46:51.000000000","tz":0},"subject":"Update Ceph repo name for dnf config-manager command","message":"Update Ceph repo name for dnf config-manager command\n\nSince change I1bc19f8198da3f9ab2ae2a8864c3349b21b0249e we install the\ncentos-release-ceph-reef package as a dependency, but some code was\nstill expecting the quincy package.\n\nThis is causing container image build failures in the Kayobe seed-vm\ntest, but it is unknown why kolla CI is not failing in the same way.\n\nChange-Id: I8ebcf815d80f3bead25e0078d69b34e17ad013bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d186cb584d96ae2a2f9e0598f58b5c350385620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d186cb584d96ae2a2f9e0598f58b5c350385620"}]},"branch":"refs/heads/master"},"88d96689da9861e799e634f91082f1eee8104403":{"kind":"REWORK","_number":5,"created":"2023-11-22 07:48:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/901634/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/901634/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/901634/5"}}},"commit":{"parents":[{"commit":"d86cba116ef686a7bad6153a5ce865ba77850cc7","subject":"Merge \"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Bobcat)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d86cba116ef686a7bad6153a5ce865ba77850cc7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-22 05:52:05.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-22 07:48:04.000000000","tz":0},"subject":"Update Ceph repo name for dnf config-manager command","message":"Update Ceph repo name for dnf config-manager command\n\nSince change I1bc19f8198da3f9ab2ae2a8864c3349b21b0249e we install the\ncentos-release-ceph-reef package as a dependency, but some code was\nstill expecting the quincy package.\n\nThis is causing container image build failures in the Kayobe seed-vm\ntest, but it is unknown why kolla CI is not failing in the same way.\n\nChange-Id: I8ebcf815d80f3bead25e0078d69b34e17ad013bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88d96689da9861e799e634f91082f1eee8104403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88d96689da9861e799e634f91082f1eee8104403"}]},"branch":"refs/heads/master"},"d82272e00eb28f595229ff0f5cf55e2746341959":{"kind":"REWORK","_number":6,"created":"2023-11-22 09:11:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/34/901634/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/901634/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/901634/6"}}},"commit":{"parents":[{"commit":"d86cba116ef686a7bad6153a5ce865ba77850cc7","subject":"Merge \"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Bobcat)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d86cba116ef686a7bad6153a5ce865ba77850cc7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-22 05:52:05.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-22 09:11:33.000000000","tz":0},"subject":"Update Ceph repo name for dnf config-manager command","message":"Update Ceph repo name for dnf config-manager command\n\nSince change I1bc19f8198da3f9ab2ae2a8864c3349b21b0249e we install the\ncentos-release-ceph-reef package as a dependency, but some code was\nstill expecting the quincy package.\n\nThis is causing container image build failures in the Kayobe seed-vm\ntest, but it is unknown why kolla CI is not failing in the same way.\n\nChange-Id: I8ebcf815d80f3bead25e0078d69b34e17ad013bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d82272e00eb28f595229ff0f5cf55e2746341959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d82272e00eb28f595229ff0f5cf55e2746341959"}]},"branch":"refs/heads/master"},"33c988a6ce9fc3fe86d32d0bb97f6cb5130bf2f2":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-11-22 10:06:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/34/901634/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/901634/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/901634/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/901634/7"}}},"commit":{"parents":[{"commit":"d86cba116ef686a7bad6153a5ce865ba77850cc7","subject":"Merge \"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Bobcat)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d86cba116ef686a7bad6153a5ce865ba77850cc7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-22 05:52:05.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-22 10:06:39.000000000","tz":0},"subject":"Update Ceph repo name for dnf config-manager command","message":"Update Ceph repo name for dnf config-manager command\n\nSince change I1bc19f8198da3f9ab2ae2a8864c3349b21b0249e we install the\ncentos-release-ceph-reef package as a dependency, but some code was\nstill expecting the quincy package.\n\nChange-Id: I8ebcf815d80f3bead25e0078d69b34e17ad013bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33c988a6ce9fc3fe86d32d0bb97f6cb5130bf2f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33c988a6ce9fc3fe86d32d0bb97f6cb5130bf2f2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
