)]}'
{"id":"openstack%2Fkolla~901756","triplet_id":"openstack%2Fkolla~master~Idc9fae4a64206410ae1935e3338fb96cf4faa14f","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-11-28 10:58:53.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-11-28 11:04:20.000000000","reason":"Change was abandoned"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-11-28 11:04:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idc9fae4a64206410ae1935e3338fb96cf4faa14f","subject":"use sha sums files for prometheus node exporter and msteams","status":"ABANDONED","created":"2023-11-23 12:01:28.000000000","updated":"2023-11-28 11:04:20.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4dde92189433c386d7da49920917c1a65d3d5026","_number":901756,"virtual_id_number":901756,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-11-28 10:41:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-11-28 10:42:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-11-28 10:42:04.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-23 14:16:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-24 11:43:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-11-24 11:47:46.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"0a5dd97163e7d7830b97e9a65bdef30313514d64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-23 12:01:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fd12a89b54b87f37f80c29005ac82fac491972f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-23 12:13:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nThe following files contain Git conflicts:\n* docker/prometheus/prometheus-node-exporter/Dockerfile.j2\n","accounts_in_message":[],"_revision_number":2},{"id":"2c3cf73f17199d4ca0a974988a7bbec859fde22c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-23 12:15:07.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"568dc66e1a702a2be4fc85905a8f154ba7a3e3f0","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-23 12:15:14.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"748720d99f0af72cbe048610f221efec301ffea9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-23 12:33:23.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"0d6ce6e7c11a405cb5daab6ad6718ced7e6d6d12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 14:16:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/53fa3ab3c1764f0caddc458a2f70ca0b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6523ea8f177340129c613129abf49411 : FAILURE in 19m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/068fca3c171f4300a6311d11673a5849 : SUCCESS in 49m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8f4073353e954f61bde066a45296c7a3 : FAILURE in 16m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9b1a996ceb8a4927b137ee31b8402872 : FAILURE in 26m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ebbd4614819e4297848c1d74157c4a97 : SUCCESS in 1h 01m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/68ee143cf84c4603a28d6d5a02d40ad1 : SUCCESS in 1h 36m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6e3f740b1849b7aa3bb46147bbe582 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a39da276a61c402b9254cd000c752610 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc11dad7435f4f5488a396fec643ad77 : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3ca5908d67d40459ac3d5bb69593665 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba2acaca695943938c67e82ab3bff862 : SUCCESS in 4m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0e3c2afe6cc24c3a9589f30015825cda : FAILURE in 22m 55s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5353ee701f3f4b248901c07356c1e115 : FAILURE in 30m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/77a527f381844326800549afb45e5b39 : SUCCESS in 1h 08m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ebbafc9ed14240c69caf3186d1ed1b11 : SUCCESS in 1h 33m 12s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/27c48815732041f0b8d521de6e9de646 : FAILURE in 17m 34s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/55469822b8b34263b6eeb3954f21c617 : FAILURE in 18m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b457a270722e440d811527dc937fe3a6 : FAILURE in 26m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c3632543d378487b8b9fe56f5258212b : SUCCESS in 1h 00m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7fae08a23c1c4cdbb4a224ffa9b75022 : SUCCESS in 1h 32m 56s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3c439d69683a4b8c84faf6e26f9aa477 : FAILURE in 13m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d2cbd5d1442fd28fdfa8a58ae9f6ba6f91f1898f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 14:17:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a3ca3c6a85d476d8decc9b8018d1aed\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/37cd36413ca647409f8fd9e18c5134d6 : FAILURE in 27m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b27c9fe07f34486184420d0f77a5cf2a : FAILURE in 1h 41m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/96edfdc592e14156aa5ffa3423e7f0bf : SUCCESS in 1h 11m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81bf0c72f51d40e8a59b862e5a35bd56 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3f3b647a4086404d8b2ac34689658e76 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bea22b5d3c9548e980e462ba42d31259 : SUCCESS in 5m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ae8b1ac7ec4247ca90ba0347b12269c3 : FAILURE in 24m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3205ab7de4bb45a1acef9c015b82933b : FAILURE in 1h 35m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fe1ff7c7adcee688f3e63874fca1454b300e92c1","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-23 14:23:54.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"0a54a3bc6a811ca296b1e2b37717c122fbc774ff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-24 10:14:50.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c0dfefd0d2c6cad681c30d50dc8f9b2fb3b4597a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-24 10:15:29.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nThe following files contain Git conflicts:\n* docker/prometheus/prometheus-node-exporter/Dockerfile.j2\n","accounts_in_message":[],"_revision_number":6},{"id":"7f2ffda0d1aad45c909408da49de9e1e948952c0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-24 10:16:10.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"abc3768861bf3b43c5adf4af66773175dd668b5f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-24 10:17:32.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"7bda6cc17d152e110690c249919dce7ffbf60a4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-24 10:52:15.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0a218d7dbcb46ff0a174fcb81140ccbc08053f86","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-24 10:54:24.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nThe following files contain Git conflicts:\n* docker/prometheus/prometheus-node-exporter/Dockerfile.j2\n","accounts_in_message":[],"_revision_number":9},{"id":"13f75b02b427c6f23200e582a9c723acc20eb4cf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-24 10:54:53.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"b256b8329fcf8b87b307dadf0e713f4340ad7252","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-24 11:22:53.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":10},{"id":"66afe30e5cd65a328bbdbd08b35b269faa8f41c8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-24 11:43:03.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"14d64aa4dac12c5f342a41eb01877fddb104ec69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 12:28:20.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21419f676e1c4812af8d3248fc73c43d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7c3d544310e04c0fa860cc0bdefa7bf3 : SUCCESS in 21m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/10a72bf9f3974ea283aaccd6dec1d464 : SUCCESS in 1h 01m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7ae5a759c59d454bab8ce0fdda7a39f4 : SUCCESS in 21m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/72c882f0f0e44cb09414015ff72a2b27 : SUCCESS in 26m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1d3689c4faed4e03be2794543858330e : SUCCESS in 1h 04m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d035a240f3c046b28721f810229547b3 : SUCCESS in 1h 28m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/990b6b2f340743459feece2b995819b0 : SUCCESS in 2m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b3dfb5131284a44a62c66f2d45b87b1 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3eeb51c3940144bf99e4e3b86f028432 : SUCCESS in 3m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc102d512c8948cdb6d84c06c8654777 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac7b3f5fb0b4346b99664eb2c39b34d : SUCCESS in 5m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/33409cb457384c96ae60fc4a74ae441c : SUCCESS in 20m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/25acf692b4984216b96c5a4612c5b24f : SUCCESS in 20m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4c2ea993f1094837b5f56b7542d5d13f : SUCCESS in 1h 05m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/129b2c6ed8df4672ac7841a05ae800d8 : SUCCESS in 1h 27m 20s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/06a15149eb324811a12f396da5293699 : SUCCESS in 24m 39s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3e451d26534d4da89b43866b98447565 : SUCCESS in 13m 38s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1314043c8b274e9e9db72fbc117333da : SUCCESS in 18m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/24414790e4ce4be5885e48fcb504636c : SUCCESS in 52m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/174cc45fedda49a598e63b1b68006da5 : SUCCESS in 1h 23m 34s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/97a24927cf5d480282178f6a6252a945 : SUCCESS in 28m 52s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5ab5b3714a1e7e4a6e91e9f3387067209257e794","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 13:04:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cdcb1f322ef40488c77dd2cefbeed37\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fd671b39cc50466e92dc48a96c8a1823 : SUCCESS in 1h 19m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f77867d6d10240d49ce625a8f5dd8847 : SUCCESS in 19m 49s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2d73b4fce3d84520841822a929b3fb48 : SUCCESS in 2h 07m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03ea57af354e4b048e756157a1fa76b4 : SUCCESS in 2m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/def1ca9b12624f2d907602074ffd0665 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1a45b48104524e459c55c0932c9dab13 : SUCCESS in 2m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/697b751c7fc04183a0984aa42da2f87b : SUCCESS in 1h 09m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d29df630dd15438abfb5e5013c56bfcc : SUCCESS in 13m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a4b761aea37059e6d16bca26c29e36a300913835","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-25 08:26:31.000000000","message":"Patch Set 11: Patch Set 10 was rebased on behalf of \u003cGERRIT_ACCOUNT_23084\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"_revision_number":11},{"id":"36d9ae7b58ecbd7ae0559429b17eb9c9dd3d0178","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-25 08:26:49.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"bf82771c87c9be19bb662f25717abef028fdd803","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-25 09:41:09.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d5ca36a54404c64a928f9bc36b4760d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a1caa27d15214a24a7314a3418176d51 : SUCCESS in 59m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b29ceb753f7845af8536e70662bcd032 : SUCCESS in 21m 22s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7500c4e0a86d42a59edd1eb3b9633380 : SUCCESS in 1h 11m 38s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3a5e0a30e7046c898198bae89ad2a5d : SUCCESS in 2m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5cb1743aa4e746fb90a7dfd27e9346fb : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/14a0ae2bf6dd4543b440d25fa036aef9 : SUCCESS in 2m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5fe078dd5c6746559e22e560de1a9de8 : SUCCESS in 16m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c8166d5395c9482982f10265b32a2132 : SUCCESS in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"4e61f6f4122a751c682c2376a8ec4451dac82958","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-25 10:08:31.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/977b91857be141b191f99c0f22815c56\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4bfd1d78a0054de3850ddad5ae1dc541 : SUCCESS in 18m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/38e1480d0e6d456983562f28a7c17340 : SUCCESS in 57m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/92fbe1bd101740a6909c45cbf82dd0f6 : SUCCESS in 18m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9f2a799c422448f7bd06a741091e3a9f : SUCCESS in 27m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c231a57297c4425db1b37386e38f050a : SUCCESS in 1h 08m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/af595d7d40424683b42705fd7e250409 : SUCCESS in 1h 21m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8595313fa3fb4a6cad95ef98bc643da6 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d03787001de4ac486c51fe39d851b22 : SUCCESS in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2c0bc23ef4e478ca53dddf3ed9871c2 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db03e9f03e214c43b9a71b2eec3ae9cb : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca3f51b04cf54017bd2552f9cca122b5 : SUCCESS in 4m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a96dea30783d4b0085d3dd44894e6a62 : SUCCESS in 19m 55s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4198a4b137eb493fa6b3cece45966716 : SUCCESS in 23m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/62f263f8479446fa8ae2e384fd49d2d4 : SUCCESS in 1h 02m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a941d1044b3145258ef7578e59b350aa : SUCCESS in 1h 37m 03s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1cb4f2f3e95d429991a2f3f9f4e15952 : SUCCESS in 30m 42s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/33891f48cb50445d8291d5d67463c6ba : SUCCESS in 17m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/044f0a3e8f66408685b2c1322603de34 : SUCCESS in 24m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/853ac2a71a2c4278a7df348f7d1d5e58 : SUCCESS in 52m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1c2c08f3ef574fb2adf923759ed43f7c : SUCCESS in 1h 31m 02s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ec9a182d258c47e199333e7bfd757df5 : SUCCESS in 21m 47s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2a5f7edb0ec4771bf5bce71af8d953bc1562360c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-25 10:08:54.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"ca756df85bf1e76fb66abefd60390e0777a88699","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-25 11:53:37.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/663b083953cc4f82a4051658a923d0b8\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/633c01fcc8344480be1f7dd9f98aa497 : SUCCESS in 14m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1f745b94b8a44134bd9c460625c51c65 : SUCCESS in 20m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/13f35d7b8b5e4b60b3046ef4a01cff3e : SUCCESS in 45m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ddd689efd109475c84e5379b3448fba1 : SUCCESS in 1h 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/723601b57091498382868f19b2a0ead1 : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c709c69e5e0473bace0845d0707c353 : SUCCESS in 2m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/509c6bdb146a440ebb4d22f59fdf72e2 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840030a89e7549efb90b89eff60bda8c : SUCCESS in 4m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6f053dcc91584953a9d11f1d3dc7f95c : SUCCESS in 19m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b0a83e4ac2404edc891abc741447c794 : SUCCESS in 27m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cd3ab3fd0fe641a0bd0ef16f894167b0 : SUCCESS in 1h 07m 47s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d0ebeca093d744f693a13c7628b6c63f : SUCCESS in 1h 40m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7458fa63851a42dca7e17f618be22a22 : SUCCESS in 19m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9f3d702982514e80a0e4786f6ed318a2 : SUCCESS in 22m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/215ba48900ea43dc9f6e7cfeffb1cd3c : SUCCESS in 42m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b168de6c18524e9f8e1dcf75bafefba2 : SUCCESS in 1h 14m 01s","accounts_in_message":[],"_revision_number":11},{"id":"b3150ebcc892e7ea1410b50f177002cd55074a60","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-28 08:39:39.000000000","message":"Patch Set 11: Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"df1987c3a7ebec123bf4e8f4a780993cc6a22b40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 08:40:00.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"4fbbe141854f9485c007cc799d76c1f557b5d9ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 10:33:26.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5495cfa863a34f3893761e293b4546e5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/310722f74e2e4a64ab385340da8dcf6b : SUCCESS in 20m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/63c27a7b31de4973ac862ef1555eba76 : SUCCESS in 31m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6156c8a7e7aa41d2b8118819dbf03caa : SUCCESS in 1h 15m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b660d21c537f4e9da2164665cd6ddce2 : SUCCESS in 1h 46m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7d4167e6e5945788b3e26ebd4b39cb6 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/526605f416e6448aa925910be5847fd1 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aeb541f7cacb4d1382f5bb2ef51e2390 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/208f005f53d14c24b4906352c2b57280 : SUCCESS in 5m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/27318e5bbc9a4da5b6c7a13f744d42d7 : SUCCESS in 20m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/627d2c3be1404ad78ae88abc0e42b2f7 : SUCCESS in 26m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d34031f9f3544889a43963a85bf509d6 : SUCCESS in 1h 12m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d013ccc78490461baaefaee6cb0b76ab : SUCCESS in 1h 44m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e33f201df3604fd79bc94bb7c82d53fc : SUCCESS in 20m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b2cd48046fe34e4aa7b2ea2341813ecd : SUCCESS in 35m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bc47c1e1348449b8b49a1298da66c9a5 : SUCCESS in 55m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e9e02c78c79b423fa2ea300bb87d159d : SUCCESS in 1h 38m 04s","accounts_in_message":[],"_revision_number":11},{"id":"3b4099ee530ae7c2adce2b00d7967ce060870e94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-28 10:41:27.000000000","message":"Uploaded patch set 12: New patch set was added with same tree, parent tree, and commit message as Patch Set 11.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":12},{"id":"4e997783d323e4b1c566aadd7e660adb27b099d5","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-28 10:42:04.000000000","message":"Patch Set 12: Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"1826c1c925c180d881d15f83d494588cda0604e4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-28 10:58:53.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"4dde92189433c386d7da49920917c1a65d3d5026","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-28 11:04:20.000000000","message":"Abandoned\n\nlet\u0027s rethink approach in more general fashion in next cycle","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"d3ff7266bade63aa9b079a8b8e7ddb58d211c091","revisions":{"458816bc0aa54f566ae2a0da748023da10616b97":{"kind":"REWORK","_number":1,"created":"2023-11-23 12:01:28.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/901756/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/901756/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/901756/1"}}},"commit":{"parents":[{"commit":"2123435f188dca2ce123960c628692f1e73e5fdb","subject":"Update prometheus services to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2123435f188dca2ce123960c628692f1e73e5fdb"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-23 11:57:25.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-23 12:01:23.000000000","tz":0},"subject":"use sha sums files for prometheus node exporter and msteams","message":"use sha sums files for prometheus node exporter and msteams\n\ninstead of hardcoded values\n\nChange-Id: Idc9fae4a64206410ae1935e3338fb96cf4faa14f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/458816bc0aa54f566ae2a0da748023da10616b97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/458816bc0aa54f566ae2a0da748023da10616b97"}]},"branch":"refs/heads/master"},"3e9c533276eedd6ba8497e3368fc6049c5be66d1":{"kind":"REWORK","_number":2,"created":"2023-11-23 12:13:23.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/901756/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/901756/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/901756/2"}}},"commit":{"parents":[{"commit":"a620d8f39372e18f50c41ee44d58dd4c0f789064","subject":"Update prometheus services to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a620d8f39372e18f50c41ee44d58dd4c0f789064"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-23 11:57:25.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-23 12:13:23.000000000","tz":0},"subject":"use sha sums files for prometheus node exporter and msteams","message":"use sha sums files for prometheus node exporter and msteams\n\ninstead of hardcoded values\n\nChange-Id: Idc9fae4a64206410ae1935e3338fb96cf4faa14f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e9c533276eedd6ba8497e3368fc6049c5be66d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e9c533276eedd6ba8497e3368fc6049c5be66d1"}]},"branch":"refs/heads/master","description":"Rebase"},"5a4e5dfa857a75e12cca4ce67d079670e9efa1c0":{"kind":"REWORK","_number":3,"created":"2023-11-23 12:15:07.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/901756/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/901756/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/901756/3"}}},"commit":{"parents":[{"commit":"a620d8f39372e18f50c41ee44d58dd4c0f789064","subject":"Update prometheus services to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a620d8f39372e18f50c41ee44d58dd4c0f789064"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-23 11:57:25.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-23 12:15:07.000000000","tz":0},"subject":"use sha sums files for prometheus node exporter and msteams","message":"use sha sums files for prometheus node exporter and msteams\n\ninstead of hardcoded values\n\nChange-Id: Idc9fae4a64206410ae1935e3338fb96cf4faa14f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a4e5dfa857a75e12cca4ce67d079670e9efa1c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a4e5dfa857a75e12cca4ce67d079670e9efa1c0"}]},"branch":"refs/heads/master"},"988ec8d6013015874b081b582cc0fd603c43c9a5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-11-23 12:33:23.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/901756/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/901756/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/901756/4"}}},"commit":{"parents":[{"commit":"2a4c3b9ce2c7e5b38a54b3cba18fe8593a43a7f5","subject":"Update prometheus services to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a4c3b9ce2c7e5b38a54b3cba18fe8593a43a7f5"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-23 11:57:25.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-23 12:33:23.000000000","tz":0},"subject":"use sha sums files for prometheus node exporter and msteams","message":"use sha sums files for prometheus node exporter and msteams\n\ninstead of hardcoded values\n\nChange-Id: Idc9fae4a64206410ae1935e3338fb96cf4faa14f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/988ec8d6013015874b081b582cc0fd603c43c9a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/988ec8d6013015874b081b582cc0fd603c43c9a5"}]},"branch":"refs/heads/master","description":"Rebase"},"0c907e05b4c6e45c453b8f1a6afb16d6abdf9dfc":{"kind":"REWORK","_number":5,"created":"2023-11-24 10:14:50.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/901756/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/901756/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/901756/5"}}},"commit":{"parents":[{"commit":"2123435f188dca2ce123960c628692f1e73e5fdb","subject":"Update prometheus services to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2123435f188dca2ce123960c628692f1e73e5fdb"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-23 11:57:25.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-24 10:14:41.000000000","tz":0},"subject":"use sha sums files for prometheus node exporter and msteams","message":"use sha sums files for prometheus node exporter and msteams\n\ninstead of hardcoded values\n\nChange-Id: Idc9fae4a64206410ae1935e3338fb96cf4faa14f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c907e05b4c6e45c453b8f1a6afb16d6abdf9dfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c907e05b4c6e45c453b8f1a6afb16d6abdf9dfc"}]},"branch":"refs/heads/master"},"114667eec1484a735752deba36ac5b9246cdd3bd":{"kind":"REWORK","_number":6,"created":"2023-11-24 10:15:29.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/901756/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/901756/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/901756/6"}}},"commit":{"parents":[{"commit":"ead8aea601bef6e98add4f064a20cb1ecb9c0f5e","subject":"Update prometheus services to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ead8aea601bef6e98add4f064a20cb1ecb9c0f5e"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-23 11:57:25.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-24 10:15:29.000000000","tz":0},"subject":"use sha sums files for prometheus node exporter and msteams","message":"use sha sums files for prometheus node exporter and msteams\n\ninstead of hardcoded values\n\nChange-Id: Idc9fae4a64206410ae1935e3338fb96cf4faa14f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/114667eec1484a735752deba36ac5b9246cdd3bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/114667eec1484a735752deba36ac5b9246cdd3bd"}]},"branch":"refs/heads/master","description":"Rebase"},"a95504e24b85f2a0fca50db14febf8e26dde21dc":{"kind":"REWORK","_number":7,"created":"2023-11-24 10:16:10.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/901756/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/901756/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/901756/7"}}},"commit":{"parents":[{"commit":"ead8aea601bef6e98add4f064a20cb1ecb9c0f5e","subject":"Update prometheus services to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ead8aea601bef6e98add4f064a20cb1ecb9c0f5e"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-23 11:57:25.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-24 10:16:10.000000000","tz":0},"subject":"use sha sums files for prometheus node exporter and msteams","message":"use sha sums files for prometheus node exporter and msteams\n\ninstead of hardcoded values\n\nChange-Id: Idc9fae4a64206410ae1935e3338fb96cf4faa14f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a95504e24b85f2a0fca50db14febf8e26dde21dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a95504e24b85f2a0fca50db14febf8e26dde21dc"}]},"branch":"refs/heads/master"},"ab93c515d620c8e1a80cba16de36fcf26c1b8914":{"kind":"REWORK","_number":8,"created":"2023-11-24 10:52:15.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/901756/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/901756/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/901756/8"}}},"commit":{"parents":[{"commit":"2123435f188dca2ce123960c628692f1e73e5fdb","subject":"Update prometheus services to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2123435f188dca2ce123960c628692f1e73e5fdb"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-23 11:57:25.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-24 10:51:05.000000000","tz":0},"subject":"use sha sums files for prometheus node exporter and msteams","message":"use sha sums files for prometheus node exporter and msteams\n\ninstead of hardcoded values\n\nChange-Id: Idc9fae4a64206410ae1935e3338fb96cf4faa14f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab93c515d620c8e1a80cba16de36fcf26c1b8914"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab93c515d620c8e1a80cba16de36fcf26c1b8914"}]},"branch":"refs/heads/master"},"0152c0f8c376ec2e921646f5dddd9f21be6bde21":{"kind":"REWORK","_number":9,"created":"2023-11-24 10:54:24.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/901756/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/901756/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/901756/9"}}},"commit":{"parents":[{"commit":"ead8aea601bef6e98add4f064a20cb1ecb9c0f5e","subject":"Update prometheus services to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ead8aea601bef6e98add4f064a20cb1ecb9c0f5e"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-23 11:57:25.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-24 10:54:24.000000000","tz":0},"subject":"use sha sums files for prometheus node exporter and msteams","message":"use sha sums files for prometheus node exporter and msteams\n\ninstead of hardcoded values\n\nChange-Id: Idc9fae4a64206410ae1935e3338fb96cf4faa14f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0152c0f8c376ec2e921646f5dddd9f21be6bde21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0152c0f8c376ec2e921646f5dddd9f21be6bde21"}]},"branch":"refs/heads/master","description":"Rebase"},"9be4b21684e7dec7eb54c5129cfe43dcc86f32f7":{"kind":"REWORK","_number":10,"created":"2023-11-24 10:54:53.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/901756/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/901756/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/901756/10"}}},"commit":{"parents":[{"commit":"ead8aea601bef6e98add4f064a20cb1ecb9c0f5e","subject":"Update prometheus services to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ead8aea601bef6e98add4f064a20cb1ecb9c0f5e"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-23 11:57:25.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-24 10:54:53.000000000","tz":0},"subject":"use sha sums files for prometheus node exporter and msteams","message":"use sha sums files for prometheus node exporter and msteams\n\ninstead of hardcoded values\n\nChange-Id: Idc9fae4a64206410ae1935e3338fb96cf4faa14f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9be4b21684e7dec7eb54c5129cfe43dcc86f32f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9be4b21684e7dec7eb54c5129cfe43dcc86f32f7"}]},"branch":"refs/heads/master"},"e4aff9ec4a10739be797c61d357e8e21498305e7":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-11-25 08:26:31.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/901756/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/901756/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/901756/11"}}},"commit":{"parents":[{"commit":"773c6963e2405af3d535c41d7794361a079b71c4","subject":"Update Prometheus services to latest releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/773c6963e2405af3d535c41d7794361a079b71c4"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-23 11:57:25.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-25 08:26:31.000000000","tz":0},"subject":"use sha sums files for prometheus node exporter and msteams","message":"use sha sums files for prometheus node exporter and msteams\n\ninstead of hardcoded values\n\nChange-Id: Idc9fae4a64206410ae1935e3338fb96cf4faa14f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4aff9ec4a10739be797c61d357e8e21498305e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4aff9ec4a10739be797c61d357e8e21498305e7"}]},"branch":"refs/heads/master","description":"Rebase"},"d3ff7266bade63aa9b079a8b8e7ddb58d211c091":{"kind":"NO_CHANGE","_number":12,"created":"2023-11-28 10:41:27.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/56/901756/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/901756/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/901756/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/901756/12"}}},"commit":{"parents":[{"commit":"773c6963e2405af3d535c41d7794361a079b71c4","subject":"Update Prometheus services to latest releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/773c6963e2405af3d535c41d7794361a079b71c4"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-28 10:41:20.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-28 10:41:26.000000000","tz":60},"subject":"use sha sums files for prometheus node exporter and msteams","message":"use sha sums files for prometheus node exporter and msteams\n\ninstead of hardcoded values\n\nChange-Id: Idc9fae4a64206410ae1935e3338fb96cf4faa14f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3ff7266bade63aa9b079a8b8e7ddb58d211c091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3ff7266bade63aa9b079a8b8e7ddb58d211c091"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
