)]}'
{"id":"openstack%2Fkolla~901857","triplet_id":"openstack%2Fkolla~stable%2Fyoga~If4e346c2db841aad9b2ddac049dbbbc1ba5782ec","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-12-21 06:44:43.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-12-21 08:22:19.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-12-04 10:11:17.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-12-20 10:09:53.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"If4e346c2db841aad9b2ddac049dbbbc1ba5782ec","subject":"Swift: Remove unused rootwrap","status":"MERGED","created":"2023-11-27 11:56:09.000000000","updated":"2023-12-21 08:23:31.000000000","submitted":"2023-12-21 08:22:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901857","cherry_pick_of_change":886380,"cherry_pick_of_patch_set":3,"meta_rev_id":"886daa339bae4f37b1167c5fb79b8f9269ab090d","_number":901857,"virtual_id_number":901857,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-21 08:22:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-12-21 06:44:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-11-29 21:46:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-12-04 10:11:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-12-08 01:41:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-12-20 10:09:53.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-08 01:31:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-12-21 06:44:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-27 11:56:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-11-27 12:19:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-03 12:40:38.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-12-07 22:52:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-12-08 01:31:53.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-12-08 01:41:50.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-12-08 09:18:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"f84c66bbe347c0f225f7123649f3fc2cc7ce6e6d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-27 11:56:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/swift/swift-base/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"95da26c07e0d903d18de0e33896d1047aea9e4bc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-27 12:00:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"820f7df319f101df9b8d8abb317c26f20c62ec6c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-27 12:02:02.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"04d77ebe0c548dea7e2bc033355c80aca0342f4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 12:19:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db9774103f2146de84673d1bbd93e99a\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7388114ef4124bc79d401c5da2686f44 : FAILURE in 7m 12s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ceee666fdb8b4463a8588900faeaf99f : FAILURE in 6m 28s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/eeaeeb25bebe411ea6a97456409b4467 : FAILURE in 5m 35s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0210ce42a9e240749f1074e3ad3e8e6a : FAILURE in 12m 35s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f529d7eecb6246ec9a06b30722556882 : FAILURE in 12m 46s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/5a46cb60d3564012bd712d5cb3dc909e : FAILURE in 10m 31s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e160913b3bfb45eca07bde12c9497e71 : FAILURE in 3m 16s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f9ec9971d3ef4ef1896d40d43b3f074d : FAILURE in 4m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8e9b3d9d15564746b36eed5f663f6d0f : FAILURE in 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/089fd0da517d48739ad4c6ee82fc3329 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9ddf1f499ea45e6a5b43c5512946f27 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36795bb0b0b147aba3bfb8f9e7704d42 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66868233bec7422886c1123f44c75f32 : SUCCESS in 5m 38s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/05ee9947f0ae46989ee077e170299c9d : FAILURE in 5m 48s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/86789d576480473cbe002cb09e0df664 : FAILURE in 9m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/281cc120ea734c6db8bba4c6ea6d70cf : FAILURE in 3m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f196908248624046912c47369b5a4ef3 : FAILURE in 3m 22s\n- kolla-ansible-ubuntu-binary-focal https://zuul.opendev.org/t/openstack/build/477515e6023f45f8af20ed955c5dc474 : FAILURE in 10m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b2db2c6c5a07dde747613e4531bf94436bdb8b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 13:06:54.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d53d34a09c174619b9de070ffd3223cb\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1863f0cd9c144262ab408e1b4a7e8ccd : FAILURE in 6m 30s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e8dcaeb3b1b84d4dbaa45058663a9f45 : FAILURE in 5m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0c69c6c8b4ee4643b3b7a09573ff568d : FAILURE in 3m 58s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/807fcef119b142aca52d3f44b40e39fb : FAILURE in 17m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6602b901fde40d38f8aa704f6993791 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9abe0aa785f246aabaf11aadf6d2dd2b : SUCCESS in 7m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a209e3c54dca4e32ab863102f0bc4739 : FAILURE in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f236a9ff8c704bab9888fe52584694de7b7307f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-27 20:23:19.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fb74926c1fb01bdcf75e8d7356a2a3e1b9f6ebe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-27 20:24:55.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"eddab614cf63fd78f1a3ad16b3d445676e0907de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-27 20:28:10.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"4ac677ad70dc612af0bb3cd15c1b6597d43444e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 21:55:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4be6ade34ca2410a97e6d72aa4563ac9\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/178a0789b6304ba19214b663d04df2a1 : SUCCESS in 39m 20s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e8f66b328fdb407fbb62562324eae17c : SUCCESS in 38m 46s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/a6c898c2c376427d87b1a60a7ffbafa8 : SUCCESS in 31m 16s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5e3bc0f57121470baf4aa98e26834bf2 : SUCCESS in 1h 20m 19s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4cb6738c989d493191d66da6fc59a52a : SUCCESS in 1h 06m 09s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/c590664990944778aa3f74e417f33d3a : SUCCESS in 1h 02m 49s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/726f3798eea74da8b95fc3c8cf9bd85a : SUCCESS in 18m 24s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/181621e3311f407f80d49f6ff013629b : SUCCESS in 25m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/30402ada5f654f499f9b48f2b3bb25a9 : SUCCESS in 1h 05m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/356d5e3d2d8c4dfda0683761bab3f952 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc439a3c8d2f4011a8556fcdcb6904c0 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89762594bae64dadb7fa01311b82d7e8 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b13e052e12614102995d9b9b0fa28a3c : SUCCESS in 4m 45s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/a8f5a0251dbd4792a5bb69bab94dd9e0 : SUCCESS in 25m 17s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/91c43a9520f7481c8bfb0a8263dd8fbe : SUCCESS in 1h 05m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/79e217d1fb314549801bd1adeddcf738 : SUCCESS in 17m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/946742714df84a208b8c1a68636fc2b0 : FAILURE in 25m 52s\n- kolla-ansible-ubuntu-binary-focal https://zuul.opendev.org/t/openstack/build/f014a43b087b494b82570a467ed24bf7 : SUCCESS in 52m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c67cf4d0c3b474916613cf7e16bea4047c7b14f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 22:36:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12f5bd3602e04e2a86f985c9263e8c60\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9c656c59f04243408d1e0207c98b5fc6 : SUCCESS in 28m 08s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/aeb967cfd84d412d934c8181da17d7d3 : SUCCESS in 22m 15s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/36cf65f1e53c446eba391ad3e5aaa839 : SUCCESS in 22m 24s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d1051ec0de54f3a84f4a4d03c6483bd : SUCCESS in 2h 08m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b6a5e7b9b1c42c2a9a562c3d9e76cf5 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e1d2e58dc61741819eeec1d808699a5d : SUCCESS in 4m 58s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4163a98f694a4471b2b5ac90947be241 : FAILURE in 17m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"aeb7c26a0c4467bcae939560491d653960619fe8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-29 21:46:10.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dfbea83c1e048fe29cdab647e502988c8dc30e3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 23:07:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c2027ec39804419b3c8d5ed755806c0\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/09692fe3abd74ee78bfeb297ba859f09 : SUCCESS in 32m 58s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/04fb01bfc0b640e1bb3546d3deb22341 : SUCCESS in 23m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/87ef29a38fe14d8f949c00b792ad39e5 : SUCCESS in 24m 04s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/910b32343bc14204bf407bca7fb1ee31 : SUCCESS in 1h 20m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/215e10fbedf34f3b9675b190666d2653 : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3e20c180d2b4f09a63477fdbd14f2b1 : SUCCESS in 6m 03s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fc01ad522c054c6dad2e5d0ddef48fb5 : FAILURE in 20m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7d7e46212d8c017df6d3eec0c845195d93f20845","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 23:16:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47337ca529ee4ecfb09180b0cf93ebc5\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c175698880c8486b992ce5563ccdcd5e : SUCCESS in 44m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a9c53018c86449f7a80b61d88d51179d : SUCCESS in 33m 56s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/9222df92f30e429a8f9793d4ac54a0ad : SUCCESS in 27m 35s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b48202b192344e4bbf3bcf479694031b : SUCCESS in 1h 24m 47s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/28632aaa004647d7bf886ea716eb87db : SUCCESS in 1h 03m 11s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/5f38368e780342a7809f544713dd0f0d : SUCCESS in 56m 30s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c8c8673bbd90421f81a956a008ba13f9 : SUCCESS in 25m 29s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/146dfc78c3504ebf9578c753ebf6f12f : SUCCESS in 16m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a40931864dc94dbb9e0cb8046352c668 : SUCCESS in 52m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae949e85693e4208bf790f9970570fef : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99cb23e3f902445a892ad70ae076da4f : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/873641ce0b124e5f9dd8e48dcfabc6b7 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fac04f6dbd6a412e8b26ff6637d8877b : SUCCESS in 5m 19s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/a0f1ae8de5ce487c8867106c1241b6da : SUCCESS in 16m 24s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/2cc76ecfe2cc47ec9ea6526f49845ef3 : SUCCESS in 1h 05m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/acf363679c09452d87588f1e8378ffdb : SUCCESS in 19m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9ce034754c414c41a57faad8cb990af9 : SUCCESS in 26m 13s\n- kolla-ansible-ubuntu-binary-focal https://zuul.opendev.org/t/openstack/build/b0b07f71cbdd4c48855354d5cfe37d27 : SUCCESS in 56m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dad794b1a1cc6feaf4a9d92984b048b9b38afd4b","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-12-04 10:11:17.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"eef8fc3166fb10533ea65e6a09f3d1711f9eb146","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-12-08 01:31:53.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"956224d1c7f9225cb27a66f37e721894d6e6f9f9","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-12-08 01:41:50.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"ee2a7a2b26b2618fe1e5a57433dc91794803d631","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-12-20 10:09:53.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0a83d3d86b3acebe119e7a24adcba01f2e4d8c1c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-21 06:44:43.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4daf6a98337cff0c04bdba0023b82a62a224cce1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 06:45:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8f1629aa734b54679cef184a7de648bd6dea0d0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 08:22:18.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b99de6a06c4b4c288645c26fef50511d\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/85b465c9396f43bda760a8532d238d0a : SUCCESS in 29m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f94e5da335dc45db9bb91fda48f2d15c : SUCCESS in 1h 31m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bc1ec0763f8d445abe49a32fdadf7889 : SUCCESS in 30m 29s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6b669f0ddede4485ae159b50c0545fea : SUCCESS in 1h 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67e5a586c1f04d33858837ac03e80f26 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4db7a1419bd24ed1b2dc8dc6468d0361 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b136b471809440afb806821d680747a6 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a909dc12bf54e059ba45418f9b2e6b3 : SUCCESS in 5m 16s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a33d1479a13c4c62baddec19ffc6b05d : SUCCESS in 1h 09m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/51838aa5e1a2409580e6fcbed92fbd87 : SUCCESS in 23m 47s","accounts_in_message":[],"_revision_number":5},{"id":"24fdb5e2bd372278dfcb392d8674c67d94195a3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 08:22:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"886daa339bae4f37b1167c5fb79b8f9269ab090d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 08:23:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cbf0d94dd2d4fe1aa26444d0d3ced33\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22d66081220449c295206e68d6b192be : SUCCESS in 51s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3a484edcd621126f149c474c768f17741f9eb315","revisions":{"c6558bd7395eff14f3aa9833705fc897368cdaba":{"kind":"REWORK","_number":1,"created":"2023-11-27 11:56:09.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/57/901857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/901857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/901857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/901857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/901857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/901857/1"}}},"commit":{"parents":[{"commit":"8304b9456cc38686feb28ace6a44404d4a60f8fb","subject":"Fix missing grafana opensearch datasource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8304b9456cc38686feb28ace6a44404d4a60f8fb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 15:11:29.000000000","tz":540},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-27 11:56:09.000000000","tz":0},"subject":"Swift: Remove unused rootwrap","message":"Swift: Remove unused rootwrap\n\nSwift does not use oslo.rootwrap. Thus the command and its config file\nis just useless.\n\nChange-Id: If4e346c2db841aad9b2ddac049dbbbc1ba5782ec\n(cherry picked from commit d7e497ce52cc109c9f7b4a83d85f0bebfa05298e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6558bd7395eff14f3aa9833705fc897368cdaba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6558bd7395eff14f3aa9833705fc897368cdaba"}]},"branch":"refs/heads/stable/yoga"},"c7a9e41128d9e50502b4d6bbfaf11fdb1d3d214c":{"kind":"REWORK","_number":2,"created":"2023-11-27 12:00:36.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/57/901857/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/901857/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/901857/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/901857/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/901857/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/901857/2"}}},"commit":{"parents":[{"commit":"8304b9456cc38686feb28ace6a44404d4a60f8fb","subject":"Fix missing grafana opensearch datasource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8304b9456cc38686feb28ace6a44404d4a60f8fb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 15:11:29.000000000","tz":540},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-27 11:58:14.000000000","tz":180},"subject":"Swift: Remove unused rootwrap","message":"Swift: Remove unused rootwrap\n\nSwift does not use oslo.rootwrap. Thus the command and its config file\nis just useless.\n\nChange-Id: If4e346c2db841aad9b2ddac049dbbbc1ba5782ec\n(cherry picked from commit d7e497ce52cc109c9f7b4a83d85f0bebfa05298e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7a9e41128d9e50502b4d6bbfaf11fdb1d3d214c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7a9e41128d9e50502b4d6bbfaf11fdb1d3d214c"}]},"branch":"refs/heads/stable/yoga"},"62420f0f655d0f823fa584e885d16a1e155850da":{"kind":"REWORK","_number":3,"created":"2023-11-27 20:23:19.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/57/901857/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/901857/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/901857/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/901857/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/901857/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/901857/3"}}},"commit":{"parents":[{"commit":"8304b9456cc38686feb28ace6a44404d4a60f8fb","subject":"Fix missing grafana opensearch datasource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8304b9456cc38686feb28ace6a44404d4a60f8fb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 15:11:29.000000000","tz":540},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-27 20:23:19.000000000","tz":0},"subject":"Swift: Remove unused rootwrap","message":"Swift: Remove unused rootwrap\n\nSwift does not use oslo.rootwrap. Thus the command and its config file\nis just useless.\n\nChange-Id: If4e346c2db841aad9b2ddac049dbbbc1ba5782ec\n(cherry picked from commit d7e497ce52cc109c9f7b4a83d85f0bebfa05298e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62420f0f655d0f823fa584e885d16a1e155850da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62420f0f655d0f823fa584e885d16a1e155850da"}]},"branch":"refs/heads/stable/yoga"},"caa7361d94aff717d2e85b60940d05ed3788d4d1":{"kind":"REWORK","_number":4,"created":"2023-11-27 20:24:55.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/57/901857/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/901857/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/901857/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/901857/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/901857/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/901857/4"}}},"commit":{"parents":[{"commit":"8304b9456cc38686feb28ace6a44404d4a60f8fb","subject":"Fix missing grafana opensearch datasource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8304b9456cc38686feb28ace6a44404d4a60f8fb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 15:11:29.000000000","tz":540},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-27 20:24:55.000000000","tz":0},"subject":"Swift: Remove unused rootwrap","message":"Swift: Remove unused rootwrap\n\nSwift does not use oslo.rootwrap. Thus the command and its config file\nis just useless.\n\nChange-Id: If4e346c2db841aad9b2ddac049dbbbc1ba5782ec\n(cherry picked from commit d7e497ce52cc109c9f7b4a83d85f0bebfa05298e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/caa7361d94aff717d2e85b60940d05ed3788d4d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/caa7361d94aff717d2e85b60940d05ed3788d4d1"}]},"branch":"refs/heads/stable/yoga"},"3a484edcd621126f149c474c768f17741f9eb315":{"kind":"REWORK","_number":5,"created":"2023-11-27 20:28:10.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/57/901857/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/901857/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/901857/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/901857/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/901857/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/901857/5"}}},"commit":{"parents":[{"commit":"8304b9456cc38686feb28ace6a44404d4a60f8fb","subject":"Fix missing grafana opensearch datasource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8304b9456cc38686feb28ace6a44404d4a60f8fb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 15:11:29.000000000","tz":540},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-27 20:28:10.000000000","tz":0},"subject":"Swift: Remove unused rootwrap","message":"Swift: Remove unused rootwrap\n\nSwift does not use oslo.rootwrap. Thus the command and its config file\nis just useless.\n\nChange-Id: If4e346c2db841aad9b2ddac049dbbbc1ba5782ec\n(cherry picked from commit d7e497ce52cc109c9f7b4a83d85f0bebfa05298e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a484edcd621126f149c474c768f17741f9eb315"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a484edcd621126f149c474c768f17741f9eb315"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
