)]}'
{"id":"openstack%2Fkolla~901900","triplet_id":"openstack%2Fkolla~master~I28d0ba5f92a658a5a183f8936a785e2f4989ee19","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-03-28 14:47:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I28d0ba5f92a658a5a183f8936a785e2f4989ee19","subject":"DNM: build bifrost 2nd approach","status":"ABANDONED","created":"2023-11-27 08:10:58.000000000","updated":"2024-03-28 14:47:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"2dd7079734cea371c7f8bd6df3ef3c9aee5f135b","_number":901900,"virtual_id_number":901900,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-27 10:25:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-11-27 09:46:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-27 10:25:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 12:54:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"09acd647c4d4fd390532776c6ccd85f569d96999","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-27 08:10:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c38ee86df4ab00692c94bd24c27e57ba3d6d3c0e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-27 08:34:23.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"50769892fbb5b9b0b9d0ff8e02b3dabf15234f9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 09:46:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4feb04f30874fd4bc56f038706407a8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b5af3e4f8d7a4236a0a2a4e20448b6db : SUCCESS in 20m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/250c1f608b034f9d89539576c1cdd74e : SUCCESS in 15m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2bb6f7b49769463e8b0426b8f3730637 : SUCCESS in 1h 11m 10s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6647b4b16dd4f15b752522c0cf821f7 : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/20c5fcd736cc4e7a8f2e2b26fb0ac1e7 : SUCCESS in 2m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/74d7bed736ad429f835e22539d6556bf : SUCCESS in 5m 20s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e0204ae7e1a94eff9a2ec23c17f2e8b5 : SUCCESS in 1h 02m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/44b719c9665a459e8d38191428fb2545 : SUCCESS in 14m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cdef51d5a269b2c148956eeaedc4e1ad972ee3ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 10:25:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5f0c9d9243f440f8b9af73d502975ad\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/06e39162ce4042bb980ec7c7a0cb07cc : FAILURE in 28m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/10038a2027eb434ea06a4a71e0e776a3 : SUCCESS in 47m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4f5b3641a22b4ac58d0abcdee6bef247 : SUCCESS in 27m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fb5548a2e8cb4945bb6bc4d5b6dcc3cb : SUCCESS in 24m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1370269913504a1c9eac9d9e68bef687 : SUCCESS in 49m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a800b2da2984408d9f918628b9ebbaa4 : SUCCESS in 1h 21m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/739ea49f005c47a9a1528ea056805923 : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13823f4298f34370b6621e4ce03dc6aa : SUCCESS in 2m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f2b1841e4b64adf8b4f9a285febf4fa : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af0dd4653b034d11b18179e95dd9392f : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70f4a63c96964915a7f1395cb394e321 : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b812b849d8274710a2dcdba043f1fde3 : FAILURE in 31m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8d061d0354ee4a118a783c34034be172 : FAILURE in 40m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/07d9476def34464d95f7fdce1e9e139b : SUCCESS in 1h 04m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3bcc514bdc044ce48a3a0b4e6c6e1604 : SUCCESS in 1h 42m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3dbcf964094b4bc18f1ed585f9623d9e : SUCCESS in 20m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/db2ba734a496473c80c1d3feceeacc81 : SUCCESS in 24m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8a28239fc1194f929852c1bd5cc02c3f : SUCCESS in 43m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3abbf520f74242db83c8fab1ddbc8891 : SUCCESS in 1h 24m 36s","accounts_in_message":[],"_revision_number":2},{"id":"2dd7079734cea371c7f8bd6df3ef3c9aee5f135b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-28 14:47:15.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6a2a4c3d35d1040c79246fb7c4cb6f73ef013718","revisions":{"712a74621cb144fbd4c3a79cc414fd5e0ea7f097":{"kind":"REWORK","_number":1,"created":"2023-11-27 08:10:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/901900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/901900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/901900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/901900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/901900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/901900/1"}}},"commit":{"parents":[{"commit":"a9b4c5361f09b4ebac96c4c8eb5afb3abe26d344","subject":"Merge \"Swift: Remove unused rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9b4c5361f09b4ebac96c4c8eb5afb3abe26d344"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-27 08:10:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-27 08:10:55.000000000","tz":60},"subject":"DNM: build bifrost 2nd approach","message":"DNM: build bifrost 2nd approach\n\nChange-Id: I28d0ba5f92a658a5a183f8936a785e2f4989ee19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/712a74621cb144fbd4c3a79cc414fd5e0ea7f097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/712a74621cb144fbd4c3a79cc414fd5e0ea7f097"}]},"branch":"refs/heads/master"},"6a2a4c3d35d1040c79246fb7c4cb6f73ef013718":{"kind":"REWORK","_number":2,"created":"2023-11-27 08:34:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/901900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/901900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/901900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/901900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/901900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/901900/2"}}},"commit":{"parents":[{"commit":"a9b4c5361f09b4ebac96c4c8eb5afb3abe26d344","subject":"Merge \"Swift: Remove unused rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9b4c5361f09b4ebac96c4c8eb5afb3abe26d344"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-27 08:10:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-27 08:34:23.000000000","tz":0},"subject":"DNM: build bifrost 2nd approach","message":"DNM: build bifrost 2nd approach\n\nChange-Id: I28d0ba5f92a658a5a183f8936a785e2f4989ee19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a2a4c3d35d1040c79246fb7c4cb6f73ef013718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a2a4c3d35d1040c79246fb7c4cb6f73ef013718"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
