)]}'
{"id":"openstack%2Fkolla~901909","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I07be614846be9305f0775df22a6d931ceab1720a","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-12-21 06:45:04.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-12-21 08:22:24.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-12-04 10:22:01.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-12-20 10:07:52.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"I07be614846be9305f0775df22a6d931ceab1720a","subject":"Add rsync to nova-compute","status":"MERGED","created":"2023-11-27 11:53:31.000000000","updated":"2023-12-21 08:24:14.000000000","submitted":"2023-12-21 08:22:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901909","meta_rev_id":"5b12f823dd5f75a2edb0ff1a0bba99708ec5e130","_number":901909,"virtual_id_number":901909,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-21 08:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-12-21 06:45:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-11-29 21:43:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-12-04 10:22:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-12-11 01:38:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-12-20 10:07:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-12-21 06:45:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-27 11:53:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"state":"CC"},{"updated":"2023-11-27 13:14:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-03 12:41:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-12-07 22:51:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-12-08 09:18:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-12-11 01:38:44.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"e0765fbd27679b66733e6656f2eb97914c0faec8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-27 11:53:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1608dce00d47417b96002a0b869c39417e8887dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 13:14:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0c690c4b3f74d3c9620bd3bce57aa15\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/51b5da477ac14c459d5f491a320d467d : SUCCESS in 32m 12s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c1b1dc69deb34fdaa5bedb7359bf8a2b : SUCCESS in 34m 49s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/c44cf780adff46b2bcc01fe636c4ea64 : SUCCESS in 29m 42s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2b3eaf4bac9946fc8d8cb5cc91fc68f8 : SUCCESS in 1h 16m 11s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0f80226130bd46889ccebf37bb8a5017 : SUCCESS in 59m 25s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/3a5198bf196e4fd1918f11ce196858d6 : SUCCESS in 1h 08m 33s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c641ed3526054e13a9fdcbb8eeb580b9 : SUCCESS in 24m 11s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/abc84a6a82124cfe86e6b49fb8f993ca : SUCCESS in 22m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6263b0205a6549768c34528a5e675110 : SUCCESS in 54m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/482a7dccdd6b4f09b9215693d0bba0e9 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89fa953f049242898e1f843a7be4e5f3 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a7bde3f9d0e4edabe3ea3b5aa6211f3 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/447de9fe148f44ad8275cb5bfc380594 : SUCCESS in 5m 16s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/534611cf4cb84c9ea317c3bf0bf58951 : SUCCESS in 27m 43s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/77f460bd9b964a0f87410b4754e21584 : SUCCESS in 1h 06m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/49cf20f0a1ee4299bfc17392f7873730 : SUCCESS in 21m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6fe3e874ad38436d85de460dfc77ffa3 : SUCCESS in 26m 24s\n- kolla-ansible-ubuntu-binary-focal https://zuul.opendev.org/t/openstack/build/30dd148fa62849e3a80b625f80908b46 : SUCCESS in 52m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03778cacda55a3bf32209f6a75712780b2d64460","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 14:27:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f13ea6bbd8fd473aae70d470c054194e\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/65cae7b8ca67429691cfa90b577811dc : SUCCESS in 30m 04s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6adb6128234841758b8aea022bae725f : SUCCESS in 31m 29s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e576fc1bd05d4bd0891a2f94e8e7a3dd : SUCCESS in 26m 58s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f426732a2cc94fd194baea46e269f19d : SUCCESS in 2h 31m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9009a4ac10da4059935c56d7f74a6c3c : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c389bb020b0b4d6f9c24e3c123f25f25 : SUCCESS in 6m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c0380a340dc4a19863517b45564246a : FAILURE in 24m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3dc377a06e51c2faa3f8931bdc8a7ada71d31cb5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-29 21:43:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e348e9ee7154a5a96b3968a0275288922342cc78","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-12-04 10:22:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fcdecf4d6d743e1e0498ab1bee56f95e324a1d0c","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-12-11 01:38:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"129a840a92df383be1dd3cb4feb9ab4a3f89f3b1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-12-20 10:07:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a3a27e067c247ca07faa5bb1ccf4812cde580f8c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-21 06:45:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aefcc604bb7ba7603f84868cca045d6e63fc5745","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 06:45:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"11d3c1d5ebc4ef92f4191ae6b898035674497cd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 08:22:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85d39e4405bb4f14b97aaed299a9805b\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/17bcdced6ba546718c628d0bc768f09f : SUCCESS in 31m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/68f6b383d16642be9a07b9dfcfcd9a87 : SUCCESS in 1h 19m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cf1d27baecf44bad968a253efe46f5cf : SUCCESS in 26m 50s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/80444862e12343baa025a4d8ca98566b : SUCCESS in 1h 01m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b47a4c0f9ac4a46aa865164659174b6 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c42a759578aa40ffb0f35069fcde6aff : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fce871e95454b78b2a0a124c76a7437 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cbf4896a28d4d1390e96340a7462e8e : SUCCESS in 5m 39s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f70bb889f78d4d20a4738424619a5461 : SUCCESS in 52m 24s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/10e0de57370e403985ee87b212cfff7e : SUCCESS in 31m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2cfb4c83fa7e95ac223c8df35d50cd13044f55bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 08:22:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5b12f823dd5f75a2edb0ff1a0bba99708ec5e130","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 08:24:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98bf2954462a4319921f2ffc73ec2a07\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85f2682b05e44f60862bea5af2466ecf : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be43974c5f1d37d91ba0b3360a24547860a9a485","revisions":{"be43974c5f1d37d91ba0b3360a24547860a9a485":{"kind":"REWORK","_number":1,"created":"2023-11-27 11:53:31.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/09/901909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/901909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/901909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/901909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/901909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/901909/1"}}},"commit":{"parents":[{"commit":"8304b9456cc38686feb28ace6a44404d4a60f8fb","subject":"Fix missing grafana opensearch datasource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8304b9456cc38686feb28ace6a44404d4a60f8fb"}]}],"author":{"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","date":"2023-10-22 07:41:30.000000000","tz":480},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-27 11:52:47.000000000","tz":180},"subject":"Add rsync to nova-compute","message":"Add rsync to nova-compute\n\nwhen configure `remote_filesystem_transport` to rsync, rsync binary is\nrequired.\n\nChange-Id: I07be614846be9305f0775df22a6d931ceab1720a\n(cherry picked from commit 92cedcd2b155907ab65884a0ba4dbae480900d28)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be43974c5f1d37d91ba0b3360a24547860a9a485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be43974c5f1d37d91ba0b3360a24547860a9a485"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
