)]}'
{"id":"openstack%2Fkolla~902053","triplet_id":"openstack%2Fkolla~master~I473ea1c68d2fd7f15fa0545d77892276b1accfeb","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-11-29 08:43:45.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-11-28 16:11:24.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-11-29 10:19:47.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-11-29 10:19:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I473ea1c68d2fd7f15fa0545d77892276b1accfeb","subject":"Handle the case where docker is not available","status":"MERGED","created":"2023-11-28 11:03:15.000000000","updated":"2023-11-29 10:20:58.000000000","submitted":"2023-11-29 10:19:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902053","meta_rev_id":"a15ff146783fefa871e4ecca2a030f79998f3115","_number":902053,"virtual_id_number":902053,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-29 10:19:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-11-29 08:43:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-11-28 13:58:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-11-28 16:11:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-11-29 06:08:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-11-29 08:43:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2023-11-29 08:08:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-28 11:32:28.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-11-28 12:51:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-28 12:57:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-28 13:58:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-11-28 14:45:37.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-11-29 06:08:10.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-11-29 08:08:34.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"5c28fda84cd34b1f13d66036d82675022e41aa68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-28 11:03:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be95109d8da2046bb71881449efccf12ca22fe67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-28 11:07:42.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a69b345c9ab39c4bee097b0f7c3e0944efe3e16e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 12:51:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b839dbe3618e4151a7797d90d644ee6f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a3f9b872cfbd45e9902690b2c391e983 : SUCCESS in 26m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8556f3aabf6a4b7f9da83718c275ed58 : SUCCESS in 25m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8efa047192174499a4a6613c4e0cb12b : SUCCESS in 1h 15m 19s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fbb6913de69e4c5baec7d8c0224763a9 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3d4e6d7564ae47ba950c3b7a48072e11 : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5db5fa460c5c43cb95f4503ff1b6eb21 : SUCCESS in 3m 03s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8fd2c09c75ee4c9b9656b745bf774ad3 : SUCCESS in 1h 40m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5919e4079e4a40e29fef66c9b5184bff : SUCCESS in 23m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5a6424cea9e9db78d0d353dec427f39dbd53a3aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 12:57:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eef5abcf1ac401ab41dd215899fbabd\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/400e041c503142e791736e1b43a264d0 : SUCCESS in 16m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/51069620e8b24f979b7ec4c57c7fc882 : SUCCESS in 53m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/69458a7a3d76426a8bb531860cf48218 : SUCCESS in 22m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f0ce4c289c024141bbb9c6eec721a818 : SUCCESS in 24m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3afc2dabee5c43ef9b8a3a9b1bf0c626 : SUCCESS in 52m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/aa158a9c600c4963a37c91f26df85909 : SUCCESS in 1h 21m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61d23ca70c2745bc9674b8ba93db82e4 : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efee9d15583245d883f114111c135488 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d0ad6e8cfe14f208b40ddfaed8d27d3 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcc88348a0aa4041a4e735859f9e1eeb : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/141f540f2c6f4a9e8dc8b17fec3a9aa7 : SUCCESS in 4m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/55e003f1bf8649e6907a1a3e33ebd320 : SUCCESS in 17m 10s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f2eef2b467354615a1313dae4adbb614 : SUCCESS in 26m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9098cdb9fb044ccb942aa04f5ea31a48 : SUCCESS in 1h 10m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6857445a2cd54a3b991baf90dd89676f : SUCCESS in 1h 42m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6566253ad8be41fb843e8e51f6c5660d : SUCCESS in 19m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a389858953aa47bbb0d42a5161af5a6a : SUCCESS in 25m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/edb9bf822aae47819df6da1baf6e156e : SUCCESS in 53m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9337dddaaf0a40a5a3e274860c2966a8 : SUCCESS in 1h 25m 20s","accounts_in_message":[],"_revision_number":2},{"id":"17cfff4cab1226b099639e3af4ab41a4fb923b0f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-28 13:58:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"06b63b6a4431be2f74ed90fe3918917290831282","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-11-28 16:11:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4aab2d2d5059d38e231165f975b88b86a4ee1173","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-11-29 06:08:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3d1c474534615763e783d2f5e7a8f4e15700cb63","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-29 08:08:34.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"28c48b767f7c30201d32cc790b07575a092a4855","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-29 08:43:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"98ce3e868480a0e3a4c00ade2701501943fda923","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 08:44:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"75a7465306f94a5c5022912a177cbfb87b417cc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 10:19:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0c74ab0820b4ece8499402c5a4c598b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/44645456951444ac9a43ef15d8ffe1bb : SUCCESS in 18m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4a671f786ee24f3e8c8c2427bd024f2e : SUCCESS in 26m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/23e9fdcea9eb431080624520b340c257 : SUCCESS in 1h 02m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b41680b49b4d4428b8691ae9aebe1a94 : SUCCESS in 1h 31m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5149bd511142a0aa7375488505d25d : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a58c1bf1c19e4338b5757b1964362c4e : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65e182a9a768427fb1a014f8fd9b8a9a : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c623669615e7445abe2cb583cfe8e8a7 : SUCCESS in 4m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2d417a0a5ba448c6a79d43a4f1db17fc : SUCCESS in 19m 10s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e0fac1a58f834f39b268072046c36eeb : SUCCESS in 28m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8cd5d9410aa547e9a9ba1ffa3e79df36 : SUCCESS in 1h 03m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cb27919b752847aeae6164ae1b8d61b0 : SUCCESS in 1h 29m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2facfd9a79454edab0b22cb34aab8ff0 : SUCCESS in 27m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f831f8c203ce499296b5442a20a60322 : SUCCESS in 39m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/42fb6d18d03d44679d68d5dea13fd2d6 : SUCCESS in 54m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/957b287b5ebd4157a5f0dce432777be0 : SUCCESS in 1h 27m 07s","accounts_in_message":[],"_revision_number":2},{"id":"1f2878918e713ac83240cca303628b2e116f759c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 10:19:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a15ff146783fefa871e4ecca2a030f79998f3115","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 10:20:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dadd7cc13a04a1ca02344bb06fb10dd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ebcdf687044ec0beb533dae33893a2 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ef4bd9bbcd2b8b1df067004b806ee6cbcf2f6467","revisions":{"48f433b2768611092df13e7aeb0429db7893b2f2":{"kind":"REWORK","_number":1,"created":"2023-11-28 11:03:15.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/53/902053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/902053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/902053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/902053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/902053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/902053/1"}}},"commit":{"parents":[{"commit":"a9b4c5361f09b4ebac96c4c8eb5afb3abe26d344","subject":"Merge \"Swift: Remove unused rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9b4c5361f09b4ebac96c4c8eb5afb3abe26d344"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-28 11:02:41.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-28 11:02:41.000000000","tz":60},"subject":"Handle the case where docker is not available","message":"Handle the case where docker is not available\n\nChange-Id: I473ea1c68d2fd7f15fa0545d77892276b1accfeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48f433b2768611092df13e7aeb0429db7893b2f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48f433b2768611092df13e7aeb0429db7893b2f2"}]},"branch":"refs/heads/master"},"ef4bd9bbcd2b8b1df067004b806ee6cbcf2f6467":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-28 11:07:42.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/53/902053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/902053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/902053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/902053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/902053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/902053/2"}}},"commit":{"parents":[{"commit":"81dc7a6b74a47e8d026d094024fccda1adbb2e3c","subject":"import packaging only for docker container engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81dc7a6b74a47e8d026d094024fccda1adbb2e3c"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-28 11:02:41.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-28 11:07:42.000000000","tz":0},"subject":"Handle the case where docker is not available","message":"Handle the case where docker is not available\n\nChange-Id: I473ea1c68d2fd7f15fa0545d77892276b1accfeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef4bd9bbcd2b8b1df067004b806ee6cbcf2f6467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef4bd9bbcd2b8b1df067004b806ee6cbcf2f6467"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
