)]}'
{"id":"openstack%2Fkolla~902057","triplet_id":"openstack%2Fkolla~master~Ic64eb51325b3503a14ebab9b9ff2f4d9caec734a","project":"openstack/kolla","branch":"master","topic":"bug/2045558","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2023-12-21 17:43:20.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-12-21 15:53:42.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-12-08 08:45:07.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-12-21 17:43:20.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-12-21 14:41:28.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"Ic64eb51325b3503a14ebab9b9ff2f4d9caec734a","subject":"Sync only local cell in nova bootstrap \u0026 upgrade","status":"MERGED","created":"2023-11-28 13:22:58.000000000","updated":"2023-12-21 21:41:05.000000000","submitted":"2023-12-21 17:43:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902057-bug/2045558","meta_rev_id":"7a07f85828c265b9c071f3613985227aaadbedbc","_number":902057,"virtual_id_number":902057,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-21 17:43:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-12-21 15:53:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-12-21 21:41:05.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-12-21 14:41:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-12-21 15:53:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-12-21 10:47:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-28 13:25:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-11-28 14:31:51.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2023-11-28 15:22:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-02 12:29:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-12-08 08:48:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-12-20 15:55:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-12-21 15:53:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"d1688b647d9e935d837bf2799b9a736c77b2059e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-11-28 13:22:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c4443ab7100045a5b2001fd4c4c8dc34ee6e994","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-28 13:25:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ffad77b3d78180ad50ebbe3e20fd6d9d721d9ea","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-28 13:25:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0fcbf458d38a47532b6ab34f8a4baf3fa019f7f5","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-11-28 13:28:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07be9ddf598b31d51fa89a5853f0c7b36ca0f091","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-28 14:31:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53402ee3b8ade7915d14ee80f462e68fa589b4f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 15:22:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f154b236d364828a273a618d032d59e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/040a338296544ef8a1f85114294ef144 : SUCCESS in 23m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/aa1b8422d84a4d85b510cb3a6d7b0dd1 : SUCCESS in 1h 08m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e8b26ac1c7014f17a4dc079c058f3c1b : SUCCESS in 23m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a8e17586ac7d4c46a350f5a8996b0d32 : SUCCESS in 35m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/90b186fa01824cceb6d6de7b80b7f214 : SUCCESS in 1h 03m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/894858a7eb0e4f0ebb14fd21e19dff8b : SUCCESS in 1h 34m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/393481ffa81348d0b5e33730bd181e95 : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59bf1f2f60774545a2d7acaa57674ef7 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c0c1a6b6708947aa9f6ff2c6e146106b : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cf184848fb44ef3bba5f101ae6b3520 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d8eb0567b5d4d3995807d2bf14f538e : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d4390152b96349f29b466aa9968bd99e : SUCCESS in 14m 34s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4eea08e812ae4a9ab7b1c64b0043b12d : SUCCESS in 28m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c503956c3ac843519cad2b434a403a0f : SUCCESS in 1h 04m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3a941ab61b784d819aedbdb1ce62992d : SUCCESS in 1h 35m 04s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/7cad058be36d40e1a18358f130ac13f2 : SUCCESS in 1h 01m 01s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a71cdbe270bc4740838303ce91841fc8 : SUCCESS in 25m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ef8b7ec1cc6f47cb8c082b1177d7642d : SUCCESS in 23m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/20f1d9601d164ae7885b6728a0c15d7e : SUCCESS in 50m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c1da9a68cf7f49df9d5fd2efd96b6465 : SUCCESS in 1h 30m 19s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/7ec157955d5440398d2a883a8830d33c : SUCCESS in 46m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a47c1445838d410d681b10698c106746a816587","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 16:32:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bed0529125d49ef929c52536c495b9e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/eb1796606e674d7aa5a941cfa566e1a7 : SUCCESS in 35m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/96da8dd9f82f415d9adf9ed23fbb073c : SUCCESS in 2h 19m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eed1f65858754fe49cf000d594e8e689 : TIMED_OUT in 3h 03m 21s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0155b184d169455396cd41811f02673f : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c319455979fc4626be67bbcf56f3ac7b : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ae1493c40ec944d38c7b99d21f6d3529 : SUCCESS in 7m 16s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c3f05778a0f543caade17098a434f164 : SUCCESS in 30m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5efe2f07cbe8497cb34ec949595b0058 : SUCCESS in 2h 17m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b65ce6ef8fcb66e62ad1f5b078feb836ede9de2a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-29 09:19:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b07f93648f1b781b43a92532666e1b5aacced0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-12-04 13:35:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ae55984de5fac7f2f5ba9db86d85363a2f4961da","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-12-04 13:37:20.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f09704a64a2daf995cb239d7565eecc292a01d63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 15:52:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/724d1959e40845b692b99a20d4596b7c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/73d367e5a9e949c384cd057851b39113 : SUCCESS in 18m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/664111951b714ed8ad4deec839c9c0b1 : SUCCESS in 17m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4575515753d04496b0901f9c09c3c7db : SUCCESS in 2h 14m 50s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f475674da6d94579bd7ad137e3becf55 : SUCCESS in 2m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/056cbfc549784ceeb9cf89e8aa461271 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fc69c1a75bc2489eb5bf1f8a54ea6d78 : SUCCESS in 3m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/346537d5d208447a9249fa27def23170 : SUCCESS in 1h 27m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/eed9c562652a4e519629a92e91d3f1ac : SUCCESS in 14m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"45017829fa9836cf78c819cdfa24563e834c10e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 16:39:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/92e1c0031fb74276945852d2ed0934a3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/027f52e1d1454e28871386d02e2579dc : SUCCESS in 19m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b448729bb4f04672889689f21ea09053 : SUCCESS in 1h 04m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/50356499005c478eb44465542096814c : SUCCESS in 20m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bada6488b5154d35805e600d6d7e0e91 : SUCCESS in 32m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/42d618f9474f42399cf3245d85c15a07 : SUCCESS in 1h 12m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/03c722cb2dc840678fa0b2dfb8cca070 : SUCCESS in 1h 41m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f57ee7f868b64f84926d8583829e41d0 : SUCCESS in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfafb161f26b427eabaaf28dfe539ec1 : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3819f15a9eec43c58f2109110718683f : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f753065c7ec0406e8fc6d14697c9079f : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfbd089de86a483e9a662aa70eeeaa3a : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/459b4409453c4575950f7fddc6021e44 : SUCCESS in 3m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0ea02b1d05234761b437493d91ec33e0 : SUCCESS in 18m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/324b4ce6c62147989a917a91d1f9f34c : SUCCESS in 28m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0f5db9f4c59d4d308d61567e53c2d701 : SUCCESS in 58m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5f64fdcc2b514b62a79ec08a9fa0767f : SUCCESS in 1h 34m 00s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/20bae37c4add4b7f8617ca805c8976ad : SUCCESS in 53m 11s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/42a721d8f9f549b9b66f3bd52c9a6cdf : SUCCESS in 18m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c190bf00e0174105b14a540eabd4f568 : TIMED_OUT in 3h 01m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/905207c40c3c47d9a97a41dc533d50bb : SUCCESS in 49m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c442395739de4f4182643234622812ab : SUCCESS in 1h 35m 33s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/50e1ee4a47fa40be8c99318dd725d5c1 : SUCCESS in 49m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"63bab006213b643ab42e33605302837a8fdba9aa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-08 08:45:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7295b56ce8fdf94c3ed7250cf535b117914ab611","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-08 08:47:54.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"5c5b77d6fdb06d2eb3e81b5ab4c5aa44764f9b65","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-12-08 08:48:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7827fec4f7b56ebd3835c0ddd1322320d81e52d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 10:13:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c184dcfc4c484230b9996d9a8369d5d4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/367911622a2f4d11bf1e979c390bab6d : SUCCESS in 1h 19m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/118fb7e93f844b18ad4edc4fe21da810 : SUCCESS in 15m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5bddc8fe62994797b003411a63c78851 : SUCCESS in 1h 01m 29s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63ca0739b4c0429f87cf4bf8e0458e31 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e29962c802854228845812cc0db756f8 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/781e4f4565dc4fcab312986667b483e1 : SUCCESS in 9m 16s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b2e037685ae0452295ad417d5355cea2 : SUCCESS in 20m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e70fd79e94a94e3d921267ee3af809ac : SUCCESS in 1h 24m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c9cab64373009dee7898d27bd5ea14460fdbd206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 10:28:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bad123f4547441c9d08aa3bb5cdc849\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/577f1fbb61134727a7eb03a8a2872116 : SUCCESS in 21m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4283a80c990e4ff4bdd06e642c4fd099 : SUCCESS in 1h 03m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5d55febb0b5d4645beef81e4a1fb70dc : SUCCESS in 20m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6f88e324fd494ddca36c042a6529099f : SUCCESS in 33m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4764f761cb8b4cceb557eb8bee9aa4f4 : SUCCESS in 1h 08m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/aff19c48bd0943df8cf2a9eaa37d0a93 : SUCCESS in 1h 36m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0068d25dd97e41a487b48005c62847fe : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3656c3b078014e5eac375055e134f140 : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/14d339e4bab648e58fecb220131ea33e : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3afbe8e7da5b4e8d9cdfd7f16d3c156d : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcc7c5798c14453a92202c4f70d0fee0 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5edcf7e7a724305adf82814506fbb3f : SUCCESS in 3m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0c255bf6191641deadc53b0c3af78bf3 : SUCCESS in 23m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a9dc74bf5e7348d186a4f612f97e0082 : SUCCESS in 29m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4a4e8ffe9b3d4d4093bb0cb0bd8666ad : SUCCESS in 1h 11m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f172594503744cb58cc4e19157365036 : SUCCESS in 1h 35m 38s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/f5c86263469a4025a811243e99d12385 : SUCCESS in 56m 41s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/06b9c48108bf4ad38c12091d4748ad99 : SUCCESS in 18m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0c42573647354044aaae26af20afdc32 : SUCCESS in 22m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/534cee630c4b4926b604cad22ff1d70e : SUCCESS in 57m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4490d9db9d854bb8a027bb952f7ae151 : SUCCESS in 1h 35m 46s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/ed15e80e09264cbe9214ad2255cc6be5 : SUCCESS in 49m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"63111cf5fd3adfa3d2ae0d2482e125f449b2910a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-12-12 09:43:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"21ef1c8a5ad938e110e167ae879fb5b8ca433914","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-12 10:27:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e2e40d12938ace7f994bf22c9b6d4e9076ca4b9f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-20 15:55:30.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"277b884ca13094593ff98b862da8d64f5f60e2dc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-12-21 09:04:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2daabe70ac893f6b136f254a906ab7c664f68151","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-12-21 10:47:26.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"07e869ccf24e3260d9dd1fb0aaa4708811a0b377","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-12-21 10:50:08.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0223b27f0d00e5cf6628570b39dbd880ae245769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 12:36:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b94450bc761c4e7f88e086641f467568\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/087fe5d174164ad4b6fced4dbe7add33 : SUCCESS in 20m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/559d7bba59a3452fa9620f87b6783511 : SUCCESS in 1h 12m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7011a063752e453290e448fa21dfdf64 : SUCCESS in 17m 54s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e7c47ff8c7df486f9fdfdb50a8825ccb : SUCCESS in 29m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/68793ed7188640f080e0cb7a356445c1 : SUCCESS in 1h 02m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8bcd397957db498b88acf503e5c3ae58 : SUCCESS in 1h 31m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c998dc464d24ff1bda7e5215f3ec0b4 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b022b3678a9b498ab9a2f0a85933266f : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0e23ce54caf4b7390a02bd0319d4b15 : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3563fb652683493d976b5bc8d6669bd6 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9944916b70a64566bf982a9ef4962619 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be09629644634adabb8086eadf623d82 : SUCCESS in 4m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6a02d45de7b4440f9672469ba43c7d04 : SUCCESS in 20m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/219824eb90334a0bafa9946bf7b1b5bf : SUCCESS in 29m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/af2f8272bd7247a8ac111b8da556ad83 : SUCCESS in 1h 03m 37s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/61e0665395f94162b5186719fc31988c : SUCCESS in 1h 40m 58s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/34948f37d0ca4511a3500fc8fe3024ea : SUCCESS in 1h 00m 27s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ea92fd8c95bf4fe88aa10ea616a7469a : SUCCESS in 19m 27s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bf35800e47cb4edba40087e5694f2d9a : SUCCESS in 1h 01m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5c814a7a774d4dd4ae2154081981fc21 : SUCCESS in 1h 02m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/021039059b264cf88eaf083149194ab8 : SUCCESS in 1h 11m 01s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/46e3ed83f29549bdb649f6d86626e7f7 : SUCCESS in 46m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0e9f22ccff0558b68afaa1c9b1f6dcd8e8f8941a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 12:57:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58b47801090147ad8632f6353f9295b6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/09a997cf187144a586fd590ff6dde9b5 : SUCCESS in 1h 40m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fec47a007de1449783a7614f71770e12 : SUCCESS in 1h 30m 32s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ac7ea54fc7fb4798b4c3d05c243fba10 : SUCCESS in 2h 07m 32s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/34c2ac7412c64987b465cc4c6564e572 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f0cd930b00848969bc2bb119898e48b : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/92786df5975146f5937c46cde56e5d9c : SUCCESS in 5m 25s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f615fdd8efad4ba3811ac195a7afb51e : SUCCESS in 1h 25m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1adff316409a41a3ba16d3d94c4f91f1 : SUCCESS in 1h 32m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ff765f952ce06047753ea7a8e9319af4d46272b3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-12-21 14:41:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"460297ce4fef66ff7bf35fcaf1a53d40eeac9520","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-21 15:53:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82dcb39cf073bfc3e7acf438ed2c15b1edd820e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 15:54:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e76bf462e2d102edb662a6e881b812d0677ca5b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 17:43:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ac9f1f77111439d973a9b9425525907\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3d8f1aed5d904992be7b5f59de998a6f : SUCCESS in 15m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4cabc7fca3fa431898c4df75dc8157fe : SUCCESS in 23m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0b9ee269d9d147bfabdeae599f9ed072 : SUCCESS in 1h 07m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c37ad7304bf8466c93efde64b7287976 : SUCCESS in 1h 43m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5570ee7eb11d4cedb6ee511fca4a9a44 : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53361a54eb4a4a3080ed0924bec4648a : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6bd26e8a13b245dba8a68bfe8cbcc881 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22a573597ab749c29f53d1ac8be96077 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64870b59f4bd4772b669a0ea24edd0f4 : SUCCESS in 4m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5e142a69fda54fd2932163e934ee3482 : SUCCESS in 21m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/eb6a67fee52a493480969a5eb46328e9 : SUCCESS in 25m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/23d9278676f449089983c04bd490c2d8 : SUCCESS in 1h 11m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/221902cbf60c48ea895be73dc2889c77 : SUCCESS in 1h 43m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/25c4e932785147acb72373fc143c36e5 : SUCCESS in 22m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b769d5a2c96541f88fe650fd989e47a9 : SUCCESS in 1h 16m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/da845ffae14348f7b8787d25db04b86f : SUCCESS in 1h 00m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a47dd64c44f8435e82482f017d6d2d5b : SUCCESS in 1h 38m 09s","accounts_in_message":[],"_revision_number":3},{"id":"3252eebb30a40b558b45592c912d186c7639cfe2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 17:43:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0466c86f4afa9d13e2bb077acc90430766a08269","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 17:45:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7171bf64ad74a979dcce427d6b0c1c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2d6b3076d5449a8baff985907a8e5bd : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6fe09e1b590411eb10a2842ca894d99 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3},{"id":"7a07f85828c265b9c071f3613985227aaadbedbc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-21 21:41:05.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"de1487f05187387aea3f113dcbb6926734b1ce6b","revisions":{"50b75f1f48b6a012436b8c130e79ee8268622c6c":{"kind":"REWORK","_number":1,"created":"2023-11-28 13:22:58.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/57/902057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/902057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/902057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/902057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/902057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/902057/1"}}},"commit":{"parents":[{"commit":"a9b4c5361f09b4ebac96c4c8eb5afb3abe26d344","subject":"Merge \"Swift: Remove unused rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9b4c5361f09b4ebac96c4c8eb5afb3abe26d344"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-11-28 13:08:58.000000000","tz":0},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-11-28 13:22:57.000000000","tz":0},"subject":"Sync only local cell in nova bootstrap \u0026 upgrade","message":"Sync only local cell in nova bootstrap \u0026 upgrade\n\nAdded the --local_cell argument to nova db sync commands during\nbootstrap and upgrade.\n\nThis was previously thought to have no effect [1], but has since been\ndiscovered to fail when rotating the nova database password in\nkolla-ansible.\n\n[1] https://opendev.org/openstack/kolla-ansible/src/branch/master/ansible/roles/nova/tasks/bootstrap_service.yml#L2-L3\n\nChange-Id: Ic64eb51325b3503a14ebab9b9ff2f4d9caec734a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50b75f1f48b6a012436b8c130e79ee8268622c6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50b75f1f48b6a012436b8c130e79ee8268622c6c"}]},"branch":"refs/heads/master"},"055059e13c4bfebed9f488023e7538244fa0a105":{"kind":"REWORK","_number":2,"created":"2023-12-04 13:35:56.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/57/902057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/902057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/902057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/902057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/902057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/902057/2"}}},"commit":{"parents":[{"commit":"a9b4c5361f09b4ebac96c4c8eb5afb3abe26d344","subject":"Merge \"Swift: Remove unused rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9b4c5361f09b4ebac96c4c8eb5afb3abe26d344"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-11-28 13:08:58.000000000","tz":0},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-12-04 13:32:55.000000000","tz":0},"subject":"Sync only local cell in nova bootstrap \u0026 upgrade","message":"Sync only local cell in nova bootstrap \u0026 upgrade\n\nAdded the --local_cell argument to nova db sync commands during\nbootstrap and upgrade.\n\nThis was previously thought to have no effect [1], but has since been\ndiscovered to fail when rotating the nova database password.\n\n[1] https://opendev.org/openstack/kolla-ansible/src/branch/master/ansible/roles/nova/tasks/bootstrap_service.yml#L2-L3\n\nCloses-Bug: #2045558\nChange-Id: Ic64eb51325b3503a14ebab9b9ff2f4d9caec734a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/055059e13c4bfebed9f488023e7538244fa0a105"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/055059e13c4bfebed9f488023e7538244fa0a105"}]},"branch":"refs/heads/master"},"de1487f05187387aea3f113dcbb6926734b1ce6b":{"kind":"REWORK","_number":3,"created":"2023-12-21 10:47:26.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/57/902057/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/902057/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/902057/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/902057/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/902057/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/902057/3"}}},"commit":{"parents":[{"commit":"a9b4c5361f09b4ebac96c4c8eb5afb3abe26d344","subject":"Merge \"Swift: Remove unused rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9b4c5361f09b4ebac96c4c8eb5afb3abe26d344"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-11-28 13:08:58.000000000","tz":0},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-12-21 09:46:59.000000000","tz":0},"subject":"Sync only local cell in nova bootstrap \u0026 upgrade","message":"Sync only local cell in nova bootstrap \u0026 upgrade\n\nAdded the --local_cell argument to nova db sync commands during\nbootstrap and upgrade.\n\nThis was previously thought to have no effect [1], but has since been\ndiscovered to fail when rotating the nova database password.\n\n[1] https://opendev.org/openstack/kolla-ansible/src/branch/master/ansible/roles/nova/tasks/bootstrap_service.yml#L2-L3\n\nCloses-Bug: #2045558\nChange-Id: Ic64eb51325b3503a14ebab9b9ff2f4d9caec734a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de1487f05187387aea3f113dcbb6926734b1ce6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de1487f05187387aea3f113dcbb6926734b1ce6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
