)]}'
{"id":"openstack%2Fkolla~902179","triplet_id":"openstack%2Fkolla~master~Ic0d9b2c26fe5474f579b72d6779243b7ad39b420","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-11-30 05:41:28.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Ic0d9b2c26fe5474f579b72d6779243b7ad39b420","subject":"podman: install \"rich\" dependency","status":"MERGED","created":"2023-11-29 18:07:34.000000000","updated":"2023-11-30 08:59:16.000000000","submitted":"2023-11-30 08:58:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902179","meta_rev_id":"ea59f331e38322a17328528d0fefee654b1021c0","_number":902179,"virtual_id_number":902179,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-30 08:58:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-11-29 20:04:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-11-29 20:04:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-29 19:46:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5197fb99442544e997f37dd3b54dd8c30da98131","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-29 18:07:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fccabc2abf09be9ec84777f2817c8dd160a39fd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 19:46:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b181754f0194034927bd3ec6208011d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b69f8f73b4f24262be236d14b0a48c99 : SUCCESS in 19m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f869282f6a4141c4b502e27b2d9b5cdc : SUCCESS in 1h 05m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7ddf9152647845dcbaa9df1b0408f669 : SUCCESS in 21m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/039f2858390d419d878e7daf639711e4 : SUCCESS in 28m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4a63410932b4482db1e0f8baba7b7407 : SUCCESS in 49m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2962a07002244807b49036820560f0cd : SUCCESS in 1h 32m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9441dca746f0494aa25a1518362a6862 : SUCCESS in 2m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae7b919e71b44049b2c1249e426885fa : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5192c2ef40bb4824bf649ecf09b7237d : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c4c571e770d42d9a010e638fcd9875c : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e9c8d3cfacb48b1a33298d05022900a : SUCCESS in 5m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/52f411f19b46457c9fa89b5ff5fa54f9 : SUCCESS in 19m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9565e6d9badd47d2ab50450f17def686 : SUCCESS in 31m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4a62321ba0ec4b37b49d8f0a2701ebb7 : SUCCESS in 1h 05m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c35929cc3ce146faa3889fff3a741d1e : SUCCESS in 1h 15m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/aa8cc39095e94c1eb083d8f14cc54751 : SUCCESS in 23m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/16d2a8740a0c4879862eb42550b547e5 : SUCCESS in 34m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c07a6241d5cd4b44aeb0582bd988d9b2 : SUCCESS in 55m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a2c343c1282241f5b52027bdcbdd7554 : SUCCESS in 1h 25m 51s","accounts_in_message":[],"_revision_number":1},{"id":"65d0c69d51751b6afda92f9fb72f5aa82915b1bd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-29 20:04:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"afb19f9747963eb77ddd3ec8213783b34b6b0fb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 20:04:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09158cee9a4507d7d742c4b2dc80c4647bc9d7be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 20:43:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0fdd7039a4f4204b01e7b337f0df8ba\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c82646581dff467b92c2eb9cb810ff0a : SUCCESS in 1h 56m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0423a0460eb5414a87cf5face8ed6655 : SUCCESS in 2h 13m 26s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ef41c499a68f4784b950e38be43f4ec5 : SUCCESS in 2h 19m 10s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f1d4a3523c04b71b864241a8fcc20dd : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b8324f5a248a43ad8a9c09c345fe18e7 : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8b80184b4989425388d6bb81c2135cbc : SUCCESS in 5m 11s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5c11675015914adc91c9f4c80e13171b : SUCCESS in 2h 06m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/15ba0836df184a468373e1183bc0b94f : SUCCESS in 1h 55m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3fb117b593b9f2120dcb6eb43e549d29b5ebd8fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 23:43:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0cc3d3cfa034e7795928ffa910e7e2e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c0a78bed19754675a4269569646cbb73 : SUCCESS in 22m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f20bd03c87a645068eaacb20abc00651 : SUCCESS in 31m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e37d39d9db83409c87c9ff2c55e708c7 : SUCCESS in 54m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5c9ca1a9289040dfa21ea0353da3a033 : SUCCESS in 1h 30m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8cf7e9864bb4f22a1bee1f80e4bf0ab : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9072f7c8db04e34baa3467e4c403dcd : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3d3b06e926c4bae8fe1d70ca21c4b47 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d65e9ea816c04d169feddad01c74dccf : SUCCESS in 4m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4b14bf60f5f940beb3a360d5f64eb761 : SUCCESS in 18m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/62dd3450995b4a7698b478f0c7099aab : SUCCESS in 25m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5c7ef0baf7f6400e90c76ae31fb29a85 : SUCCESS in 1h 05m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9899938d205a4fa9b07412f802dfe77d : SUCCESS in 1h 27m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a25711604e204c92ba15c2e2a4965aa8 : TIMED_OUT in 3h 01m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/48cd30e5f9fa49a9b392b254a6b9c010 : SUCCESS in 26m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/adcfa0658e8142528607e75dac548226 : SUCCESS in 52m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1f6be8a41ca04b95a4fc159de02f50e6 : SUCCESS in 1h 23m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a31bb40ec377550211b22900bd0bcab55109c293","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-30 05:41:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1cf17f9285f6aa42a4ce95529034e7519c5d0a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 07:12:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/339e2bd473ae42a8ae6120f8c58756f2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e243eae95d1a4d26844d46f0fbe7f110 : SUCCESS in 20m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/358df46315754cf998aaa94c879e4a13 : SUCCESS in 1h 13m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/36d0f80570b145a48e7556a4585a664d : SUCCESS in 20m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/de56ac1b279842228d18c199415402e4 : SUCCESS in 27m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d439df88dd7845489ed6442f8d94a3e4 : SUCCESS in 51m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8f3408dead18411d90a9f1d3dde4b90b : SUCCESS in 1h 16m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50e6358826d34d82a97f8b35782b76c2 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dac282259ac04a599f394a1514bff4ad : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0cacc03ec59148a9a0fc479dc7515805 : SUCCESS in 3m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d61a4f9aee864c19a95905b831cfd3f8 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e55acda6460245dbb5f7c11354c11a80 : SUCCESS in 5m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e6f54902e3bb4604ac145679aa95affb : SUCCESS in 17m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/919a913d27394fdc81c90beea8445626 : SUCCESS in 21m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/317973bdcb5c402baae523569ec95de5 : SUCCESS in 59m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dca2160370454262a258990ead7620e5 : SUCCESS in 1h 25m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/df6e99c3e5584bb6b534681a6c6c2b28 : SUCCESS in 19m 01s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/97cddc0650a6412da362aa98365363d6 : SUCCESS in 24m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/036938ee750549e2866374beb1f44da9 : SUCCESS in 44m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ee629aa16ad44e5caf745a1e8abe7a6c : SUCCESS in 1h 19m 54s","accounts_in_message":[],"_revision_number":1},{"id":"7f06977f5a60073010571dee23dd9394e07496dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 07:12:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f01e04bc2984be0fd25c601812bc1312c0b1475","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 08:34:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6602a0c681da4d27bb12b5941330c363\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/361eb50018094da1acd231bb223ef013 : SUCCESS in 19m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3ed1f6a594de41d3b6f1039624695ac9 : SUCCESS in 15m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/db8eaa1d09b44d1586f2dd3a78bfb930 : SUCCESS in 2h 49m 10s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70d95466d3ef499990387a02b964c5ea : SUCCESS in 2m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7658206685a44a40b7215ba084853895 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/650de712b3ad49eab0ca211c370ed46e : SUCCESS in 2m 20s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a5feeeba609545e2b8e0429d7038e05b : SUCCESS in 2h 08m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fc3ecc24a4974fbaa04e530802d83a92 : SUCCESS in 14m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94bb47718f4c065f50d465d1db2c441a16903394","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 08:58:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1da517b50bb345dd807fc273e5ecf1cf\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2fc71f5efbcb4fd480b2bcc0856f58b1 : SUCCESS in 23m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c1d3e02206934e79974bab5d77045be3 : SUCCESS in 34m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1412954b6c104a0a9c63f1f0297f4236 : SUCCESS in 1h 04m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/249efdc5bc544800a8a32df41dcfe8cd : SUCCESS in 1h 36m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa2ab3bbbfba420c8c8f63f6185774f2 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/656251bddd4e41da8402019a33ea29d4 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d44bd736db014ab482f6022ede375f83 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/780a9e4367584e6e993f6b5a317d8bf6 : SUCCESS in 4m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8a635b5c2bf34fb592474de771a24d86 : SUCCESS in 14m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/edab42d20bb84647994a6e9f2bdde942 : SUCCESS in 30m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f5b2205069674ba68e16c789c64ebc40 : SUCCESS in 1h 11m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3923c6a68d804868b26420fcb6f4f064 : SUCCESS in 1h 39m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7867d71d300742fcb74450f3b3078289 : SUCCESS in 32m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/97951833c2a74b94969c790252b4ae40 : SUCCESS in 35m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e17ca04e7fb1408e9c285531ffb27bb8 : SUCCESS in 52m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e5196c4529a9454da0c6b0b593ea7452 : SUCCESS in 1h 28m 24s","accounts_in_message":[],"_revision_number":1},{"id":"e651e2662822e277b4d57c1faffc507f47a66228","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 08:58:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ea59f331e38322a17328528d0fefee654b1021c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 08:59:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fb69c3d33d944308201c52389f6e240\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0593b48bf0e44f284fef21fe43137ef : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0cfc5c7c0992a0187f6b92d64a2a831e0394eeeb","revisions":{"0cfc5c7c0992a0187f6b92d64a2a831e0394eeeb":{"kind":"REWORK","_number":1,"created":"2023-11-29 18:07:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/902179/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/902179/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/902179/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/902179/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/902179/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/902179/1"}}},"commit":{"parents":[{"commit":"54e05af7e57c381aef244d8d39395e058e951394","subject":"Merge \"Drop docker-py from requirements.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54e05af7e57c381aef244d8d39395e058e951394"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-29 18:07:10.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-29 18:07:31.000000000","tz":60},"subject":"podman: install \"rich\" dependency","message":"podman: install \"rich\" dependency\n\nThis dependency was added to podman-py in version 4.8.0, but not added\nproperly to their requirements. Install it explicitly for our tox\nand integration testing as a workaround.\n\n[0] https://github.com/containers/podman-py/issues/350\n\nChange-Id: Ic0d9b2c26fe5474f579b72d6779243b7ad39b420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cfc5c7c0992a0187f6b92d64a2a831e0394eeeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cfc5c7c0992a0187f6b92d64a2a831e0394eeeb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
