)]}'
{"id":"openstack%2Fkolla~903161","triplet_id":"openstack%2Fkolla~master~Ifce506b49571950ee16e26aae06f20aa94567722","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-12-20 17:09:51.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-12-13 13:00:01.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-12-08 14:34:14.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}},"hashtags":[],"change_id":"Ifce506b49571950ee16e26aae06f20aa94567722","subject":"CI: Use newer podman/buildah on Ubuntu Jammy","status":"MERGED","created":"2023-12-08 09:55:48.000000000","updated":"2023-12-29 03:52:28.000000000","submitted":"2023-12-20 17:09:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903161","meta_rev_id":"09f8689d5e2e94c569dcafbf2245122416296372","_number":903161,"virtual_id_number":903161,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-20 17:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-12-20 13:38:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-12-14 08:08:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-12-29 03:52:28.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2023-12-14 08:08:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2023-12-14 08:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-12-14 08:56:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-12-20 13:38:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-08 10:25:59.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2023-12-08 10:26:11.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-12-08 10:31:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-12-08 10:34:55.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-12-08 12:47:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-08 12:53:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-11 08:43:27.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-12-14 08:56:09.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-12-29 03:52:28.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"46d389c330f9ecb26ffce496f258d30d065f7e8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-08 09:55:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39e9e5372b674dc416c0bba630603e2d52e14c52","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-12-08 10:25:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc69f975ad950711ac2912da05676495b65ff19a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-08 10:34:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4fca378da7f51fe2bde7ea0c182e113ce8e30b9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-08 11:13:23.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d2e7928a822fe45353d77d56441f4669a20e3759","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-08 11:13:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e64dcad2484113ee94cad39e48f71700437ed397","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 12:47:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0200565c09f54aafbf014c7201eadb57\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/bc1d28d1b4d040f6bcb257fae30a016d : SUCCESS in 25m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ebda67174eb8444099b35a3c9f2843eb : SUCCESS in 1h 30m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d6d031f0efc240ccbea17c0d0a4dacd7 : SUCCESS in 1h 18m 25s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3baed78388314503897fe7e9a14d5b40 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/adc9b4c3a87a4c49b4447ca99d314535 : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/59140bb296884acda6c06fce9ec9ede9 : SUCCESS in 4m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/296e0c5415d643d78d2ba8bff1dc491f : SUCCESS in 22m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/cb6ea541b5984bd7b278461e6fe38fb6 : SUCCESS in 1h 19m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c0942b3b9a5838c74d41878b49e2a3bb455ec2dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 12:53:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38d0ed7dded44b70ace7b037c752986d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0bc13a1193884e4abc2dce6db21abb57 : SUCCESS in 21m 48s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e5b4996a3c5c47889cf30101ed30f94a : SUCCESS in 1h 09m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a57c32769b6a4bf785e7c97643fe1284 : SUCCESS in 21m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/69e519f0e8db4c46aea2af910e11ba45 : SUCCESS in 28m 21s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/86e3200cbb64471e8bc4e9f51079953d : SUCCESS in 1h 10m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/886c5126471c4f18b10813a57134bd2e : SUCCESS in 1h 28m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d8a88e567de4d05b410995508d575ac : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1fa117197ea46249ef02ba6f6571881 : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/57f84cacbe9742ada89048c75726ad28 : SUCCESS in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a873ddfa623643a1bd157ee46401df28 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adedca6523b4448dab7cb0cab5bec8ea : SUCCESS in 5m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3dd23ab5968d4ada80303f2c4173b134 : SUCCESS in 24m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b0232f3957524e8983dbbfad523e80f2 : SUCCESS in 26m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ce65643c319e483292dd03dc78c2c800 : SUCCESS in 1h 08m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b909c0fc822e4a20b8160d8ab8409801 : SUCCESS in 1h 37m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/892d4d6b644d427080704a0db972e751 : SUCCESS in 17m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7bd8adef06ee4826a1cdbb8672d36aa7 : SUCCESS in 47m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d16c24384e7e404a8dcb21708048dcae : SUCCESS in 47m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cd7f8ae601af40248fd8f19fa9df9946 : SUCCESS in 1h 30m 44s","accounts_in_message":[],"_revision_number":2},{"id":"fa07cb9866eef0fa8203699a7020b3a7e4330ea8","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-12-08 14:34:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9b0f569802126628076b91295c5156c60d44e702","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-11 08:43:27.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e4a99e022e8fca8f1bde5ff8291bdf2442785c43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 09:38:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63fb97d7307d4258b395d14e37082775\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ddf632ddcdf841d7abd330acc119ae65 : SUCCESS in 22m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6633281db21b4364b3bc808ff3f8ddeb : FAILURE in 11m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4b6b1f4692bb48a29a8fdbf9d85a7291 : SUCCESS in 19m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/651684b31cf4466f8db5fa3871fe3beb : SUCCESS in 29m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6eaf1fca2681428fb525f5df1cebad80 : FAILURE in 7m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/90dc02a42e2b488eb79ae6b2a47da765 : FAILURE in 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28de5f8d8ec5442f86483bd8a36bdb2e : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2def66a3f1740fa95ae9f8caeca6746 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eee6515c7bba43598eabba4d261c969d : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20a9267ff0374f7c96dced7448a81c7c : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d31d5bd4788a428a90e1b035874853a7 : SUCCESS in 4m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/78e2b8768aa04a4594d3036bf2fc9bc7 : SUCCESS in 19m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a4a8666f8c984d618f8b593847e03543 : SUCCESS in 25m 13s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f76d1d55a10b4d6b8a949fb6a8e8fdd3 : FAILURE in 8m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/836aabc3767645ebba721193f48b6751 : FAILURE in 12m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/081f175d9c1547bca22e5d4f23757700 : SUCCESS in 16m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9c8f0a7b6f0b40508e2cac0820d6ab3e : SUCCESS in 54m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c7d5cc80f2fe4bf7bc54a5f7b8c273b0 : FAILURE in 7m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9e63c82ce4f74b289eda2d17433a0ef9 : FAILURE in 8m 59s","accounts_in_message":[],"_revision_number":2},{"id":"5e87c2d7dfa9b8ee2760fda41f647528258a87df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 09:51:15.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e05e0ff26a7488098d9ba074b4f7058\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/40f2c4539a0a4d8faa97902df5b8e073 : SUCCESS in 18m 19s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dd34072bcd7f481499865490788e5796 : SUCCESS in 16m 06s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ac6c95281eab4f8bad340c0bd6f6a147 : FAILURE in 7m 12s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a2ed80426754ecb941674505073d45d : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/24a49bdc075f4c44bead8f06cc1aee2b : SUCCESS in 2m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c39de4e23a7f4af8b527cb77e10388bd : SUCCESS in 5m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7d1a9bc7f1cd4f4fb58c1e4ba227163f : SUCCESS in 17m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1ddcce6959f343aeb39c2d6077ad2a41 : SUCCESS in 1h 05m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b968887bad6ecaae0cfd82d77e3e4514b3cc835","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-13 13:00:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d75cccda0d01a0ef4b4c53a08ec984924534b61a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-14 08:08:03.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_22629\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"_revision_number":3},{"id":"c5f59d7c89e6b9f0d14ceab81667c76af1084a87","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-12-14 08:56:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3781490c78c171292c8036aea7a7ed26242ca08f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 09:48:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3f67276eb944a67ba49e1dc2eb6df48\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2eb6c750380a4af595e3768c097ce713 : SUCCESS in 23m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/325c230165ce4d17863be593fcdc94f2 : SUCCESS in 1h 09m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/beab367a20784103a8a20f7ce6992f09 : SUCCESS in 19m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/09ca7202b13c4cbcb8b3fa5d669a9e98 : SUCCESS in 30m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ca40912b45434511988d57943b9a0354 : SUCCESS in 52m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/64aed5e352ed4417950fe5572f3f1345 : SUCCESS in 1h 30m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61240ae20a814104837be460908cea67 : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e805813c4664fa09e41a10c0376bce5 : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6d7a0035fd74784938dcfc38917271c : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7238cd2bbcea40d6a5e250559601f4b1 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4be8ae16025749038f05772d809f97c2 : SUCCESS in 4m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bfc85b104c7848e5a5fc7f6842ebec12 : SUCCESS in 21m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b327995c64ab41129121aac90396074a : SUCCESS in 28m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6b6ad27333f848ccb127eb009eba2303 : SUCCESS in 1h 02m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ed358d20fa55429ca2ef817da70c2362 : SUCCESS in 1h 21m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/366315da507e4810b9ee01ea52f7347a : SUCCESS in 20m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1bf20aa142084eb7a6942fa412940f21 : SUCCESS in 1h 16m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2de81b33ff164f0ba10fa9e2db11be98 : SUCCESS in 53m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/61264e8553d84cf097b113c4d4b84e3b : SUCCESS in 1h 27m 15s","accounts_in_message":[],"_revision_number":3},{"id":"d2caa54035d72c252f422997ddce1a4bc3416e31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 10:12:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba0086d5813d4febb7d2f931c7a16ce1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ec9c62cde4ef4f6fa4f13c65bc0a9e01 : SUCCESS in 17m 13s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/87c0423dc6c64aaf9780081126a0082e : SUCCESS in 16m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b7fcad993af54c88b14693ef10778b49 : SUCCESS in 2h 02m 23s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca743ca532204384afa5c919e82df94d : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc4f8915366d4d08958bcd252b64fbc2 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/38283ae7a5984f0b9539d69903d0ca97 : SUCCESS in 2m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/59b786f8a708462d84ca4e04e3101baa : SUCCESS in 1h 11m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8a8df0436e9940b0bed19d4ebc1972ff : SUCCESS in 15m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ec2677bb4e259c064148363e12565ac28987939f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 09:58:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ba60e927041d30fb547ae3b3713961e15c7dbfa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 13:08:28.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/903363 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"13aaad9ae48ca16cb96b36f6c8591d35a1d53cf6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-20 13:38:15.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"83fd0e4ae119560050ac729db3744c9e691a21e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 15:22:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/190a2f0cba584cafb4123ed3927c8b43\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/dbf93d5a07784eb1a0351027b537126e : SUCCESS in 20m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d208d3a08eb34b968eaa18b23fd2f3d5 : SUCCESS in 1h 05m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c225a4f10f604df292e6c432a225ffb9 : SUCCESS in 24m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/859a27ba1776459ebb0617734397cc05 : SUCCESS in 37m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/996871cd0b1b4d4381ff8df0399a230f : SUCCESS in 1h 06m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/092e688ac17b4509a38cb870748450f8 : SUCCESS in 1h 30m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e45b0d262f4f9a8048f8c092d9647e : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27707b3158224108988e93c39597e696 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb54fbc021654f6b84a938f594ba6190 : SUCCESS in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cb18bc7a5814ed19b6c294506467cc0 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7568031ffb8b448b9074db228afd401b : SUCCESS in 4m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/092e38def0bb4ceaadd429493dbf07e9 : SUCCESS in 26m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0257577ff2a34112bc1a73bae0f2139a : SUCCESS in 23m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dc76cb2848f646dc812630e53af3acf1 : SUCCESS in 1h 11m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/89f088a866fc4912b3f5f7942a14739a : SUCCESS in 1h 35m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fa3813cc092f4926a3443066619cf7f8 : SUCCESS in 22m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0dd1467b99444e33a2ac5cbb9e7e2b9d : SUCCESS in 47m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/aa46c6272f004e01a7da4b1a951fe899 : SUCCESS in 47m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cfbf9899b50441949c745d3339f5dd89 : SUCCESS in 1h 39m 42s","accounts_in_message":[],"_revision_number":3},{"id":"381858e393cfdfba3fcbf4464446f4c30e84e81b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 15:28:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"04e122cd35bacacacd22d807bfaaab7d2149ec57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 16:00:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81dbce76d35742ce813d7aadc79d040c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/888922ad110842c49f4a140ca412b394 : SUCCESS in 1h 41m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eb48360d90e14509aa19c5696cf0c1dc : SUCCESS in 1h 32m 18s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/aefef1513bb342579cc411de77003121 : SUCCESS in 2h 19m 24s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/64db8b31e8f54051ba7449ee238044e1 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f1f51ee8060f4facab032e73eaf6f2cf : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/74f9daf8c10b408d9c8632163bc6cb52 : SUCCESS in 4m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4c2d6100a9c140b0b684c211238f9441 : SUCCESS in 21m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9361db2b06a7491b9f40cbcb0baee596 : SUCCESS in 1h 38m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b2402a2862b991966524af292f07217b2a710c05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 17:09:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ff6ba74091d4328bdc93dcf6bc880b1\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/07287dee9e6e4b56a8c032a9a7f5b134 : SUCCESS in 21m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/472d8e9e9bd8485892f71b627ccabb84 : SUCCESS in 23m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bd539bf8a70b4a49b824b50586d2c538 : SUCCESS in 1h 04m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/10fcd1da3a9d48008e01f1f8cb455e42 : SUCCESS in 1h 33m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3037bb97ba704d82b90172525ec288bf : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5e41110e0d34babaf864fdb0073d0eb : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5dae8c809c1e456294b07247a6408448 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2506c6ad66114f8494622e674dbb47e4 : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/09acaf3477804f8cb3c0310c3ddfb62d : SUCCESS in 18m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f466d0add96f467ab47ec8489cc2d743 : SUCCESS in 27m 19s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a9032ed0b1f34f2096ce0b92529af4d0 : SUCCESS in 1h 07m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/73a80160cb6f4fb2bc07febca26ea4e2 : SUCCESS in 1h 36m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1ed7dcd3b2c14fe0b7cf574eeff7acf9 : SUCCESS in 16m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fa13f9afc1654bad8b5265fea3d96dd2 : SUCCESS in 1h 13m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/36fba1e2fe5743e1b14ecd7a233875b8 : SUCCESS in 52m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/605dbfa1358b4613bc106f8dd5f5e304 : SUCCESS in 1h 20m 27s","accounts_in_message":[],"_revision_number":3},{"id":"110d13c1895c67a97edd3382c7b6f2ea1eb47a48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 17:09:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"df6c46fb1491a27c056064eb686f0ed306fdd08d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 17:11:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ee2157c8c5e44c58b49e1fe4d79957e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d4fe38f13ac40a095d9a51c42d60168 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3},{"id":"09f8689d5e2e94c569dcafbf2245122416296372","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-12-29 03:52:28.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"316bafaf075ee85f555fce5e5c4df72f88e52450","revisions":{"240c717bccf0d3eecda64d8b50d5de779177867f":{"kind":"REWORK","_number":1,"created":"2023-12-08 09:55:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/61/903161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/903161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/903161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/903161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/903161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/903161/1"}}},"commit":{"parents":[{"commit":"4c8298de9a41899ac902dd42548986ba1591fe8e","subject":"Revert \"Disable build of Gnocchi containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c8298de9a41899ac902dd42548986ba1591fe8e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-12-08 09:55:33.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-12-08 09:55:46.000000000","tz":60},"subject":"CI: Use newer podman/buildah on Ubuntu Jammy","message":"CI: Use newer podman/buildah on Ubuntu Jammy\n\nChange-Id: Ifce506b49571950ee16e26aae06f20aa94567722\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/240c717bccf0d3eecda64d8b50d5de779177867f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/240c717bccf0d3eecda64d8b50d5de779177867f"}]},"branch":"refs/heads/master"},"2ae668c039b73ab97c98c536a76fb19038316db8":{"kind":"REWORK","_number":2,"created":"2023-12-08 11:13:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/61/903161/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/903161/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/903161/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/903161/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/903161/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/903161/2"}}},"commit":{"parents":[{"commit":"4c8298de9a41899ac902dd42548986ba1591fe8e","subject":"Revert \"Disable build of Gnocchi containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c8298de9a41899ac902dd42548986ba1591fe8e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-12-08 09:55:33.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-12-08 11:13:23.000000000","tz":0},"subject":"CI: Use newer podman/buildah on Ubuntu Jammy","message":"CI: Use newer podman/buildah on Ubuntu Jammy\n\nChange-Id: Ifce506b49571950ee16e26aae06f20aa94567722\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ae668c039b73ab97c98c536a76fb19038316db8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ae668c039b73ab97c98c536a76fb19038316db8"}]},"branch":"refs/heads/master"},"316bafaf075ee85f555fce5e5c4df72f88e52450":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-12-14 08:08:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/61/903161/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/903161/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/903161/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/903161/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/903161/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/903161/3"}}},"commit":{"parents":[{"commit":"8f863beb45ead74d5a7b4c956acad0c7e0f22978","subject":"CI: Pin docker installation to \u003c7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f863beb45ead74d5a7b4c956acad0c7e0f22978"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-12-08 09:55:33.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-12-14 08:08:03.000000000","tz":0},"subject":"CI: Use newer podman/buildah on Ubuntu Jammy","message":"CI: Use newer podman/buildah on Ubuntu Jammy\n\nChange-Id: Ifce506b49571950ee16e26aae06f20aa94567722\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/316bafaf075ee85f555fce5e5c4df72f88e52450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/316bafaf075ee85f555fce5e5c4df72f88e52450"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
